)]}'
{"id":"openstack%2Fdesignate~902506","triplet_id":"openstack%2Fdesignate~master~I94d789d4aa58ffcd4ea05079f7d7015de232ce19","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-12-07 19:31:46.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I94d789d4aa58ffcd4ea05079f7d7015de232ce19","subject":"Cleaned up quota implementation","status":"MERGED","created":"2023-12-02 21:31:41.000000000","updated":"2023-12-07 21:36:19.000000000","submitted":"2023-12-07 21:35:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"902506","meta_rev_id":"7426249ed6ae45324c980c90c7f6811cd0166f40","_number":902506,"virtual_id_number":902506,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-07 21:35:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-12-07 04:30:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-07 05:55:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2023-12-07 18:16:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-02 23:03:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-07 04:30:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-12-07 05:55:01.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"65e6934ed5df6e3a2099fc1954e174c00583b2e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-02 21:31:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4600456a2540d1349a1e437335c3bd38ea9a9180","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-02 21:51:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5946a317c2e5fe72b3804a84c812d44b96cf9b5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 23:03:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e286f2ef54ca4507b14895b40eaac7ff\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ca3b15ce40424ebcb50150e93e5ab60a : SUCCESS in 45m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/facf25d57e9146309f77746733b25c1d : SUCCESS in 42m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/60df4c032591456e820872201a155db4 : SUCCESS in 54m 50s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cbc3b1d6c84941818ee00c7175351fbc : SUCCESS in 46m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/02591b1a94f74097bf6871a25b22e1a4 : SUCCESS in 49m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/19a0d92c6c8e4a8baf3432e62fffc482 : SUCCESS in 1h 06m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cefab1cebe3d44a991e376cadf9fd38a : SUCCESS in 55m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a8ed021af2b64be88d07e3d817b86cc6 : SUCCESS in 32m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/300fb75c646e4bd3b3567e3268269bc4 : SUCCESS in 53m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6b3c42eb85142dcb0087b1db82027f6 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7128b6dc5d2f4aec980b6d572fe7a00c : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3efc01be1a842bc88c872bbf885123b : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d28afb38af924712ab61a7490dc00d92 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0e15fb9e5004896a80a2d5c12bb5362 : SUCCESS in 6m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/522e75d6e1114179a58035b9a48328f3 : SUCCESS in 44m 31s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/05cf6b64f7a84a93951188344cbbfafe : SUCCESS in 7m 19s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e72e5f1ad3d64f20a72cda1a3ac9d499 : SUCCESS in 35m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/50bb4aacbdda496584b6e5d1b44fc1f4 : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"95857629290c6363739a0b65190d3c239cec5a1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-05 11:28:19.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d819c3ca3170b703dd5c14dbb0f8ea8c0b8e3f57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 12:31:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a3bfadec4cb4bfe964bdab83a531a4d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8942fd75d91c4cb4955c72793da7faf2 : SUCCESS in 49m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e29da61bd19f4842882de993e765f19d : SUCCESS in 58m 12s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/988fe2c5abef43319ad579fdf5f73fb8 : SUCCESS in 54m 28s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a0882b0adace423bb9591170a715a9b7 : SUCCESS in 34m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/23d653fcf2f749718148f068a3e5e003 : SUCCESS in 52m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bfaf112e6f13461598de9d28e09891b9 : SUCCESS in 54m 03s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7a438816aaf34227a578950f4ea2275c : SUCCESS in 56m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/caf56961682d40f58bfe961a61ee07ab : SUCCESS in 37m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b1b2fa51504d419c9b2b87c7619e5dda : SUCCESS in 35m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89a5d38f1ec741be83db570a1ac177d4 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63fc55f9cbdb417589df19d00721ed72 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06c3eadfdef94583ad8d333d67b101ad : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff0669ed1f4b41d895207380a50e2ee4 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91c3533803d342f7acf9772e3189c295 : SUCCESS in 9m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bc6c99c3e02e49d9bfdbe87365ee60b3 : FAILURE in 22m 09s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fac821017b5f4606bcbd27097813aaff : SUCCESS in 5m 52s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f6bc4b3ecd6b4b21b210707d9799a20d : SUCCESS in 51m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c325ecb7244d45fa83217497071fee69 : SUCCESS in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ea3ef03fda1869a035013facc5616e00225f7dca","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-05 12:41:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"761e367045f0d49af4e03a09711a9207e55b03bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 14:13:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75b5dd99fe894af5b8a5ca19ce981517\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b4bc78b9611f4f1ba71b96000fe12f32 : SUCCESS in 50m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/988d01c81d7d4190a3601538137d00c3 : SUCCESS in 44m 57s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcbedf851b8149a799c5df7936b2b669 : SUCCESS in 54m 13s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0ca1f19a437449c9ac537f865dffd10b : SUCCESS in 58m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d6cce821d0ac465cbaafb19590dcae23 : SUCCESS in 56m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a993ffaaa13b4af9beed7af7a711755c : SUCCESS in 1h 06m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/60aca04cc97c46cdbdcc9623bf49afd8 : SUCCESS in 1h 18m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0582d67c377440099f8c8078af95d1f7 : SUCCESS in 49m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/42dc97c363b34531a55cf822efc01131 : SUCCESS in 50m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54a09b894ba5405ba7885a343b1a49a3 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55662efd79e7423a9670832855590bc6 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23e9169fc05a4fe6ad0e83a861f56f2e : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6bcf2e1ca895488783fe463082f853cc : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c4acbdb6364e9788721a71a2f7b8d3 : SUCCESS in 8m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7793d4475f0a4d0a8ed4fc62f9d07f1a : FAILURE in 19m 19s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6fcb4a3b5b0645fe968e7a8ef6dea5c2 : SUCCESS in 9m 06s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/781f53a63c8e4b0ea478306c363560b5 : SUCCESS in 49m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0d300be70d2d42289aa75ea53317cf9c : SUCCESS in 5m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"756b761af2be82008873dd526e3bf96390af7faf","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-05 14:28:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"17591f2e3af116b8dc939c51bba0999c1db3ef70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 15:31:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d5cc1e5d5494c06823c8e6e4e8243cb\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/42f23334fea34efd8f9632c7a9fbeac2 : SUCCESS in 38m 59s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/955447482e3b44a697037e493e5278f0 : SUCCESS in 46m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8fd2bfa0651041e8a7a9a0fcce62c739 : SUCCESS in 47m 17s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/54ef2be02d1a4e1499ab36c397dd9e6e : SUCCESS in 35m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5635c5883b964d6582f6abe27e072f6a : SUCCESS in 39m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/95bc2e54c6ca434083a3b71bf89af3c2 : FAILURE in 56m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c1b6b7d79a634aae81ab5c3761912b46 : SUCCESS in 1h 00m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dc74b3cfd6784976a36fcf36503fdd1d : SUCCESS in 54m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d61be594a62b4f8cbb7a2403f35c3317 : SUCCESS in 50m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/362c75fa2a384a449c1f6f18b9924de7 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfbd3bb638ba44a5b9ddf31935f4cc54 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c79463da80e54c4297ee54693d307c21 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98c7239dbed44e5eb73cdc42cbecc7b3 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fb70b50c4154583bd44ff019f53dea9 : SUCCESS in 7m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b92dcfb6ad364f258a4ba67bdfa1d06d : SUCCESS in 33m 22s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/206ee13b59b640789f933538531217b4 : SUCCESS in 7m 29s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c49d9fe66fd5454797045628bac8503b : SUCCESS in 47m 58s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/188363097ef34ff6937bfc7ef2db2fb3 : SUCCESS in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9eb8e232deb99b55fd4a9e2c63482182b4885926","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-07 04:30:46.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"09c529e80ba0c77df335d6c79ee5d7554272e349","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-12-07 05:55:01.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ea88963068a7429dedbfb195b064ebd0d07dc0d4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 18:16:05.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6c6485c3ab4835b88bcb9e9c82c0fee9f66dda17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 18:16:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"94f7db10361503d4c0dbefdcc22a1eaf7c0905b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 19:11:20.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/56d5da49baee4fb184ea48b935e98603\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/be10942d540b470aa550fea4dbff1d3b : CANCELED\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ea6d3c0062ab46b69096615814a9b2e8 : CANCELED\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9037dfd9dfb545dd8092ae396b30f015 : CANCELED\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0ecc819fcdac4869b09cbd86e81c7f38 : FAILURE in 53m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/46d1bd70bf99417eb557f21dc0f8fa8d : SUCCESS in 37m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2b65e9720fb14ff0a313c7a0d0c3c05b : SUCCESS in 33m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6346b5a12d94605817b6e1ea943c079 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e751be149224837a1044a3d4ce4bf75 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e4b144913e34ab8b2734056a6634b7c : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d50ca6c66f6746bab2ade7bb72f1be31 : SUCCESS in 8m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1dabafca6a5e4bd197e7c3c55dade75d : SUCCESS in 30m 31s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b32139a54497484aab7e7591119e60de : SUCCESS in 7m 01s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c0b26fb608c742089cd84ce6ddcc3114 : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":3},{"id":"2c952cadc246d4ab0ae9ccb827337d4ee5a7b1c1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 19:31:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c8d1461787d804afe954ed7cbe2f41f02c532c6e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 19:41:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"03f20e18980b0a5bbd3624bbbe0caa7602798a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 20:41:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f4ee85a100f447abe9ca9a87267a9ac\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7b0a4690ab264035bb9c1de86e537faf : SUCCESS in 46m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/10ada94ce9ca4d8cb3237857ce8bd660 : SUCCESS in 41m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/768e2f90e1304732a73ac6f038b2e713 : SUCCESS in 48m 06s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1b3d3df411464e1d9b805dbddeb8d916 : SUCCESS in 50m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ee71593be558484e8cb96f5fa02010e0 : SUCCESS in 49m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ad238c3c1c6940d5a6c59584047fd376 : SUCCESS in 49m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0d33e982008b41ee85207713da1c73f2 : SUCCESS in 1h 08m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7e680e5c133f4bd7b1d7932413af036e : SUCCESS in 54m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/664a5e27a44f4dd39de317082d0634b0 : SUCCESS in 38m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6de4b55cff1e4562ab5f388495a1684f : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce553f0566544a0aa4bb3d98d62f6d04 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4be1789b07e94a22ae7d9de5f7178238 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1040bcca65034dad8224ffbe413ea102 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0237382b777f4eb39c06513d08267d86 : SUCCESS in 9m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ce6b078bcb204fc980cf86a8bf4b525b : SUCCESS in 43m 17s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/310ffa7684b14eb28161193efa2de4e7 : SUCCESS in 8m 31s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0011b48b1ad04489809e2bed93b5e737 : SUCCESS in 41m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c35f0a383746406da0665ea9bd8553d1 : SUCCESS in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"09828606fb4dc5074d5537bd3ceb83d6c4fe4924","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 20:41:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"042d9c37def6420545c0e5504ad8adf9164d1621","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 21:35:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25be97d6881e4f1e8c63c2a37bdaa0a3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8cc82bd28ba74d9cbc85866a55881645 : SUCCESS in 35m 36s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/47d5ab312f7e4a2e8817eeb2850ab879 : SUCCESS in 35m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5d8acf20e9294f5e96b649628a4bedc2 : SUCCESS in 34m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c7a80367324546eeaa12ea79a73f0931 : SUCCESS in 53m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6545f5da51804f6ea298f744eef90ea6 : SUCCESS in 36m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8f2b1983431a4a7bbf9cbfd630c4b6c9 : SUCCESS in 45m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/659c0fb1b218441996993d5ef9fcb593 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29fdac91e7ea4876939a2cf72e59f786 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8283ac4829df4388af106549e16e9e2d : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47c6a1c3cf1349019c41060ca3b91822 : SUCCESS in 8m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6b25d723a2664150982cd83cde3fea87 : SUCCESS in 30m 53s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3de29e97d24a469b8d29a01ceae39fd0 : SUCCESS in 7m 03s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b9c8327b211242839e5cc8183de45037 : SUCCESS in 50m 26s","accounts_in_message":[],"_revision_number":3},{"id":"ac8643f78a2662dedc98169386bd4539019e1484","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 21:35:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7426249ed6ae45324c980c90c7f6811cd0166f40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 21:36:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bc876783eb84422a9f755717b7730e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f70391560cca4ecc9300132b112f158a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"df57620f77c2d60106ee64c9adb15bf0f737be1f","revisions":{"70c4910a2c6ec99059c122a228ea201443eb0319":{"kind":"REWORK","_number":1,"created":"2023-12-02 21:31:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/06/902506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/902506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/902506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/902506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/902506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/902506/1"}}},"commit":{"parents":[{"commit":"ddad658882caa713c9a89aac7afcb1081ef18707","subject":"Remove unused ZoneLock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ddad658882caa713c9a89aac7afcb1081ef18707"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 21:30:00.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 21:30:00.000000000","tz":-480},"subject":"Cleaned up quota implementation","message":"Cleaned up quota implementation\n\n- Added additional test coverage.\n- All core methods are now abstractmethods.\n\nChange-Id: I94d789d4aa58ffcd4ea05079f7d7015de232ce19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/70c4910a2c6ec99059c122a228ea201443eb0319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/70c4910a2c6ec99059c122a228ea201443eb0319"}]},"branch":"refs/heads/master"},"a275baa5a1203afed8c5b65d0962673ac0187e7e":{"kind":"REWORK","_number":2,"created":"2023-12-02 21:51:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/06/902506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/902506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/902506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/902506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/902506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/902506/2"}}},"commit":{"parents":[{"commit":"ddad658882caa713c9a89aac7afcb1081ef18707","subject":"Remove unused ZoneLock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ddad658882caa713c9a89aac7afcb1081ef18707"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 21:30:00.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 21:51:14.000000000","tz":-480},"subject":"Cleaned up quota implementation","message":"Cleaned up quota implementation\n\n- Added additional test coverage.\n- All core methods are now abstractmethods.\n\nChange-Id: I94d789d4aa58ffcd4ea05079f7d7015de232ce19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a275baa5a1203afed8c5b65d0962673ac0187e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a275baa5a1203afed8c5b65d0962673ac0187e7e"}]},"branch":"refs/heads/master"},"df57620f77c2d60106ee64c9adb15bf0f737be1f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-12-05 11:28:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/06/902506/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/902506/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/902506/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/902506/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/902506/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/902506/3"}}},"commit":{"parents":[{"commit":"98f1ad2856a0acb9081c5449ab13718714b42920","subject":"Merge \"Improved MDNS Handler test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/98f1ad2856a0acb9081c5449ab13718714b42920"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 21:30:00.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-05 11:28:19.000000000","tz":0},"subject":"Cleaned up quota implementation","message":"Cleaned up quota implementation\n\n- Added additional test coverage.\n- All core methods are now abstractmethods.\n\nChange-Id: I94d789d4aa58ffcd4ea05079f7d7015de232ce19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/df57620f77c2d60106ee64c9adb15bf0f737be1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/df57620f77c2d60106ee64c9adb15bf0f737be1f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
