)]}'
{"id":"openstack%2Fdesignate~902507","triplet_id":"openstack%2Fdesignate~master~I4a3901512b31d1d13db9e75d3de26bf2dcb5ba0d","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-12-08 15:27:56.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I4a3901512b31d1d13db9e75d3de26bf2dcb5ba0d","subject":"Improved producer task test coverage","status":"MERGED","created":"2023-12-02 21:49:23.000000000","updated":"2023-12-08 17:53:52.000000000","submitted":"2023-12-08 17:52:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902507","meta_rev_id":"abd086056f29a4fc40d65e02cb2707b213292b15","_number":902507,"virtual_id_number":902507,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-08 17:52:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-12-07 23:27:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-07 23:27:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2023-12-07 23:30:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-02 22:50:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-07 04:15:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-12-07 05:55:18.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"8c3924da7aa4698af600a6c7366bfb542c538f4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-02 21:49:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a5fc52fddc24cfcccdc8ff24b6205da94195018","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 22:50:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f69745f3caca4ff69bc911ebe2c2e5cc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/215b58e1a2294f1081ddbed4110a9111 : SUCCESS in 54m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ac0f04263afa43c382942ba181cdabf0 : SUCCESS in 41m 15s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2afee2207191415dbfa0b21c35a9bb81 : SUCCESS in 39m 40s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5c29c7b75d794a55925aec065905477d : SUCCESS in 38m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7ff7da2f2db0477d87941c2584f7079b : SUCCESS in 34m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/99244716db2f47548dedf41daf6aa73a : SUCCESS in 51m 22s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3c40e2b305744fee89a45176f6f74cc1 : SUCCESS in 58m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cc16ace53357431f802843df014823c9 : SUCCESS in 31m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dc79e8edbf754236bf88e5432ef4b268 : SUCCESS in 36m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5a4058c39394241ba40d040a4f490b3 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ade3a2c150374fe58ed9ee789ac48545 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa6a7c7432904237b93142bd4d354bd3 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1917ebd484024448873dd40621e1e592 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da2be3027bad412e9f771df023a666ee : SUCCESS in 7m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c70f190b773847c2a7814c2de312aee7 : SUCCESS in 34m 33s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9a41aed8961340bc88c9907189aa90d3 : SUCCESS in 6m 26s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/016c3f11148041eba6b6bd013994d658 : SUCCESS in 53m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/caeb25e3943a4037a384a6303a370cf7 : SUCCESS in 4m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a6522149c51d713f681fd2dc4f059552678afb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-05 11:34:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6b85ce596ce6011a038e7fe6255a1c472ea17b4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-05 11:40:40.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ede948a9b84d44e449d5a7e56003592b5b27d7f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 13:10:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5e03efee79042bdb20130bc212a94d4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e10de1fb4a5946849f4ebd0ade15640e : SUCCESS in 47m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a9b8f48370a74f05b63354230440049a : SUCCESS in 1h 03m 50s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7aa0c0aab667497291a2f4c269ab530f : SUCCESS in 58m 32s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5ed827b3cf604e1783bdc45826b6731e : SUCCESS in 40m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3aeae5062d4946c3a4edbc93cc07459f : SUCCESS in 50m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/daf22506a0dc489c82a23ce0b64058e8 : SUCCESS in 1h 13m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f649f3857c694bfb95b48e1edf194c46 : SUCCESS in 1h 21m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/078937dd94ac4572a74cb9d5a3cfc142 : SUCCESS in 38m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/112949aea43b44f490f82b696295c0db : SUCCESS in 34m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78b6fc57ba7c498d863f5d31ec82bdc9 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f997bb0547dc4659b0f6fcf403a075ed : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2e09df7c8634df1b8fe8460d2aaf6f0 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c644b900016d4dfb8b80600515c5436a : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54a30f14c6cc4a3488ad68ec44d4f416 : SUCCESS in 8m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ebb255b044ed4917aa56bcc6a503daef : SUCCESS in 32m 43s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/416e396fdf544154a15f733ea10e0b84 : SUCCESS in 7m 45s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9e1b2a34483e49f494353a7606baac77 : SUCCESS in 39m 15s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7758dd40524747c29e801c8dd891066a : SUCCESS in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"45257ae65ed92a544ee8159eb3e31cd845994fa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-06 13:03:57.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b97a9e94db2a2d1d1fa9f8ad8575fb0d19e28fa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 14:12:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/986f57a4f9da4b73a418bb6cbb2fb59b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/08c7732e00a847b4b298e01cc2274473 : SUCCESS in 39m 50s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6c9debe104dc44529852c1ab4a3f4478 : SUCCESS in 1h 02m 30s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/81317275f50a493195774003f9a9eceb : SUCCESS in 54m 13s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3642a050d5a2438093e46dfe41fe1e42 : SUCCESS in 37m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/38508cfedcc44a9ab0a5ff4f9d064c63 : SUCCESS in 48m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/66a4edef617844a39b2fbd2d319806a0 : SUCCESS in 1h 03m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/799edbde42d84904a24c6c657fa72cef : SUCCESS in 1h 00m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/958f0d44a0cb472ba0f03f860a77d66f : SUCCESS in 49m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1ef1379caca040a791e8983ef75f04f1 : SUCCESS in 41m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c8cd4f5c6064c84aa7e480b4235e9db : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2657e4ce4af94a60b9f9dea743c156f7 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e41a353a4e664273b35d7def88ba5abb : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/834db062f8e04db3b46593ab036cb596 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cbbe0013f6346649ae2da3cd58ff7ce : SUCCESS in 7m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/982023060d7e4cc0a72913ede5c0be0d : SUCCESS in 32m 45s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/399e5b1c3ed94dc8bd2edfb5475c7b2b : SUCCESS in 6m 59s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1200de09b4ab4ec8a254212308548a52 : SUCCESS in 41m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9b314dbaaa264e5eadeba5c6ef069562 : SUCCESS in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"53e9474c16ca82fc0665e02cfdbcca064d432227","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-07 04:15:57.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a508602382987bf0717d7a415273e8d6749b9949","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-12-07 05:55:18.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"973b8ae2227a0631d2b311543b96cfebf802a07d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 15:50:13.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"582ce46e4031f74c8d8142fe083ad856006314ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 15:50:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b735a7714fc4e3b64154f38bd89e5adf48a40579","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 17:13:09.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/902501 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"9432eda74e2a4b25b63e9d2d3a91b79194fb43f7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-07 17:24:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd0837079f549595802e283aa46966d778f3023d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 18:26:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f477092b5ac44328111ee5d35701294\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d6a4357eca034fef94d8f4d316fd04bf : SUCCESS in 50m 54s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e48031af938244109bb5abd3902d3c55 : SUCCESS in 57m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/10cdb41423ee42e28e6833c2261041bd : SUCCESS in 53m 29s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7e09779574b34c50aa99ec02f00412fa : SUCCESS in 51m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/db0234b4b6f24bc3b74a990e0526c0ec : SUCCESS in 44m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/78bade36553b4c91a3a8478fceaf6c93 : SUCCESS in 59m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fd0f07c9bda54c968d592056b9f3f8f2 : FAILURE in 52m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/feb2f29f28904123a769b1552c096c6b : SUCCESS in 55m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6ed088f0477b4151829e821929192726 : SUCCESS in 40m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/584a1868f6ae45ef9489e7ffb6079a29 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3791413508642419ce2ba9d2e30ec04 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d83265bd5bb427ca364978f3260ab96 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e04810e66a140f3b3bd32ac2fe2d241 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc340e28425b4f4f971c82a453f5b797 : SUCCESS in 7m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7b532f3b39484a908d03fe3ece60d53f : SUCCESS in 31m 16s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bb710ad43490430393a49cce7830b344 : SUCCESS in 7m 56s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9ebf9ddca89f4a04a23776ae7638cfa3 : SUCCESS in 40m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/373f2e2a00ab4fd3a0a3dcda7abe1858 : SUCCESS in 5m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ce1af3932a5e9942e4f9758a3b93fc13727f57d2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 20:04:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7d2b7d8e03fbc146a40b1560c3fbeaf95e353057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 21:27:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d988565755c40a8a775794185d4cd52\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/93c6300f2d0948b688d3077f59478f37 : SUCCESS in 53m 21s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/541e135c27704bb7b072cc35cbfdfe9a : SUCCESS in 1h 06m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3745c8f368944225a0c560c691b67635 : SUCCESS in 54m 50s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3508e0419aee485a9a07f94e59b4ce2d : SUCCESS in 34m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d6fbf7d0d6ab4681a5a750fecc9207a4 : SUCCESS in 33m 54s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cefc5738c9034da3a5775addc9d34c83 : SUCCESS in 1h 11m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dafd5520646d406892d20bfcf75c7c62 : SUCCESS in 1h 21m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b4cc97b4df924759b5f776fa6167cf42 : SUCCESS in 32m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bccb571dbe394c7b86ed49b68191e275 : SUCCESS in 31m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa9f0f7753ab49c2b52af6ca0feb10ac : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d42446fe26574da4a57397eeea5a38aa : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa8e3f56ecef447d9dae803cfe036e47 : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/797185252185454183d4c7ccc0339998 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43769fdd0fe146198fc7008531c2a674 : SUCCESS in 7m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e3a3b5be227d4ec58cce11096d0ec8cd : SUCCESS in 32m 41s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/196f7735227e412e8de9cbf7138857bb : SUCCESS in 7m 19s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/38b14cf825d343fc9c01c957da931778 : SUCCESS in 49m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d860ec3eb0724c2fb94d2791e06f0f7b : SUCCESS in 4m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"46ed6f58ce1168e8b411bd2560ae0e6e08358795","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 21:27:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c204560971afb0809b056fa5d118d33f91b8d6de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 22:34:31.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/232fd779cfe64d679d08d5b4e9fbdbb5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7d448b57079d4adeb5dc509fde551599 : CANCELED in 2m 42s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/30c7b93a2c1e48be90e1be9c2ff5914e : CANCELED in 2m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/22491b1037794028b49024451fd97023 : CANCELED\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/243eed5194e6417ebd274f8d2b83a45d : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/abbbe21d8b9c4e0781b1efd64b6f7521 : CANCELED\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ed8cfb839ddc4d5eba46547234b5ca8d : CANCELED in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7505c8d0944466981ca74cc31b9e93f : POST_FAILURE in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41066e0fb6634c599d7cb7234f4384f9 : CANCELED in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6cf33c7cc6f6400d9ef4f88d14733df8 : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb003323bb9b42dab1ff6552b29ee31e : CANCELED in 3m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7c8fad07fd3245cdae29a22f1289ae38 : CANCELED in 31s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8049e344300a41efb34fcd8081b53c4b : CANCELED\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cc894af7544d43359ee128e24b73b8a0 : CANCELED in 3m 45s","accounts_in_message":[],"_revision_number":4},{"id":"99774958cb22b29838c9711a4d0f3d207470a988","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 23:02:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"04223bcf648a9d42b99653f7e72a742b972ddbfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 23:27:44.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d67c636efb45f106de3d5b4b27d7a730c4581834","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 23:30:10.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"add0304608239a7ae60a6b0f81803fc1e5c4d895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 00:49:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b072bc62b641409c9a0aa476aa9ead8a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f009c74d4f014fceb847d2dd801cd190 : SUCCESS in 38m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/77106ff602184ef9a721003735119202 : SUCCESS in 37m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b33be1a439604a18b3a1caf676cba640 : SUCCESS in 40m 32s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/580e667abf9b489a96e93a5ee7520a20 : SUCCESS in 37m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/05309b26a9f24871a610f931bc20d85b : SUCCESS in 52m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d50a076af0c84db9814d549e8ede070a : SUCCESS in 53m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c3bfc0bfdba8407fb0deb9be36219b8b : SUCCESS in 1h 15m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f764bbbbd63b40059725330e9c0e9bcd : SUCCESS in 36m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/88c269eb05d4460c8fed73de298a5a69 : SUCCESS in 46m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cbfef9905f64a70b4525e94ee2e7cd6 : POST_FAILURE in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baa362dbeb194cb3aa257785301d9e3e : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/640820dc21984413834d2fdac2d8495a : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e99846870f041d4a6dca2b586eac211 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d34865e1f1d47608da35bf16dcab275 : SUCCESS in 8m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/71b19dc92cd3449f9c6686bec606efe0 : SUCCESS in 34m 42s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d7af4297dcce493985891c958fdd2e20 : SUCCESS in 6m 47s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6b7993bd5db7487cbe8950a7ef2c412e : SUCCESS in 36m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ca92be970e4d4ee1b92b094bfcfbfc2e : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0f7d0fd9f19b9f669218b4a6d5c1369dcb8cdd2d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-08 12:03:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7b10f5b13199d8c482452f141b36af17a32a7c5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 13:09:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92aa0453ea1d452d8fba61e5e025fdf6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/733f5a689bc24993bd020478d236960b : SUCCESS in 38m 08s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d6001e12db1c4a11bb07b4ce30b6719c : SUCCESS in 45m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b498475e5789437497de3a61f55d9a79 : SUCCESS in 39m 09s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/db736f2fdc94493ba0c466b6c922aa3b : SUCCESS in 40m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e827edcc965e4ed2a33ba372fbbaf068 : SUCCESS in 41m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/60e77379df3142bfa99eea5b6974e366 : SUCCESS in 1h 04m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1c4ad483401748d596711cc760cd2ce9 : SUCCESS in 1h 02m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3f5437ab5c97487e9af4e81a5394c9f7 : SUCCESS in 59m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/65b2e68b06cd45fc82d5d75d53b6edbd : SUCCESS in 36m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce610afacfb64c989f375a815d5a8e16 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0597971f167a438f9e606fbba3a49ddc : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17565ab62c5740908b1355b8fab213f0 : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c380682929654bbe81cc4e2c45328ead : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60019233aeca4ffd8d97bb778caf7fc4 : SUCCESS in 8m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4154c8806fe1420eb21b57db3182ced7 : SUCCESS in 49m 03s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c5b5163fc1cd49539dd577e5b4093892 : SUCCESS in 7m 23s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1295d74e9ff94e91a8054d433e0ddcf7 : SUCCESS in 53m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e105a013192e470cb5cf61c709cfd98b : SUCCESS in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"67451577925d7d8428879c45f1ed93da949474c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 13:10:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"355932368ecc130cc935fd1befb8cd75a5c6387d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 13:27:25.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c78e8fd66ca4a70a5fb70048b2b2835\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5bcd2a75b87c473e8d35e60eb09dcbe9 : CANCELED\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/90bfd7934f2e461380af53aa9af25177 : CANCELED\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1c9a731bd6cb46c4a1a80b8a4677ccad : CANCELED\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8a86c3802a88481e888a3344a435f4bd : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a136f7a1ba7b4b5395eece22a66cecde : CANCELED\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/35a0153dbb4742f7bf9ad1966d1f8d53 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1ad13302c4f4f5b93a5e435d53bec06 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b658a5ae04f4ae58e71b68ca2e09592 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7c8201e3f99452f9d5bf05ba04ff87e : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc7ea35f20c74e82b0cc2c1cdf538c2e : SUCCESS in 7m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4068f23a50bc4127994ac545369f9ea6 : FAILURE in 16m 47s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c45839d276f64aacbc69e10f86aefec8 : SUCCESS in 5m 57s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7a9b395aed63411daa6e020ebf2264a7 : CANCELED","accounts_in_message":[],"_revision_number":5},{"id":"f55d032da2faa93029176bb5afa1c79db13e8e15","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-08 15:27:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6e673678cb0413e0a32835306d6e54814f62cfc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 16:56:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02ac6731884e43eeac8117eb608ad25a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e741e7bd320149c78d3dfce1dd01f934 : SUCCESS in 48m 19s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/636e8b94154d4b0cb3ca5350cf50a2dc : SUCCESS in 57m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbbe8a91a6174147ae1b42741fc325d7 : SUCCESS in 45m 18s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/530e4eb82e0f49f4a0308063d030b348 : SUCCESS in 49m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1ec1f71183684f26aa1a2d2dc165de4c : SUCCESS in 50m 10s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/318442a0e50344fea821f4f2b1ce1b45 : SUCCESS in 1h 21m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/13d56d2596664e5ab7912f4f150705ee : SUCCESS in 1h 17m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0e9f358f50ef4b4abaad92eb33264b24 : SUCCESS in 48m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/14170b40bd3e4c2abc0344118ec04a18 : SUCCESS in 57m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2441368ca31a4929903a9e9670b32b2e : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a981ddddda5446a9afd87b95d7ebad4e : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c4ff57dfedb45d69a54be67cfb5920e : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a5cb0ede7df48128f532edaf59a09d2 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/094e4238db3940e7a07241cb86632e5e : SUCCESS in 7m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/72b54c1abb0b4ef5993cda674634d76d : SUCCESS in 31m 46s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2f1b1de1a28d497d87512d53d652b5e8 : SUCCESS in 6m 33s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c15f0bbf0e154a0b8d3b8f1e9a09107b : SUCCESS in 54m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/153e9e0560564737a3775b0f53d03439 : SUCCESS in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ac70fb7b6541196fe5a51915918d5861e5051a98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 16:57:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1e3865863d0e86b61b6d12643a59d8b90da4aaaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 17:52:34.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6671aa063a4410f8f09ba07001baf6b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/27d8049960a3462fab2bc71a94b30e63 : SUCCESS in 48m 07s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b42c99bc8e914ea5b0ee4889b7a93319 : SUCCESS in 35m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/893f4d5a7a2b459bbf820ac25392a3b7 : SUCCESS in 36m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fdf8982e962841518d79ad008a7d69f1 : SUCCESS in 54m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d373b3842c204908aa43c3f5d46413e4 : SUCCESS in 38m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cfd6285400404f07b2d5b5b265b1181d : SUCCESS in 48m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00020f074394498db8a3ca5863e5ef1f : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1271fbdfe1734b49aa5f6e0b01914644 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a83e46be46164faf93b45ea68d739624 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eca0ec40b87842aa99d72361220f3515 : SUCCESS in 7m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/af329bec1bd7430a8473586a4031a0c1 : SUCCESS in 33m 04s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d087e211dcb1451b95ff4a311bd395be : SUCCESS in 6m 35s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a48ba9fa91dc4bb78abb03fcde91276c : SUCCESS in 43m 58s","accounts_in_message":[],"_revision_number":5},{"id":"0cb8c3f72ed4085d9c2cb325fd1b426ebc0faf4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 17:52:34.000000000","message":"Change has been successfully merged\n\n4 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":5},{"id":"abd086056f29a4fc40d65e02cb2707b213292b15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 17:53:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9296122a84bf4050ad5242cb7fefd93b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c908ffb4b44c7ca35a84541bec5bf6 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f424245d56e8310c79b0b9a603b64b886244b95b","revisions":{"e941c8265618982aabeb9236ccc5bb77f20da9fb":{"kind":"REWORK","_number":1,"created":"2023-12-02 21:49:23.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/07/902507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/07/902507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/902507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/902507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/902507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/07/902507/1"}}},"commit":{"parents":[{"commit":"e264fda84851146c17db898d825e4c5923b20d39","subject":"Cleaned up rpc loggingutils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e264fda84851146c17db898d825e4c5923b20d39"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 21:49:07.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 21:49:07.000000000","tz":-480},"subject":"Improved producer task test coverage","message":"Improved producer task test coverage\n\nChange-Id: I4a3901512b31d1d13db9e75d3de26bf2dcb5ba0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e941c8265618982aabeb9236ccc5bb77f20da9fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e941c8265618982aabeb9236ccc5bb77f20da9fb"}]},"branch":"refs/heads/master"},"e7fc62c5bc20743e4a134d4c58a3c56eafca74ee":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-12-05 11:34:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/07/902507/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/07/902507/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/902507/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/902507/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/902507/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/07/902507/2"}}},"commit":{"parents":[{"commit":"98f1ad2856a0acb9081c5449ab13718714b42920","subject":"Merge \"Improved MDNS Handler test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/98f1ad2856a0acb9081c5449ab13718714b42920"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 21:49:07.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-05 11:34:30.000000000","tz":0},"subject":"Improved producer task test coverage","message":"Improved producer task test coverage\n\nChange-Id: I4a3901512b31d1d13db9e75d3de26bf2dcb5ba0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7fc62c5bc20743e4a134d4c58a3c56eafca74ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7fc62c5bc20743e4a134d4c58a3c56eafca74ee"}]},"branch":"refs/heads/master","description":"Rebase"},"71ce7d07924870b58e1904ccf486c822b2ade05c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-12-05 11:40:40.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/07/902507/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/07/902507/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/902507/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/902507/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/902507/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/07/902507/3"}}},"commit":{"parents":[{"commit":"1cf282f50b81bb095b32303afdafed632ac951ad","subject":"Cleaned up rpc loggingutils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cf282f50b81bb095b32303afdafed632ac951ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 21:49:07.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-05 11:40:40.000000000","tz":0},"subject":"Improved producer task test coverage","message":"Improved producer task test coverage\n\nChange-Id: I4a3901512b31d1d13db9e75d3de26bf2dcb5ba0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/71ce7d07924870b58e1904ccf486c822b2ade05c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/71ce7d07924870b58e1904ccf486c822b2ade05c"}]},"branch":"refs/heads/master","description":"Rebase"},"b5ec3f8e296d40fcb8bb3b391af947841bd64767":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-12-06 13:03:57.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/07/902507/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/07/902507/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/902507/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/902507/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/902507/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/07/902507/4"}}},"commit":{"parents":[{"commit":"b5b9e9902c000f1dd94474e61861956e2f3c2ca2","subject":"Cleaned up rpc loggingutils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b5b9e9902c000f1dd94474e61861956e2f3c2ca2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 21:49:07.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-06 13:03:57.000000000","tz":0},"subject":"Improved producer task test coverage","message":"Improved producer task test coverage\n\nChange-Id: I4a3901512b31d1d13db9e75d3de26bf2dcb5ba0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b5ec3f8e296d40fcb8bb3b391af947841bd64767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b5ec3f8e296d40fcb8bb3b391af947841bd64767"}]},"branch":"refs/heads/master","description":"Rebase"},"f424245d56e8310c79b0b9a603b64b886244b95b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-12-07 23:27:44.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/07/902507/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/07/902507/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/902507/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/902507/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/902507/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/07/902507/5"}}},"commit":{"parents":[{"commit":"55cc882706668482e6a3538964b9609addc74964","subject":"Merge \"Cleaned up rpc loggingutils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/55cc882706668482e6a3538964b9609addc74964"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 21:49:07.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-07 23:27:44.000000000","tz":0},"subject":"Improved producer task test coverage","message":"Improved producer task test coverage\n\nChange-Id: I4a3901512b31d1d13db9e75d3de26bf2dcb5ba0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f424245d56e8310c79b0b9a603b64b886244b95b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f424245d56e8310c79b0b9a603b64b886244b95b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
