)]}'
{"id":"openstack%2Fdesignate~902509","triplet_id":"openstack%2Fdesignate~master~Id508f9dfd776a0017bd02db461331983e1f82646","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-12-12 23:51:57.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Id508f9dfd776a0017bd02db461331983e1f82646","subject":"Improved Notifications coverage","status":"MERGED","created":"2023-12-02 22:42:21.000000000","updated":"2023-12-13 01:07:47.000000000","submitted":"2023-12-13 01:06:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902509","meta_rev_id":"99fcff027bb19fb189ab7fc53c45eed6efcbd40e","_number":902509,"virtual_id_number":902509,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-13 01:06:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-12 18:43:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-12-12 23:51:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-02 23:58:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-12 03:54:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"3a32fc1ae9988589597466931526fb4807a6908f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-02 22:42:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bd823a15c3d8a154d78716e44292b6469c6cc82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-02 22:43:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6ae0f7984026b6ff9c745c94ce8bf2996582ab0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-02 22:44:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b367f5ffbf35a2a53cd797ab81f691f0a3aabd75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 23:58:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1ef4c2213c448338e6008dc0462dfea\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fb6dc6ec271d4d939244bbb8cc2f2512 : SUCCESS in 45m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ca1e3de33c634c728e3c01f09d61b2e5 : SUCCESS in 1h 03m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/84254e85004a46d389992c413d80b65f : SUCCESS in 53m 29s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/950bbd17a09142798b942d2189c44766 : SUCCESS in 31m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8f1e08adbacb40828c1856940c905bfd : SUCCESS in 36m 54s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/19b01467d25c4b58816b908ba16e179a : SUCCESS in 45m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7c6f1666a9524f9495845989375e3ae0 : SUCCESS in 1h 06m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/01de786afe6c4a33b85b3a6d81b27e2c : SUCCESS in 43m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/20db79aa83c049ada5712f0583da6b7e : SUCCESS in 30m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/babb4cbe57a14e1ea994aec948f8c5cc : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fba4cef4b23245fa914a5f9371f60c18 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e8838e6cc114ae6a8e7070f72dc7e48 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/49da5ab45a1843db9776f53259053b51 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/596c668299ec4c15aa8d9911043b7baf : SUCCESS in 8m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e90cbf90dffd43d0ba7f80e0ee2259fb : SUCCESS in 36m 35s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4db41fc38bb64484b996160c3501ad10 : SUCCESS in 5m 50s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/50cda45000af40289ec0abb2b9bcc0d2 : SUCCESS in 31m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/116c1ac1d4d64a8ebc208f8b792848d0 : SUCCESS in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b79af6fe1a10fd9ec616fbda9d795fdf32e4fe9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-03 01:32:34.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"fbc526bd13dbe0328cf07fd4bb86cb79e817266e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-03 02:49:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/709590a16c02478b8f8903417dd7b8c5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6aa748ba2f164801beee0ec279bc6d35 : SUCCESS in 31m 10s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/65419107db0944f49fceae17416d2b47 : SUCCESS in 36m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ffde06b033a49959e9891af19616d1c : SUCCESS in 38m 11s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e2d7e3efd07d493db791fd89f688a311 : SUCCESS in 30m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3e1d6b4043bc4db39c7cda391076943c : SUCCESS in 48m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/82b8eadaace04744a778235cf9f9bc69 : SUCCESS in 51m 56s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0ed97a85c3174e6ea0c3e90350cb058d : SUCCESS in 1h 10m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ab0197e233844a3fa1463d09d5ad8667 : SUCCESS in 28m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d16c06e751af4aaa9b8b1e118cb717f5 : SUCCESS in 32m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae975fb2bf2f4245a005da0b646e8cfa : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09ec1ca9173648f4808df1d6ff6c5f83 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97071c7c81d34a48b3500ffdaabb297f : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b1f15b6afc34683ad85782380c932d1 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/266142a36c93428e8485bd132137df7a : SUCCESS in 6m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cf644810d88a41dbbe3220060acab5e0 : SUCCESS in 31m 11s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c4f43458fadb4f1689c8ef9d87b48a8c : SUCCESS in 6m 46s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/31eb8a9ccd774f23af102e70ecccb27f : SUCCESS in 49m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ed05091a0e2341068d5c55fb34496824 : SUCCESS in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"be78f8f1ed0e87efc74343358778ad0b3d03df32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-05 11:53:41.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"bddfae5ba328b0e666e781bf974d6f304f17174a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 12:50:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/107496af7aad4d4daf5b01b5759fa30f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f7f4765de6d644bb90ab7312367a31ec : SUCCESS in 35m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0608ce7f4e4642169df7bac53af8f1c2 : SUCCESS in 40m 42s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f747906bf756498d88914bdee4073a96 : SUCCESS in 44m 56s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2bec4d31f20d400597e6b30038ace83a : SUCCESS in 41m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/226ae9726a914c7bb46481a10751e72e : SUCCESS in 37m 14s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b79c33f6826b4f4e9e4d2d92abea0d11 : SUCCESS in 54m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ae469336b22d43c98442fe4b8b2b669a : SUCCESS in 53m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e14606a499494c90be28dd52272fb06a : SUCCESS in 37m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d85009c0c5a74becb14511a2dc41b662 : SUCCESS in 35m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0f429fff4794801936415067d40961a : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/835aa2349b3d4c3e93d14f627fd96958 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/353a97ab529143958e6c431a10af1706 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0caac70af79a48e4a7eb67edd35f9684 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ef676f3623449c7bcf36379990d564b : SUCCESS in 7m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cb18ad01789c4ce8afcd96ea3718d8aa : SUCCESS in 33m 45s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/85872741eb2b42c7924db01964bc2381 : SUCCESS in 8m 05s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9169631ea1934b0995b90f67e853fac7 : SUCCESS in 50m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e548005ddb1f4761b13cc2ae4fb07116 : SUCCESS in 6m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2331f279b71a86a927f2443ead7febe979196ab1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-06 12:48:56.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"cff9cceced4d620e16170be192c587071f58377d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 14:05:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c1ba3fe15c94456960b608dd756bc14\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1510e5f501034a9ba379f9e068bd1ac2 : SUCCESS in 47m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/284bccab8d3e4359b44ee79e80b38f2c : SUCCESS in 58m 20s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5493e0a405974fc0bf5c42c92e813bc4 : SUCCESS in 56m 22s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/810366817581413b8adb1717fd7c0105 : SUCCESS in 49m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a9a4a3f934fb4534a0bfbaef60261e2f : SUCCESS in 38m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/42dae9bb240745169b16ff0f15899104 : SUCCESS in 1h 02m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3d5c66b3a06d43e59b3220caa9abc0a0 : SUCCESS in 1h 08m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b19db9e1e62a483a846b241c75702d6c : SUCCESS in 48m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bd3c7f47b20c453b837e35957df5518e : SUCCESS in 50m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e4316387198422bb6f844d8ec453c07 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8af6c469de2045df9ba9833ce98491b9 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4231608a97ab4d5289afd0481d26441a : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/99274ac98785406089a2b55425660288 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8628ae8d1dbd408a8e91ede89f3514d4 : SUCCESS in 7m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d10a4f7afe19463d944bd91a916e552c : SUCCESS in 50m 14s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/70b9e9ae4a6e47979357fbd08ad53347 : SUCCESS in 7m 44s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/14f8e893923c4a1bb980d7a55766453c : SUCCESS in 40m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3f094a7dbe71489a8db9d5f6d63b0977 : SUCCESS in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"510d82b399466a6b85eaba155f593554aa3c628a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 23:28:17.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"fb70ebd39b627f4a627f33dddb9bf8173883dd10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 00:42:27.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b071a0b6e12415586a17fefed484d1b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e31c19c9ddcd4c55bdacfe1e77d931ee : SUCCESS in 48m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3ccb3eb6aaf2451ca06363e41c6adebf : SUCCESS in 1h 01m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1582fec4483e4403841c4b80bc6cd2ec : SUCCESS in 55m 12s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b1cf1d40b199443a80063eeeb4df8c05 : SUCCESS in 37m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fc424d3be8fd4a2c8912498f42a2c0bb : SUCCESS in 36m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/dcf649b8faac405da64e9156aaf71698 : SUCCESS in 46m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b968fc4d9337451aa1b43a150f8e7960 : SUCCESS in 1h 08m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/beef95d75f174e0d94ad85f587d0066f : SUCCESS in 49m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/730f3f88a7464a0e9625848551f4da11 : SUCCESS in 42m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81ba98e9387f494585f95e13d40e2db2 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9974ba2e59eb4ab6b9c875847b65b6e6 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46541ef564a54d188695e943b7c48084 : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d4188fdb5fd42eb9dc92629c103ff22 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cab5f1013a8c49dd9a6626f5c0e96dd5 : SUCCESS in 9m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9d6a952d4dd24ce09feb17661435b4fc : SUCCESS in 32m 37s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/00b85d3387084e8bac46796fcb7d3a47 : SUCCESS in 8m 19s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/af297b4d5af7466c869aef7651e15e71 : SUCCESS in 31m 11s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6d18c38f9bdf49e58c5a9a729a2260f2 : SUCCESS in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7df5a60094fca91f6f990866d8fc14a0facfcdae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-08 13:42:26.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"abd49252a70a9e1778a1cb262fca62f43c252279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 15:01:32.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78aa284b00254f4e8e1823ca6781ac58\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6f72faca521540ddadcb96801be97e0b : SUCCESS in 37m 56s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e7436c5e11cd49799bb92c936933342c : SUCCESS in 40m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cf08a8953f54ffd8b1786aec43f0a77 : SUCCESS in 53m 28s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/588280f2d3224d59abec55778353f8ab : SUCCESS in 1h 01m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/15848bebb4d2491f8bb66660b3ae2c86 : SUCCESS in 38m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/75d95d0182d248a29b859993251385ce : FAILURE in 1h 03m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1b2d93370cf341df95c8a10c7de76065 : SUCCESS in 1h 12m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d5391ce5028c43fb8308afcb81ecd224 : SUCCESS in 38m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5b2b80ea9d7f44f398fb195aa39377ff : SUCCESS in 49m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cdf0db27baa414f8afdaa2b39e6ac61 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3225c37c0db4f8780f716035316d041 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07ae0d3c72844736854505486b4889be : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8afa195aba0445d29aba0cf54458d572 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bdc4451343d447481a98d335f547607 : SUCCESS in 10m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0a47b9fda9f14b6f93431ca9368af1d7 : SUCCESS in 32m 04s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d15613037d304dc29ac50282234c9e82 : SUCCESS in 6m 37s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/07decac4a8a04c57bd37010e75902744 : SUCCESS in 48m 32s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9c866b23f51843a39e8230462739ecc9 : SUCCESS in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8ec15437550b6b21b2cbab85776cd4169bdd5f58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-08 16:31:01.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"a232c556b61c90f56d2fcf31b424eac9d9e4cdc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 17:56:52.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd9540b5d41c42988cd504c6da2875e3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bfea3489403b4a2499db1cb4357ba3cb : SUCCESS in 44m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/76ada18c04ad4ea3a8cd1a4c8ebfddf8 : SUCCESS in 1h 00m 06s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a79855607e0d411eaa7f04ba5218f24a : SUCCESS in 55m 53s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ccc44a2da9b44c7eb6d69827e23b4cc7 : SUCCESS in 37m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1eedd9c6b12f40158499220057e7fbc5 : SUCCESS in 38m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0e00371108d449c2830f490c4ebeed61 : SUCCESS in 1h 17m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0731685b7d374cbfb359b7fc4b6a223f : SUCCESS in 1h 14m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a200e52df3b54f6e9ab709055d8f947c : SUCCESS in 52m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b91a207a1a1043cc91622aa139351690 : SUCCESS in 42m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e3517e5e6444bad815dbfd125fa54f9 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5139a529861498a9204850276e33ddf : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ba37d246e494824b554aaec5d9f2318 : SUCCESS in 6m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/186520a581714f7288a8c121f3ae3768 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698ee0ec6e7042bd9550d282d031ae06 : SUCCESS in 8m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/62f373e08aac4c709780e8172fa54b54 : SUCCESS in 46m 06s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f80f34ebcf4c48aa8ee7d92455795410 : SUCCESS in 8m 05s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f1d58cbb4a634807a0b753188cc3eb38 : SUCCESS in 49m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8415697e8eb84a23838cec0814631eb4 : SUCCESS in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a40380cf815e32697bd3e26edb6f7c48f3d6ac0a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-12 03:54:01.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8fa2e7663ba077c34cd12e0c6714e2a40dd0a816","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-12 10:14:34.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"87119d08aafc9ffb8edf26f394eae53e135bb069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 11:32:01.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4b7b74e2d2246d2a3ac711d3f54e0d1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/812b8164e4384d0c87f133994ed6d8b3 : SUCCESS in 48m 48s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a0325816b78b435e95ebba882d943353 : SUCCESS in 58m 12s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a81602845ddd49959abe9391af946ba1 : SUCCESS in 57m 15s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a853c9ed542240bcbc3b17146917d61a : SUCCESS in 49m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5ea84876281e44f88f0c8b30ebb5505f : SUCCESS in 36m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2f050e692d4b4565a996a75c8af652f7 : SUCCESS in 1h 09m 43s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e392411dd0394a6ea4ff504c3a394f3f : SUCCESS in 1h 07m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/abb995c9d3304568b217c049d43f1e47 : SUCCESS in 40m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3fb6d5ea99eb40eea861e7d51395e34a : SUCCESS in 44m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf0072c4407445b3bda159ce24bed728 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db3481f6ef9a473ab5c870dd09fe7aaa : FAILURE in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/617adf67339a48d5b9b4788e33a9bcae : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6fd7492e7b64e70b502316ce3266aef : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7517a7af029c4d30bfd4400351127d2c : SUCCESS in 9m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1665d20d7d6c4dad94ac6390bc8f7e15 : SUCCESS in 37m 40s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f6796fdad38b45ab8564093505a828fa : SUCCESS in 6m 45s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ef254ac5331b41d78b7271190288c155 : SUCCESS in 47m 17s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/55db3bedda774aeb9bcba88a4280a6d8 : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2967437b87437552d097f368e0ccc36dacc5a84c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-12 18:43:20.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"c6384bf2cb9f26caadc09ce0ba10df2da2d5358b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 20:06:26.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fab745b7991435caf3302103adbc665\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6eec9b926efc43f09f442299abb9bcbf : SUCCESS in 35m 42s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0c912cec1ec24cbbb160799518096a53 : SUCCESS in 56m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2416bbcbe2b94573bf61180773fddbb7 : SUCCESS in 54m 55s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/bd52ad0bad254d6a995357a95c37a58c : SUCCESS in 49m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7c90b56a643b463ba0985fd7f31ed0f4 : SUCCESS in 49m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/50a6d3eae98b470197c221452e003c8a : SUCCESS in 1h 12m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2a1c46371a974ae88370b6d61a0d30c7 : SUCCESS in 1h 02m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/abb27ce2f144432288ea2c1389d977d9 : SUCCESS in 47m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3aa1a3b228b54f99ba5c286c985c012a : SUCCESS in 40m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02ffa5279537486888b2c557ae9cfe94 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45310e68a02b4a5e9f07ca704ea1eec6 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e552cb62a1f343b398e4c6374fcc6774 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0755da818a449b48f3a7b5049b6c907 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18de7da1c61f4cb7a26dbb7368d446e6 : SUCCESS in 8m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/27affa95b96749a88be35a5c13e49785 : SUCCESS in 31m 34s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7ac33a5aa91a411397684ae5dd75d171 : SUCCESS in 7m 23s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c559dcea91604dc396a8cb4758a9dc97 : SUCCESS in 49m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f5e0de6eccb34c0488a696a4467c8b11 : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ffb82811bc729509e0719991cd8ff70f0bda0dfd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-12 23:51:57.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"94dc54e47a053703a54d5749d4cadca72d0c4aa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 23:56:49.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"671bb1f3c6b3faee3dc8dbcd543e5e8f1dbac0b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 01:06:37.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11a94988afc042689800009abe7ba8bd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b9f1ea0dca4f421a870d5dbb34a7e0a1 : SUCCESS in 39m 58s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3adfd634a22b4c9d92da22ac7294b0dc : SUCCESS in 34m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f5f37261f6ec401b8f858b3a2e021b96 : SUCCESS in 33m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d7c1b9fb56fa49a0b6c1faf7b8b15495 : SUCCESS in 1h 04m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f4d73aab10094a72b826c3cbb278e7d8 : SUCCESS in 51m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ab4f7b9dd6f34cd186fc7e6e579aac44 : SUCCESS in 53m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9170904b70a4d9badfb79283bb42004 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59607da9e0594bcd81d4122aa09e3f60 : SUCCESS in 7m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/000bc0638cdb44bb9d3eb5a8d0865304 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88b5225c4192426ca59fc09587d4ff35 : SUCCESS in 7m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fc9c4d9658834808838a7dba7edc68ff : SUCCESS in 32m 36s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/63a5a087256645e3a70dcee211cc7f57 : SUCCESS in 7m 03s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/913fa93c7af64bbd82fc0341f137005f : SUCCESS in 35m 39s","accounts_in_message":[],"_revision_number":11},{"id":"c9bb0fbce1c44f8aaa106149a469f14cb899ce3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 01:06:37.000000000","message":"Change has been successfully merged\n\n9 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":11},{"id":"99fcff027bb19fb189ab7fc53c45eed6efcbd40e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 01:07:47.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79d14e99510b4c64adfaddd9ffc53897\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cfa941250fe43848c8f1ed853704e78 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"7067155a7d9b30a5155b5d9cbdd6089a6b2a44f3","revisions":{"4509664e9397768d3cc94eb3e524f3e2f0317fd1":{"kind":"REWORK","_number":1,"created":"2023-12-02 22:42:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/902509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/902509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/902509/1"}}},"commit":{"parents":[{"commit":"ced07eeebf00e5774aa8c64a39eea8645e52979b","subject":"Improved MDNS Handler test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ced07eeebf00e5774aa8c64a39eea8645e52979b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 22:41:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 22:41:36.000000000","tz":-480},"subject":"Improved Notifications coverge","message":"Improved Notifications coverge\n\n- Minor cleanup to code to make it easier to read.\n- Restructured tests.\n\nChange-Id: Id508f9dfd776a0017bd02db461331983e1f82646\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4509664e9397768d3cc94eb3e524f3e2f0317fd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4509664e9397768d3cc94eb3e524f3e2f0317fd1"}]},"branch":"refs/heads/master"},"0b195378d868ceb8ddbbb133d712cf42d585b92b":{"kind":"REWORK","_number":2,"created":"2023-12-02 22:43:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/902509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/902509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/902509/2"}}},"commit":{"parents":[{"commit":"ced07eeebf00e5774aa8c64a39eea8645e52979b","subject":"Improved MDNS Handler test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ced07eeebf00e5774aa8c64a39eea8645e52979b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 22:41:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 22:42:55.000000000","tz":-480},"subject":"Improved Notifications coverge","message":"Improved Notifications coverge\n\n- Minor cleanup to code to make it easier to read.\n- Restructured tests.\n\nChange-Id: Id508f9dfd776a0017bd02db461331983e1f82646\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0b195378d868ceb8ddbbb133d712cf42d585b92b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0b195378d868ceb8ddbbb133d712cf42d585b92b"}]},"branch":"refs/heads/master"},"9ffb430624f9914b47f9cffbfb6cd582703ff138":{"kind":"REWORK","_number":3,"created":"2023-12-02 22:44:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/902509/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/902509/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/902509/3"}}},"commit":{"parents":[{"commit":"ced07eeebf00e5774aa8c64a39eea8645e52979b","subject":"Improved MDNS Handler test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ced07eeebf00e5774aa8c64a39eea8645e52979b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 22:41:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 22:44:07.000000000","tz":-480},"subject":"Improved Notifications coverge","message":"Improved Notifications coverge\n\n- Minor cleanup to code to make it easier to read.\n- Restructured tests.\n\nChange-Id: Id508f9dfd776a0017bd02db461331983e1f82646\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ffb430624f9914b47f9cffbfb6cd582703ff138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ffb430624f9914b47f9cffbfb6cd582703ff138"}]},"branch":"refs/heads/master"},"2dd37cc92f0dc92f4495a1e7d0e33311e06ca020":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-12-03 01:32:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/902509/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/902509/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/902509/4"}}},"commit":{"parents":[{"commit":"ced07eeebf00e5774aa8c64a39eea8645e52979b","subject":"Improved MDNS Handler test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ced07eeebf00e5774aa8c64a39eea8645e52979b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 22:41:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-03 01:32:33.000000000","tz":0},"subject":"Improved Notifications coverage","message":"Improved Notifications coverage\n\n- Minor cleanup to code to make it easier to read.\n- Restructured tests.\n\nChange-Id: Id508f9dfd776a0017bd02db461331983e1f82646\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2dd37cc92f0dc92f4495a1e7d0e33311e06ca020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2dd37cc92f0dc92f4495a1e7d0e33311e06ca020"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fbae0a92f173a2d9a5720cef6babee744adff9f1":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-12-05 11:53:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/902509/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/902509/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/902509/5"}}},"commit":{"parents":[{"commit":"98f1ad2856a0acb9081c5449ab13718714b42920","subject":"Merge \"Improved MDNS Handler test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/98f1ad2856a0acb9081c5449ab13718714b42920"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 22:41:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-05 11:53:41.000000000","tz":0},"subject":"Improved Notifications coverage","message":"Improved Notifications coverage\n\n- Minor cleanup to code to make it easier to read.\n- Restructured tests.\n\nChange-Id: Id508f9dfd776a0017bd02db461331983e1f82646\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fbae0a92f173a2d9a5720cef6babee744adff9f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fbae0a92f173a2d9a5720cef6babee744adff9f1"}]},"branch":"refs/heads/master","description":"Rebase"},"32b026652856285f9fe9fba437eccfee88e0befc":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-12-06 12:48:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/902509/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/902509/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/902509/6"}}},"commit":{"parents":[{"commit":"4a84b9c0c2b91d154b85b5c5c1a2fff3612949bc","subject":"Merge \"Fix RPC tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4a84b9c0c2b91d154b85b5c5c1a2fff3612949bc"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 22:41:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-06 12:48:56.000000000","tz":0},"subject":"Improved Notifications coverage","message":"Improved Notifications coverage\n\n- Minor cleanup to code to make it easier to read.\n- Restructured tests.\n\nChange-Id: Id508f9dfd776a0017bd02db461331983e1f82646\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/32b026652856285f9fe9fba437eccfee88e0befc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/32b026652856285f9fe9fba437eccfee88e0befc"}]},"branch":"refs/heads/master","description":"Rebase"},"86205eeb8bdb51f9f46dfc8b20046b55d8f23714":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-12-07 23:28:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/902509/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/902509/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/902509/7"}}},"commit":{"parents":[{"commit":"55cc882706668482e6a3538964b9609addc74964","subject":"Merge \"Cleaned up rpc loggingutils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/55cc882706668482e6a3538964b9609addc74964"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 22:41:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-07 23:28:17.000000000","tz":0},"subject":"Improved Notifications coverage","message":"Improved Notifications coverage\n\n- Minor cleanup to code to make it easier to read.\n- Restructured tests.\n\nChange-Id: Id508f9dfd776a0017bd02db461331983e1f82646\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/86205eeb8bdb51f9f46dfc8b20046b55d8f23714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/86205eeb8bdb51f9f46dfc8b20046b55d8f23714"}]},"branch":"refs/heads/master","description":"Rebase"},"ec63b86ccf3ece2af5f02e187ecfe4365617741a":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-12-08 13:42:26.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/902509/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/902509/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/902509/8"}}},"commit":{"parents":[{"commit":"060a17ac655825d71d62b25ac891ee62cfc27012","subject":"Fix occasional failure in Service tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/060a17ac655825d71d62b25ac891ee62cfc27012"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 22:41:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-08 13:42:26.000000000","tz":0},"subject":"Improved Notifications coverage","message":"Improved Notifications coverage\n\n- Minor cleanup to code to make it easier to read.\n- Restructured tests.\n\nChange-Id: Id508f9dfd776a0017bd02db461331983e1f82646\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ec63b86ccf3ece2af5f02e187ecfe4365617741a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ec63b86ccf3ece2af5f02e187ecfe4365617741a"}]},"branch":"refs/heads/master","description":"Rebase"},"0ba66a9d0353eaf0a655140d0ab35f58b26a05a6":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-12-08 16:31:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/902509/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/902509/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/902509/9"}}},"commit":{"parents":[{"commit":"bdbe0715d85cdd658abc7d596ebba92cc1d4eff3","subject":"Standardized Service unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bdbe0715d85cdd658abc7d596ebba92cc1d4eff3"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 22:41:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-08 16:31:01.000000000","tz":0},"subject":"Improved Notifications coverage","message":"Improved Notifications coverage\n\n- Minor cleanup to code to make it easier to read.\n- Restructured tests.\n\nChange-Id: Id508f9dfd776a0017bd02db461331983e1f82646\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ba66a9d0353eaf0a655140d0ab35f58b26a05a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ba66a9d0353eaf0a655140d0ab35f58b26a05a6"}]},"branch":"refs/heads/master","description":"Rebase"},"19e82537edcf29208abaa7ee9e85d3d406b1559f":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-12-12 10:14:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/902509/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/902509/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/902509/10"}}},"commit":{"parents":[{"commit":"c3688cbea4f0a2893fab36ab3c953c1e8a7830e2","subject":"Standardized Service unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3688cbea4f0a2893fab36ab3c953c1e8a7830e2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 22:41:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-12 10:14:34.000000000","tz":0},"subject":"Improved Notifications coverage","message":"Improved Notifications coverage\n\n- Minor cleanup to code to make it easier to read.\n- Restructured tests.\n\nChange-Id: Id508f9dfd776a0017bd02db461331983e1f82646\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19e82537edcf29208abaa7ee9e85d3d406b1559f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19e82537edcf29208abaa7ee9e85d3d406b1559f"}]},"branch":"refs/heads/master","description":"Rebase"},"7067155a7d9b30a5155b5d9cbdd6089a6b2a44f3":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-12-12 18:43:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/902509/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/902509/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/902509/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/902509/11"}}},"commit":{"parents":[{"commit":"db5766e883339d11af399c29d4350f242afdb197","subject":"Standardized Service unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/db5766e883339d11af399c29d4350f242afdb197"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 22:41:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-12 18:43:20.000000000","tz":0},"subject":"Improved Notifications coverage","message":"Improved Notifications coverage\n\n- Minor cleanup to code to make it easier to read.\n- Restructured tests.\n\nChange-Id: Id508f9dfd776a0017bd02db461331983e1f82646\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7067155a7d9b30a5155b5d9cbdd6089a6b2a44f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7067155a7d9b30a5155b5d9cbdd6089a6b2a44f3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
