)]}'
{"id":"openstack%2Fdesignate~902517","triplet_id":"openstack%2Fdesignate~master~I0dd7c77def9a9d41a518b92b9a662a0e1fce1e3e","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-12-07 14:06:11.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I0dd7c77def9a9d41a518b92b9a662a0e1fce1e3e","subject":"Remove unused rpc calls","status":"MERGED","created":"2023-12-05 14:40:43.000000000","updated":"2023-12-07 15:29:30.000000000","submitted":"2023-12-07 15:27:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902517","meta_rev_id":"1f53912add883958513fd173f06f857ec2abad89","_number":902517,"virtual_id_number":902517,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-07 15:27:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-07 04:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-12-07 05:54:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2023-12-07 14:06:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-05 16:00:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-07 04:32:48.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-12-07 05:54:48.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"4989b1bb02e490a749cc2d25c0dac01bdbd8419c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-05 14:40:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"147a16df456d25c679806c6d35372c212a05348c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-05 14:42:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c6b7b3caf5fb5d8454e8c0dc5be416c4de591f32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 16:00:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40174240165b4573a2ab40fe079751e7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/486c708084154621980ddf3c71485998 : SUCCESS in 42m 37s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e855af17d8394dc199c0cec039c6061d : SUCCESS in 50m 00s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d94a83ab6d049da93d4d4e7d1bb137c : SUCCESS in 45m 40s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0405b0430a34442d8b1a021c7ea5819b : SUCCESS in 50m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f7a162abde2c42c99a9153f4982eb78d : SUCCESS in 31m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e73b79084d9f43558866f86458ca57e3 : SUCCESS in 1h 12m 25s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/11c9447c336144da9a93ebd70e8d2544 : SUCCESS in 1h 10m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/926846b9c9154c7c9dc1d96e921d2080 : SUCCESS in 38m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/620bb8752f344f6294ad7031dc6e64a7 : SUCCESS in 32m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/390797339b3246b1a8e1eee625030f1e : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc18b5202dbe4e008d83442fce96e87c : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be5d7c93e96c449489db20983975be10 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ed45dc8301d435c8b16d41b0d895135 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/824e37114b0246698fccee4096f221aa : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0140a79500bc47ffbfdcc5ab65b13b93 : SUCCESS in 6m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b59deef8a24c42db979ec5a0560f0a7b : SUCCESS in 41m 51s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f8326433cdd24eebb06885dbd978fd74 : SUCCESS in 6m 02s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7cf6b51fdb2542db8ac0c2a9a483b436 : SUCCESS in 50m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e336d09b80694314b9884aa888f473f5 : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"37139420fa3dce27e8cb4db7297566ed365f0e67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-06 12:49:07.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9ddf24fda243a57f5ffb7120d5192ac149959326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-06 14:00:57.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f4b9be13de781040cc641a0cf05a2d7a4fa8f725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 15:11:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/197641269d074177b9c457eb877f34c3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f0f795e0786e449fb1bca15687c6a193 : SUCCESS in 37m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c17e934f04734612b8e7a9e088580b9f : SUCCESS in 56m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2420a8d46b14866858340fdcc154781 : SUCCESS in 59m 28s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e3843fe90188446889bd3b4d35f798db : SUCCESS in 50m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/956d7ba5467e47b09f892e324ac97e7d : SUCCESS in 37m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d951ec6ce62d4e4fa8eaf15a09ae08dd : SUCCESS in 1h 04m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5682e29e68c948bf9561d4043eaf15e4 : SUCCESS in 58m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6f638553fc44497e8332cf9e393d3286 : SUCCESS in 48m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e5b1c8bcf0644703bf1ec2c167fb52bf : SUCCESS in 33m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b8c064b11494088973adeb20330ca2e : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b392f135cf464c81a4a2854686a582ed : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42230d4eda7042e78ecd095fec38b409 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a11b8ad0fc64ed099bf3b0a34be1bd8 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a50ae731e6c144bcb101b97d6e986a84 : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a7d87d98199457c90b32730c51233ba : SUCCESS in 5m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5edae2b9d16d478c83ed2dc2fc3a3e00 : SUCCESS in 42m 01s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b1735b94782948acb08f5553ef4dc5fb : SUCCESS in 6m 51s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7580b7d60941468c90c91b76eb2c7ed0 : SUCCESS in 50m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2802f778cd9c40fd918b49d67d3bfd5b : SUCCESS in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cbeaf200e51753f6903b3e4cc189a33539de6069","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-07 04:32:48.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9485dc014abd0a6b43046c999a680c0bc7da8ef8","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-12-07 05:54:48.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a2736a4d691eac0fbec4f2a94a43f31e88963266","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 14:06:11.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3c530634d5ac3a5c4895533cab5d9f917194ce26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 14:06:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4e3a4e022344971094b76e9e44b56eec10c42772","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 15:27:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b4c6b27cc1749198f16086118c9d835\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7a88a8e0163f4f2dab81680398d9fe9c : SUCCESS in 52m 37s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1befc01f3bfc427fa2920c1f86ee2306 : SUCCESS in 47m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/55971ffc50054b40b46daf7111699f96 : SUCCESS in 1h 00m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/af7737a1df8b4bc3a871512ebec12964 : SUCCESS in 1h 09m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fac5b17c435d4a748ee4bda5400d4da5 : SUCCESS in 49m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/437d62d2ee6f479eb5a827d9c9f0b929 : SUCCESS in 55m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d92ecbc03844823aeb7b09352958526 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b9e1c7453024093a26e7215efce006f : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf6334c45a194ba2a7faec88e9774186 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6baa98b710843e6bb5513913cd8fb0c : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/daabf24b374d4a77b87ceb2cf10548cd : SUCCESS in 4m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/87de1f0f3d2c48f69294b6891dd4155a : SUCCESS in 46m 38s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0c4dd7099c3848c589d540ce3adb8b11 : SUCCESS in 9m 22s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/379c0ec5c35e4c97beab1b21f8884f7f : SUCCESS in 54m 16s","accounts_in_message":[],"_revision_number":4},{"id":"94d5225abd3bad2aae5246eaa819dcb17d1b6a58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 15:27:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"1f53912add883958513fd173f06f857ec2abad89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 15:29:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acf23784afb04bf1b26b399f000adbde\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08938a617f444a7c8187b4ce72eff39d : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fca957c4a544d43878d10a370715e35 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f83ebe6a0edc1aa1ea762635c3df6bf166075464","revisions":{"232c8fce8edea7078260bfd3e4fe709228b38db8":{"kind":"REWORK","_number":1,"created":"2023-12-05 14:40:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/17/902517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/902517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/902517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/902517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/902517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/902517/1"}}},"commit":{"parents":[{"commit":"c08eeadba630faa0cb2edfcd2458efc6ef1bc9cd","subject":"Cleanup Central tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c08eeadba630faa0cb2edfcd2458efc6ef1bc9cd"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-05 14:37:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-05 14:40:17.000000000","tz":-480},"subject":"Remove unused rpc calls","message":"Remove unused rpc calls\n\nThese RPC calls were only used internally and does not\nneed to be exposed as RPC calls.\n\nThe following rpc calls were removed.\n- find_records\n- find_recordset\n- update_zone_import\n\nChange-Id: I0dd7c77def9a9d41a518b92b9a662a0e1fce1e3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/232c8fce8edea7078260bfd3e4fe709228b38db8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/232c8fce8edea7078260bfd3e4fe709228b38db8"}]},"branch":"refs/heads/master"},"a0322c036c8f83ea546a0b28074201f4b6d92f37":{"kind":"REWORK","_number":2,"created":"2023-12-05 14:42:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/17/902517/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/902517/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/902517/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/902517/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/902517/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/902517/2"}}},"commit":{"parents":[{"commit":"c08eeadba630faa0cb2edfcd2458efc6ef1bc9cd","subject":"Cleanup Central tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c08eeadba630faa0cb2edfcd2458efc6ef1bc9cd"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-05 14:37:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-05 14:42:02.000000000","tz":-480},"subject":"Remove unused rpc calls","message":"Remove unused rpc calls\n\nThese RPC calls were only used internally and does not\nneed to be exposed as RPC calls.\n\nThe following rpc calls were removed.\n- find_records\n- find_recordset\n- update_zone_import\n\nChange-Id: I0dd7c77def9a9d41a518b92b9a662a0e1fce1e3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0322c036c8f83ea546a0b28074201f4b6d92f37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0322c036c8f83ea546a0b28074201f4b6d92f37"}]},"branch":"refs/heads/master"},"ca53677e630faa3165eff3f26f5a844d62ecb9e3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-12-06 12:49:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/17/902517/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/902517/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/902517/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/902517/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/902517/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/902517/3"}}},"commit":{"parents":[{"commit":"83615b6908c670a0c24daea042e0e20468bf886e","subject":"Cleanup Central tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/83615b6908c670a0c24daea042e0e20468bf886e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-05 14:37:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-06 12:49:07.000000000","tz":0},"subject":"Remove unused rpc calls","message":"Remove unused rpc calls\n\nThese RPC calls were only used internally and does not\nneed to be exposed as RPC calls.\n\nThe following rpc calls were removed.\n- find_records\n- find_recordset\n- update_zone_import\n\nChange-Id: I0dd7c77def9a9d41a518b92b9a662a0e1fce1e3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca53677e630faa3165eff3f26f5a844d62ecb9e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca53677e630faa3165eff3f26f5a844d62ecb9e3"}]},"branch":"refs/heads/master","description":"Rebase"},"f83ebe6a0edc1aa1ea762635c3df6bf166075464":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-12-06 14:00:57.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/17/902517/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/902517/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/902517/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/902517/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/902517/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/902517/4"}}},"commit":{"parents":[{"commit":"4a84b9c0c2b91d154b85b5c5c1a2fff3612949bc","subject":"Merge \"Fix RPC tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4a84b9c0c2b91d154b85b5c5c1a2fff3612949bc"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-05 14:37:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-06 14:00:57.000000000","tz":0},"subject":"Remove unused rpc calls","message":"Remove unused rpc calls\n\nThese RPC calls were only used internally and does not\nneed to be exposed as RPC calls.\n\nThe following rpc calls were removed.\n- find_records\n- find_recordset\n- update_zone_import\n\nChange-Id: I0dd7c77def9a9d41a518b92b9a662a0e1fce1e3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f83ebe6a0edc1aa1ea762635c3df6bf166075464"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f83ebe6a0edc1aa1ea762635c3df6bf166075464"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
