)]}'
{"id":"openstack%2Fdesignate~903110","triplet_id":"openstack%2Fdesignate~master~I791295e7500a75a3a1ee28a7d796163a77591a9d","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-12-16 21:55:25.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I791295e7500a75a3a1ee28a7d796163a77591a9d","subject":"Increase minimum test coverage to 96%","status":"MERGED","created":"2023-12-08 20:02:52.000000000","updated":"2023-12-17 00:45:02.000000000","submitted":"2023-12-17 00:42:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903110","meta_rev_id":"47b69930e262bd23f4d6821fc6328376a9f86a71","_number":903110,"virtual_id_number":903110,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-17 00:42:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-16 11:32:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-12-16 11:32:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-08 21:22:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-15 23:48:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"5f47a50c9317bc385c4d31b7514a1696f4ea185f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-08 20:02:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4627988755253b8cf29ce5022aeb1a1de5b2b9fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 21:22:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e05a76bb3c9469fa7389c3f94d6c52a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/83108bbacfbf43ecb3041fb63306e253 : SUCCESS in 4m 10s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/87048fe8a838461690e59c7365971028 : SUCCESS in 46m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c75f7890b13f4779b9eda39a50151de7 : SUCCESS in 59m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/79855e39c3994c6192a903bdb364a8a3 : SUCCESS in 55m 06s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9fdcf13dd6eb4db4a6ccb71c1a22c457 : SUCCESS in 53m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c538412e8df44e359d87389bf774327f : SUCCESS in 55m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d1041b7c7d6c4bd29c4aa6684acfeef7 : SUCCESS in 1h 13m 29s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/94178ab96b7340d49d1bf856efb47597 : SUCCESS in 1h 03m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6de78c71258b45baa60e1ea15838cb48 : SUCCESS in 45m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/610f8a7d4af14f4abbacfd3be6df3425 : SUCCESS in 48m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd1bda985f0e4654a0080aa769a4a3be : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78e11d77812b492fb88f42e9c56425d3 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a1f1613eca5401cbc9ae6c78c302197 : SUCCESS in 7m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd3a035efc9448d29509d6294039b4a3 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3dffa39dce4691bf8fed9c666b5e42 : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/944293b483384b3f94c0d5363e66d567 : SUCCESS in 4m 52s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/dc698f6986434898bb4c3ed181ad107a : SUCCESS in 8m 48s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9fd203b1d09341ec8b4c6b489c88206c : SUCCESS in 39m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e5756943c9b14f7385b0f06b893842c3 : SUCCESS in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bcbea1fc37f69ebe1766590cd3913dbfbe786e4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-09 19:06:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3dd84ff3887205ba7a48afd96f747a614bcbafe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-09 20:29:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88ac6084e5e242b4a676bda7593fdf1a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b9a4ee3b1d83466b8a1555f76d3f585c : SUCCESS in 4m 56s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/355492869a114f5287868e3d846ff694 : SUCCESS in 31m 07s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/77fc48c1ae6f4e379188b14985f60d36 : SUCCESS in 39m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/594185f0021f4ebcb38eb8d359fe6b58 : SUCCESS in 40m 47s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/633578d4d20a4adea3984800cff51efb : SUCCESS in 34m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ec405eb093e247d4af6778f81f9221ba : SUCCESS in 30m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/993d1faa61f64f8daf0ec35fef2ea8c5 : SUCCESS in 50m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c8dd64503a5347db92a60df28eede7ae : SUCCESS in 1h 14m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4855405372e746cdb3f26ff779ed75bd : SUCCESS in 56m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9408bb3ed15d4894957d3c2ba8136df0 : SUCCESS in 45m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbb23a1da5c34cb89f89558509d58bde : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e07562401d4a4ec58a92c7bd1ec59a2a : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4e43814ec3c47ff9877a14b21cf356a : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ccaef27e392140cd9f5047ca4a67968d : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eef451f31af4a8486b14810ccc6ddce : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcdcfd7698524b59a4f1763c92589ba6 : SUCCESS in 5m 54s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f7fc3263005c4797aba0b83a2683c00f : SUCCESS in 7m 14s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5212959a71c645209d96b27119127cad : SUCCESS in 43m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/912ac980f11541e3aa97b770b6753c41 : SUCCESS in 6m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ea2aa11f99baf61592bbe245b19703e5cf89274b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-12 10:15:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c81cd9f04bf59eb6a0e33ae9ecb656d8b4dc44ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 11:31:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02ca89c819e3429f9e595e7186248d69\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8aa60bbfc8904a57b6620c1d6b3f8746 : SUCCESS in 3m 57s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d8b1308cb44442f187d30d75af11b243 : SUCCESS in 42m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/569257056e5641f59da71e924f9f3974 : SUCCESS in 48m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca0427f475cb4f7fa2196d7896bb882d : SUCCESS in 55m 19s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ae09daf57bb04e889122e977f1faff65 : SUCCESS in 38m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4a4ea7f9ac1b4b2ea899f26e9ce890bd : SUCCESS in 38m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f0d686e9222a414899ad777484707156 : SUCCESS in 1h 05m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ce6842f7df8447eb9c5c63381e452684 : SUCCESS in 1h 09m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7de35609b6824c709cea947620f5877b : SUCCESS in 49m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a1425ad0fd66435db6c3cfa6c9afcd99 : SUCCESS in 40m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8c3240d13194fb788d7994079bb90a9 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9e26e62881e4b94b67bff5d09cad596 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fdc4662e4614518bf27a52f68a99558 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c1a104f22bc437ab4657cf1cd99f443 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df8c5f32e5fb4366b30ef7886cb5b46f : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f90304155504159a537354aa1f9ddc9 : SUCCESS in 5m 08s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/56b0e411686c4935ac5c6b2429ce4b24 : SUCCESS in 9m 29s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9e569e9a73dd46648b8687b325cdaa69 : SUCCESS in 33m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a3bff19fcfc44f6b952874ea0f99f89c : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7fc4c525eabeb225d5517e3d83c0f483db5abc09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-13 11:57:43.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b505011cccb8c5aac898d91df9043bfbaf49f076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 13:02:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48a0ee426cab413d9586dff59a64910e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/15aaab02cac349b898a75c3484871073 : SUCCESS in 6m 06s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8bb4b65931914ca9ab3d79a16ad4f4ed : SUCCESS in 47m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0d381efc91874c92a34595de9a6f5a40 : SUCCESS in 55m 37s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/230c5a95ec384bbfbbeb0eec9b37c959 : SUCCESS in 42m 28s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/67adcebc4e834f8d8d6574f1b6f4e5a5 : SUCCESS in 47m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b32bee52d7ba4d68b64cf6f1855751c5 : SUCCESS in 34m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e608892fc7f84a80bd15b86890610310 : SUCCESS in 1h 00m 43s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2d532009b73341f6847ce8f7f16ab2d8 : FAILURE in 1h 02m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1413fce08181419c86e5f8fac9aaf150 : SUCCESS in 39m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e4381f1ba916459ea1899595d09aaba8 : SUCCESS in 45m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5079b27c044c4b91a578d13ba975f1ba : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a7c58b2ab2d4ac0b6891ec0bb567df8 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe306967c25b4cf2950f8ea173f61e2d : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6982983f858e465a9e2041c874944e47 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c158fd9c8aaa4c759d8de3d583f42eaf : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc75087dbd804133b4d84d7d5e63fd65 : SUCCESS in 5m 40s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/aafbbbbba3fb49a0a768a4e665e4f671 : SUCCESS in 6m 37s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3c27f4eb027741d7b81e83445968296e : SUCCESS in 39m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/447f077196e34fec98619fc50b754d94 : SUCCESS in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5491e3d4a0a3e0fd63a1a57df60a00eccc73f2a4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-13 15:22:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"57c94bde944f926645788e0d0bc959790f42dd56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 16:39:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cd15cab1d534d3aa85d57ccc94348e5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d0dff19e4c5f49de8a1c621f599ac669 : SUCCESS in 4m 10s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/66ed6b8da9364cbc906627a2476ee5b4 : SUCCESS in 48m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/256f28cb045b4f61bb8b7fe2548ba0b8 : SUCCESS in 57m 14s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/170b1c749a17424fba77463aaa8ea798 : FAILURE in 40m 32s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/09579302e3034bfaae91622a28e3ada2 : SUCCESS in 39m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f7e547e0ef5b41e6b0dba01617f09fdb : SUCCESS in 48m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4c06c444604247aebc75d146e0498b03 : SUCCESS in 1h 10m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/20d2fe49df4f47259222147a5fabe11d : SUCCESS in 1h 05m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7f3e00959be14940aa8a45ae1a06b4dc : SUCCESS in 42m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3166faad2678482fa2eba8b9c5c1bbdd : SUCCESS in 42m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3c13fcce86840d3b0ce5cc32a96f0af : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fddc9b45683f442b90f267c8aaa0ec89 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d492cc3a08f8420d9443b160153fc350 : SUCCESS in 6m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94b096554dff4447b2141abd6561afd3 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33c780217c0d4d5a9b00b81e2a8ac05d : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85bd86a8d77d4f90880fc5525f032e23 : SUCCESS in 5m 49s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/182e95554a5c4ca79b1e82a2828fee95 : SUCCESS in 9m 29s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/77dc17015ac04637b2b53df69d4e3e35 : SUCCESS in 39m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/71a94d199480498ea1090de8d045fa6a : SUCCESS in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6be01f12789798eeaeaf2989feff9e94da865726","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-15 23:48:44.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"582b42f8ffd6678e7cf7cea8da52b41ccf595eab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 11:32:50.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"1ad41a6ec3b0515b493274a89878205356920782","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 11:32:54.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"737b4e1c89c81ae10b168871b70a4d4759aa8f3a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 11:39:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8ca59bb7b854341ff7c94b49415e5bac8ee8b7bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 12:53:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/33ea73b5305d4bd3ae3df092e26c2d4d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2fafb77f0559418db1d73b64536cd391 : SUCCESS in 4m 27s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c139b54c44dc4c65aea8cd7b2f67fa39 : FAILURE in 31m 42s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a8dd35ab9b50465fbcec822d05ef5105 : FAILURE in 59m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f5fb5da9da54572b2485a28e49c4548 : FAILURE in 56m 32s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f9ebf2878fcb4386be33a0b60b07a41d : FAILURE in 50m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/228235acb3864116ac7ae27c578f1eb8 : FAILURE in 48m 54s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/01c40eeea7774573a0afc6fe111e41fe : FAILURE in 48m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ebb148ee5a19491aad6d83b001c0423c : FAILURE in 48m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/37d8328a5c5f4e0f8d67be297542f68c : FAILURE in 46m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/95b7bace61604df4b97528006d6e7aa4 : FAILURE in 49m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39e6b0fbbc7945099ab857de1f4d3bdd : FAILURE in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b825fe4293c47649fe550dec922ac8e : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25c1c0a027b54f6ba0ed0f047e63b77c : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d0fa9d5f0fa4cc895d4d1bc768fda48 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153aedcfe03b4547811dda4b2c5ab2f7 : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05ab359d0edc4c7ca54b3ec4e7ec08ec : SUCCESS in 5m 11s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/70d25c44f99c4b0a8617f19ed68c4f12 : SUCCESS in 7m 53s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8332aaa987274048a5ddd52f0950589e : FAILURE in 49m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b281987cf0a04e93845018e638361165 : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6361f84386205a5ee27f22c79b5862a7ff16b8c2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 17:28:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6bf4c7ebec9f8b59a27c001325efd1984a46645c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 19:11:17.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/26f1ab7eba3143bca9b45c83de546336\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/31029abf0f954c5e96240c546ff2a5ba : SUCCESS in 5m 19s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cdd582bd182b4a5e9a16b73201438eaa : SUCCESS in 51m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0518a7e8e42d4ba3964d213327ac36d8 : SUCCESS in 49m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a1792ecbbff4862ac2e146b713742e1 : SUCCESS in 49m 06s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e5bb7e796ffb4174ba8d8818aea0dfdb : SUCCESS in 55m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e3e134ba9d3f41eeaa1aae1e4357f721 : FAILURE in 47m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f4571cae89b14c4db3ea6feb0433f594 : SUCCESS in 1h 09m 25s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/08040caf00b44290b9aa1888d7653a5b : SUCCESS in 1h 21m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/75fe38d513ba4b08b349c704fdc33179 : SUCCESS in 43m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8f8e53b8fb7541bdb85430d3d81efc38 : SUCCESS in 45m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3f36fd7b1294b969bba556a30f2352b : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/becb74b1bd334ff791fb4a2f10dd2e67 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71ab536d14f048ccb16df4e446216733 : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6610ee1e5bc4d7dbe014eeffbabd067 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50d0adc7fbb147aeb268544c137e97ca : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc56c6ea8d0e4196acc477df82409d1c : SUCCESS in 5m 17s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fd6a0360f1d94f13aecfb921861e0350 : SUCCESS in 7m 59s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/70b3d9200f7145f5afa5e202965f3d33 : SUCCESS in 48m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1a881d80a7234f4aa4de5b05c4c77289 : SUCCESS in 4m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a227400e29f8bb5b87d9d93443c259b1e7130584","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 21:55:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b1723948d7ae7d1c6f70e288e7fb94a5e2c10ac8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 23:20:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/245cfb3a0a4b4cdfa6ea80242dfa7439\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/48aa82046eac4b0e91b2b3e321d6497a : SUCCESS in 3m 49s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e7bc46308b9a4d87bdfde69d150ff182 : SUCCESS in 53m 37s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/dfbf4e5cc2f8402dbd251f507fdde8e7 : SUCCESS in 1h 04m 14s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0dce1a428f143bcbdad211e5855801f : SUCCESS in 49m 17s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/bda909ab26974e999f7a45136620bd44 : SUCCESS in 57m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4d4e04b30af54fee8fc9fc0990dfceb2 : SUCCESS in 47m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7954ed51737d48159caaf4fe6a5956e0 : SUCCESS in 1h 13m 52s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4948fc953b974443a019454db6bf37ac : SUCCESS in 1h 12m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/30e4890d9a384d7daaf620bf2bfa2315 : SUCCESS in 36m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bece1f565e8246dea2622a5156c03330 : SUCCESS in 42m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f04822aabe1746c399e0d267a2a0066b : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/356df80a5fea449d8c5a074968e7475b : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e421dcc5315c4665a4b8b48da843e11a : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8300b645baea42a6be1ac4f1d296ebeb : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e683328917844efaa8d0ce05bf3fea9 : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae69a0ed4e154fdabf5c3fb31abfbfab : SUCCESS in 6m 25s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6972f43b15994de68f363d65f6ec3653 : SUCCESS in 10m 00s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a6d639ea712b47908f6adabe305bca34 : SUCCESS in 52m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c6e1b03ce5454d949aeef8017bfefc6c : SUCCESS in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9c1f95b070dc0798742f2fb07a65f85374c5477a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 23:21:10.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ad3c152c062f6d8704f89a0947e8f44c594d9ef0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 00:42:21.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b10c263d61d4e44b5efc3745c47629c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/97bb28c06b62466b87994cd16d2305c5 : SUCCESS in 4m 04s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d4c673c658084cb7b3a26019473ebd07 : SUCCESS in 43m 18s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1ec5ff5fc95e43a29f5e05028bfb1b54 : SUCCESS in 44m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6f2158ce75054cd48779f1b541c33bec : SUCCESS in 50m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/77583f59143043fc920c952bfd3c8d29 : SUCCESS in 1h 07m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/07f95e68bb8c4461b47c944bd0dfeaad : SUCCESS in 48m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e3e61b3b74b2494ebeea23276615c42c : SUCCESS in 53m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ff21151bef74f7c88061709f679c498 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e0bd2d4841d4bc2beab9ad8947fa69f : SUCCESS in 6m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2dc499a89dec4e1b8394a5836338cc88 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/810a0a17ed654187bc91e3ef5b579be2 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be6ffce72e6d4c639a8ad00db63d421d : SUCCESS in 7m 02s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c32f9d072d58498d8a7d9c77e4b2730d : SUCCESS in 8m 31s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b2a0ddce250a4811af27a8201b9bae7e : SUCCESS in 51m 28s","accounts_in_message":[],"_revision_number":5},{"id":"9a567de9f29c8c5f9291ec5dc7f2726e6f986a68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 00:42:21.000000000","message":"Change has been successfully merged\n\n4 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":5},{"id":"47b69930e262bd23f4d6821fc6328376a9f86a71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 00:45:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a4be817c6c5426aab5a725ffd68df14\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/04136a59fdac49a3838cc003f55e4b66 : SUCCESS in 1m 22s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e8a925c369a44a3b66df95a557854a0 : SUCCESS in 1m 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01a2854127434987951b7b3127eba294 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"af940fb6e4431c9eb9799cc983a85465ec865011","revisions":{"0c15817959eec93671de87d2e297ee7712670b5c":{"kind":"REWORK","_number":1,"created":"2023-12-08 20:02:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/903110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/903110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/903110/1"}}},"commit":{"parents":[{"commit":"e2db8c14636f36fdeb41e8211d898dddbcd69d66","subject":"Improve API Controllers coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e2db8c14636f36fdeb41e8211d898dddbcd69d66"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-08 20:02:28.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-08 20:02:28.000000000","tz":-480},"subject":"Increase minimum test coverage to 96%","message":"Increase minimum test coverage to 96%\n\nChange-Id: I791295e7500a75a3a1ee28a7d796163a77591a9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0c15817959eec93671de87d2e297ee7712670b5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0c15817959eec93671de87d2e297ee7712670b5c"}]},"branch":"refs/heads/master"},"dbc530489d0272604a60ed7d2943909f1247b744":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-12-09 19:06:35.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/903110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/903110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/903110/2"}}},"commit":{"parents":[{"commit":"a3d95ee5e0a258d2368d5fc764d7c3965e7d8688","subject":"Remove unused infoblox code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a3d95ee5e0a258d2368d5fc764d7c3965e7d8688"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-08 20:02:28.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-09 19:06:35.000000000","tz":0},"subject":"Increase minimum test coverage to 96%","message":"Increase minimum test coverage to 96%\n\nChange-Id: I791295e7500a75a3a1ee28a7d796163a77591a9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dbc530489d0272604a60ed7d2943909f1247b744"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dbc530489d0272604a60ed7d2943909f1247b744"}]},"branch":"refs/heads/master","description":"Rebase"},"f37b21ab7546c430be062b9bfbe5213103185a46":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-12-12 10:15:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/903110/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/903110/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903110/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903110/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903110/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/903110/3"}}},"commit":{"parents":[{"commit":"20fb21c6e3c3e5e3811e06fc6155d85ac6e27b9d","subject":"Remove unused infoblox code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20fb21c6e3c3e5e3811e06fc6155d85ac6e27b9d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-08 20:02:28.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-12 10:15:10.000000000","tz":0},"subject":"Increase minimum test coverage to 96%","message":"Increase minimum test coverage to 96%\n\nChange-Id: I791295e7500a75a3a1ee28a7d796163a77591a9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f37b21ab7546c430be062b9bfbe5213103185a46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f37b21ab7546c430be062b9bfbe5213103185a46"}]},"branch":"refs/heads/master","description":"Rebase"},"c2282f1e47cd4ca376d164f47e0db57c092b6838":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-12-13 11:57:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/903110/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/903110/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903110/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903110/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903110/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/903110/4"}}},"commit":{"parents":[{"commit":"d90691f3a29416dbb209b52aae288a5916bca632","subject":"Improved service coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d90691f3a29416dbb209b52aae288a5916bca632"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-08 20:02:28.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-13 11:57:43.000000000","tz":0},"subject":"Increase minimum test coverage to 96%","message":"Increase minimum test coverage to 96%\n\nChange-Id: I791295e7500a75a3a1ee28a7d796163a77591a9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c2282f1e47cd4ca376d164f47e0db57c092b6838"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c2282f1e47cd4ca376d164f47e0db57c092b6838"}]},"branch":"refs/heads/master","description":"Rebase"},"af940fb6e4431c9eb9799cc983a85465ec865011":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-12-16 11:32:50.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/903110/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/903110/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903110/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903110/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903110/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/903110/5"}}},"commit":{"parents":[{"commit":"8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78","subject":"Add new API for zone move","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-08 20:02:28.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 11:32:50.000000000","tz":0},"subject":"Increase minimum test coverage to 96%","message":"Increase minimum test coverage to 96%\n\nChange-Id: I791295e7500a75a3a1ee28a7d796163a77591a9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af940fb6e4431c9eb9799cc983a85465ec865011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af940fb6e4431c9eb9799cc983a85465ec865011"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
