)]}'
{"id":"openstack%2Fdesignate~903111","triplet_id":"openstack%2Fdesignate~master~Ic428d77d9773f5c1f9d5516a9ea2e5a56c5a101e","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-05-17 19:35:56.000000000","reason":"Change was abandoned"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-01-07 13:03:35.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ic428d77d9773f5c1f9d5516a9ea2e5a56c5a101e","subject":"Remove unused infoblox code","status":"ABANDONED","created":"2023-12-09 19:05:59.000000000","updated":"2024-05-17 19:35:56.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9956b71eb8efa47b468367c00ca7bbc6e57b8046","_number":903111,"virtual_id_number":903111,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-04 16:47:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-04 15:24:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-02-04 15:24:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-09 20:09:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-02 07:31:36.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2024-01-05 22:55:32.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"9a69f54d4af5fb3932b75aa99600ff6c29c3c30e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-09 19:05:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c203139dff8f1cf8f7b8ef3b207320cf6031469","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-09 20:09:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b69e5d9a4854cb29134e8a86310081e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1834ced032c7407a808fdc93b2d560b9 : SUCCESS in 51m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/aad46877bc954c6a8ca513cf8e0097d2 : SUCCESS in 58m 10s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/85635cff8d414c3fb44f7ccddd394fdb : SUCCESS in 55m 21s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/45fc8605044842e6acd0f3982b07c01a : SUCCESS in 32m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2d098f3b93d54d8b914594be6d8aa2dc : SUCCESS in 32m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3b9e2e0e5c3b47e2b38cc4cae4aab53b : SUCCESS in 49m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cc365cadb7474462855a37ac452c7800 : SUCCESS in 50m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/99ba9d85690b48e7bcd1d6394e66eb2a : SUCCESS in 32m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ca2c789e7b3540afa917fd8618026a1f : SUCCESS in 31m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac8cf87eb2244051922cee98b3231d8e : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaaed9424d5c4b38ab8c561f25ff90c3 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b5b8a63e64a4cdfbc7312af0ff5a713 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44b5660d32754ead974eb5d7190acfa5 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a08bceb138442be81eb665bf2ee3985 : SUCCESS in 7m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aff01b92f7364372aafec57ba151c1e9 : SUCCESS in 33m 03s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ae49a8d79467448e88059ab1146c9823 : SUCCESS in 8m 13s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9f8bdbc6c7b44066a1c85676e40eaa1d : SUCCESS in 49m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d20b880f9f6a443e92f1e598b39059a7 : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e95d9ff3d071ac52f08ec9f0e437db97149012c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-12 10:15:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e0eee7aefc21b4c38742dc36097d2ca49bf3198b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 11:36:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee3e5740c12a4579b8a233d1a6293b0a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7d37ac38251848bbac94732b752b7d57 : SUCCESS in 45m 52s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/776421dec3854bceb697ebaef009ae03 : SUCCESS in 1h 04m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c2b9a5f6a0c47399e8c92c2df48c408 : SUCCESS in 54m 16s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4b2cc14e00e84e65899c8b98fb7bc9c2 : SUCCESS in 44m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e64aa7d6e7164121b00c02652eca6dc7 : SUCCESS in 47m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/703902bd44184ab79a25461354d3fe1d : SUCCESS in 1h 15m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3c71eec1ba1343dc9faac104a7c24861 : SUCCESS in 1h 07m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/23304048fc4e4939a740d9888200fd27 : SUCCESS in 43m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/31597f9e9d824fb9a3858acaa0b81bcf : SUCCESS in 40m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4ce4543b023409285508db5ec2ff76e : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fa00ab1e47b4212a6b558a9e012fcdc : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/872f5e15904b4ec0993fc25d72131c7e : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/09ee00e16ef649e98cb83f9cf76c11de : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9c334c03dc74b1a8ea1d6e7fdcc60bf : SUCCESS in 8m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fe3db5397b34444da79cefcfa29c600c : SUCCESS in 30m 14s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/156af18bad6f4098ab7fbe81646c6ad6 : SUCCESS in 7m 16s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d6f6e760310f49c6adc69460f5d96019 : SUCCESS in 48m 07s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fb10cb5f50ad430cb59f62a9314c9830 : SUCCESS in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f9b112a9c3b8bceb11ba34db12d0010e922e1be5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-13 21:38:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f7d3b49a858543da9932c8bfedfbba06cb10b61f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 22:50:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b518ebdc93d42b9bbe9238b8b0e620b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a6a3b52b2b1445dfaa03427ea809b1e1 : SUCCESS in 31m 33s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b1f89852fc784f64bd05cacc2a296779 : SUCCESS in 36m 12s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/867c2d7b91234281aab6ff6c242608f0 : SUCCESS in 55m 27s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b5a89aea484b4aa59e19b6561f0b3347 : SUCCESS in 52m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a176ac05d55243a995065bc48a57d2d0 : SUCCESS in 46m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f3bb6634ce834eaf9aa3d479e0b356a4 : SUCCESS in 47m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/adc816cb714d46f4aa815039e0b83735 : SUCCESS in 1h 06m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/61a89fd4bf87428daa5a5b60e13ed403 : SUCCESS in 54m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/704649e52c514668bb2f11ae1deb4f83 : SUCCESS in 50m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/010ec7e6c4e24e35a57a78da3c177c19 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2cf0cb3318d4f2ebc939d953d7e95a9 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/851b7c7a03994ea1bad3b2b3acc28dee : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0b76c26ea714e3faead186c2c1d1ed4 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b891d644f84c4bf3b73830b219a1588b : SUCCESS in 7m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ffd590e83dac44a6973aa29f50a0792e : SUCCESS in 32m 37s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c57966e7951243d08d12b52b559143c7 : SUCCESS in 7m 09s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/87c31301aeaf4ec48374b5984f93a274 : FAILURE in 40m 46s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/79fea8bb99bb4eb5ba53ab3a5e500864 : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d611d7c6af37f301906efd0e2d58c41b35316684","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-13 23:54:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fbc9c9a118ca85bd1bc0065e480112b8988a6294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 00:58:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/831ec85f8cce43308ad55e5553752f6e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/00a8f5359a1d4b60a9da742ddf6bb911 : SUCCESS in 50m 59s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e5c80d49a9ea43e2b60469819a7d6157 : SUCCESS in 59m 09s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbaf1b18fbe74e2495e5d66d51db0eda : SUCCESS in 52m 47s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e15a1e207d11480981921d458e145746 : SUCCESS in 48m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a183c7399cc8429d92294b3b7b7f2450 : SUCCESS in 47m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4a109cf54b7c4bbc81ba09c36b2921a9 : SUCCESS in 49m 13s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/29f7a515987d4903bba1613516674413 : SUCCESS in 45m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6de2e298c4cc460599c5ad455bf01182 : SUCCESS in 30m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a3ee4bf75b514dee86f8829e70a3860e : SUCCESS in 31m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80ef702c201e4af683661d56cdd1d478 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c6a445655024b05adcf6ef45b95869b : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/400fd3730cba467eba1995aa2cf87935 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86a52e8a8b7e4a1abfb7ba2f4bef6441 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2642730ad3340b88e3c1abdc95b60d0 : SUCCESS in 8m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c8068cf2ad2f4913b2e1876442cc1d6f : SUCCESS in 32m 23s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/67e1241ee9ee44cfb7d67faee1a4a213 : SUCCESS in 5m 51s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f202d6a3df82471eb57cc078f261718e : SUCCESS in 49m 16s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7aed7ae8621d4477b2fe5e5d74437d2d : SUCCESS in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d485e08c5953c9b1ed29670d5ffd0e104433cb29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-17 17:13:59.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"0880f79530906160e51b964bfbfeeab14ec0c41b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-17 17:16:50.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"0f3873e427a426f596bcfc2848f1923a389207d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 19:18:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/736ba648e4984c8fbf47aa81edcf4503\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/72ecc222903347069f04aa72a7b151f0 : SUCCESS in 43m 57s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e6ce7bc6e89a47a08335d7c1e93b3a34 : TIMED_OUT in 1h 14m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/36cb586379024f29a1e80a53c82dca47 : SUCCESS in 1h 04m 55s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4ed20771532d477e923359cbfccb8c7a : SUCCESS in 47m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1bd340fe8b6545bf9b63efe81c63558c : SUCCESS in 49m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/256954c12c0b4c459d70f922f077e9c6 : SUCCESS in 1h 01m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/514ce1eec9ad43118d3cb0a76051d242 : SUCCESS in 1h 00m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a5cfd1b20a2e4d48929a804e2148cd52 : SUCCESS in 39m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/56aadd890d3b42968c77b9ed6079168b : SUCCESS in 53m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13065f8841c44e80a3036ec6a257b8f8 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aa227aade584462ac94c58ac120bcbd : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/604966fd7bf5403d8e572c9a80c88f20 : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/981164422a854f0eaa981cca2d566dd9 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4753b71a09a7429db4b24fbc963c9b28 : SUCCESS in 8m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/42f74a21820148948e171eacd93d285c : SUCCESS in 32m 51s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/acc537104ebc488dbd2e8fb34bdda2a4 : SUCCESS in 7m 45s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cdda17f673f74d6b80e0012e8966be03 : SUCCESS in 1h 00m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ccfb87b0cf734b4a9f75d5d28ee571ba : SUCCESS in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c3436d8ff4b9535cc448666d3c24c67ad9870bd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-31 00:39:28.000000000","message":"Uploaded patch set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"36b0bc7fb12f68f20522d582a23ab4c46dd167a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-31 01:58:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cf4a18176da47c2a786f151e571a166\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/393298e4c6b044fd981a31ed0a821e94 : SUCCESS in 31m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/71a40f4668ab43718bb8ac16187db530 : SUCCESS in 35m 01s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c636e9a6d8094f32afb989f33c38a46c : SUCCESS in 54m 59s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/22771826b33a4bf8ac8ba7420c26aa91 : SUCCESS in 45m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c3ae862f407b425e8d056c1b08749aeb : SUCCESS in 46m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ba35de2fbc2a45e1ae56b8ee828edfc7 : SUCCESS in 1h 12m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/08cecf9102174b1cb606fbf60e180a7a : SUCCESS in 1h 09m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/12fa3edae251471a8876af85fb8890b1 : SUCCESS in 51m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c86df1bd2a3849ae89928d03a37d3e25 : SUCCESS in 30m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cae4145d88b444cda2e31f0ce336be83 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8599d3095d644eb2ba4f2fd389de6cfc : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ee3c9ef236d4589a8e79de9532e24ba : SUCCESS in 9m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fac6faf3425544bbb62b8b5721b048d5 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9889d246822a4327b1915166a21f3ba6 : SUCCESS in 6m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cf95c8167aab470489e9bbfe209a82d4 : SUCCESS in 34m 10s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/09b3651405ae477084a59bb6898280bf : SUCCESS in 6m 34s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a39119f55a574a5faf021724b994bfed : SUCCESS in 46m 37s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/dd4025ae24844ec19a96b4b507278035 : SUCCESS in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"47b013f02cfc15aa182974f140334bc839f9d869","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2024-01-02 07:31:36.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"d6d1d17dce3206ec77217e0e40cfb44f9eee509a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-01-05 22:55:32.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dc2f9b203240cb983a7a2046ef40d7da27b09a3a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 13:03:35.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dde53449ab8381db157eecd96f6b77385733cfc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 13:06:39.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"95d381485535a3baefa27940ade86c1f5e1d9ac4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 14:24:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c34563e610d4a8ca28517ffb83badb5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/622fc96ae8f840b1985dd068b2e2ed36 : SUCCESS in 52m 10s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b9ae5e42433b4aabb2b8c5745f835a72 : SUCCESS in 1h 00m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf8977585de24043a9811d35c1e99d2f : SUCCESS in 54m 49s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c4826d4eac9046a9a2627348deb76385 : SUCCESS in 35m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/438894229e2548ecbd7742ef6a7a7c6f : SUCCESS in 40m 49s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/30001a000c4f4d85956ca3cecda2068a : SUCCESS in 51m 08s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ce336cd38167433ea161670e91928b47 : SUCCESS in 1h 11m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3d18422905c04d679c7653c3ff745616 : SUCCESS in 52m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/593d42d51abb4d8085a30552d14ec3fe : SUCCESS in 49m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a77d8ee96c44ced8df2046734c023c8 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f228df01d8148a4bbe59fc01b60c92d : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76ff546919ae4835b769fd7f3e09d7f0 : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb0f4027589f4d5bbd91b18b27dc8bcb : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d91fa784f3094770b0ee592d30124bd7 : SUCCESS in 7m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/182143054db543059808f717ceb5e542 : SUCCESS in 32m 28s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/dc6632e20f79469ea18e81b09a0342c4 : SUCCESS in 7m 43s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9738d17d39c54c079560d27da7df5d70 : SUCCESS in 51m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/555a995156a545a49360374807633f36 : SUCCESS in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6364c94663e3f559dc9064fcb09960165254d9db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-08 17:55:49.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"1fa8f1fb76d990f39c29fcfb6c3de2ff541e546f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 19:27:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6af0cc165c14bf3a112d61162652afa\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0bffc2d695464d5d9d8eb087d18ba168 : SUCCESS in 49m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/38497a1bf68e43838941ed9fdbdcdfee : SUCCESS in 1h 03m 15s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7dd09d30f4484584b1238cbb2c39a274 : SUCCESS in 52m 15s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/eed972f4ff704c88a263301f1286de61 : SUCCESS in 46m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d254ec84164146f88b5b17b444a72f23 : SUCCESS in 50m 10s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d77d4c8db32c4dd78cac65eef6cc81d8 : SUCCESS in 1h 09m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d7776f4b752949b7b281ed29ff6cf03a : SUCCESS in 1h 02m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4a0a41494e2d400686d43787660df9f3 : SUCCESS in 50m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e64b35dc482746d4bf7df866aa8cb7b5 : SUCCESS in 54m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bade9cbec0b04a149264537531cd7696 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/871972799f104fcdb369f0bf310dfedc : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98fab7f79ecd4e40a0bbb73d7e426d9a : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4bb866123e6f4e19bdd6ee937e0316ab : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c38a691fc5bc465fa68e0d7c2119a6d9 : SUCCESS in 8m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/30bfd23ff40e4ba387a99a72189b25bb : SUCCESS in 34m 27s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d50e91584844484a84c1faaac6ef27f2 : SUCCESS in 7m 16s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a3b077cd9842407ba53fd98c199f2128 : SUCCESS in 46m 38s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e88e04ccf035429fa14cecb37d0a1522 : SUCCESS in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"14900b988d7a1f2d57af43e94f97b3dc2135288a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-09 00:37:48.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"a03eb5486496151a6636d49ce655b4736f8c2f32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 02:07:12.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f198f5010334cf182f4ae53eca184a2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e1b972701307476f8702701a98d4f87c : SUCCESS in 49m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2a77fdfed0f8463f9fd91aea708cb18e : SUCCESS in 52m 37s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/654a48fcc6e64bb287314e929e24608e : SUCCESS in 57m 54s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ed972fc8dd4941269f38be3c3f0cac7f : SUCCESS in 48m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c34d6e6425d948d7a5f5c104c0e1e786 : SUCCESS in 45m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/254df45431ee468b8eb07e595bd26c78 : SUCCESS in 55m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c9524afe647f469e913317eff0cdd8ed : SUCCESS in 1h 09m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4fe896f45e6d436c92e871a1ec18d5a2 : SUCCESS in 39m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/12d2b7c291e149d792d881178f0a06e4 : SUCCESS in 47m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d70ea8893d647a69fe663de6e826280 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d2291ef0db74a98aab9d2f5acb2aa84 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c00db6f198ed42ca876a18036ab06b25 : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8fae535bd22e4c8d9964ba1a3df2307f : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/609246d0e4e34a9cb71bd3f028be9f5d : SUCCESS in 8m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/51ad25457f354f64b4b39e9387ca384d : SUCCESS in 33m 09s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a124963c357a41e18cf6f6676d075ea1 : SUCCESS in 6m 43s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/996214ebd46e40a7b9e69393e24c219f : SUCCESS in 34m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1f8cb9d464e040fdad54d7b710e7fcbd : SUCCESS in 6m 01s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"029599b253ac47322cc3b05f4bff8791cd7d6fe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-02-04 15:24:57.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"d83bae0cba7d89c2d8f91e8f8871939e319e50b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 16:47:52.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/762f6eccc44746ce982a4d519dfadf09\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6cdfb306d89b478ab11f7a52b8deeafb : SUCCESS in 48m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5e3dddd69a27408a94ccde00f17fdf64 : SUCCESS in 46m 22s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/39110f16f95c412ab9797e4e3fce3679 : SUCCESS in 45m 32s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/de2b7db055994867b67390ffaaf042b6 : SUCCESS in 38m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d57058d843c14a6fbc786ea525c3694f : SUCCESS in 31m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3bb1643a00ba4a6f94eaf548f938b42c : SUCCESS in 1h 08m 25s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/66688093f2d843009ec32634430d5033 : SUCCESS in 1h 12m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9b4b93b7589446fd87444614356f18de : SUCCESS in 32m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2c62a3fa06b54201adafca56b003d628 : SUCCESS in 37m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/210247738e484b5f86c8baa84926380c : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8296282f19a48cebb9db62523eb92cc : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7255389dbedd4535940da84e728f04da : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8fb89741a3e34834b5e9b84dbef25bb2 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/029b0269286743f8ae5ce07d8a266fa7 : SUCCESS in 8m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4e0599f87ecf4cdfbf65e3e6c69bd415 : SUCCESS in 31m 05s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/263ad0b6887c46fd9c3dd50f29fef913 : SUCCESS in 6m 52s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5fbba9fce81945d38ea5da2bee2444be : SUCCESS in 31m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/38d665893404472dba1b93118d3c9402 : SUCCESS in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9956b71eb8efa47b468367c00ca7bbc6e57b8046","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-17 19:35:56.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"090039dcc1667547469179645be5d571e4614416","revisions":{"a3d95ee5e0a258d2368d5fc764d7c3965e7d8688":{"kind":"REWORK","_number":1,"created":"2023-12-09 19:05:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903111/1"}}},"commit":{"parents":[{"commit":"ca4b0ef515aecdbbc2ab431180a11054386595ec","subject":"Merge \"Cleanup context code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca4b0ef515aecdbbc2ab431180a11054386595ec"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-09 19:05:42.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-09 19:05:42.000000000","tz":-480},"subject":"Remove unused infoblox code","message":"Remove unused infoblox code\n\nChange-Id: Ic428d77d9773f5c1f9d5516a9ea2e5a56c5a101e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a3d95ee5e0a258d2368d5fc764d7c3965e7d8688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a3d95ee5e0a258d2368d5fc764d7c3965e7d8688"}]},"branch":"refs/heads/master"},"20fb21c6e3c3e5e3811e06fc6155d85ac6e27b9d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-12-12 10:15:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903111/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903111/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903111/2"}}},"commit":{"parents":[{"commit":"b68451c076f1b70a642414808bc697e6f9006edc","subject":"Merge \"Improved worker rpcapi coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b68451c076f1b70a642414808bc697e6f9006edc"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-09 19:05:42.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-12 10:15:07.000000000","tz":0},"subject":"Remove unused infoblox code","message":"Remove unused infoblox code\n\nChange-Id: Ic428d77d9773f5c1f9d5516a9ea2e5a56c5a101e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20fb21c6e3c3e5e3811e06fc6155d85ac6e27b9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20fb21c6e3c3e5e3811e06fc6155d85ac6e27b9d"}]},"branch":"refs/heads/master","description":"Rebase"},"2d18c6755cdc2a42c1c90bcfdcb214ce09e582f6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-12-13 21:38:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903111/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903111/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903111/3"}}},"commit":{"parents":[{"commit":"d90691f3a29416dbb209b52aae288a5916bca632","subject":"Improved service coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d90691f3a29416dbb209b52aae288a5916bca632"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-09 19:05:42.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-13 21:38:41.000000000","tz":0},"subject":"Remove unused infoblox code","message":"Remove unused infoblox code\n\nChange-Id: Ic428d77d9773f5c1f9d5516a9ea2e5a56c5a101e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2d18c6755cdc2a42c1c90bcfdcb214ce09e582f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2d18c6755cdc2a42c1c90bcfdcb214ce09e582f6"}]},"branch":"refs/heads/master","description":"Rebase"},"69516f4f93f62fe522d1bc928b48fd27ad519768":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-12-17 17:13:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903111/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903111/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903111/4"}}},"commit":{"parents":[{"commit":"70a76b5d8bdd3f394ab31e1aafaac73b514a34a5","subject":"Improve Scheduler coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/70a76b5d8bdd3f394ab31e1aafaac73b514a34a5"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-09 19:05:42.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 17:13:59.000000000","tz":0},"subject":"Remove unused infoblox code","message":"Remove unused infoblox code\n\nChange-Id: Ic428d77d9773f5c1f9d5516a9ea2e5a56c5a101e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69516f4f93f62fe522d1bc928b48fd27ad519768"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69516f4f93f62fe522d1bc928b48fd27ad519768"}]},"branch":"refs/heads/master","description":"Rebase"},"c8dd55f2ab588439c31ab5474a94a47158bba512":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-12-17 17:16:50.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903111/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903111/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903111/5"}}},"commit":{"parents":[{"commit":"fd7532831bb7b530cab816a357b95d5a8b84c517","subject":"Improve Scheduler coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd7532831bb7b530cab816a357b95d5a8b84c517"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-09 19:05:42.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 17:16:50.000000000","tz":0},"subject":"Remove unused infoblox code","message":"Remove unused infoblox code\n\nChange-Id: Ic428d77d9773f5c1f9d5516a9ea2e5a56c5a101e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c8dd55f2ab588439c31ab5474a94a47158bba512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c8dd55f2ab588439c31ab5474a94a47158bba512"}]},"branch":"refs/heads/master","description":"Rebase"},"7f44dfb5120ff4228d63d893d0b490d07d2c2ce9":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-12-31 00:39:28.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903111/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903111/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903111/6"}}},"commit":{"parents":[{"commit":"fd7532831bb7b530cab816a357b95d5a8b84c517","subject":"Improve Scheduler coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd7532831bb7b530cab816a357b95d5a8b84c517"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-09 19:05:42.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-31 00:38:15.000000000","tz":-480},"subject":"Remove unused infoblox code","message":"Remove unused infoblox code\n\nNone of these code paths were reachable. If we want to expand\nthis implementation and make use of this code we can bring\nthem back in the future.\n\nChange-Id: Ic428d77d9773f5c1f9d5516a9ea2e5a56c5a101e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7f44dfb5120ff4228d63d893d0b490d07d2c2ce9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7f44dfb5120ff4228d63d893d0b490d07d2c2ce9"}]},"branch":"refs/heads/master"},"170de0c532be4e6be5a0bac4e4fff5d6f908a4fe":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-01-07 13:06:39.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903111/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903111/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903111/7"}}},"commit":{"parents":[{"commit":"21ae2fb4ff434d591eab7befef462fcbdbe1222d","subject":"Improve Scheduler coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21ae2fb4ff434d591eab7befef462fcbdbe1222d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-09 19:05:42.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-07 13:06:39.000000000","tz":0},"subject":"Remove unused infoblox code","message":"Remove unused infoblox code\n\nNone of these code paths were reachable. If we want to expand\nthis implementation and make use of this code we can bring\nthem back in the future.\n\nChange-Id: Ic428d77d9773f5c1f9d5516a9ea2e5a56c5a101e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/170de0c532be4e6be5a0bac4e4fff5d6f908a4fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/170de0c532be4e6be5a0bac4e4fff5d6f908a4fe"}]},"branch":"refs/heads/master","description":"Rebase"},"7b0350ef0d223aff3f22a11989389f7dbff91849":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-01-08 17:55:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903111/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903111/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903111/8"}}},"commit":{"parents":[{"commit":"2e2b652419f2caef75cec5ae1ea4b8bd6a6fe54a","subject":"Increase minimum test coverage to 97%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e2b652419f2caef75cec5ae1ea4b8bd6a6fe54a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-09 19:05:42.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-08 17:55:49.000000000","tz":0},"subject":"Remove unused infoblox code","message":"Remove unused infoblox code\n\nNone of these code paths were reachable. If we want to expand\nthis implementation and make use of this code we can bring\nthem back in the future.\n\nChange-Id: Ic428d77d9773f5c1f9d5516a9ea2e5a56c5a101e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7b0350ef0d223aff3f22a11989389f7dbff91849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7b0350ef0d223aff3f22a11989389f7dbff91849"}]},"branch":"refs/heads/master","description":"Rebase"},"d54a79eb8e01c376428d41419a97cf7fc8b07329":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-01-09 00:37:48.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903111/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903111/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903111/9"}}},"commit":{"parents":[{"commit":"9d96f5f955a48953cf15ce1ce40f163813265c07","subject":"Merge \"Improve Dyntext coverage and removed unusable code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9d96f5f955a48953cf15ce1ce40f163813265c07"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-09 19:05:42.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-09 00:37:48.000000000","tz":0},"subject":"Remove unused infoblox code","message":"Remove unused infoblox code\n\nNone of these code paths were reachable. If we want to expand\nthis implementation and make use of this code we can bring\nthem back in the future.\n\nChange-Id: Ic428d77d9773f5c1f9d5516a9ea2e5a56c5a101e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d54a79eb8e01c376428d41419a97cf7fc8b07329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d54a79eb8e01c376428d41419a97cf7fc8b07329"}]},"branch":"refs/heads/master","description":"Rebase"},"090039dcc1667547469179645be5d571e4614416":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-02-04 15:24:57.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903111/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903111/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903111/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903111/10"}}},"commit":{"parents":[{"commit":"dd5993ad1b0633cee3269c4b2476d4acc9b672fc","subject":"Merge \"Removed unused function and improved Zone Master coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dd5993ad1b0633cee3269c4b2476d4acc9b672fc"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-09 19:05:42.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-02-04 15:24:57.000000000","tz":0},"subject":"Remove unused infoblox code","message":"Remove unused infoblox code\n\nNone of these code paths were reachable. If we want to expand\nthis implementation and make use of this code we can bring\nthem back in the future.\n\nChange-Id: Ic428d77d9773f5c1f9d5516a9ea2e5a56c5a101e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/090039dcc1667547469179645be5d571e4614416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/090039dcc1667547469179645be5d571e4614416"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
