)]}'
{"id":"openstack%2Fdesignate~903114","triplet_id":"openstack%2Fdesignate~master~I9386d1bfbfee1fe8ff41859520cdbe94381ee3fd","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-12-14 17:23:14.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-12-14 18:37:30.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I9386d1bfbfee1fe8ff41859520cdbe94381ee3fd","subject":"Add upgrade test coverage","status":"MERGED","created":"2023-12-11 13:24:29.000000000","updated":"2023-12-14 19:59:35.000000000","submitted":"2023-12-14 19:58:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"903114","meta_rev_id":"e0052420da1d21ef40fb5f123119875487be5b93","_number":903114,"virtual_id_number":903114,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-14 19:58:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-12-14 17:23:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-14 18:37:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-11 15:12:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-12 23:25:50.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2023-12-14 17:23:14.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"649962d2c8f094f021df238e6bb859d178217afe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-11 13:24:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4f6e6206595226df86bf01d83f6bd029a5d557f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-11 13:26:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"06eb5c3dfaa21bad874824565277db6de94733c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-11 13:32:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6ecde3a3f52515a99e21f7a9628eba6c84675142","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-11 13:40:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5465f3295e11bdf2d444a3faedb88437017e25e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-11 13:45:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6fc9f76ade52f80ad86be0c5a8af32f1f8411b12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-11 13:49:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"049a336e0df4888b0fab035986c21f30155327b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-11 13:54:38.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"baaf8a4539030897ee91ebc96d53a6f3b563fb50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 15:12:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f0f2c4751a74ccc87693ac8c50143e5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/777e4e8fc1d34550884e944d20713e42 : SUCCESS in 5m 10s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b77f5c97a87744869f97a109d92b09cb : SUCCESS in 40m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2598c6ebd9fb46f7aec9ed4b45ca965d : SUCCESS in 48m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7183133a9d404f668619855ee35623bd : SUCCESS in 1h 02m 38s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/740ec32580ca476682ce94abaddac52b : SUCCESS in 1h 00m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bfe51866b76446a987077cdd58042873 : SUCCESS in 48m 10s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/447b9ea32c414332a0b53cb7d80d5b6f : SUCCESS in 1h 04m 05s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/36ac51286cca4e4b96d90b4edaff6a69 : SUCCESS in 1h 16m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9e74e9d4565e443b90933228d3c7b99c : SUCCESS in 42m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/345c558230f14a45a8fcf00bf031c36f : SUCCESS in 50m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b93d0c0fb8a244c0accee1478bcaebd9 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/128b110eff864046888f56e8e2e46c2c : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fee5927d8378487e8faab0891f3d475c : SUCCESS in 6m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5085e71c60e46c4bf88d9ed65a32cc0 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1271d8a25c90433c8701fc6dc5a09f43 : SUCCESS in 9m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/caa7b954188e49a2a2d8b588b894d2dd : SUCCESS in 2m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c130f1c55b44f34ba5d22474c03eb46 : SUCCESS in 5m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/82777fa20230494b80a1077a0fc91c97 : SUCCESS in 36m 02s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/866a88b8d42940c09fe964ed600f281f : SUCCESS in 6m 42s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d721ac89d58046dfa002cf5afc691aa4 : SUCCESS in 52m 29s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/dbacbadfed744837b1a4647d247edcde : SUCCESS in 6m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f5b25f758f1b4335e17a37a6b6318e6b09efca24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-11 19:48:16.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"5c04f0d54b02146a50b2a3bdae34175c3fd927f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 21:05:50.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0da15ab70fa644858595d0dcda51e424\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bcb5438a4be54050bf0256e63ab3f5f9 : SUCCESS in 4m 10s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/783146fa7e744875b3630a218fa627e9 : SUCCESS in 35m 35s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/443b593b8ca7439c93e660b2c0e0273b : SUCCESS in 48m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd6a533353764c46950f09bb2eba5bea : SUCCESS in 45m 43s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1993ff38c05c41c4bc08d94b4530e7bc : SUCCESS in 51m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a16fbde1a9b74cc2975fc994431ddf9b : SUCCESS in 54m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/407669922bdd4af8a52c74e19367e6c1 : SUCCESS in 1h 11m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4e66050075d94f398866d73df940bb38 : SUCCESS in 1h 10m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a9c730470d3e40ac892620639ac93c9b : SUCCESS in 33m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d34bc4a69092465eba2af7b99d3fcfd3 : SUCCESS in 36m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b72b1ea4b1994c338952b0c36662e0cc : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/078cf7e717904c258eab8cb8b2743a0b : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/799712d3fcca4bbc8df984d527418ad5 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c233af048e7a4285ba66d1fcfaa0d10f : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e61144795954fd09504e428b39ebeff : SUCCESS in 8m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c48e9fe58b964a57a1a82c17becb1dcc : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d62132796034655bd15509bc2ad6a24 : SUCCESS in 5m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cf79a84211384a0fa14f74e422b625b4 : SUCCESS in 44m 37s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/67b36b5d4c9c473885ecb0b3ab7c0daf : SUCCESS in 6m 32s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4a21481281c044178c065d5069aa8c0b : SUCCESS in 37m 17s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f03d2e5e12b94d03a44d82ba3b506476 : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8a45d516d75820b6a35871fac73a7626299732f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-12 10:15:13.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"7a80987c546faa0d219d958b8e4987872890d35d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 11:18:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7202724ddeb14da8a4351c6c06ebe766\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/75a8a6a496684daf89c83ee1f4f29ba4 : SUCCESS in 4m 23s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b6d548be84244558944c672696963a7c : SUCCESS in 44m 37s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/be9f8d2c79b84e62889b631607879ee3 : SUCCESS in 46m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ba2c62820da48d0aa9d7f823168a8bf : SUCCESS in 54m 09s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/136489a8789f4515b7e050f44cf1f458 : SUCCESS in 51m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0a40de6147854aec80b62390d925b291 : SUCCESS in 52m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/431443e0c412483a9221b10e26f8f2c5 : SUCCESS in 53m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d73546c4f33e47feaa3dad5e501def5d : SUCCESS in 1h 01m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e3acafd6d42241c28c644b5bbf0a2f30 : SUCCESS in 50m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f8bee91fca6d4da6a1ba806935150b57 : SUCCESS in 48m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b2f81b7528040da8e237ed6eeb3c914 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b69c4cb3ce842a48de7e7f74a08af8d : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5fb36a762c54477b4e5d3c1cce0f6a4 : SUCCESS in 6m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c90955f96dfb4a6ab4b5b6783638cc46 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/939d7ff245d244b2aed68ff3d9ff1cd3 : SUCCESS in 8m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f6e4a0daa0774d2bb9bb905b21a3b502 : SUCCESS in 2m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5d1b8ababe741d0a3d73c96c11b90af : SUCCESS in 7m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/89301839b8854e3fbf5e227aa42c4a9f : SUCCESS in 31m 40s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/007958a8c0034d27808ca487e447dcfe : SUCCESS in 7m 45s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5b584009ae6846d3977150fa941af328 : SUCCESS in 40m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b3d4a424c1fc4e7cb882caea55407cf1 : SUCCESS in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"72b654e0b7f74c09fa2ee47cb86e0f4b30a77561","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-12 23:25:50.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"96bd60cfb08e8b8ef42f0f2e9ed769b90d50e675","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-12 23:50:24.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"457becf5a3975ec215e1985b8849f023342ed13c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-13 21:38:31.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"7b6d5bdff2139bedd7f1fa9fd7bd5531bf879a24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 22:58:30.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/934aca5d828f4d5389be418229e41bcb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/87b9910deb1f462f81391af1f2bbb90c : SUCCESS in 4m 25s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/06d040dba5b04eb79cc574eca0b0aba5 : SUCCESS in 47m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d972ffe6ba164e94892b8e134fb97eb8 : SUCCESS in 1h 00m 59s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/705f4f9e208f408482499a6e644a804c : SUCCESS in 58m 55s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2ef0f50c52cb462887abe5fdb2cc7def : SUCCESS in 30m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/138438efd7794245a0129c98e9de984c : SUCCESS in 49m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/adb9b3cfda524847a026cc79c414cb69 : SUCCESS in 52m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a51269b7c0f2467cbde35e6dc9a59744 : SUCCESS in 1h 14m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/65184ade91fa4b4789e4a001b1ad27f4 : SUCCESS in 34m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b24e68fea4ce432e9d4f535ce8feb4bf : SUCCESS in 37m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efcceaf41da44489bb9058d1e8f9ddb0 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a35e6837cd02446d8f4ac2e457ca1946 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6285ca997d7d44ddb88b709aead4f74c : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97fe1800f82f41e6b3319b08b0e91043 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84336bd993d2411faa37d47201013cb3 : SUCCESS in 8m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/26bed67331ef45e88da52fde5819d053 : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8064dd2c8e44167af770f8ccb9621e0 : SUCCESS in 5m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/03b2acc6738648aea66448ebdc029c41 : SUCCESS in 35m 06s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3a1b6ea1c9754ae09c864523777a0ae4 : SUCCESS in 6m 49s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/91e2146c6ead419bbd384421f3662cad : SUCCESS in 48m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e70ba462030b421caf1885af4eb91561 : SUCCESS in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"68726087eed9ef4ee1dde9e9f7ad756ae12a2e7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-14 13:46:53.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"8d71c9f9229bbb9e6fbfdba472f7e9820a7b4f8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-14 13:54:00.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"32bf9e216b6b408bf3eb110bdf55c7d84e28f2aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 15:07:03.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adc26c52da824c24bdba2daeafb2495d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/577f27f8a5c3464c8d7bcb6156e00555 : SUCCESS in 4m 14s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dbb4bf8467df444fb03a653bf6ba66bf : SUCCESS in 38m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8df20816dd1d4c97a4655c28e8d5990d : SUCCESS in 1h 00m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d8aea6bd14d486f9c24ea0b8da75261 : SUCCESS in 47m 31s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0e6c36e536bc4baf8afc3395a8718937 : SUCCESS in 36m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/12509560690548e88c945dfbb21da9bc : SUCCESS in 53m 14s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d3e941694b5d41d8bc4424494cf9bb9d : SUCCESS in 1h 07m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/68aa11ebc1a1459680595f2bded57d36 : SUCCESS in 1h 02m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3a31019053934d9aa4cac81c4b9edb1b : SUCCESS in 34m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4274eb9ac2ef44a3a978a121a30bc966 : SUCCESS in 35m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ac06ce8b7cf42f881b8415346c1dcd9 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a8fc0e906a8482a9f601efc8f2d5563 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2529c75d68124e388feef5311871ab93 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aebe738a50924c71ab2f3a3f45ee7316 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/754f664fc481407993c452df7c0d9ed0 : SUCCESS in 8m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/56a55b48760d4c119aaf7cb3f69cb340 : SUCCESS in 2m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4459b06076d4acb9209c9ddf20ccfe9 : SUCCESS in 4m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ea0b12c2511c48958f22c60780c3d0a8 : SUCCESS in 41m 46s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e68361dc10e4443586e1140e2ebe6c9a : SUCCESS in 6m 59s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3b03739ac5114089b240b2f40b4259aa : SUCCESS in 36m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3cd3de81f88341a6939eb0a2e13a89a3 : SUCCESS in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"86af138d344921bb3fffebe7a7e4403bf639e02c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-14 17:15:31.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"d45e590e2c0bf5025ea4301850cbc59b6640b1af","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-14 17:23:14.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"6b1287b34e67879116331645294ba1c930ac5595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 18:29:13.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c1a5e95ac8c4fc792862ea4570496fb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a69b40b6d716488691d75b39bb957d43 : SUCCESS in 5m 43s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/722959ad39d546bd9260486cd35a5bb1 : SUCCESS in 47m 08s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8242489803c24c21888bbac05f475435 : SUCCESS in 45m 42s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0823f79e9ee549d7b80aaae46c79c7e9 : SUCCESS in 56m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e902a3a1f2df488484990b969528937e : SUCCESS in 33m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d9f52b14fe364b9088f1821a7a9ee48b : SUCCESS in 41m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/89bfb0905a76481e855db0d5d4c433d0 : SUCCESS in 1h 08m 05s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1b3b572c07c945ffaf6038535c8bfb87 : SUCCESS in 52m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4c1bff37f9ee4d38994e1fb42521b90b : SUCCESS in 48m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a20a089589e847a8beee52b65c107917 : SUCCESS in 52m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e544b6eea6224b5592dc740133ffc05b : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f336f1188af1498994a136cb666a3397 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/116cdd21247841c9a239c1a807ea0590 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df19957b24a04fec98990e7d93348689 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd062d0fc3341c082ee143c82bbc5ad : SUCCESS in 7m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fb5c2b8a990b409099129ec3f0afabfd : SUCCESS in 2m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39dc1e301d244f309b568d487c4e50b8 : SUCCESS in 5m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5613bf3bfa1e44dc9980aa994955a94a : SUCCESS in 44m 08s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/11cbfd277218444c8c8dd4fef6099699 : SUCCESS in 8m 28s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/02a2a4b4c2bc4b6c9093781df28b1ed4 : SUCCESS in 44m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c642c49d0d484944960a4d2822ead5ed : SUCCESS in 7m 43s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"742b49d49c66d6172287221d68835059efe2a6ba","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-14 18:37:30.000000000","message":"Patch Set 13: Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"b37847ab690c543aa6d49e180a9c304f3f5fc6c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 18:38:06.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"48d2caabff823ea8920ef7f16ea07aef2515c4f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 19:58:31.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aacf9a6cdb69495d9bfb7697060fd39a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dce14eea32a34c93b1c6b59f0b2343d3 : SUCCESS in 4m 58s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/adc03fc5aaae4b71a47bad9586582c5b : SUCCESS in 36m 22s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b62dccc283f44982a4e32c96604ebd26 : SUCCESS in 38m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4b4900cb73e14e1ab4f2593444998acc : SUCCESS in 48m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a1aad258184d449fa54712a746b68aab : SUCCESS in 1h 12m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c7cb430ffcd641e38614ea50dccf375d : SUCCESS in 47m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c53fd9fb2fb04b0088e9e29ec7540a67 : SUCCESS in 33m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b9ecc81072d4057a1b9c94f5e0fb44a : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a613c3bb98304112a51d2324e4733d55 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/83f232406e9f4c19b46f0dbf54d41152 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f05b366c3e9f4fff82a10b3904e6a3db : SUCCESS in 6m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c8b7da6b68054996888e72570dbc978c : SUCCESS in 2m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94d19e3bfda849de9e825915327ded32 : SUCCESS in 5m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aa3cf298a8a648be9035aab8778eaa64 : SUCCESS in 40m 59s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9680be8ad7e24e9f85c6437b09fce894 : SUCCESS in 7m 36s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a051d37f89774fc7a17be5b0a9e19d16 : SUCCESS in 38m 45s","accounts_in_message":[],"_revision_number":13},{"id":"a3e7253ca8d7f9a38d2ce6a7729dfe7c4546feca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 19:58:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"e0052420da1d21ef40fb5f123119875487be5b93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 19:59:35.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6306a4b2dea94ab084815554ad1bc95c\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6267260319bf4f64a103e6e56899d904 : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3892ddfb6c4d5fad8bd0be8bd4303a : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52c33c38f6b9432bbd11b8b056387f3c : SUCCESS in 43s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"f5a034272ddeddf80dc858396cf31ab581f4ae5e","revisions":{"894e764c6bd3786612af14a65dd638396c014f0b":{"kind":"REWORK","_number":1,"created":"2023-12-11 13:24:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903114/1"}}},"commit":{"parents":[{"commit":"ca4b0ef515aecdbbc2ab431180a11054386595ec","subject":"Merge \"Cleanup context code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca4b0ef515aecdbbc2ab431180a11054386595ec"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:23:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:23:36.000000000","tz":-480},"subject":"Add mysql upgrade test coverage","message":"Add mysql upgrade test coverage\n\nBased on the nova testing implementation. Also added\ntest for is_migration_needed.\n\nChange-Id: I9386d1bfbfee1fe8ff41859520cdbe94381ee3fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/894e764c6bd3786612af14a65dd638396c014f0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/894e764c6bd3786612af14a65dd638396c014f0b"}]},"branch":"refs/heads/master"},"78f971189d48effe467bdc75f73f7cad8b50222e":{"kind":"REWORK","_number":2,"created":"2023-12-11 13:26:00.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903114/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903114/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903114/2"}}},"commit":{"parents":[{"commit":"ca4b0ef515aecdbbc2ab431180a11054386595ec","subject":"Merge \"Cleanup context code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca4b0ef515aecdbbc2ab431180a11054386595ec"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:23:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:25:54.000000000","tz":-480},"subject":"Add mysql upgrade test coverage","message":"Add mysql upgrade test coverage\n\nBased on the nova testing implementation. Also added\ntest for is_migration_needed.\n\nChange-Id: I9386d1bfbfee1fe8ff41859520cdbe94381ee3fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/78f971189d48effe467bdc75f73f7cad8b50222e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/78f971189d48effe467bdc75f73f7cad8b50222e"}]},"branch":"refs/heads/master"},"e2359b497d99616c5e7d69af4a97759660d2c36b":{"kind":"REWORK","_number":3,"created":"2023-12-11 13:32:23.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903114/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903114/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903114/3"}}},"commit":{"parents":[{"commit":"ca4b0ef515aecdbbc2ab431180a11054386595ec","subject":"Merge \"Cleanup context code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca4b0ef515aecdbbc2ab431180a11054386595ec"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:23:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:32:17.000000000","tz":-480},"subject":"Add mysql upgrade test coverage","message":"Add mysql upgrade test coverage\n\nBased on the nova testing implementation. Also added\ntest for is_migration_needed.\n\nChange-Id: I9386d1bfbfee1fe8ff41859520cdbe94381ee3fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e2359b497d99616c5e7d69af4a97759660d2c36b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e2359b497d99616c5e7d69af4a97759660d2c36b"}]},"branch":"refs/heads/master"},"c340be9bdab5026f91977f39083723d711c7dee6":{"kind":"REWORK","_number":4,"created":"2023-12-11 13:40:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903114/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903114/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903114/4"}}},"commit":{"parents":[{"commit":"ca4b0ef515aecdbbc2ab431180a11054386595ec","subject":"Merge \"Cleanup context code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca4b0ef515aecdbbc2ab431180a11054386595ec"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:23:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:40:36.000000000","tz":-480},"subject":"Add mysql upgrade test coverage","message":"Add mysql upgrade test coverage\n\nBased on the nova testing implementation. Also added\ntest for is_migration_needed.\n\nChange-Id: I9386d1bfbfee1fe8ff41859520cdbe94381ee3fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c340be9bdab5026f91977f39083723d711c7dee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c340be9bdab5026f91977f39083723d711c7dee6"}]},"branch":"refs/heads/master"},"2c5bc89a86187f3fdc9e7ee8c5c9d7447576474f":{"kind":"REWORK","_number":5,"created":"2023-12-11 13:45:25.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903114/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903114/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903114/5"}}},"commit":{"parents":[{"commit":"ca4b0ef515aecdbbc2ab431180a11054386595ec","subject":"Merge \"Cleanup context code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca4b0ef515aecdbbc2ab431180a11054386595ec"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:23:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:45:16.000000000","tz":-480},"subject":"Add mysql upgrade test coverage","message":"Add mysql upgrade test coverage\n\nBased on the nova testing implementation. Also added\ntest for is_migration_needed.\n\nChange-Id: I9386d1bfbfee1fe8ff41859520cdbe94381ee3fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c5bc89a86187f3fdc9e7ee8c5c9d7447576474f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c5bc89a86187f3fdc9e7ee8c5c9d7447576474f"}]},"branch":"refs/heads/master"},"d96509f1f756469d8afa9a521886bf9ef2897011":{"kind":"REWORK","_number":6,"created":"2023-12-11 13:49:18.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903114/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903114/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903114/6"}}},"commit":{"parents":[{"commit":"ca4b0ef515aecdbbc2ab431180a11054386595ec","subject":"Merge \"Cleanup context code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca4b0ef515aecdbbc2ab431180a11054386595ec"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:23:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:49:13.000000000","tz":-480},"subject":"Add mysql upgrade test coverage","message":"Add mysql upgrade test coverage\n\nBased on the nova testing implementation. Also added\ntest for is_migration_needed.\n\nChange-Id: I9386d1bfbfee1fe8ff41859520cdbe94381ee3fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d96509f1f756469d8afa9a521886bf9ef2897011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d96509f1f756469d8afa9a521886bf9ef2897011"}]},"branch":"refs/heads/master"},"6c46e62da8ae5fb4c84ccfa40ae14855bc532b71":{"kind":"REWORK","_number":7,"created":"2023-12-11 13:54:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903114/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903114/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903114/7"}}},"commit":{"parents":[{"commit":"ca4b0ef515aecdbbc2ab431180a11054386595ec","subject":"Merge \"Cleanup context code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca4b0ef515aecdbbc2ab431180a11054386595ec"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:23:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:54:34.000000000","tz":-480},"subject":"Add mysql upgrade test coverage","message":"Add mysql upgrade test coverage\n\nBased on the nova testing implementation. Also added\ntest for is_migration_needed.\n\nChange-Id: I9386d1bfbfee1fe8ff41859520cdbe94381ee3fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6c46e62da8ae5fb4c84ccfa40ae14855bc532b71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6c46e62da8ae5fb4c84ccfa40ae14855bc532b71"}]},"branch":"refs/heads/master"},"a0fc0ad4993194b5fe010bc4a906ae254b1b9a1f":{"kind":"REWORK","_number":8,"created":"2023-12-11 19:48:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903114/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903114/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903114/8"}}},"commit":{"parents":[{"commit":"ca4b0ef515aecdbbc2ab431180a11054386595ec","subject":"Merge \"Cleanup context code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca4b0ef515aecdbbc2ab431180a11054386595ec"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:23:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 19:48:07.000000000","tz":-480},"subject":"Add mysql upgrade test coverage","message":"Add mysql upgrade test coverage\n\nBased on the nova testing implementation. Also added\na test for is_migration_needed.\n\nChange-Id: I9386d1bfbfee1fe8ff41859520cdbe94381ee3fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0fc0ad4993194b5fe010bc4a906ae254b1b9a1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0fc0ad4993194b5fe010bc4a906ae254b1b9a1f"}]},"branch":"refs/heads/master"},"fe1d03ca0cefcc064ed0f26af3ac253ace14dd99":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-12-12 10:15:13.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903114/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903114/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903114/9"}}},"commit":{"parents":[{"commit":"b68451c076f1b70a642414808bc697e6f9006edc","subject":"Merge \"Improved worker rpcapi coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b68451c076f1b70a642414808bc697e6f9006edc"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:23:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-12 10:15:13.000000000","tz":0},"subject":"Add mysql upgrade test coverage","message":"Add mysql upgrade test coverage\n\nBased on the nova testing implementation. Also added\na test for is_migration_needed.\n\nChange-Id: I9386d1bfbfee1fe8ff41859520cdbe94381ee3fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fe1d03ca0cefcc064ed0f26af3ac253ace14dd99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fe1d03ca0cefcc064ed0f26af3ac253ace14dd99"}]},"branch":"refs/heads/master","description":"Rebase"},"b1ef84bacba8ceb4ef9294e90ee53a859aa5e317":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-12-13 21:38:31.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903114/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903114/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903114/10"}}},"commit":{"parents":[{"commit":"d90691f3a29416dbb209b52aae288a5916bca632","subject":"Improved service coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d90691f3a29416dbb209b52aae288a5916bca632"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:23:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-13 21:38:31.000000000","tz":0},"subject":"Add mysql upgrade test coverage","message":"Add mysql upgrade test coverage\n\nBased on the nova testing implementation. Also added\na test for is_migration_needed.\n\nChange-Id: I9386d1bfbfee1fe8ff41859520cdbe94381ee3fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b1ef84bacba8ceb4ef9294e90ee53a859aa5e317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b1ef84bacba8ceb4ef9294e90ee53a859aa5e317"}]},"branch":"refs/heads/master","description":"Rebase"},"c8eb7948d1e52043e3961fffba6ea020e470fc04":{"kind":"REWORK","_number":11,"created":"2023-12-14 13:46:53.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903114/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903114/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903114/11"}}},"commit":{"parents":[{"commit":"d90691f3a29416dbb209b52aae288a5916bca632","subject":"Improved service coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d90691f3a29416dbb209b52aae288a5916bca632"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:23:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-14 13:46:53.000000000","tz":0},"subject":"Add mysql upgrade test coverage","message":"Add mysql upgrade test coverage\n\nBased on the nova testing implementation. Also added\na test for is_migration_needed.\n\nChange-Id: I9386d1bfbfee1fe8ff41859520cdbe94381ee3fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c8eb7948d1e52043e3961fffba6ea020e470fc04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c8eb7948d1e52043e3961fffba6ea020e470fc04"}]},"branch":"refs/heads/master"},"d5aa99d993de21649405e56fbe45a9e986c1b202":{"kind":"REWORK","_number":12,"created":"2023-12-14 13:54:00.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903114/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903114/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903114/12"}}},"commit":{"parents":[{"commit":"d90691f3a29416dbb209b52aae288a5916bca632","subject":"Improved service coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d90691f3a29416dbb209b52aae288a5916bca632"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:23:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-14 13:54:00.000000000","tz":0},"subject":"Add mysql upgrade test coverage","message":"Add mysql upgrade test coverage\n\nBased on the nova testing implementation. Also added\na test for is_migration_needed.\n\nChange-Id: I9386d1bfbfee1fe8ff41859520cdbe94381ee3fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d5aa99d993de21649405e56fbe45a9e986c1b202"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d5aa99d993de21649405e56fbe45a9e986c1b202"}]},"branch":"refs/heads/master"},"f5a034272ddeddf80dc858396cf31ab581f4ae5e":{"kind":"REWORK","_number":13,"created":"2023-12-14 17:15:31.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903114/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903114/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903114/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903114/13"}}},"commit":{"parents":[{"commit":"d90691f3a29416dbb209b52aae288a5916bca632","subject":"Improved service coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d90691f3a29416dbb209b52aae288a5916bca632"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 13:23:36.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-14 17:13:11.000000000","tz":-480},"subject":"Add upgrade test coverage","message":"Add upgrade test coverage\n\nAdding additional test upgrade coverage. This implementation\ntests things like the is_migration_needed flag. In the future\nwe can also add support for MySQL functional coverage as well.\n\nChange-Id: I9386d1bfbfee1fe8ff41859520cdbe94381ee3fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5a034272ddeddf80dc858396cf31ab581f4ae5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5a034272ddeddf80dc858396cf31ab581f4ae5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
