)]}'
{"id":"openstack%2Fdesignate~903115","triplet_id":"openstack%2Fdesignate~master~Ib1a907f6377967901b1c079580a10f99d408eb12","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-12-12 06:41:38.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ib1a907f6377967901b1c079580a10f99d408eb12","subject":"Improved worker rpcapi coverage","status":"MERGED","created":"2023-12-11 16:40:06.000000000","updated":"2023-12-12 07:52:24.000000000","submitted":"2023-12-12 07:50:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903115","meta_rev_id":"b39ab1560f67431229b61e36795c3fa771124a4f","_number":903115,"virtual_id_number":903115,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-12 07:50:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-12-12 04:15:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-12 06:41:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-11 17:42:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-12 04:15:08.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"37c2f8037889db41ffd742181129e1aa56435fa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-11 16:40:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffebc60a5e18c80701fc1286dc1f1fc959c1be0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 17:42:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52a120fb362e403f99b85b2b65f2649b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d91238c4de6f4def84229224abf0a2d8 : SUCCESS in 36m 56s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/83de343fbe554b9b8d7f7a4bea68dd9a : SUCCESS in 49m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1760ca918e9434e9e06b3c3523fe761 : SUCCESS in 55m 22s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/278d4af0dfd447d2bc36947f58f9398e : SUCCESS in 44m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/73104e68ece24030be9c449fc2118148 : SUCCESS in 34m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2f8d5045360a41ab87e87008629fd14c : SUCCESS in 1h 00m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2d0c110d89444ef09d5c57e136633d13 : SUCCESS in 56m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e0faad55d8444c81bb3e04f10c0d23b5 : SUCCESS in 53m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6dcf2bde9b5449638d6f3801e7d27744 : SUCCESS in 46m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f794b6e144aa4beebd76978015794528 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/473abc6212654e18a541aa6673998a76 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/533e66520fe745e1a762e1ba1673efe0 : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70bb68e882da4db4975fec8e61e6e651 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69c94875cee244d48e2d0316eb4dd11e : SUCCESS in 6m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0fdf38aa5ef54f41ae59ea1751124dd8 : SUCCESS in 31m 24s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/060118d12e244e50bef61a934a9e7b5f : SUCCESS in 7m 36s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ad4ab9fe8b3d4a59a37dbf3bba959296 : SUCCESS in 38m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1d753ebd8793419084088da145cf6952 : SUCCESS in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3f70140e647dce95e83925383667063ab814324","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-12 04:15:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d566383f1f352925256b858cb3edae406b4d0cfa","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-12 06:41:38.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"61db42818a4e72b236bd0944920a51d8e5746b0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 06:42:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5287898232321116568f81a529d1be69de6add45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 07:50:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c0291d391254a558f1884e1595c0dbd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/72f48ebdec3a4c1d9248d023342fe77d : SUCCESS in 31m 13s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3b17aa543040463f9c7848b398d57a72 : SUCCESS in 40m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8de866ac64194e18b0d8488c733a97c5 : SUCCESS in 37m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9e4aadbf618c48f48a27225c69a9fc3f : SUCCESS in 1h 02m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/659121b9118f44b3970a3c064def6ddd : SUCCESS in 32m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/27bde8d1edc242f88703ff981b1d3978 : SUCCESS in 38m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc224ba2577d4945bee827f215abc400 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2e7c64ed7d44e4fbdb23cb9636aadfd : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5bea8b8e576a469587856890d7fbbb21 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb499fe038d043f28318e6b22c101d68 : SUCCESS in 7m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2b62bbac1e9548409b29f7a23c2da846 : SUCCESS in 35m 52s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7f61c55294764459ad68a3e26f2944fa : SUCCESS in 6m 49s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/37af9076d2f849b4b2e128ba54aadda0 : SUCCESS in 51m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7202b6053a91aff2fae01291c9fa4bf6229a7f34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 07:50:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b39ab1560f67431229b61e36795c3fa771124a4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 07:52:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06dd38dd2b8244688ac8186b0a803230\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c49f820abc04a03ba84dc7f2db7cbfd : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21f95128751dd8b1bb90acad00521b58d8042daf","revisions":{"21f95128751dd8b1bb90acad00521b58d8042daf":{"kind":"REWORK","_number":1,"created":"2023-12-11 16:40:06.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/15/903115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/15/903115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/903115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/903115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/903115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/15/903115/1"}}},"commit":{"parents":[{"commit":"ca4b0ef515aecdbbc2ab431180a11054386595ec","subject":"Merge \"Cleanup context code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca4b0ef515aecdbbc2ab431180a11054386595ec"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 16:39:19.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-11 16:39:57.000000000","tz":-480},"subject":"Improved worker rpcapi coverage","message":"Improved worker rpcapi coverage\n\nChange-Id: Ib1a907f6377967901b1c079580a10f99d408eb12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21f95128751dd8b1bb90acad00521b58d8042daf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21f95128751dd8b1bb90acad00521b58d8042daf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
