)]}'
{"id":"openstack%2Fdesignate~903405","triplet_id":"openstack%2Fdesignate~master~Iaded56fbdfd06d66b906617259ccbaf7cc21e910","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-01-07 13:06:36.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Iaded56fbdfd06d66b906617259ccbaf7cc21e910","subject":"Improve Scheduler coverage","status":"MERGED","created":"2023-12-13 12:04:30.000000000","updated":"2024-01-07 15:51:07.000000000","submitted":"2024-01-07 15:49:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903405","meta_rev_id":"5d7c1ad628250bda49ae10f38449a5a76ae766b3","_number":903405,"virtual_id_number":903405,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-07 15:49:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-07 13:06:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-01-07 13:06:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-13 13:39:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-05 22:39:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"d06fb62ad660248d1100972768a8f8f55a2b64fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-13 12:04:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c985042aaaff4783f2fee8af1ec4942c48ca4c51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-13 12:23:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cf45e12161dc44729fdd40bd4ec9d794c7149627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 13:39:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b2b346d89af443584dcffdbe8ccce2e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8cf479e341de4797ba508f55959b1fa8 : SUCCESS in 50m 12s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3fcca4b450e845af9ead0be7276d5bf9 : SUCCESS in 57m 43s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe4ba6d7d93c46e0bc4f297fdc4e90c9 : SUCCESS in 55m 32s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/00691f1b96d04afabf9f238bba667baa : SUCCESS in 47m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e44055d416bc4b46852c8455b5ef4dd9 : SUCCESS in 47m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/65655ec0eed34f8db4b2e5623c4a0d32 : SUCCESS in 1h 09m 56s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bca84b8c3dd74cf5b34edee4b7f49d79 : SUCCESS in 1h 10m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bb6a2952925447e589c82aa59e752ac7 : SUCCESS in 44m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/44f62ba5df304edcb5212a23fb00664f : SUCCESS in 48m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ccf7b71ad7d4479903c6c0b0b315c50 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55608ac6808f49b382779d62a3dab388 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40eba4ab9e99439fa6c334f8d536f489 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a320386854154b2da18a34244e92dcc3 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79e8ad53956f4a109c4150a48f641a80 : SUCCESS in 7m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a948418da623481e9af68a3d772e264f : SUCCESS in 34m 33s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c0573111154c4e07b058cf0c8f534e16 : SUCCESS in 7m 25s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/18bef4335ea64898ab7f876ac6f3d0d7 : SUCCESS in 32m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9db9c6ab6536473aa5484f8c65c81e06 : SUCCESS in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a5201c7c680c30ce9c488a44cb4ea64434a6b958","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 11:36:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"66f169dc141ae2046b0eff74c9c3b9a378e3d6b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 13:06:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b66f2c211a344268bbdfbc94a7e7504f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0daeb0ca75e144fc9e1c13e34752f1de : FAILURE in 38m 07s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/17037cbd43984bef8e61cb9cc24d063e : FAILURE in 1h 04m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3263190ffb4747c38ca8b7cf02860640 : FAILURE in 52m 18s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/858d0400bffa437793946b1bf964cd1c : FAILURE in 39m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2b4b0842ab654b1396db471430aee274 : FAILURE in 40m 06s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/882f2d42f3e24d95949bdf16cba4b274 : FAILURE in 49m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2e9a0ceb8dc14e0c86a09377c72f5726 : FAILURE in 40m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3153b33b178a44f1a7584cdd838f3016 : FAILURE in 43m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/09dde3b78f2840baa324715172e0f768 : FAILURE in 42m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/113382533b3540c180d3ebcd48f0f993 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c30ccd4135a2449fb8308a086edc4950 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a72cf2a47ddd48599a28a0823b10ad11 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6fc0273b88f4942a9c14044d411b7a3 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/684dc0143cdf4d6c84d6a834ceeadae3 : SUCCESS in 8m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fe6b7743a4644eaaa22f7192c8278db3 : FAILURE in 41m 05s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/be1adca868a742559f007f638adef83e : SUCCESS in 6m 56s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/035cc6457fec4ccea93a1414a098afac : FAILURE in 51m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9782d229eaa34ea9adaf7bf2d8bf9961 : SUCCESS in 6m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3cc0e8677f40745f812a4f9112502be5fed08cc2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 17:27:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d2aeabda96584c4b5612c83eac58ee26a8d9c5ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 19:05:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e3fba96dae44419bb9f0497da474ad7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a872b75cb14741a2a0e759fe5e860542 : SUCCESS in 46m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/78275f98e24b4ff39515cce101c8aa5f : SUCCESS in 42m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/695b7df48a1c4fd7aa06994742b04b95 : SUCCESS in 1h 01m 59s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/eec7b6de23134eddaf4a7004ac6a2cb7 : SUCCESS in 1h 01m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bc9367985bde4cf4ac39fb56d2cd0852 : SUCCESS in 51m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a929749e95f14b8bb99f45f508971f3f : SUCCESS in 1h 09m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5f586988eef5428486eaf8f1ef389074 : SUCCESS in 1h 09m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0397d152a01548ada9d4163bd6bd8719 : SUCCESS in 51m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f80367c58dad45f4b42bea1bdb539053 : SUCCESS in 55m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c09f169f269f480691ea3fbf801e5f64 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41688f86c2304f0ab621ef37f912ff24 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90d37801605f4ff0adb79f6d00f02d62 : SUCCESS in 5m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c45bf496710949258e05ca2febe62d86 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5237902d8efd43c68694b4ec32791576 : SUCCESS in 8m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/098eb68b2a8045389462f7d30f1022cc : SUCCESS in 32m 54s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4e52accb9c2d4cdeb0e9e5eeed0de32d : SUCCESS in 7m 33s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/db0c2acb1e534bcc9514efa68477b7ad : SUCCESS in 54m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f62c115eb5b54b25bfbe733968855b51 : SUCCESS in 7m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dff5176e5093de3f5d20e9b5ca34b4a5eeb94129","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-17 17:13:41.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f2242ee14411595204ca0b7d23bce783d74021d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-17 17:16:47.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"561ca31b286f491e6f07758e8b12b4663b02fd77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 19:22:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed4a08956dd04aedbad0056ee589344a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c0a3f5ff6f954caa91f17eb93a293668 : SUCCESS in 55m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d954e4ab312e485e8fbfd61e6deeaf12 : SUCCESS in 1h 12m 01s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/292b0bcc977c44099a554f3dedceb9b8 : SUCCESS in 52m 05s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3fdfd9355252429089a501c9a0ed29c3 : SUCCESS in 41m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7a064b5c64d845598babe07696250e89 : SUCCESS in 47m 24s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/faf08862124842e5bb489b9268d67f43 : SUCCESS in 1h 00m 45s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ef5ad285530c46f6b150f7b5e6da3bfb : SUCCESS in 1h 10m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fd3f18fac87940e0ab780333c9588db8 : SUCCESS in 50m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9a364a48096a4877909740c4d7b4f87e : SUCCESS in 47m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36b1980d18e54b2686ff254988644846 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e340150e34487fa7e336cb4c8a35f0 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d72b21044f6546c9b378184f87204b3b : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/148873a068664ffb90531ac433abdbd8 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52831c7be29343b6936b2e1d355790df : SUCCESS in 8m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5c99e85cacf4ca383f0e6ddea532f31 : SUCCESS in 31m 37s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d539e8c7d28446cf9ec06cd74b5bc62a : SUCCESS in 7m 32s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/66e813180fcb468aaa5079d1419c8dc3 : SUCCESS in 39m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/05fa5319571a45e5a9ce39ee02693c1a : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"10716c1ad8bd04240bd726cd48c2884ca6139893","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-01-05 22:39:39.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b3f0239d3c61db6b637faf9a9dd495cf841fbe77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 13:06:33.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"c1ee79c3aa6cb34132aa14eba1fc73f28ac0b469","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 13:06:36.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"99d63519586362914079f69ac393249a1848cb29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 14:12:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81f8cd4efa8b4b41a42096d724222700\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/469a99296c8d4ce69d36737c0535ac4c : SUCCESS in 49m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7d706321f4f04c39aa57dec7069f5a13 : SUCCESS in 39m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f32f59f6a422462fa69c7ae10d75005a : SUCCESS in 54m 43s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/71c69dbbf55a4077925d4b1ee7455444 : SUCCESS in 45m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/aae1d91c5a91472483c1b871e204dc8b : SUCCESS in 34m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c57352f289fe4c20a918984052bc871b : SUCCESS in 51m 09s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c5191f6b3e114e6f9af1e217d1f9adc8 : SUCCESS in 1h 00m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ae8de327046443d7ae48be08ccecd6ff : SUCCESS in 34m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a825c3c37578414ab71fd3399cd880b5 : SUCCESS in 49m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e55d91c542284f99b86770025553a1b1 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e215906fb5ec446b8b93062f6e7925dc : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fc51f585c0740269744b1a729be4bf3 : SUCCESS in 6m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ebf1e22c9de4046ba18507f520be1d5 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a54b126d212441289510813a37e9b3ca : SUCCESS in 8m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/03590de9583244958057e7bf28e071d9 : SUCCESS in 32m 52s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b7a6a2f1175449d695e2cf56fcbd5782 : SUCCESS in 6m 42s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5d90039c6de9437a841cbf70320be112 : SUCCESS in 52m 37s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/750ecf9a18254a00ba977691e47f9b74 : SUCCESS in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dc661a3f3fb0df376b6d651b4003071d86a14c07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 14:27:08.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e0171448c8cd94ddb4842bd580a5bde5d0759cca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 15:49:15.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d493f8f0e4442c292606ed2e14f4a96\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6ff47d0bca3e4d8fb07fba493209cdab : SUCCESS in 32m 03s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/65688bd9c4ec4aafb53712f2be8d0965 : SUCCESS in 41m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b8552464515040cc887d9c865a437388 : SUCCESS in 54m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/57831b3e45e44a18817e8fec2805cab6 : SUCCESS in 45m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bbdbf1044cd84754b516564bda10fb38 : SUCCESS in 39m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d02d6181e8584d7995db4a8d77da0f90 : SUCCESS in 34m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc8b08506ffc44758b66c0c45a9a1dc1 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a67fea6544f747a4852a0b60c7627de0 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eebca9433f944bf8b28851eab057c67b : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c5beee3aab4057994ac8b778fdc84f : SUCCESS in 7m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9cdfd2934de449c2859d5488b63f03f7 : SUCCESS in 31m 12s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0ea763e1937245a2b5331d202730e67e : SUCCESS in 6m 59s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/134446509265431ea191ccf96f3386c0 : SUCCESS in 50m 03s","accounts_in_message":[],"_revision_number":6},{"id":"c48ca3e2cc9832ce65778782d6294ba7226b0d6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 15:49:16.000000000","message":"Change has been successfully merged\n\n5 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":6},{"id":"5d7c1ad628250bda49ae10f38449a5a76ae766b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 15:51:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f83ae67bb94d57ae70c4c836eee45f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed6418cf6014a77987ec7e8a399c27b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"21ae2fb4ff434d591eab7befef462fcbdbe1222d","revisions":{"1f91c91742f1c441c7b75aea58546b72ffe354ee":{"kind":"REWORK","_number":1,"created":"2023-12-13 12:04:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/05/903405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/05/903405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/05/903405/1"}}},"commit":{"parents":[{"commit":"d90691f3a29416dbb209b52aae288a5916bca632","subject":"Improved service coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d90691f3a29416dbb209b52aae288a5916bca632"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-13 12:04:05.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-13 12:04:05.000000000","tz":-480},"subject":"Improve Base Scheduler coverage","message":"Improve Base Scheduler coverage\n\nChange-Id: Iaded56fbdfd06d66b906617259ccbaf7cc21e910\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f91c91742f1c441c7b75aea58546b72ffe354ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f91c91742f1c441c7b75aea58546b72ffe354ee"}]},"branch":"refs/heads/master"},"f193b4712533f3183556043b4fa7dcf1e55128ab":{"kind":"REWORK","_number":2,"created":"2023-12-13 12:23:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/05/903405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/05/903405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/05/903405/2"}}},"commit":{"parents":[{"commit":"d90691f3a29416dbb209b52aae288a5916bca632","subject":"Improved service coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d90691f3a29416dbb209b52aae288a5916bca632"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-13 12:04:05.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-13 12:23:29.000000000","tz":-480},"subject":"Improve Scheduler coverage","message":"Improve Scheduler coverage\n\nChange-Id: Iaded56fbdfd06d66b906617259ccbaf7cc21e910\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f193b4712533f3183556043b4fa7dcf1e55128ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f193b4712533f3183556043b4fa7dcf1e55128ab"}]},"branch":"refs/heads/master"},"9a3d274d4a1577a6d8569999be1540377dcf8af0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-12-16 11:36:26.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/05/903405/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/05/903405/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/05/903405/3"}}},"commit":{"parents":[{"commit":"8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78","subject":"Add new API for zone move","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-13 12:04:05.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 11:36:26.000000000","tz":0},"subject":"Improve Scheduler coverage","message":"Improve Scheduler coverage\n\nChange-Id: Iaded56fbdfd06d66b906617259ccbaf7cc21e910\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a3d274d4a1577a6d8569999be1540377dcf8af0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a3d274d4a1577a6d8569999be1540377dcf8af0"}]},"branch":"refs/heads/master","description":"Rebase"},"70a76b5d8bdd3f394ab31e1aafaac73b514a34a5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-12-17 17:13:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/05/903405/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/05/903405/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/05/903405/4"}}},"commit":{"parents":[{"commit":"6194b53de9fee36f4be5c7cfab1883cdfc8d9834","subject":"Added basic fake network coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6194b53de9fee36f4be5c7cfab1883cdfc8d9834"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-13 12:04:05.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 17:13:41.000000000","tz":0},"subject":"Improve Scheduler coverage","message":"Improve Scheduler coverage\n\nChange-Id: Iaded56fbdfd06d66b906617259ccbaf7cc21e910\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/70a76b5d8bdd3f394ab31e1aafaac73b514a34a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/70a76b5d8bdd3f394ab31e1aafaac73b514a34a5"}]},"branch":"refs/heads/master","description":"Rebase"},"fd7532831bb7b530cab816a357b95d5a8b84c517":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-12-17 17:16:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/05/903405/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/05/903405/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/05/903405/5"}}},"commit":{"parents":[{"commit":"0e7b7acd4d3fd4977c3a600cf3f091a3ce2037d2","subject":"Added basic fake network coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e7b7acd4d3fd4977c3a600cf3f091a3ce2037d2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-13 12:04:05.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 17:16:47.000000000","tz":0},"subject":"Improve Scheduler coverage","message":"Improve Scheduler coverage\n\nChange-Id: Iaded56fbdfd06d66b906617259ccbaf7cc21e910\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd7532831bb7b530cab816a357b95d5a8b84c517"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd7532831bb7b530cab816a357b95d5a8b84c517"}]},"branch":"refs/heads/master","description":"Rebase"},"21ae2fb4ff434d591eab7befef462fcbdbe1222d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-01-07 13:06:33.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/05/903405/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/05/903405/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/903405/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/05/903405/6"}}},"commit":{"parents":[{"commit":"16e2433d25756bee9e069d725ac852895f43d7ed","subject":"Added basic fake network coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/16e2433d25756bee9e069d725ac852895f43d7ed"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-13 12:04:05.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-07 13:06:33.000000000","tz":0},"subject":"Improve Scheduler coverage","message":"Improve Scheduler coverage\n\nChange-Id: Iaded56fbdfd06d66b906617259ccbaf7cc21e910\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21ae2fb4ff434d591eab7befef462fcbdbe1222d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21ae2fb4ff434d591eab7befef462fcbdbe1222d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
