)]}'
{"id":"openstack%2Fdesignate~903410","triplet_id":"openstack%2Fdesignate~master~Ie8c5591dd12a91350173bec2b418b6c9ddea79fe","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-01-07 03:18:40.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-01-07 12:50:47.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ie8c5591dd12a91350173bec2b418b6c9ddea79fe","subject":"Restructure testing","status":"MERGED","created":"2023-12-15 18:10:09.000000000","updated":"2024-01-07 14:10:46.000000000","submitted":"2024-01-07 14:09:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903410","meta_rev_id":"c1431bebd12aecdaf573e9c73987a2943504dda2","_number":903410,"virtual_id_number":903410,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-07 14:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-07 03:18:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-01-07 12:50:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-15 19:43:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-05 21:51:27.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"fb2460c5c08cb71c027bdf0a00d540edafe652dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-15 18:10:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60fcc80a8261be9afb093dd81f611a5080b9f049","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-15 18:23:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cfbe48dc2b3a1ea80b9536ddfb1578f1d112b43f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-15 18:24:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f8cb5f13975d8cd7eebfe7028a1ddafa3c373033","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 19:43:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66bc3c44757b41409e1998b8261e2ba0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7fe7a99f35384689bf7bc300da875881 : FAILURE in 40m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ee9db6655b1f4ab497ca0da37f77594a : SUCCESS in 40m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/433c3b169a6d4debbb116a0d270ff725 : SUCCESS in 51m 48s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/83529e8a2b9144eeaf3a820f7994ace6 : SUCCESS in 43m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/40eb677f44a549cc850385b988cc44c1 : SUCCESS in 37m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c7af481849924e368aed2e1cad4ab726 : SUCCESS in 1h 10m 56s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/37ecb3639c9d456b8e400708993ee06f : SUCCESS in 49m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e15bcbe8e24e43a98f216da30e6b4ca2 : SUCCESS in 52m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fa9db210908b491eb2491d24df433c8b : SUCCESS in 50m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0643f31b8044b9287949c5029390233 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a888edc141414e5e926fed82ec172fd7 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae42c914e23c40b38ff033cf913f5e3c : SUCCESS in 6m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df0112365cdf4789901dff92edea33fc : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a338f039ecf74f87a6845ea05ac2276f : SUCCESS in 9m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ac452463491f4a8bbffbb01a08413c67 : SUCCESS in 31m 01s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/15adbde9b9394c59be6cda38202ca082 : SUCCESS in 7m 25s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ee400781e1354dcb99754335456b71e5 : SUCCESS in 50m 29s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3b206f00086c4126b670f38f1388d98e : SUCCESS in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"38ef337aea6dd3b9d664263ad1aa999e50331ef1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-15 20:09:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5ddf87ee097ed6788c80a7defea50860d2b0a111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 21:32:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7704998f0b4f48a48faabd3a58183f0a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4fc790d35fb94f9796ea496865ba455a : SUCCESS in 41m 04s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/73b099f892f94eb891cae17d5fe4dd20 : SUCCESS in 40m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4630b4682cdc4ce7923f657eda164397 : SUCCESS in 37m 03s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/12b1b35f13c948b19cd82d65907549cc : SUCCESS in 34m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/33179daba99a4261b1705a210e32affc : SUCCESS in 40m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e62eefa1d40c42e9bad4e8410bf7cf70 : SUCCESS in 53m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5c45d504680e46b69123cc947c19507b : SUCCESS in 1h 15m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5cb1547b5d384dff8e66f8f118aa666f : SUCCESS in 37m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/79733e25d33246ae8962aea6cd83d5fc : SUCCESS in 53m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0100da6942c1452eb876e8633cdaa526 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf262105c60f4dedb842dc6438debe65 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/001c97dea3f743eb865cbdc54e90af91 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/984c5423a76b4a169f01c995a54a7bb6 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c021823693ab45a5bec81b39d0c5cf13 : SUCCESS in 7m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aa0baa89f0bc4185ad2f4c1c3583c4b1 : SUCCESS in 33m 16s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9e979e3f28af48eeaae542ff7235f6a5 : SUCCESS in 6m 53s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6e2351b436b7473697d3432ae30c6a39 : SUCCESS in 49m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/969fc59b03174dc5991f372394f5a123 : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8d761a43bd8565062ed602e5caa176861f79ced9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 11:16:20.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c917898d1411b2a6a75fcfa494e0183a30bb3e08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 12:56:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/195df66f0686446d9477f5f72c66fcd7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d72d2069e9f74cc3bff1db90dcf086bd : FAILURE in 48m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e97d1619d89c4c96b814231e447a2a20 : FAILURE in 1h 01m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/772b17ffab9040eb973a2970232fd46b : FAILURE in 57m 20s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/dbfa659fb47c44a9a19a7a51cb2e4259 : FAILURE in 51m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fdb7b5c67e0448c08bf2cc4f3829e8a4 : FAILURE in 51m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f4415da698b2443bb3f04a3a5bf5c82d : FAILURE in 48m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/54fcdcad43394a15878a2180c9bb3e63 : FAILURE in 38m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e96226a6ab60426f9c2f69e923ab2076 : FAILURE in 49m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c00085b80e0b4286bd8895c6b8b6f27a : FAILURE in 54m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75c2038b495b4840be7a2aba979abd95 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/237422f7bd0841e1a29ab598a4185e4a : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5c51a6e5e8443de8140fd8148dfe4a0 : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/599e61b92b8842129f2ba5162d471721 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90454a28c33e4b5c9e74907525dddf21 : SUCCESS in 8m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1adb56462fb14bdb81e37d03d72a9e47 : FAILURE in 41m 03s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2abc171968044ba28d8740bd89dfca8d : SUCCESS in 7m 08s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9794d86096e249eba1291b45ab4fabdd : FAILURE in 38m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/31bfa7f871a94a68b2e0f3e9b607d30a : SUCCESS in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"be46c4e0be2d3bc82d513b59fc987463adfa3e57","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 17:27:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"51561328f8340dcab3d16d7d107288d491826790","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 19:13:38.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d866535f178149618411246a9d14afe1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/036b5b77663a46e0a375aef7a5a5521c : SUCCESS in 47m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/08259b93fc8e48b289726767723a0d76 : SUCCESS in 1h 04m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7fd79f632604827b0bf645029c7eaae : SUCCESS in 54m 47s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ff163e539b4b4a40a94c9d83f1d8089a : SUCCESS in 52m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6b7cb9dc27ec4cb4b5dbe3fecbe8e368 : SUCCESS in 57m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/32e3ad6fc4ba48079b74fa5fca60992d : SUCCESS in 1h 05m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5bcc20a6af324497a925608bcd2f3f5d : SUCCESS in 1h 12m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bda2aae230b948f1807866750f67474d : SUCCESS in 40m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4f63933fa13445f38f121bdf7d8fe0c2 : SUCCESS in 44m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7ac06c237494068885f4f8deb10789c : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e13a8aed56c4911a839d74797437b2d : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/344e1a75635347e3b705bb0467bdf3ac : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d4c57aab2094826a81214fbd8c63cff : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0efb85b5808d4586899d5ef4d76d08b1 : SUCCESS in 8m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e32e2951594b40caa530d8bcfc0272f8 : SUCCESS in 33m 58s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3bee91515d344fb8bfa07bfa91fbb140 : SUCCESS in 7m 39s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5a819ccd64eb49458a5f4ecceb4a5809 : FAILURE in 44m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/08879b2e4688430db3827dfbc25478eb : SUCCESS in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9e3cdd1b4badc617a19486f8e7d20b013351ddcf","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 21:50:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7960a31038f9b6f6698b21e88af9ab3ba43bfb56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 23:07:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86bc66b906c449599d1ce6fab3d5f22c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e5122c317bf045ad842941554ad1cfc4 : SUCCESS in 53m 12s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/41d8a8e55f074257a9a6d8218dbaede1 : SUCCESS in 1h 02m 33s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/24adf556d55f4867ad91fb48c52dbc0e : SUCCESS in 52m 59s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8981c8efcb1c407799b94e79bf5e53e2 : SUCCESS in 44m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f015655a18cf4216914c186a72dff98a : SUCCESS in 47m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/352a1bab2951456c981c896fecb4eede : SUCCESS in 1h 11m 24s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7ef54fc17766455abc217f6f2d472611 : SUCCESS in 1h 13m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/346d7fa5f50e4905b04b2e989b22abb0 : SUCCESS in 51m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f19f1c329e6b40aa8dcdec5a37a73d99 : SUCCESS in 44m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17863d4f34344389a51b2e6871b7314a : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/208738c71e6448b99ff0a137b4b0fc41 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42b8b3170ebe4e81a3e72be65fef57c4 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b54efc59d6e4ef69bee90ca9f7be4f1 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9690bf46fbb46349d419af41ed102ad : SUCCESS in 8m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8b7ee94d3f9e4d5ab0c830815d1234fe : SUCCESS in 47m 17s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0afd426426c447bfad146b2ca965ca9b : SUCCESS in 5m 33s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8eb3939db0a14756871002a3404d6f77 : SUCCESS in 36m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/670fb44410b6452eab031bd816bb5af5 : SUCCESS in 4m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9460e20ee361e0f035ff74c17a95e1377f23524c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-17 17:16:28.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6cd9319bb2dad0568f106d6a7e4a5ea9f60b6745","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 19:10:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1614f0f6e7b742cba2712e1acb81e5dc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6c6f942059b04aad95b87ed92bb17995 : SUCCESS in 50m 08s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/04866592db00422f86875b689620c6cb : SUCCESS in 1h 02m 22s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1929e8dd4eb44141a02911483f6cd427 : SUCCESS in 55m 58s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ebef9e560961462bb29af7f67987b8cc : SUCCESS in 51m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3bcc2dfaf9df43ab82f5e4bfca9117f0 : SUCCESS in 52m 06s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bfb49e3198d44310ac7584991817b4c7 : SUCCESS in 1h 00m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bf85cce1d37742cc92caedb0719ecf1d : SUCCESS in 1h 06m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/faa2331381b54860829352232fa0321a : SUCCESS in 42m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/537650703011449b9aaa6b1f46304a6e : SUCCESS in 45m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/610f9d7772c34c8c8a7b4dbca0e875de : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14a6de0f00174f8b820579c0b8f825f3 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6e155ff2c6e4833846607f512ede785 : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/96e97d31704341a886eb38a9cf5e6b61 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5060188dc474526a12475a60c49d613 : SUCCESS in 8m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2f20e2db5cb643e09e79dd1e43ae6bc5 : SUCCESS in 32m 33s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a9733e512be24199b3e164f2d62ded51 : SUCCESS in 7m 36s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f790d7eac2444493a67c69e99e4c1a7b : SUCCESS in 50m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e874a28e49394d96b77ce854df37bb5f : SUCCESS in 5m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"22bafef1936d6d7bab92069b1e7ed263962dcd29","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-01-05 21:51:27.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4ecaa62e6dd888a99133f566a1e3cf35b7dc0565","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-05 22:51:58.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"148d2c325abf9ffcd294572d1240e5a466efe90a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-05 22:52:09.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"627e9d056a62679f2aef1335d299fed615116065","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-01-05 22:54:11.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d049b82244b3a4eaca21ef961618047e7956fcf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-05 23:09:34.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"cae3acb3e7a6803e0a13b71a4616fbbf07af6efb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-06 00:12:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cce943d1ca5447bb265b7308a8e63d5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bef6f09063bd42f7a0634d256ff6e2aa : SUCCESS in 48m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/231960af1022406bbd3f565d19d38282 : SUCCESS in 1h 01m 15s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2abab55bfb5e4e95903f32b0a27f29e7 : SUCCESS in 56m 08s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/38d2dd334d314434a56240d11e2d11bd : SUCCESS in 30m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e67e4b88debb479fa58c928674abbe4a : SUCCESS in 47m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b16db4400b624412828ce7fc3c67f7a2 : SUCCESS in 51m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/32351365fc1d4178a7ffcada69f97ba4 : SUCCESS in 52m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f2e83ede4eb64fc7aef566bbee9da5d2 : SUCCESS in 46m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/62018ea8b7ac4e6f944aac8d50127b4a : SUCCESS in 50m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be770f0d44394c19a705d52b3362810b : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deef0e2c09634bd1841026f8d7fe58fc : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b44a025bfc974573a11e5de46ab8ab19 : SUCCESS in 6m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d49177a1f5d84d5d82e11b3142d3f54b : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d4f79d03a9f422db31d7aa289f11b70 : SUCCESS in 7m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/be2601364840405085ad2a7e4e6b1212 : SUCCESS in 40m 13s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/36d938e73d7b47788b29534a95d24a3f : SUCCESS in 8m 17s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d3a8cb5e3fb8470a9075b3a126244cdc : SUCCESS in 56m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ae2b64932c1848cd849a881ac56c884d : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d5b5c2ca5276220112c92c815fa0b899a04dec3d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-01-07 03:18:40.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ab9c38df38c8a4f660d25901488fc9a251b7cf75","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 12:50:47.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"c20ab281d5db7eddbbfd17c2aef7969e1f73060e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 12:51:26.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"749ed0c4b5f0e3ebebb7f35109baa5a81a9a5466","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 14:09:43.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbcfed0de7954a76af9637fd3e3cda7e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f407c6afda8d45bb9cb4e33748ee9f8c : SUCCESS in 43m 10s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0eb827b974f2462997b1fa7c8deec3a1 : SUCCESS in 52m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/13c5aa68ab494a0eafce1fa41fde15f0 : SUCCESS in 46m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cc5593e610ca4b30b253e17ffa1f201f : SUCCESS in 1h 06m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4eb8dfb99f114bb79871ce324f3c9f39 : SUCCESS in 48m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6795bdc49454463fa35b89273d589e7b : SUCCESS in 47m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99191be4944545c4be45c14a7c9ca6b4 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a68d70386544ea09ff8d8c3e65585fb : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1dbd71263fc460fbcd5a881a7e8a98d : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ab2656f12f74d69a76e688682976448 : SUCCESS in 8m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7a0cde60903c44e9b5a58c5930388fb2 : SUCCESS in 42m 22s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/53215b5ca55c4c41987cfb72e100b25d : SUCCESS in 5m 56s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5891cafa03754b5aaa61542a1f2ee7aa : SUCCESS in 48m 03s","accounts_in_message":[],"_revision_number":7},{"id":"062461a279fbda06a83b3c8e7d789778f19c208d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 14:09:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"c1431bebd12aecdaf573e9c73987a2943504dda2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 14:10:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae1991a0c65e456b86b95da8632ad40b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d700d7acf10c417fa857216109948c6b : SUCCESS in 45s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"57f2e367bf9263da87ab7238d9176d50781eab7b","revisions":{"ec6dbf982e725f8de523fa764ff2a2029af7c5cd":{"kind":"REWORK","_number":1,"created":"2023-12-15 18:10:09.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/903410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/903410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/903410/1"}}},"commit":{"parents":[{"commit":"f5a034272ddeddf80dc858396cf31ab581f4ae5e","subject":"Add upgrade test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5a034272ddeddf80dc858396cf31ab581f4ae5e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-15 16:06:57.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-15 18:09:10.000000000","tz":-480},"subject":"Restructure testing","message":"Restructure testing\n\n- Moved all functional tests to dedicated folder.\n- Updated a few tests to use faster oslo tests.\n\nChange-Id: Ie8c5591dd12a91350173bec2b418b6c9ddea79fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ec6dbf982e725f8de523fa764ff2a2029af7c5cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ec6dbf982e725f8de523fa764ff2a2029af7c5cd"}]},"branch":"refs/heads/master"},"eb008a22e2649748301b1ec8e26f2c2b3dda30da":{"kind":"REWORK","_number":2,"created":"2023-12-15 18:23:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/903410/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/903410/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/903410/2"}}},"commit":{"parents":[{"commit":"f5a034272ddeddf80dc858396cf31ab581f4ae5e","subject":"Add upgrade test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5a034272ddeddf80dc858396cf31ab581f4ae5e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-15 16:06:57.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-15 18:23:38.000000000","tz":-480},"subject":"Restructure testing","message":"Restructure testing\n\n- Moved all functional tests to dedicated folder.\n- Updated a few tests to use faster oslo tests.\n- Renamed fixtures.py to base_fixtures.py.\n\nChange-Id: Ie8c5591dd12a91350173bec2b418b6c9ddea79fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb008a22e2649748301b1ec8e26f2c2b3dda30da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb008a22e2649748301b1ec8e26f2c2b3dda30da"}]},"branch":"refs/heads/master"},"124685db7a9d107b1ea9b8638377b0e3ddf0da32":{"kind":"REWORK","_number":3,"created":"2023-12-15 18:24:27.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/903410/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/903410/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/903410/3"}}},"commit":{"parents":[{"commit":"f5a034272ddeddf80dc858396cf31ab581f4ae5e","subject":"Add upgrade test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5a034272ddeddf80dc858396cf31ab581f4ae5e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-15 16:06:57.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-15 18:24:22.000000000","tz":-480},"subject":"Restructure testing","message":"Restructure testing\n\n- Moved all functional tests to dedicated folder.\n- Updated a few tests to use faster oslo tests.\n- Renamed fixtures.py to base_fixtures.py.\n\nChange-Id: Ie8c5591dd12a91350173bec2b418b6c9ddea79fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/124685db7a9d107b1ea9b8638377b0e3ddf0da32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/124685db7a9d107b1ea9b8638377b0e3ddf0da32"}]},"branch":"refs/heads/master"},"7676d38cd80ef32f74c23167c5cf165998adee90":{"kind":"REWORK","_number":4,"created":"2023-12-16 11:16:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/903410/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/903410/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/903410/4"}}},"commit":{"parents":[{"commit":"8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78","subject":"Add new API for zone move","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-15 16:06:57.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 11:16:07.000000000","tz":-480},"subject":"Restructure testing","message":"Restructure testing\n\n- Moved all functional tests to dedicated folder.\n- Updated a few tests to use faster oslo tests.\n- Renamed fixtures.py to base_fixtures.py.\n\nChange-Id: Ie8c5591dd12a91350173bec2b418b6c9ddea79fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7676d38cd80ef32f74c23167c5cf165998adee90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7676d38cd80ef32f74c23167c5cf165998adee90"}]},"branch":"refs/heads/master"},"be0aabc82c7b7697ee6c4cf1603e709680e4abb1":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-12-17 17:16:28.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/903410/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/903410/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/903410/5"}}},"commit":{"parents":[{"commit":"af940fb6e4431c9eb9799cc983a85465ec865011","subject":"Increase minimum test coverage to 96%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af940fb6e4431c9eb9799cc983a85465ec865011"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-15 16:06:57.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 17:16:28.000000000","tz":0},"subject":"Restructure testing","message":"Restructure testing\n\n- Moved all functional tests to dedicated folder.\n- Updated a few tests to use faster oslo tests.\n- Renamed fixtures.py to base_fixtures.py.\n\nChange-Id: Ie8c5591dd12a91350173bec2b418b6c9ddea79fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/be0aabc82c7b7697ee6c4cf1603e709680e4abb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/be0aabc82c7b7697ee6c4cf1603e709680e4abb1"}]},"branch":"refs/heads/master","description":"Rebase"},"937bd4529930d0f4acc0d59873c10e4994ad11df":{"kind":"REWORK","_number":6,"created":"2024-01-05 22:51:58.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/903410/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/903410/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/903410/6"}}},"commit":{"parents":[{"commit":"af940fb6e4431c9eb9799cc983a85465ec865011","subject":"Increase minimum test coverage to 96%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af940fb6e4431c9eb9799cc983a85465ec865011"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-15 16:06:57.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-05 22:51:44.000000000","tz":-480},"subject":"Restructure testing","message":"Restructure testing\n\n- Moved all functional tests to dedicated folder.\n- Updated a few tests to use faster oslo tests.\n- Renamed fixtures.py to base_fixtures.py.\n\nChange-Id: Ie8c5591dd12a91350173bec2b418b6c9ddea79fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/937bd4529930d0f4acc0d59873c10e4994ad11df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/937bd4529930d0f4acc0d59873c10e4994ad11df"}]},"branch":"refs/heads/master"},"57f2e367bf9263da87ab7238d9176d50781eab7b":{"kind":"REWORK","_number":7,"created":"2024-01-05 23:09:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/903410/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/903410/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/903410/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/903410/7"}}},"commit":{"parents":[{"commit":"af940fb6e4431c9eb9799cc983a85465ec865011","subject":"Increase minimum test coverage to 96%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af940fb6e4431c9eb9799cc983a85465ec865011"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-15 16:06:57.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-05 23:09:29.000000000","tz":-480},"subject":"Restructure testing","message":"Restructure testing\n\n- Moved all functional tests to dedicated folder.\n- Updated a few tests to use faster oslo tests.\n- Renamed fixtures.py to base_fixtures.py.\n\nChange-Id: Ie8c5591dd12a91350173bec2b418b6c9ddea79fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/57f2e367bf9263da87ab7238d9176d50781eab7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/57f2e367bf9263da87ab7238d9176d50781eab7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
