)]}'
{"id":"openstack%2Fdesignate~903411","triplet_id":"openstack%2Fdesignate~master~I59926b701e375b01dd46fc0fad0ce2db6dd73fe7","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-01-07 13:06:03.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I59926b701e375b01dd46fc0fad0ce2db6dd73fe7","subject":"Improved pool move test coverage","status":"MERGED","created":"2023-12-16 14:08:44.000000000","updated":"2024-01-07 15:50:20.000000000","submitted":"2024-01-07 15:49:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"903411","meta_rev_id":"a556e84a59ffaab2a0e6c72e01147945d35fd954","_number":903411,"virtual_id_number":903411,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-07 15:49:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-07 13:05:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-01-07 13:06:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-16 15:50:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-05 22:02:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"f3ff60b122f956b7d53d3dfcb182e702bc8f788b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 14:08:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e72b1a0199e6b97196a90dde4bb5c54c17ae4df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 14:10:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2342f513c188c57e4eac48ceaf5f777372d2e171","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 14:12:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"35142d63171fb39cf2fa29969c492a8fbe1177aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 14:21:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5227b4caa58fb04d12e633ba744ba3e847365415","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 14:23:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"dc1bbe28cdb5890517426d6f433996efff5e15b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 15:50:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b5d516a918145038e9b72baa4ced1c7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/be8efc255f154a12987a3a5438be3390 : SUCCESS in 48m 41s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/fa7667bb65c5478fbee730a2dc973418 : FAILURE in 35m 21s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9feac3fb2bd94ee78e03d717778ecfd7 : FAILURE in 49m 52s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e817b0cb96f84fc1b4aa72b1ad40bb52 : FAILURE in 27m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4424bd1dd1314e6b8f4d5dfcd5ffbc24 : FAILURE in 35m 12s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/067e3344e9fa4eac9f1c433c0985d05e : FAILURE in 45m 43s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/70b8622451bb41599449ee1975188379 : FAILURE in 42m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/862c4f765a3f45eb98e05cb26f9fad51 : FAILURE in 34m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5ae7a7e146774e3981c6736c0a5b1028 : FAILURE in 31m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6783001c094466c8bed1ed4b8b8a18d : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/064d79cc83e8430897f9f0a7f762aa8a : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28bcdbf4ceb74e9e9e3cf61924f2d8c0 : SUCCESS in 6m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/192c124fd01d44a4935cb819668ce6ae : FAILURE in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a34c3bf6b87646e3948f67fddcbd97bd : SUCCESS in 8m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f66dadbb30664190b3539442f487fbad : FAILURE in 34m 27s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/df229059990345518e5c60d1e49de6bc : FAILURE in 7m 11s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/142990405209403c94bb4b9390ca1bf6 : FAILURE in 27m 32s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1b1f3194d21e4f96b8474e22b739036f : FAILURE in 6m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2865ffbb4bb7ebc20335274c52b4d6e05d794a04","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 17:27:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c76070ae02db86e495954107bf1d2a788b7f7364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 19:09:00.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbf4a63d97c445d4af0b3b8e44003156\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a57cf9ec9f7c46678343b034a632e0d4 : SUCCESS in 40m 27s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f940741d5c514de9bc9578fb78cca881 : SUCCESS in 1h 08m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0550e467a95a41cdb27ba290c76ce6d3 : SUCCESS in 57m 49s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2d205332b21f4037943d52a472a5d79a : SUCCESS in 49m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5a0af1574db546f29cb43c6195ad0b52 : SUCCESS in 53m 52s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e2dde8ce53d9469ca21add0343e91313 : SUCCESS in 1h 12m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b9a710ee1834434d8d5aa1f8b7b8586d : SUCCESS in 1h 18m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c40b0c5af2a143d3920b9e988662ba22 : SUCCESS in 41m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d628d54409bb46f283d8cb32c0f80260 : SUCCESS in 40m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29850323dfff4d56bef428d2e12ed7d2 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1c38cc41b674cef9221553ac5bdd5a9 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ad737ac5f674744b48c1bb8b9906abc : SUCCESS in 7m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/684ef5eb92f64c1092c6c8ae42547406 : FAILURE in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4905f9307c5b41a094007107a29249e8 : SUCCESS in 10m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f35cced92db147ae9a2e1085cd309936 : SUCCESS in 33m 18s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b0d74c01785040c38c88c4afd97ef43c : FAILURE in 8m 11s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9311090dc9cb42e9997e348c0034accb : SUCCESS in 49m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/eab818779ca1452c9b3977b8a370e0ba : FAILURE in 7m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"117de2494cc92e7c80e32e232dd3b0592dc3c848","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 21:55:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ecf0e04f8aef77279e81869b7360ddc1a1b789f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-16 22:39:34.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"2affe0293ffbe1a33704114e21990d6c683b0424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 00:46:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92567e56b8ce49fe8eaadb7117643e33\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4e340a057c8141afaa7d297769841426 : SUCCESS in 46m 52s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/cf5299db1a3045ceaafbdf4af0b80b44 : SUCCESS in 1h 03m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/518711023c1742bf8812194bc0e05e0c : SUCCESS in 52m 18s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/72d197dbb2ac490899fc04b5df08f548 : SUCCESS in 39m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f574183b5c1e43efa9e2ff562842bc6b : SUCCESS in 39m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6c3a420605fc40cdbfabfe125b9fe20f : SUCCESS in 1h 03m 10s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/43fdc8f0054345139384c40247b2b165 : SUCCESS in 1h 10m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4c795393daec4366b5f7f3778756051d : SUCCESS in 53m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dabc0f0e6bef4f0a9333cbd59d7fde0b : SUCCESS in 36m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30a02034a82a40429548f6c8ceb2da17 : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3378dc825f88472fbf0d0eb7843a7d39 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50b28f376eba42e6b453c4b6459296e8 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a77aad5c589474481ec6b643afe8a29 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed48d70fd96e4250a5f8d24e32d27687 : SUCCESS in 8m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/55887a0a81284dce995167c7c3ec8694 : SUCCESS in 35m 34s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5a12256c09714716b170d654b2982c24 : SUCCESS in 7m 00s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2ea4de3dc495467580e6de3498134be8 : SUCCESS in 54m 15s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5e020110e4a84498b7c4c45b356822be : SUCCESS in 5m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5fc39cd6e67054b99360e44198d0bc858d8854a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-17 17:16:33.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"139f81deed39ef237896bd87144d6e681aa82723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 18:49:57.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/465180ef29e04cd1af6a74c33f360a26\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e9dd7e6892eb4e67acaf35223d0591fa : SUCCESS in 46m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/650c037071c546b1855f34fa638265e4 : SUCCESS in 52m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/165c601a4da74dc7892226f54772f075 : SUCCESS in 52m 04s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/01a88b84cab6468f9f88774172273ca0 : SUCCESS in 40m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1edc002c57ae44c0ade1454a0db5217d : SUCCESS in 50m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d9c09af3258843edbb764a02e0504a7d : FAILURE in 1h 12m 13s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c951fa3c465747c09c766dc48d2c141c : SUCCESS in 1h 07m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d459e65c60ff4b299adf2ca2ad7941ce : SUCCESS in 45m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c4fbd483e8b4457bb923ee108d11054f : SUCCESS in 50m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58083af0a9f04c35acd6964e0f354c29 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fedb6456b25647c9ba95bcf9e647054f : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9568f65b12c442838728f0223e1abed5 : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3fd685b842f8497c90a9a02d00937ef0 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eabcb442b53493bba8bda3080ab3095 : SUCCESS in 10m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a3bd8ebee3bf49ec9d2509b3ad340002 : SUCCESS in 32m 59s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/504db439a6a54d33b32a8dc1e8d81993 : SUCCESS in 7m 30s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/17f4c5dd9886434a9ca406d4e71dce70 : SUCCESS in 49m 35s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0e2ab6cdba8443be98691e442e4e3a3a : SUCCESS in 5m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a9a750e41975381c61d705c314aee3eba4ff60b4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-01-05 22:02:43.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"97558b621acb902450e4e36f1939cbbd318aea30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 13:05:59.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"89fe326e2eb171841310028f1f6176a901b64227","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 13:06:03.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"4371edd384f856acba412ad764077c2208c45ebb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 14:24:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a33776028e9c4937870992a503d24c0c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0d5876a2ec624d0aaee61115352d761f : SUCCESS in 38m 12s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/abc8d2d0bd1248d180470cef75e7e0cd : FAILURE in 48m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/97b39f9760284087a77e79c742644d8a : SUCCESS in 39m 22s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7fe9985cb48e4c8fa195deeb967df582 : SUCCESS in 36m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bd71e39568a94c7dbabef6d8214fd535 : SUCCESS in 31m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cf29ccfd2de545758b8e88864e0edd3e : SUCCESS in 1h 13m 09s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ed9fa4f3420d4197ae5074f496aea549 : SUCCESS in 1h 09m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/27673490bb6d44d99585a5a0fc841f9c : SUCCESS in 45m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/287d907989c3419dad3d0a748d6369c7 : SUCCESS in 50m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2fe23cf51d445d7b04ac1153bfe59c9 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1939a7e5e9394f7680df97aa86d02eb2 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b108e03dcd541d591a7a4961b156df9 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ac807bf70f8480691ac40a45dd0e995 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c01b525085c470384a7cf36a4b78bd8 : SUCCESS in 7m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a5d8e977b5164fd9b6814bda9f713574 : SUCCESS in 35m 05s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/09dae5a2cbbb4fdc9ddcda91b8862df9 : SUCCESS in 6m 17s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e40ee96e259a4a3192d1e0d602dfe8f1 : SUCCESS in 48m 17s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a21fa438cd1145a4b15a7fe869ba53b9 : SUCCESS in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0508234da74e8c82734bd803fd05ec1705b65163","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 14:25:34.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"4acb4c27e75bd0095f29cf541aa3aa29a54e3e56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 15:49:08.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e03c3adf59b349ccb220620eba812a22\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cb662e778c57447eaad0a1e865de4281 : SUCCESS in 45m 08s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/df9d8ef7e5b24c12a99bd06698fd597b : SUCCESS in 46m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4cad5706707f4d5cb396bab7b32b11c6 : SUCCESS in 30m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/687a8becf771438c8d695c7707af02a0 : SUCCESS in 1h 15m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/def2df166d884f7a9ac335f3e576deda : SUCCESS in 51m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/342196382a9b40c89c569657722da2e2 : SUCCESS in 43m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17031971df1f491295b55002bd16df58 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a490bf69681468783e2c70216227bb0 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae446681fd094252b76246918d1fb14a : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e2d0d90d5fc4c51a5a9b25c069da6a1 : SUCCESS in 9m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5b6d6118b79f430389d14c009e4c69ec : SUCCESS in 49m 54s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/41df5008a72640079bb67c73b17fe1e2 : SUCCESS in 7m 58s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b8c781092c764dea8368bc8f09568605 : SUCCESS in 42m 58s","accounts_in_message":[],"_revision_number":8},{"id":"89a11cb5b9a4baff215a92cf3c7a8626f7c0ee58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 15:49:08.000000000","message":"Change has been successfully merged\n\n7 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":8},{"id":"a556e84a59ffaab2a0e6c72e01147945d35fd954","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 15:50:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8b81565a2d7432a8c492a771f5dd28f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bd339ab387948c1acdfba7e285596c0 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"571d902c59bdcbe031d479432caf18ba2586e86f","revisions":{"2e0cd35c4416c45343fa229cf36b5997f2f66a9b":{"kind":"REWORK","_number":1,"created":"2023-12-16 14:08:44.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903411/1"}}},"commit":{"parents":[{"commit":"8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78","subject":"Add new API for zone move","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 14:08:22.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 14:08:36.000000000","tz":-480},"subject":"Improved pool move test coverage","message":"Improved pool move test coverage\n\nChange-Id: I59926b701e375b01dd46fc0fad0ce2db6dd73fe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e0cd35c4416c45343fa229cf36b5997f2f66a9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e0cd35c4416c45343fa229cf36b5997f2f66a9b"}]},"branch":"refs/heads/master"},"57c8c4966752ba208147f64fb88e5e62d792c6de":{"kind":"REWORK","_number":2,"created":"2023-12-16 14:10:05.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903411/2"}}},"commit":{"parents":[{"commit":"8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78","subject":"Add new API for zone move","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 14:08:22.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 14:09:55.000000000","tz":-480},"subject":"Improved pool move test coverage","message":"Improved pool move test coverage\n\nChange-Id: I59926b701e375b01dd46fc0fad0ce2db6dd73fe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/57c8c4966752ba208147f64fb88e5e62d792c6de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/57c8c4966752ba208147f64fb88e5e62d792c6de"}]},"branch":"refs/heads/master"},"a6e2986472bb96cf13a6ef0ac28bd823ffa6f74e":{"kind":"REWORK","_number":3,"created":"2023-12-16 14:12:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903411/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903411/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903411/3"}}},"commit":{"parents":[{"commit":"8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78","subject":"Add new API for zone move","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 14:08:22.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 14:11:49.000000000","tz":-480},"subject":"Improved pool move test coverage","message":"Improved pool move test coverage\n\nChange-Id: I59926b701e375b01dd46fc0fad0ce2db6dd73fe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a6e2986472bb96cf13a6ef0ac28bd823ffa6f74e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a6e2986472bb96cf13a6ef0ac28bd823ffa6f74e"}]},"branch":"refs/heads/master"},"0222e805067b90d2ff47fbfeb02ec854ed0de087":{"kind":"REWORK","_number":4,"created":"2023-12-16 14:21:11.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903411/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903411/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903411/4"}}},"commit":{"parents":[{"commit":"8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78","subject":"Add new API for zone move","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 14:08:22.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 14:20:59.000000000","tz":-480},"subject":"Improved pool move test coverage","message":"Improved pool move test coverage\n\nChange-Id: I59926b701e375b01dd46fc0fad0ce2db6dd73fe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0222e805067b90d2ff47fbfeb02ec854ed0de087"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0222e805067b90d2ff47fbfeb02ec854ed0de087"}]},"branch":"refs/heads/master"},"e95d26627995758a5d4ef506d7be5699c6d1949c":{"kind":"REWORK","_number":5,"created":"2023-12-16 14:23:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903411/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903411/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903411/5"}}},"commit":{"parents":[{"commit":"7676d38cd80ef32f74c23167c5cf165998adee90","subject":"Restructure testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7676d38cd80ef32f74c23167c5cf165998adee90"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 14:08:22.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 14:23:21.000000000","tz":-480},"subject":"Improved pool move test coverage","message":"Improved pool move test coverage\n\nChange-Id: I59926b701e375b01dd46fc0fad0ce2db6dd73fe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e95d26627995758a5d4ef506d7be5699c6d1949c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e95d26627995758a5d4ef506d7be5699c6d1949c"}]},"branch":"refs/heads/master"},"a901e9a40bf8da92168cf11b046f5585eeacabbe":{"kind":"REWORK","_number":6,"created":"2023-12-16 22:39:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903411/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903411/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903411/6"}}},"commit":{"parents":[{"commit":"7676d38cd80ef32f74c23167c5cf165998adee90","subject":"Restructure testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7676d38cd80ef32f74c23167c5cf165998adee90"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 14:08:22.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 22:39:26.000000000","tz":-480},"subject":"Improved pool move test coverage","message":"Improved pool move test coverage\n\nChange-Id: I59926b701e375b01dd46fc0fad0ce2db6dd73fe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a901e9a40bf8da92168cf11b046f5585eeacabbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a901e9a40bf8da92168cf11b046f5585eeacabbe"}]},"branch":"refs/heads/master"},"d511e858084de6541af991b63aa53388ab2b444a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-12-17 17:16:33.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903411/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903411/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903411/7"}}},"commit":{"parents":[{"commit":"be0aabc82c7b7697ee6c4cf1603e709680e4abb1","subject":"Restructure testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/be0aabc82c7b7697ee6c4cf1603e709680e4abb1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 14:08:22.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 17:16:33.000000000","tz":0},"subject":"Improved pool move test coverage","message":"Improved pool move test coverage\n\nChange-Id: I59926b701e375b01dd46fc0fad0ce2db6dd73fe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d511e858084de6541af991b63aa53388ab2b444a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d511e858084de6541af991b63aa53388ab2b444a"}]},"branch":"refs/heads/master","description":"Rebase"},"571d902c59bdcbe031d479432caf18ba2586e86f":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-01-07 13:05:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/903411/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/903411/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/903411/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/903411/8"}}},"commit":{"parents":[{"commit":"57f2e367bf9263da87ab7238d9176d50781eab7b","subject":"Restructure testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/57f2e367bf9263da87ab7238d9176d50781eab7b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-16 14:08:22.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-07 13:05:59.000000000","tz":0},"subject":"Improved pool move test coverage","message":"Improved pool move test coverage\n\nChange-Id: I59926b701e375b01dd46fc0fad0ce2db6dd73fe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/571d902c59bdcbe031d479432caf18ba2586e86f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/571d902c59bdcbe031d479432caf18ba2586e86f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
