)]}'
{"id":"openstack%2Fdesignate~903412","triplet_id":"openstack%2Fdesignate~master~Ib3c11d792ff1a5c0110ab2263e7a8a6d5f6baae0","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-01-07 15:49:10.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-01-05 22:54:48.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ib3c11d792ff1a5c0110ab2263e7a8a6d5f6baae0","subject":"Improved V2 controller coverage","status":"MERGED","created":"2023-12-17 00:13:07.000000000","updated":"2024-01-07 15:49:10.000000000","submitted":"2024-01-07 15:49:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"903412","meta_rev_id":"ca776af496e2a175af50b9ec06e4a7f41d05bd18","_number":903412,"virtual_id_number":903412,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-07 15:49:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-07 13:06:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-01-07 13:06:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-17 02:00:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-05 22:25:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"b381b2513939b7c1e032733329cadf8f5a38f9b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-17 00:13:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"418c8de75080df3d5e29be9f340c4c545e47e674","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-17 00:15:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"376aa34454a79610d2e4ff76ea781ed580bcd6ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 02:00:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e55cd0d5f3c4c2f92bdb1ca6fab67a7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4e9fce349c2b438d88e8f2851b16c8bf : SUCCESS in 43m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2852594cbc254f2d93958337651dfb17 : SUCCESS in 50m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/606c0ca687474eb38902b0dd5f71b26b : SUCCESS in 46m 53s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0440e69e0dc0430d82a369c91e12a726 : SUCCESS in 42m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f7eac1d4a22443bd97360eb58115bdc0 : SUCCESS in 52m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0e05ad08d12f450db9ac17fce882ba24 : SUCCESS in 1h 06m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4a144a661ac249dea58f464dabad648c : SUCCESS in 1h 02m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c272f24078644281b0f6c670c40f5ca9 : SUCCESS in 40m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/35e1652360054a839c225897e6671cb1 : SUCCESS in 52m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/325b77910b1549708cf151d5f5d59411 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/940430b1ecf041f9b2d309e9318fb7b8 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc37390ecde0444d9ede87df3faa14ae : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7ad661bb3764a9986d89594524bded0 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c25c9af29b4449f8b3d45a5279891ee : SUCCESS in 7m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ee6e9e0cc240465ca5169c8a0dcbe7cd : SUCCESS in 35m 31s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/182483cf6b8d46fcbd3f74128c94b32c : SUCCESS in 9m 23s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f45b6af58ac248bdb1073954f6128e2a : SUCCESS in 51m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ce5994e210504fe59b842690aae25e35 : SUCCESS in 6m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37fc60bef9164644084f0d7d782857bf77c355d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-17 16:45:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"987cad5dd0aecdee810b6c28c5cab00305103241","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-17 17:16:39.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"951ba8242922da9acb13a4009ee56498d6c83614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 19:06:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8497fecdfa3b4029ba06b66328998aac\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dea9fc55bc904e74a9cd2774c465a528 : SUCCESS in 49m 26s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8a509849090a4976bdad59c025e589a9 : SUCCESS in 1h 08m 14s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd7675136f5d45b38e2c097dfe293bb4 : SUCCESS in 56m 24s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8ecc33a81c824682ae64d0e9372b5ffa : SUCCESS in 40m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e2367c2becac4f91b8bd339aefd52fe5 : FAILURE in 57m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9a01cea74dcb4145a609a09065b6e0f8 : SUCCESS in 1h 11m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fe77eae48c834072a99741a5919f4ae0 : SUCCESS in 1h 08m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/95253e4927254fd584e7e94ba88b1475 : SUCCESS in 46m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/94aeaa8508a54e6fb889129e80816962 : SUCCESS in 44m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65d405b8d7364a3390392b03c17a8b50 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/409899be4d2c46bfa078b80422c05c89 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7d04c6ce94940ae86e316c0c9a7a26c : SUCCESS in 6m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8687fc547d54825a6bc170f37c7588a : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c0cf9ba45af4319b34ba89696bd3a05 : SUCCESS in 10m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f6ac89dc28eb41bfac48d160b6c40f30 : SUCCESS in 33m 06s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b65a31dba5f84ed09713b7e3a7f86e1f : SUCCESS in 6m 33s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bc7730e96ed4486e9d9345d5284c78df : SUCCESS in 54m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7fd61cb8b7164bc5bbbf5183bc1cad89 : SUCCESS in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1a78ddd2198c34da27781410d69c22bf6045d241","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-17 20:12:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ed14edf65834d845c5dd260bcc77c71579334c3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 22:13:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb737b8681704f4abc59ba08b39ca843\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f8c6ab43f64b48009dcbb5f504b07237 : SUCCESS in 42m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a3944c17f4f446a09d0ac20b9b47b5db : SUCCESS in 43m 37s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8da324c092544289a0610be816977dde : SUCCESS in 47m 29s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4890aff50a7c4f929ef6e925377a7d7d : SUCCESS in 50m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e450477ad74a42a9a53a00387ed55ac8 : SUCCESS in 44m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/455878ee759a44f0a359bc4e73cf747e : SUCCESS in 1h 09m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/00493fcb088a4cb6ab4f6dfd3fc3db70 : SUCCESS in 1h 11m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f996061797b548bf83b1ea6bfaa25a8e : FAILURE in 46m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a9c172fbe7624b5883312f3a8fad91c5 : SUCCESS in 52m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d566ebd0d024925afe2b7473dbf5eee : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26635bc8730f4abd98f39fad4b3b8bcc : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d0448eda0ca4f519c075f75088cc1c0 : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/424730b620ff4806a650a41d8d4f95a8 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03f4471a4a1f4bc497edb79aa102a2a5 : SUCCESS in 10m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ac542fa8f4ac4455a844745c2d1c0019 : SUCCESS in 30m 28s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b7b604a00359436eaf99ccf4a50bea7d : SUCCESS in 8m 12s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f769e34debe84567b2f36d422d8ec6d5 : SUCCESS in 44m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2e5562ad3c8e4dad8b513fbf2d201b61 : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"520f73a339120eb352717599f5b4211ef7ed23fb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-17 22:14:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cfa6f4b58edfecffcf31d908c02b445e99e1e694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 00:06:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3a244b3efae41479b3679418f5abfbd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6feb5a50439a41fabe500924db37f6f3 : SUCCESS in 46m 33s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/956a1d671a88491db51cd11a79bd08c2 : SUCCESS in 59m 26s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d11ac92476a945e6a7c32f478de123bf : SUCCESS in 56m 24s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ce0401981f324304a2bb712487f0111a : SUCCESS in 44m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e23d30bd21a64454972bb9711f161ee0 : SUCCESS in 38m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0f83429b21d040bfb68bda93ca9a096c : SUCCESS in 59m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1b121a0f85c440e89e002678911787d8 : SUCCESS in 1h 13m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/43cae589ad4b4717b56fee2792f78cde : SUCCESS in 49m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5ace34fc881245c5af273e5882818af9 : SUCCESS in 44m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/694c79c4b053415d8ecd81fabb9262a7 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f58131228cc0460986be2799660be790 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1ea6bf4ae2b4aa891313dbd1a9b6bdf : SUCCESS in 7m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5d2706ffc7948a791a3407d78c7dc75 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcf112ff873e4933b838e5d24855fed3 : SUCCESS in 8m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/833bb0c6e08947219731e5ed4dab2a44 : SUCCESS in 33m 13s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/16e82691de574f3696e94281c53ea8ec : SUCCESS in 6m 55s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/40c11298b90145ca9c03168a012d774d : SUCCESS in 45m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/328cdb88520e4ba08f0ac6ca40fdff54 : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f4706b51c286ff7b5cffd750227ef5f9a77f6fc2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-27 01:32:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fd4d049bcca7256d19d5e9f5760a18fe5bdd6daa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-01-05 22:25:46.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f861f1fb45d846a8203c1229692a369650cee4f6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-05 22:54:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"52a799d9f1350c2fc55e1f611597925c3022bc15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 13:06:08.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"491e4816787ede68a2a0f8810fb8fb71e5802b3b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 13:06:11.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0256e7471a99c3f0e02b1389db87311cba67f6a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 14:21:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae7b0f74e3684e22a06cfe35eec75e00\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c6a05fe35a6e48c685d75e2f47965f95 : SUCCESS in 49m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ba9f5cd474d24cc3a0302ec6d2a4759e : SUCCESS in 1h 00m 48s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/549f05f9b2644ae3a7863b9e5e544617 : SUCCESS in 54m 29s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/64ef140259274ef988d1c55f0c9d74a6 : SUCCESS in 32m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f3e37f2ee20e466285295e369ddb4cb6 : SUCCESS in 44m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bf569d058df04dea987a1c80491a3367 : SUCCESS in 58m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f6544bff703e483d9375f2e56ce868f8 : SUCCESS in 54m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4e50a62de973457ea4a60a89bba17d53 : SUCCESS in 36m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/97df0de5a8784730a6c45ba3785b3dc6 : SUCCESS in 35m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d6b290be68e4e00b6c64bc2caea8583 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13575537595a4f1a86fd2caa6fa3dad6 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79dc759b416c403ca6f2195fb28401b2 : SUCCESS in 7m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2ec8633345b4747b28770829427cb2f : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/945a4b5c3e1847c381b52836094f625a : SUCCESS in 9m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fe85aa30d7914315b96eb019fdb77e0f : SUCCESS in 48m 41s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b59ca42c86ae427a8cc6a2391fd9fa42 : SUCCESS in 7m 51s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0aa64f1ad74d46a8ae684dd3eb1b820b : SUCCESS in 34m 13s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3d210a51b0c74a98a061c0c48186d228 : SUCCESS in 5m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2f68d8ac662f3bdc5102822185a32c6bb9c6fa53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 14:25:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a4947f7be815f399c88b1fe5e0c0951e24364806","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 15:49:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4633860d94447a78935817ad0f96bcd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4c64488de9dd4cea802238db42aa4254 : SUCCESS in 45m 30s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/149fd2b822ad4ec9a5756679a9efa539 : SUCCESS in 38m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1a5c832ab26945a99b8ed7e5a0060383 : SUCCESS in 37m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f952ccc4f982412c99bd56351bc99b3e : SUCCESS in 1h 00m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/69d2a79364034c59a1b2ef6b662f2282 : SUCCESS in 42m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c9ab91521bbe4d5a8654359a5fec2705 : SUCCESS in 50m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2793ecda22c646f49d87b1159a544bea : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60a566c94d584d809a002f2a3be8e2ec : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/967d2485b56043caa42e218a57b29bbb : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a7f4d05852a4c6a92edd7001a9b840d : SUCCESS in 7m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6fd34c84a87d49f489d22587663a7f7e : SUCCESS in 35m 40s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/20b7b91cb6184882836ecc3f5b724ae7 : SUCCESS in 6m 35s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f5ba547c90fb46b1bf7bed9682088ad6 : SUCCESS in 34m 45s","accounts_in_message":[],"_revision_number":4},{"id":"ca776af496e2a175af50b9ec06e4a7f41d05bd18","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 15:49:10.000000000","message":"Change has been successfully merged\n\n3 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"39a8c8f116c05911a2ef369fafc956e3d07c7ebf","revisions":{"e594484faab371b89196aa066e72d4985f39bd80":{"kind":"REWORK","_number":1,"created":"2023-12-17 00:13:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/12/903412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/12/903412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/903412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/903412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/903412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/12/903412/1"}}},"commit":{"parents":[{"commit":"a901e9a40bf8da92168cf11b046f5585eeacabbe","subject":"Improved pool move test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a901e9a40bf8da92168cf11b046f5585eeacabbe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 00:11:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 00:12:27.000000000","tz":-480},"subject":"Improved V2 controller coverage","message":"Improved V2 controller coverage\n\n- Removed code that was impossible to reach.\n\nChange-Id: Ib3c11d792ff1a5c0110ab2263e7a8a6d5f6baae0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e594484faab371b89196aa066e72d4985f39bd80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e594484faab371b89196aa066e72d4985f39bd80"}]},"branch":"refs/heads/master"},"2bf3f05deb897aaa936da6fccc6b0f71a7b77b5c":{"kind":"REWORK","_number":2,"created":"2023-12-17 16:45:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/12/903412/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/12/903412/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/903412/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/903412/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/903412/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/12/903412/2"}}},"commit":{"parents":[{"commit":"a901e9a40bf8da92168cf11b046f5585eeacabbe","subject":"Improved pool move test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a901e9a40bf8da92168cf11b046f5585eeacabbe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 00:11:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 16:44:42.000000000","tz":-480},"subject":"Improved V2 controller coverage","message":"Improved V2 controller coverage\n\n- Removed code that was impossible to reach.\n\nChange-Id: Ib3c11d792ff1a5c0110ab2263e7a8a6d5f6baae0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2bf3f05deb897aaa936da6fccc6b0f71a7b77b5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2bf3f05deb897aaa936da6fccc6b0f71a7b77b5c"}]},"branch":"refs/heads/master"},"20a0b7660d2db1386d2b77ae828158ffbcc993a6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-12-17 17:16:39.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/12/903412/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/12/903412/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/903412/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/903412/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/903412/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/12/903412/3"}}},"commit":{"parents":[{"commit":"d511e858084de6541af991b63aa53388ab2b444a","subject":"Improved pool move test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d511e858084de6541af991b63aa53388ab2b444a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 00:11:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 17:16:39.000000000","tz":0},"subject":"Improved V2 controller coverage","message":"Improved V2 controller coverage\n\n- Removed code that was impossible to reach.\n\nChange-Id: Ib3c11d792ff1a5c0110ab2263e7a8a6d5f6baae0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20a0b7660d2db1386d2b77ae828158ffbcc993a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20a0b7660d2db1386d2b77ae828158ffbcc993a6"}]},"branch":"refs/heads/master","description":"Rebase"},"39a8c8f116c05911a2ef369fafc956e3d07c7ebf":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-01-07 13:06:08.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/12/903412/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/12/903412/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/903412/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/903412/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/903412/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/12/903412/4"}}},"commit":{"parents":[{"commit":"571d902c59bdcbe031d479432caf18ba2586e86f","subject":"Improved pool move test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/571d902c59bdcbe031d479432caf18ba2586e86f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 00:11:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-07 13:06:08.000000000","tz":0},"subject":"Improved V2 controller coverage","message":"Improved V2 controller coverage\n\n- Removed code that was impossible to reach.\n\nChange-Id: Ib3c11d792ff1a5c0110ab2263e7a8a6d5f6baae0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/39a8c8f116c05911a2ef369fafc956e3d07c7ebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/39a8c8f116c05911a2ef369fafc956e3d07c7ebf"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
