)]}'
{"id":"openstack%2Fdesignate~903414","triplet_id":"openstack%2Fdesignate~master~Id6889aafd1c1a1108ea4bb4dead27d411d05b048","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-01-07 16:26:57.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Id6889aafd1c1a1108ea4bb4dead27d411d05b048","subject":"Increase minimum test coverage to 97%","status":"MERGED","created":"2023-12-17 17:17:13.000000000","updated":"2024-01-07 18:43:38.000000000","submitted":"2024-01-07 18:42:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903414","meta_rev_id":"08e2d69bddddf8b3db47be8c557c2f57922d0762","_number":903414,"virtual_id_number":903414,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-07 18:42:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-07 16:17:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-01-07 16:17:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2024-01-07 16:26:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-17 19:27:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-02 07:31:17.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2024-01-05 22:55:56.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"e899632ff11bea535b5fa38b961077f97b7ad5d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-17 17:17:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b699abddd6698b8bb620c942989a1023d401618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 19:27:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f887c9aa67843f591d64cfe2e376e94\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d58496f548fe46c18d4900314bd76c96 : SUCCESS in 7m 45s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bd5ec88a48a442f99fa5c08c17512ff4 : SUCCESS in 48m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2c1f85c6d6bb4d449e8e97d2c05fc339 : SUCCESS in 1h 10m 30s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/83788ff2cd0c49b7a8f5aaccf86a27d6 : SUCCESS in 1h 02m 06s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7c939460b3b74c3e8a0490a42634a2b9 : SUCCESS in 46m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9dd8f12a980c4a948b5d0e0686f9cbd0 : SUCCESS in 50m 06s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8904b74699584f30b6811513dd0e3f81 : SUCCESS in 1h 13m 24s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a4fcf2c0ba744ba5b4a6066036b3e369 : FAILURE in 1h 21m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9480f4a3138840719d8aca53e962dda5 : SUCCESS in 47m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d404583f3035432e8d2f34d9fde9b1ca : SUCCESS in 44m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cc0b5c974d6436e9c1f07a5ac980c28 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c58d1924a2a4556ae5c3299ff69b865 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c2845f198144afd83caf28350ba9bc5 : SUCCESS in 6m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e234290ece14496789d1568bb5d8e68c : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e9f083d7efa40dd8afc2e39cc2a4fff : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a0dc5c3e65444f0a35d53eb49980521 : SUCCESS in 6m 23s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ddae26846c404cd5a250c399e102b274 : SUCCESS in 10m 57s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/be246791c91f4256b28a7d7bfeb98620 : SUCCESS in 43m 11s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d133f8d843c744e69418e1d22c288712 : SUCCESS in 6m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b280da61f234f1b375957f4b628048d17ce4a6ff","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-17 20:11:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9f71d735fbde1c915d3d7dc6b095a302cf51a89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-17 22:20:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9e8cd0617ed4f258421984c96c78c7a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ad3ccafda57440e0955b42b7d14c7fe1 : SUCCESS in 4m 46s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d69d4632ac6548eea2f126322ab16339 : SUCCESS in 52m 46s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7e9bbaeb3a36467f968f348bb6f5eb84 : SUCCESS in 55m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7a7a57b8fd14af79b83a230d5ea93cd : SUCCESS in 56m 59s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/fd7baaecf7514cf581a44f947c845e6b : SUCCESS in 48m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/45702b33b6e24b0ca4c5e0810596760d : SUCCESS in 49m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/119f4555e7b64db2b4091956f8762b34 : SUCCESS in 1h 00m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cf0797e101454abea26211761813f190 : SUCCESS in 1h 13m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e72660d39c5249ad862f9546d5da4872 : SUCCESS in 51m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/de3251fb17a9440484c27576769ef58b : SUCCESS in 52m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4cb5147591342d8b277f37c4a477d7a : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80b4c1c384c846c19047255751183965 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/896459fb36bb48a9bdab77654ce8c659 : SUCCESS in 6m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c94ccfecb87495bb4cffaaf98dc3560 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/276e842a0d5d49ec948cf98c193e3459 : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa6b7fac889a4d9fbbe01b28445e8f69 : SUCCESS in 4m 44s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/004bb76745d74a3b9035235e45ed58c2 : SUCCESS in 7m 41s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5a74f04c841b4ce281f9852cb20f99f5 : SUCCESS in 49m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9572a966e90d4462a57df33b3a85d8f9 : SUCCESS in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"320d6ef65f9ec8581c0db5c83d56b1718dcc3d26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-31 00:39:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f69372399649e588a372e12ddd73adc597f95624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-31 02:05:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a2c8201516e4426966372d87925ff8b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6305f170eb1a4415a31347057da2d69f : SUCCESS in 4m 09s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b040baa4513e4aa69e5ce6d3f3fd7eba : SUCCESS in 46m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/526883e0fe6a42669b1ea83e585ef156 : SUCCESS in 47m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/51509071768d4879a8fa2dffde063854 : SUCCESS in 34m 57s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b0fc5245e7e6426294aaa48e6bf3e479 : SUCCESS in 41m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ab513d675b194834a7ebcb1394bb208a : SUCCESS in 41m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/04b69fc53a9a47948e5978c16b7fd052 : SUCCESS in 1h 17m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0d4f294b1c974ad481a59d08f957e8a4 : SUCCESS in 1h 19m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/effca425d8c7413eb6ec02f247c1f837 : SUCCESS in 48m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fd1a5ffa1e8543c499e5513637237575 : SUCCESS in 48m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1fe2f9935d44b92900ee20dc2fe9f05 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af838ab6982a4a6993fba57ff82476a1 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d3da769a2b44ed6a44d6a22c55d69d2 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc4f08d63ac145f09d6c6643e770b055 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32897e4111594e6e800bcb5e7a540ab5 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d18160a01244ee2b4082d768762a8e6 : SUCCESS in 4m 09s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7ee394f89a654d1a97a2de3bfe5228df : SUCCESS in 5m 30s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6ffb5805c13b4467a5f1398a8f3877b5 : SUCCESS in 31m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/16aec8c519f14ab4843f89c2058d000c : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"21f69530b5866137161ff239c60cd7fee0857e1a","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2024-01-02 07:31:17.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7bea21525ac76349311c4f499e5b77daea283310","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-01-05 22:55:56.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fcef3bc4cf4bdbf06eb7f4995dfa122f53cb0bac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 13:06:45.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0ae4c088f9fda81e532c2981789be431d126686f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 14:23:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f57ed5dc6eb4993abf6e3118cced988\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1bee8fb4de8945829228fa85430b43a7 : SUCCESS in 3m 56s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/055ea81692ca4c5092f8f991b4be500c : SUCCESS in 49m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b3147c0316af4818b861b8122939baeb : SUCCESS in 59m 42s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/01b9742d0c304b8c97cdcc9f4f93b2a7 : SUCCESS in 39m 40s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/be2a84a5ef4a43dc9508f9857bc66ccc : SUCCESS in 33m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e77848e5b110427786dce644227f4f8f : SUCCESS in 40m 10s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/04ab74d6b3694e2a97c422df2986751a : SUCCESS in 52m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bec6c77f89d440ee89e2eabb1815461c : SUCCESS in 1h 09m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9e97a12733154107b1ed40499f611ef3 : SUCCESS in 45m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e0e1fddf53c5464ca0b1f7f5781f7878 : SUCCESS in 50m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb34ef980f104220b467ddb80d29b5f9 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e32a4339add64791abda14adc2c22084 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e0aa1ebbded4e3b80169dad6e4afd61 : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec1d147dd35c461d879a3c362b72ed46 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9358e86e3844c6eb13263c2846cea23 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee393a82a900446a92f263d694e7b0ae : SUCCESS in 4m 52s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a2c242f96e4a479987fa3daefb163d52 : SUCCESS in 7m 49s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/76cedff07b154b32a2dcd987d5244f23 : SUCCESS in 37m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7b300521d16c4b41a9507f5b61fdb0e2 : SUCCESS in 5m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"940a3710c82d56f8fdaf16f7738f50d9062359e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 16:17:15.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"870bc472cb685c949c285016518c0f9465969059","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 16:26:57.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2d3de0b20510e8918b27efb666626b9a26d4532d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 17:27:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f695dbab38c4c9fa8358ff2882cfdaa\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b875295a02e84adb9cc30e575c8895f8 : SUCCESS in 4m 01s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d458d8aa5bcb496ab1928c42299559f2 : SUCCESS in 50m 49s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/15c3c6893d084953ad9f1dcb62617120 : SUCCESS in 1h 04m 06s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a87af792de9c4d188862f3c78aab74f0 : SUCCESS in 52m 41s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/57a30e0dfe10469e9e058dcb140b4750 : SUCCESS in 47m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5aa89badff684c029bd50cacfef5cb27 : SUCCESS in 46m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/84a8c56fe50547e5a3305ab640ca0c45 : SUCCESS in 1h 04m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e013272554024e16b55bdfa86ce10fb3 : SUCCESS in 51m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5287a382cf8c41839b50c6c7a0da110e : SUCCESS in 36m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b68a616c69af470cb2fd8bb987f28530 : SUCCESS in 51m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/351529f39d6240a8a94cc07117cbff9a : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbce38b6bdd74b68adc689d4b9486852 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6443eafebbe74a7ea0b9e135a232618f : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/80da66e3075f4215938e466944bb5758 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/561826a87438470492f05258ad041bdf : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/132d45811087433b90f5bad1264d7148 : SUCCESS in 5m 43s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5cc80014105048258d7b9c1998fa3a38 : SUCCESS in 6m 33s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cdd82cd447054224807a55314fbc1850 : SUCCESS in 32m 00s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/36c26cea18584194af2de7b9f83229e3 : SUCCESS in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"64daf925dea06447124a21606c9159a4d1bd90cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 17:27:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d17db9dd0e5281daee6444a23f17353ca85949ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 18:42:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c70ec8f523fe4f2d80a3ce06a071893d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9418d4dab1a84bfaa984cb9c216df40f : SUCCESS in 4m 06s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8dbc3896262740d3870216ccf97eead0 : SUCCESS in 47m 12s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/818d641178ab466888837229b2d3540e : SUCCESS in 47m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/71b7d9e170304b5bad584f08ab4cb187 : SUCCESS in 46m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a308ce1526ff45939b1ab87366f917ab : SUCCESS in 1h 09m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7e1f60b0d8aa43738698f5b931a3bb1f : SUCCESS in 49m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3070ba3d540a4f3ebf145cb92b1a0f59 : SUCCESS in 46m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bb01821e002472a83bf246ab28312b6 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b554ac83589438c996644b7108e46be : SUCCESS in 7m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d1554c39e344d7a975b63c638b12431 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fddac3434a74ef488b6c6ed2b097800 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/712a2fa8d88d453a920453867a0980d8 : SUCCESS in 5m 22s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9f43cc32509643c4bd16d0385b3937de : SUCCESS in 6m 04s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0ff3a64f6def4370ad891ba7a6f263ba : SUCCESS in 32m 31s","accounts_in_message":[],"_revision_number":4},{"id":"70af62b8e3f5907948f0da8be2c02d2b1194502a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 18:42:35.000000000","message":"Change has been successfully merged\n\n2 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":4},{"id":"08e2d69bddddf8b3db47be8c557c2f57922d0762","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 18:43:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe0906b2d3cc4c2c95ac1c2263975af4\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e81a5fcef32548e08eacd8d00fdfef68 : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/332205f672764f67a516dc25ebb3ac0a : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0afddca03e53432ba3ee728f4b8b9eb6 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2e2b652419f2caef75cec5ae1ea4b8bd6a6fe54a","revisions":{"bc24a873f68adb6e354321a705d4912f1d81de23":{"kind":"REWORK","_number":1,"created":"2023-12-17 17:17:13.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903414/1"}}},"commit":{"parents":[{"commit":"c8dd55f2ab588439c31ab5474a94a47158bba512","subject":"Remove unused infoblox code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c8dd55f2ab588439c31ab5474a94a47158bba512"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 17:17:00.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 17:17:00.000000000","tz":-480},"subject":"Increase minimum test coverage to 97%","message":"Increase minimum test coverage to 97%\n\nChange-Id: Id6889aafd1c1a1108ea4bb4dead27d411d05b048\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bc24a873f68adb6e354321a705d4912f1d81de23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bc24a873f68adb6e354321a705d4912f1d81de23"}]},"branch":"refs/heads/master"},"54ecf467d1737db0baa7c4f17962429632884566":{"kind":"NO_CHANGE","_number":2,"created":"2023-12-31 00:39:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903414/2"}}},"commit":{"parents":[{"commit":"7f44dfb5120ff4228d63d893d0b490d07d2c2ce9","subject":"Remove unused infoblox code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7f44dfb5120ff4228d63d893d0b490d07d2c2ce9"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 17:17:00.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-31 00:39:38.000000000","tz":0},"subject":"Increase minimum test coverage to 97%","message":"Increase minimum test coverage to 97%\n\nChange-Id: Id6889aafd1c1a1108ea4bb4dead27d411d05b048\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/54ecf467d1737db0baa7c4f17962429632884566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/54ecf467d1737db0baa7c4f17962429632884566"}]},"branch":"refs/heads/master","description":"Rebase"},"cf6e15d815161b39b79cddcb764c5a7373e64f71":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-01-07 13:06:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903414/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903414/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903414/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903414/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903414/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903414/3"}}},"commit":{"parents":[{"commit":"170de0c532be4e6be5a0bac4e4fff5d6f908a4fe","subject":"Remove unused infoblox code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/170de0c532be4e6be5a0bac4e4fff5d6f908a4fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 17:17:00.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-07 13:06:45.000000000","tz":0},"subject":"Increase minimum test coverage to 97%","message":"Increase minimum test coverage to 97%\n\nChange-Id: Id6889aafd1c1a1108ea4bb4dead27d411d05b048\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cf6e15d815161b39b79cddcb764c5a7373e64f71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cf6e15d815161b39b79cddcb764c5a7373e64f71"}]},"branch":"refs/heads/master","description":"Rebase"},"2e2b652419f2caef75cec5ae1ea4b8bd6a6fe54a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-01-07 16:17:15.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/903414/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/903414/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903414/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903414/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/903414/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/903414/4"}}},"commit":{"parents":[{"commit":"21ae2fb4ff434d591eab7befef462fcbdbe1222d","subject":"Improve Scheduler coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21ae2fb4ff434d591eab7befef462fcbdbe1222d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-17 17:17:00.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-07 16:17:15.000000000","tz":0},"subject":"Increase minimum test coverage to 97%","message":"Increase minimum test coverage to 97%\n\nChange-Id: Id6889aafd1c1a1108ea4bb4dead27d411d05b048\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e2b652419f2caef75cec5ae1ea4b8bd6a6fe54a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e2b652419f2caef75cec5ae1ea4b8bd6a6fe54a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
