)]}'
{"id":"openstack%2Fdesignate~903416","triplet_id":"openstack%2Fdesignate~master~I894175c46e6290d12841e0700de3cea2829aeff8","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"I894175c46e6290d12841e0700de3cea2829aeff8","subject":"[CI] Increase number of MDNS workers","status":"ABANDONED","created":"2023-12-19 13:37:14.000000000","updated":"2023-12-29 00:04:36.000000000","total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ce12d3ddee263cb80b7cf281895ae4d6f283ca85","_number":903416,"virtual_id_number":903416,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-25 21:41:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-19 15:09:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1d336f39b7c594e728d3809f0919129d4c35d29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-19 13:37:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"562421256dc788154d140e90943d9ebf25b185aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-19 13:39:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"85704c708c21a4e09258d600c356b5680fb7fe91","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-19 13:42:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"c413b28f88a052d2e84db7c0e8651a126c0c0dd1","tag":"autogenerated:gerrit:restore","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-19 13:44:12.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"41c09c81ca52ec244ab5ee9838be262c775f8988","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-19 13:45:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d92dc0ef684fdfdbbc2d7f23e6d5ced253de50a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-19 13:48:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ed27ffaf73d6fc5553eb7045e603d3c0409d8b20","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-19 13:49:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"02d33f55a763ee071d52e088b4f943c12576b3fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 15:09:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/167dece287a74d3aa3afe25457add438\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c682edc966624d52a5f4008af72ab465 : SUCCESS in 34m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0beb0874d9a94391ad6eea07edebd7b4 : SUCCESS in 47m 54s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3859c72dbb0b4761860271f866c9441e : SUCCESS in 40m 17s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d3ae020f49924a74b4e026a9a547fdf9 : SUCCESS in 41m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fad1f5cc5146414891a5b484e91f5cdc : SUCCESS in 37m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c416e8a975ca41bda037ba20d6ec0b28 : SUCCESS in 54m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7f84f52373c348c78c12d9c724d9f6b4 : SUCCESS in 1h 08m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6e0d57781d6343dca4de53bccbdddf4d : SUCCESS in 46m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3293fbd8872b4f8f981f29273239d371 : SUCCESS in 46m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46efff19aa684a64ae02740218ed63c2 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30cbf64bcdd54df58f52cfab4e6d92c0 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba308bf1302d4c35826e7631cb3d6ac0 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf988ccaf50945ab81f25e85ee752740 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3f81004b6014a698c4b92443714e9c9 : SUCCESS in 8m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/181bb90521464b57a06a2300cfb9840b : SUCCESS in 31m 50s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1fc2995aad68476a987463bc60cfa824 : SUCCESS in 6m 47s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b93093153e5c41c398507e4e55da76ec : SUCCESS in 48m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7c47e21594eb492a99178ad96a7abdce : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"44c2ae00d81a01d2fbca12e987524f187a3cd1c1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-19 18:51:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"882b572255cb46a944a253ace86c545b865a8980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 20:06:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe852e90c60e4008823c8a002ca81ecd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fff0efa889b7414dbe4a717d0e6758a5 : SUCCESS in 45m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/512142da46224dde9cbe439c67214b4e : SUCCESS in 56m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2125baac64374feab4bdffcfdab8d1db : SUCCESS in 53m 38s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/54e393ca97fb4cf3a7251af4cd401698 : SUCCESS in 42m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/61a5dd61cabf41f284f289fe5d6decf5 : SUCCESS in 39m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ab4cccab042b4668a5cf9568563addaa : SUCCESS in 1h 04m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dafbc0464396411d9a7f29bc0e7f2e5d : SUCCESS in 1h 00m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1042f8634bee4671ab7a65fd89e8ad02 : SUCCESS in 43m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3272865cbdb54f609b6c7fb6cd2bb783 : SUCCESS in 39m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dc5817fd51049dc937a6e780196cee9 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20526ec45e444f198e1369ba2b4ecc74 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06f9af39171349b0a913de71009b81b4 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84bdcaf147464486b086a719723d1231 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc88fa271e14c238f7ba82ae8810333 : SUCCESS in 8m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fc7c56ae571d4a90a1cd25a8a3fa645e : SUCCESS in 40m 24s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/eaebfefd0db94b239546e814f32ebe0e : SUCCESS in 7m 03s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0f79ba258d3a42f3b584c3c74fab9d53 : SUCCESS in 36m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c11418bafffb4aadbd1581a1a8001a6a : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7af8ace6f2bedd971f7ed9678efbe8f92822a12f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-19 20:44:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"297fe7d7d4bbb8cf20c2d200f104dd42608b4c5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 21:57:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5daf36e769784b09bec2f8b43ed2a762\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/65a81b92cb98412386320f87ae25256d : SUCCESS in 36m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/40a789d081624092b5b1d7c800f13a31 : SUCCESS in 45m 29s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1045c9886224ecd83e88e6319b2d166 : SUCCESS in 54m 57s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1b99d356be8b4cabbda9aa7898bb899d : SUCCESS in 44m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e07f304317904d43901b4a0606e6163e : SUCCESS in 39m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4588ff84e8914323bc7e2426727afc48 : SUCCESS in 1h 03m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/87f5042bfa574759bd7471536dd7ffe5 : SUCCESS in 1h 06m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a1a81063603444dc97103d660723ccb2 : SUCCESS in 39m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4e9e8cde834b48d9879737cf403597ed : SUCCESS in 50m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/082ad57252b74949885f55d3e026b986 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/767e7885012b48679a323cf49966f42a : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f3459e0a30f41d6980e4ffd0562a922 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5e1614f4c0341838d38e5154d5d75d3 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee55e748fc141debc8ed0e7b929a483 : SUCCESS in 7m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6c4bf22164494794a26e28d827857287 : SUCCESS in 35m 44s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f8d88a5e3cf543ab856d9b8a02a3e745 : SUCCESS in 7m 02s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9e3bcb2a091c4c8db0a23a4124c04072 : SUCCESS in 47m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5441ff2f11f24cd085d8e97d721ef662 : SUCCESS in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"98a97469e542e5420a991323b71c3f335b03fb8d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-21 16:04:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8b87083524bdc685018dfe9ef7777beb356b38b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 17:28:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bc14e6e9c934ae1bafcd8ca1803c7a1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ebafb0a0e6b64a3aa82095be5a6a1b8d : SUCCESS in 50m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/32fdc3b713984fb7973287b6b7a8dee9 : SUCCESS in 54m 21s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/16435b6a5ee6415499b040dc687a8f43 : SUCCESS in 46m 17s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/fac31675986047838a0183b5dfadcb44 : SUCCESS in 49m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/197572fb28e4430a9af9fe71fa03b231 : SUCCESS in 45m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c3bc0ce8337943bfa497df613a3bfa02 : SUCCESS in 1h 14m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/615b50c6151441bcac4ea7e5468982af : SUCCESS in 1h 15m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9cecb0e8e498401aa4cdf0bcae77c763 : SUCCESS in 46m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ff30cb4d240c47408ec2e8f15f164d00 : SUCCESS in 55m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/612b6a99e6e5404aadaeb868572ca024 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/380c38d1951c45c9b44452c37fa19ca8 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34264926b763492d88b00b1bcdf87f79 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d2e0750d75946edb9b7ee72da3a578c : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac7b4f95ee5544aab3939f97afb064d9 : SUCCESS in 8m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e678901c6bc24ecb8eb9568fd9130f2c : SUCCESS in 32m 44s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0452bb9f2aab4e04bf745f2390ee5584 : SUCCESS in 6m 52s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/aed173334fa1462d839bed6815a3b33a : SUCCESS in 48m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/854ed3937b8f489ab5ca8cf4adc65ef8 : SUCCESS in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f1eb6cf53974d2fa1d01bda854e11ef836bb8657","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-21 17:47:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6a6b0f44894bd25e477ac72f8160b65cba9a0383","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 19:02:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac727ee94bc0482d8f6be4efa393fef4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e7d25e75b4bc427db4383c67926dfaf8 : SUCCESS in 42m 42s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/604bc5609d14488bad787847fa0ec655 : SUCCESS in 50m 30s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4f42878874f4fb48fa622437f0c72bf : SUCCESS in 47m 21s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9ad9f016c26745758cfbc98b47f96208 : SUCCESS in 50m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7f8d78100fda4c2ca81d3c785eaaa2b2 : SUCCESS in 49m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ace0690ed5e9477d8bb2c34e511bc654 : SUCCESS in 50m 56s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9174b699eaea498186b7831e233ef1e2 : SUCCESS in 1h 06m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2be6831eb85c4b6bb9e1692b8d153652 : SUCCESS in 46m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/19f0c25e2b764cd1999cd201bc027040 : SUCCESS in 48m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ee7aa98d8a746dfb676c44694ab89a9 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7bdce39334e4d059f03a795f6b0f423 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0850c3fdfc954466a14ae1f1069395fb : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/237552909f17457d81dff9683f14a974 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e32b117d34ee40d79e91a534ad3caca1 : SUCCESS in 8m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7633ca9c93ec4556895d456b59e0ccf1 : SUCCESS in 35m 43s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/727905d5db9842ddb3f904b0bc1ebfae : SUCCESS in 7m 09s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/be66620dd45646cb88ef95d20a0d7477 : SUCCESS in 36m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fb3bc667821b4e4ca9814a2cfaa6cc53 : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1ce3fd2a2427330ad42e82ee4532cd10331cb9e2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-21 20:23:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e05436bac3a8932490690d917f6e05d03c14a075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 21:26:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f757e88bf6574970aae9422ad4b0fd91\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7d2dad8e4ac6444fae551a1ad40bc9ed : SUCCESS in 34m 23s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/dffc191b407c433d96e745bce072f1e0 : SUCCESS in 42m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9467a5dbd1724682b89b0604b0be4561 : SUCCESS in 35m 08s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cc083d2a26374fcc882715c8dbb87472 : SUCCESS in 35m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d721f41360864a16a6a16c07342d930c : SUCCESS in 33m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/812f549e55d74f82b08aa338c1a935b4 : SUCCESS in 53m 01s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d60bd2abde864bec8dfbd22fea95c319 : SUCCESS in 46m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ffbe26b50df940789743811f078c8a82 : SUCCESS in 33m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/39371e5e77a34f9f8adbf87095005f40 : SUCCESS in 52m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcdb948048974253b701f2058c502cb1 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/267ef3d08c7d4a28af894e1de4a5e117 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beee4434ad63476fb47c8076c34eccc8 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba92660eb11845ed9d7bac8b5a837a86 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14f0a39717dd408b8b1aa2b9e2625164 : SUCCESS in 6m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ff2a4375939d41c7ae7f8691e0548052 : SUCCESS in 41m 54s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ee599716d8594981bb2e0cd8df853496 : SUCCESS in 6m 55s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e20ff0cd5ac84554bafa16b625de7d36 : SUCCESS in 54m 31s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9b570d47bf7b4d3da70ef3c9f995f787 : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ddb76035fe932d65d68d9224f43556d0de7eacdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-21 21:52:25.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"73097e7cbca27c572bb3289cde847417c5a515e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 23:07:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b043fbefe3c46278c9a92a76ae996c9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cce70265e52449219087fea292211238 : SUCCESS in 29m 34s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/54239332ed19417f92cbd237ea45d32b : SUCCESS in 37m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e71b398af9044a2a45351b653f5659f : SUCCESS in 1h 00m 08s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b3a77302f7bf4d58afa2868a50c676a8 : SUCCESS in 55m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1a1fad997c514e2997ce974d3d39eb39 : SUCCESS in 33m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4a928c44bc9545419ee53903faa9daba : SUCCESS in 1h 08m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c8f5b89d596c4e62a451757c62b4bd8a : SUCCESS in 52m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e39054c947ca41cdb16a69f9a8ada491 : SUCCESS in 41m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8927e0d3464145c8bd4f99817a9e1b97 : SUCCESS in 44m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5b58c0573954c5b9c3c773f05aebfa9 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83e7b43bd44e413d94d797f7ae999caa : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ab6b9989438418784b77e9eed198d98 : SUCCESS in 7m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3da1ddafcd444d07bafe0fa750ffd049 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6946cf1d4a3e45c49b21e40c365172d9 : SUCCESS in 9m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/090e22cd4cd64cd790ee07f02448c4e2 : SUCCESS in 40m 20s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b0a2da54307241c1905415a5a5bab82d : SUCCESS in 6m 55s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e797a9edd5ce4c81b8d42804739905e2 : SUCCESS in 48m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6bc846a3328b43448b01cba1407ccbf5 : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fa14e17a1ca238c98f9034f9dd337f262f6be904","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-21 23:58:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f711a95815655e1150f67311c7481ff5e9e98344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-22 01:08:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d882311dec434463a5b655e5a667a61e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/257e2a3c25194b09b5595db15675dcd1 : SUCCESS in 48m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/20ab653d83634b26bdd44509a62bb7d9 : SUCCESS in 1h 05m 20s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/86bf7f434f8b4364bf8ddac5d27b3af5 : SUCCESS in 53m 03s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a5ace41a7e63459ea1084451dc93f4c8 : SUCCESS in 51m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/758f9b45a3314347bc397f3c4f64938c : SUCCESS in 45m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9bea180060894d3b94012e92ee685419 : SUCCESS in 59m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8eb70e3268024da9adc9813c9d4eee42 : SUCCESS in 1h 01m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/79972bac54bf4cb397953d712266398e : SUCCESS in 39m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/64602a32353e4d128efe4ff5bceaa427 : SUCCESS in 42m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f7de88fb72c4ede992bf4cf274b3824 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/642c4eb3eb4f4de0a4a3acd75f037c9d : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c49006664f88457bb7d5318e811f91cc : SUCCESS in 6m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd338db3321c4ac7a4ed4382a1752a5e : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a62a86e73cd34b2da2d58f7ff20f2820 : SUCCESS in 7m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/053ff4777aa14446b7a5f7da62ea3153 : SUCCESS in 36m 10s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b593b42ba9cf4c2f8f8aacbff85895a5 : SUCCESS in 6m 36s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ed31579ca3724f56bc66bda3feb1f86e : SUCCESS in 35m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5ee6c9fc59c5469a84b0d4c52b9c901c : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fd1a71e638bdfe11d62cbb4d1cf86d504597a47a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-22 21:12:17.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"671c4831676f8e409d7554afa36e26b79bf20902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-22 22:29:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba311793dfbb460c923e8dd4e93cc76b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bbc8e8c787514da79426ee698df5513c : SUCCESS in 39m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a58e4a25842843e79d6d38b965af39bd : SUCCESS in 45m 06s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/345ca9ebc8c24c1d87fb764859899fa6 : SUCCESS in 52m 49s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6241cc5e3f1540da8ad232495d232ae0 : SUCCESS in 33m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e2e97ebfc94b47a6b8ef972f0b7dacbd : SUCCESS in 42m 14s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d6cbac9ca14f4a07bf6f51e9251f97c0 : SUCCESS in 1h 10m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/baf6aae4e86b4d4eabf5062b7b799c0f : SUCCESS in 52m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/583f4dd9058548a9bcc0eebbff8ca795 : SUCCESS in 38m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fddaedbea9284c2290448e073b3c0bd9 : SUCCESS in 44m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c002e346acd4262b9c796c54bce08eb : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfad098ec2614088863aef131217792c : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84e51349a32b40e5a0c9af53d27c396f : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff4d4dfedc1d483e8f461080ba650acc : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b06af7c661e40e28acf2db6ec31d941 : SUCCESS in 8m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f762b9c56e74493391afc2c05557a133 : SUCCESS in 33m 39s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7087f1b576eb4fef8f92ac40089fef15 : SUCCESS in 6m 55s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/706b8dd8e3584692b0184a9e6ff3d6dd : SUCCESS in 49m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6388d12700914130b031f8f22a2257dc : SUCCESS in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"74fc49f514ee105b818f0621cfffc172d62576d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-25 20:20:33.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"54ba0d11e404f12226695c92b1340dd150b20d63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-25 21:41:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de588e5fd2cd49b688cd044819224251\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1d910b6cff1144f9af15f4378724a99a : SUCCESS in 28m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1e5f754592d34332b87251c04e4ebbda : SUCCESS in 1h 05m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/51f28067d20a4c18ab3969132adff127 : TIMED_OUT in 1h 14m 01s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b28649b36ad947898381b032fd97f8a9 : SUCCESS in 49m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/02c3e0cacc894303b8dd0793dd41545e : SUCCESS in 47m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/79cae9487c3e4c86a1f74352104aa001 : SUCCESS in 1h 11m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b21dd11dea8a40d5a277e3b3f023f75d : SUCCESS in 1h 05m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1900e1724cd74804b39d75aafcb8978c : SUCCESS in 39m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b66d84c3ea944c2ab499acc356d08de3 : SUCCESS in 46m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c064cb381a304d89924d2f56ef515ce0 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f9a37d7bf8a46de8778f638aec31118 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce262d080af249918da5a0858adfb49a : SUCCESS in 8m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9d1f01cb63af4efdb977aa464df28d24 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e4c8015f46f4f0a8613d6a8cf2e2972 : SUCCESS in 9m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ebd95bd5855c426f8ac2b2e61caf7aa8 : SUCCESS in 42m 34s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9a7fdb2db3994dd48117ade54032634e : SUCCESS in 7m 43s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/24ee37345f68479a810933dd4e0f8dc0 : SUCCESS in 33m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/bec4326429f34b7bbac65b327c5a2dd2 : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bfafc060d8f0f0f0c89b084ebed223009f985989","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-25 21:43:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3ac86069678b263d590eb21902d366bd4d694daa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-25 22:55:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e77abf14a1574a089de7053ff1ee2b49\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9021e28d3b3f4623b9bf5a85312d4dab : SUCCESS in 47m 36s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f5f6da6595754d8184fc908c5a6b2ae7 : SUCCESS in 58m 34s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/116f4e0987b44192a0a196c80725bda3 : SUCCESS in 34m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/fe13b49fde4c48688aeb29cdc3edf1dd : SUCCESS in 48m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5107317dc331448b9fa7f67571943742 : SUCCESS in 33m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5235db2d7f0e4d0bb72fbfbf9a7dfd38 : SUCCESS in 46m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a372a20c78ca4c9d936f3aefdbb2c938 : SUCCESS in 1h 06m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4c1318a83ef54399a3b3c4813cc0ce11 : SUCCESS in 46m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/49c17204dda84260ac11c1d6bcd08b65 : SUCCESS in 44m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e537a9192ccf401fbec828231606e277 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94bc0d9ee27a4194bf5444eb4a6cf7d9 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4221b39c36749ea9364bcadd8f0f19a : SUCCESS in 6m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50517b917c7446d2bcb5c1a1cc9575e5 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc229fb76fee4112a942134ebb855257 : SUCCESS in 7m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b53dcdb4d5d049678cd5ce6e2f8139ad : SUCCESS in 45m 06s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d5f0b6a656504b55a1073ac0fbf596be : SUCCESS in 6m 59s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6f66daca8b064ca8b28699b111aacd01 : SUCCESS in 31m 37s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/18b9e8874d7e4599affb5fcb736b1e0c : SUCCESS in 5m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"12d864239905b6f4afbbb837bfefac5fc4f40836","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-25 23:33:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0df654def3550345c17b781b2292574e23f02a95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-26 00:51:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c81b42ef01b041a48fda3f0fbeeef145\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b5cb0aa90da4468193484a8294b1bd93 : SUCCESS in 34m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e812623d37e744a7bdb7b5cd8088eeb9 : SUCCESS in 40m 20s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6388c6dabb6642b8b039c93b2b1e40fd : SUCCESS in 56m 07s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/860637d1b5ee4da293bf75a6049dba82 : SUCCESS in 46m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/44fca41ca3714b58895040d93854c271 : SUCCESS in 31m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1b92c63855ac49e1a95876cd92c27915 : SUCCESS in 1h 11m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6ada8f4628ae4db0ae951ac813e1ec87 : SUCCESS in 1h 04m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/018195a250cd42dfa2f33d1f1459d1e1 : SUCCESS in 29m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b60272de39f44f9ebb975edcc30772a7 : SUCCESS in 52m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84d5aaa76acc4bd78e10f65e11d2b21a : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/154599fcb8eb41919e000a9b450a5db7 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86f717de05344a4bb8b2d025119f8244 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7b0ac722d734672b8f4c3bb62e9e3b6 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cb556ae8d754e639232af89e38cba1d : SUCCESS in 7m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8d4f969241b546888b30c6f841c16a24 : SUCCESS in 42m 01s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f5ca3297c9df402fb2812961c2a4459f : SUCCESS in 6m 30s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/adac898c4fbb451aa77a86432fc2cfb2 : SUCCESS in 40m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c93a61b4faf047839acfbf5a2ef89847 : SUCCESS in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ce12d3ddee263cb80b7cf281895ae4d6f283ca85","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-29 00:04:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7816664c38a835f6485c44fa68555bb5eac6e3d0","revisions":{"d34246218baf8def292b673ac1d2571ea0c813b9":{"kind":"REWORK","_number":1,"created":"2023-12-19 13:37:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/903416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/903416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/903416/1"}}},"commit":{"parents":[{"commit":"af940fb6e4431c9eb9799cc983a85465ec865011","subject":"Increase minimum test coverage to 96%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af940fb6e4431c9eb9799cc983a85465ec865011"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-19 13:24:56.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-19 13:36:56.000000000","tz":-480},"subject":"[CI] Increase number of MDNS workers","message":"[CI] Increase number of MDNS workers\n\nChange-Id: I894175c46e6290d12841e0700de3cea2829aeff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d34246218baf8def292b673ac1d2571ea0c813b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d34246218baf8def292b673ac1d2571ea0c813b9"}]},"branch":"refs/heads/master"},"8d58a2237e2ca927b4c6a5544d9b9366cf4da631":{"kind":"REWORK","_number":2,"created":"2023-12-19 13:39:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/903416/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/903416/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/903416/2"}}},"commit":{"parents":[{"commit":"af940fb6e4431c9eb9799cc983a85465ec865011","subject":"Increase minimum test coverage to 96%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af940fb6e4431c9eb9799cc983a85465ec865011"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-19 13:24:56.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-19 13:39:44.000000000","tz":-480},"subject":"[CI] Increase number of MDNS workers","message":"[CI] Increase number of MDNS workers\n\nChange-Id: I894175c46e6290d12841e0700de3cea2829aeff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8d58a2237e2ca927b4c6a5544d9b9366cf4da631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8d58a2237e2ca927b4c6a5544d9b9366cf4da631"}]},"branch":"refs/heads/master"},"2c6d3845c304146224a80ba0be53f8ea473becce":{"kind":"REWORK","_number":3,"created":"2023-12-19 13:48:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/903416/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/903416/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/903416/3"}}},"commit":{"parents":[{"commit":"af940fb6e4431c9eb9799cc983a85465ec865011","subject":"Increase minimum test coverage to 96%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af940fb6e4431c9eb9799cc983a85465ec865011"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-19 13:24:56.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-19 13:47:58.000000000","tz":-480},"subject":"[CI] Increase number of MDNS workers","message":"[CI] Increase number of MDNS workers\n\nChange-Id: I894175c46e6290d12841e0700de3cea2829aeff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c6d3845c304146224a80ba0be53f8ea473becce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c6d3845c304146224a80ba0be53f8ea473becce"}]},"branch":"refs/heads/master"},"f7d2e14fdffc52f563390366e183d8005efda66a":{"kind":"REWORK","_number":4,"created":"2023-12-21 21:52:25.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/903416/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/903416/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/903416/4"}}},"commit":{"parents":[{"commit":"af940fb6e4431c9eb9799cc983a85465ec865011","subject":"Increase minimum test coverage to 96%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af940fb6e4431c9eb9799cc983a85465ec865011"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-19 13:24:56.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-21 21:52:24.000000000","tz":0},"subject":"[CI] Increase number of MDNS workers","message":"[CI] Increase number of MDNS workers\n\nChange-Id: I894175c46e6290d12841e0700de3cea2829aeff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f7d2e14fdffc52f563390366e183d8005efda66a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f7d2e14fdffc52f563390366e183d8005efda66a"}]},"branch":"refs/heads/master"},"5291c6b707061e876f005ce8f90ec13142682e8a":{"kind":"REWORK","_number":5,"created":"2023-12-22 21:12:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/903416/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/903416/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/903416/5"}}},"commit":{"parents":[{"commit":"af940fb6e4431c9eb9799cc983a85465ec865011","subject":"Increase minimum test coverage to 96%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af940fb6e4431c9eb9799cc983a85465ec865011"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-19 13:24:56.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-22 21:12:17.000000000","tz":0},"subject":"[CI] Increase number of MDNS workers","message":"[CI] Increase number of MDNS workers\n\nChange-Id: I894175c46e6290d12841e0700de3cea2829aeff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5291c6b707061e876f005ce8f90ec13142682e8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5291c6b707061e876f005ce8f90ec13142682e8a"}]},"branch":"refs/heads/master"},"7816664c38a835f6485c44fa68555bb5eac6e3d0":{"kind":"REWORK","_number":6,"created":"2023-12-25 20:20:33.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/903416/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/903416/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/903416/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/903416/6"}}},"commit":{"parents":[{"commit":"af940fb6e4431c9eb9799cc983a85465ec865011","subject":"Increase minimum test coverage to 96%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af940fb6e4431c9eb9799cc983a85465ec865011"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-19 13:24:56.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-25 20:20:33.000000000","tz":0},"subject":"[CI] Increase number of MDNS workers","message":"[CI] Increase number of MDNS workers\n\nChange-Id: I894175c46e6290d12841e0700de3cea2829aeff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7816664c38a835f6485c44fa68555bb5eac6e3d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7816664c38a835f6485c44fa68555bb5eac6e3d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
