)]}'
{"id":"openstack%2Fdesignate~904198","triplet_id":"openstack%2Fdesignate~master~I741bd5342625dd0617c9d319323a343cbb2f8cc5","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-01-08 22:12:28.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I741bd5342625dd0617c9d319323a343cbb2f8cc5","subject":"Cleanup cross function / unit imports","status":"MERGED","created":"2024-01-07 18:00:22.000000000","updated":"2024-01-09 00:05:04.000000000","submitted":"2024-01-09 00:04:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904198","meta_rev_id":"442f11de278268f06270363d931dc3cf23c9b6ab","_number":904198,"virtual_id_number":904198,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-09 00:04:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-01-08 22:01:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-08 22:12:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-07 19:16:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-08 22:01:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"54cae07e92f6ba316df84c7d5cc8287540c64e2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 18:00:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7a2a1d185e245f58a642ab226878c8758bfd045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 19:16:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7aba7a8b0e9487babcaa8fd77d7aab4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aee590534dee40f1a99b3d4cce7097f2 : SUCCESS in 38m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6c26bdc6e3654055815f88f8a0a9dd95 : SUCCESS in 57m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ebf4f0e8c9b4e1985f722c247cfb183 : SUCCESS in 36m 58s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f7991804086b43bba86b5c240ea64323 : SUCCESS in 48m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9aa4a63e353d44599ff13e9ca2896052 : SUCCESS in 50m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/dee06ac78fac42d3914d7c052e7b54df : SUCCESS in 1h 07m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/13af5832bfc84e2d8acd470893a8a018 : SUCCESS in 1h 10m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/44de16a935f54041a807e549a3ddd33b : SUCCESS in 46m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/90432d8eddfa44eb8174a0425aa27235 : SUCCESS in 51m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb1e16d64f664a84bff8f48745af9388 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f56d4e719b649e29a2f77a38868bb5d : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c521cc359d474492bc9e71518b7d9638 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6856c4a2e3dd4e64b086bb20c75909fd : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6343ed1cbc8d4dd7842667e268002fa9 : SUCCESS in 7m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cf98b5ae0bdd4e9a915767ed225ab198 : SUCCESS in 41m 48s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f3f72b8910534e2599584bf16b409e2a : SUCCESS in 5m 15s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5541e9bb215944cdb80503a2e95d28a2 : SUCCESS in 31m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/190684162b4645d2a220627c1e5f2761 : SUCCESS in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a655c7e3227f9cf7fe5bde461c9aeea223ad4069","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 19:46:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"61bfe2cb96ab6b1489569be4a09dee54403ad1c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 19:52:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4c59317b8bebb9b68e264260de1089ac78cbec0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-07 19:53:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e0dff2f37610b6de02e27d3aff5d6114b0a49c29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 21:28:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a14ebedf9fe49e99fc054d4ce6d2a8a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5ecc5994e484419299dd48623b30fe21 : SUCCESS in 45m 42s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8706fd5bb98643b884e3d0edefd7375f : SUCCESS in 59m 43s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ff15d2de03e4e75a2ea99f5329d7093 : SUCCESS in 35m 13s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8a96f073f70f4c98902be310a48f0747 : SUCCESS in 41m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a28bd171c12b43ff8ee302d4361ed82b : SUCCESS in 31m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7075a32b4b39440ea385541ba0f4564c : SUCCESS in 1h 02m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/394257cbd1fe4976ad7cfadcc5266f5c : SUCCESS in 49m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/da8188138c4e40afad319169482954e6 : SUCCESS in 40m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/57e7fc4a41804e919526f83256d176bc : SUCCESS in 33m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df6ba436dc81402bb95d3813afb26ac8 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d38e927c32574a9ca64e43638f998dad : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed382804a0924d5a96acc9f80bd697df : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d409d99b6b74e26b2be6f8b1a841d17 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a3afd6a7cd54d498669cbd2df9ab1f1 : SUCCESS in 8m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/637cfc7e01b841b593cfd5133523fe9c : SUCCESS in 32m 33s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/00fd62fa5a00490dac8e1025da58552c : SUCCESS in 5m 55s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6b2d915c797d45a585e8aa4c360ab692 : SUCCESS in 50m 23s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/26bd260e2dbc43d39a2baddd504588e9 : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"61eecca95c93362fbf0d0679e77dfab53a8d42e8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-01-08 22:01:44.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d2eacf514959daec141b6b546ae085d1c4df82b5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-08 22:12:28.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"effd2939e50881bcc7b75bb07bce5c4957682667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 22:13:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a844f03f3de3e01350908577ef407752546ffc77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 00:04:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c663f535e1d242edac768d9fbe12165e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3b373477a43e460ab1993e72b6a798f9 : SUCCESS in 34m 48s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/46a654ead86b4954babcfaf354877f6c : SUCCESS in 31m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/206953880cc04218958985214048ab69 : SUCCESS in 34m 53s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cbd640fe061d416c8761da0cbe7bf4c8 : SUCCESS in 48m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4e4b2dd2847346ef91b026a5dbc2df90 : SUCCESS in 40m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cd6a34786590411ab34d6fe986e989bf : SUCCESS in 34m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dfb5207a2f5464fb963b108c31bec5b : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59577e6ea3e149809328a76370fa87dc : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fcf4c9fc254f45089d62f037cd5d8fd3 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7473a1d42eb4e69abc5c671a524883b : SUCCESS in 8m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6bf70ebd133d458cbc86ec5fe2c43afd : SUCCESS in 45m 39s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e8efa53506f64db29522205c053ab64e : SUCCESS in 6m 25s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7c10564011804b5a9c689d69661bf18d : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":4},{"id":"2b64fe1c654bd9a68ba0627a8ba575f09edc5ce0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 00:04:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"442f11de278268f06270363d931dc3cf23c9b6ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 00:05:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/922d266ded464dd5ba1d36af3e78ceec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ac3b2f163384b6c8fbe1780149eefa0 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ddef40c9b1a5118666c4a0e150b5dde84cdea707","revisions":{"621584f76b905352c04c960ed75fdb4329a8cb79":{"kind":"REWORK","_number":1,"created":"2024-01-07 18:00:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/98/904198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/904198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/904198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/904198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/904198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/904198/1"}}},"commit":{"parents":[{"commit":"21ae2fb4ff434d591eab7befef462fcbdbe1222d","subject":"Improve Scheduler coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21ae2fb4ff434d591eab7befef462fcbdbe1222d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-07 18:00:04.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-07 18:00:04.000000000","tz":-480},"subject":"Cleanup cross function / unit imports","message":"Cleanup cross function / unit imports\n\nChange-Id: I741bd5342625dd0617c9d319323a343cbb2f8cc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/621584f76b905352c04c960ed75fdb4329a8cb79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/621584f76b905352c04c960ed75fdb4329a8cb79"}]},"branch":"refs/heads/master"},"26e6ac856df1a97b55704b2f2005666fcfd459bd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-01-07 19:46:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/98/904198/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/904198/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/904198/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/904198/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/904198/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/904198/2"}}},"commit":{"parents":[{"commit":"2e2b652419f2caef75cec5ae1ea4b8bd6a6fe54a","subject":"Increase minimum test coverage to 97%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e2b652419f2caef75cec5ae1ea4b8bd6a6fe54a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-07 18:00:04.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-07 19:46:29.000000000","tz":0},"subject":"Cleanup cross function / unit imports","message":"Cleanup cross function / unit imports\n\nChange-Id: I741bd5342625dd0617c9d319323a343cbb2f8cc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/26e6ac856df1a97b55704b2f2005666fcfd459bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/26e6ac856df1a97b55704b2f2005666fcfd459bd"}]},"branch":"refs/heads/master","description":"Rebase"},"26d58f7412681cea90902ee50e19a64184319518":{"kind":"REWORK","_number":3,"created":"2024-01-07 19:52:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/98/904198/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/904198/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/904198/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/904198/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/904198/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/904198/3"}}},"commit":{"parents":[{"commit":"2e2b652419f2caef75cec5ae1ea4b8bd6a6fe54a","subject":"Increase minimum test coverage to 97%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e2b652419f2caef75cec5ae1ea4b8bd6a6fe54a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-07 18:00:04.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-07 19:51:58.000000000","tz":-480},"subject":"Cleanup cross function / unit imports","message":"Cleanup cross function / unit imports\n\nChange-Id: I741bd5342625dd0617c9d319323a343cbb2f8cc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/26d58f7412681cea90902ee50e19a64184319518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/26d58f7412681cea90902ee50e19a64184319518"}]},"branch":"refs/heads/master"},"ddef40c9b1a5118666c4a0e150b5dde84cdea707":{"kind":"REWORK","_number":4,"created":"2024-01-07 19:53:35.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/98/904198/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/904198/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/904198/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/904198/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/904198/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/904198/4"}}},"commit":{"parents":[{"commit":"2e2b652419f2caef75cec5ae1ea4b8bd6a6fe54a","subject":"Increase minimum test coverage to 97%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e2b652419f2caef75cec5ae1ea4b8bd6a6fe54a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-07 18:00:04.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-07 19:53:29.000000000","tz":-480},"subject":"Cleanup cross function / unit imports","message":"Cleanup cross function / unit imports\n\nChange-Id: I741bd5342625dd0617c9d319323a343cbb2f8cc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ddef40c9b1a5118666c4a0e150b5dde84cdea707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ddef40c9b1a5118666c4a0e150b5dde84cdea707"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
