)]}'
{"id":"openstack%2Fdesignate~904965","triplet_id":"openstack%2Fdesignate~master~Ia53230c6eaac1987c6e7cedc523079ee586dafaf","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-01-10 11:00:07.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ia53230c6eaac1987c6e7cedc523079ee586dafaf","subject":"Cleanup Heartbeat Emitter abstraction code","status":"MERGED","created":"2024-01-10 00:37:52.000000000","updated":"2024-01-10 12:14:38.000000000","submitted":"2024-01-10 12:13:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904965","meta_rev_id":"a2239f9bc1a965e5ac28a0958c69c50a87e74400","_number":904965,"virtual_id_number":904965,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-10 12:13:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-01-10 02:14:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-10 11:00:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-10 01:49:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-10 02:14:02.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"fb173805e70c2c3a646f4f1f41c9374151b6be37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-10 00:37:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f396dfe9dc9f3f4e5268faf43ea7caf056f5787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 01:49:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91ef16d946af49ca948af6e4f78f3a3a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/13b45326d5d64098aa1c70722fb98584 : SUCCESS in 47m 42s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/de77fa38cde64fe5a9903444f2fae020 : SUCCESS in 1h 06m 15s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a5e7c935f494f68951fb12dacf2516c : SUCCESS in 34m 37s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c1cfcdb440ef4656b9f41d2cb7301c10 : SUCCESS in 29m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3cb2532c988e4d7ab681aed7b3a98e6d : SUCCESS in 32m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3e197a71c96041b7a488129244570f67 : SUCCESS in 1h 05m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0f86ed412f2044fc83a1e80038681d14 : SUCCESS in 1h 08m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/14b31375d1084718971e42bef155aed6 : SUCCESS in 34m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/21cba3e91a6d4820b192bb4ff6fde9b9 : SUCCESS in 46m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fb2aa3ea8f342c2a65927410415e82d : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3362b60c8b7451ca611adcc765fc671 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2e09d556ad64b4e8c7f9ebedec8e889 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c03a1f573f74f9dbdc06f2c977d8b4c : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab7ae731fd194a62b9d4eb96a84a474e : SUCCESS in 9m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/30f58f2773e8496280cfdeee39d19331 : SUCCESS in 42m 57s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/df6a4993f92b4c4d8027748f7b080546 : SUCCESS in 8m 06s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/789ba132017a4480a9c4570bea120996 : SUCCESS in 39m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/074e75d19c6648329b2e3a7772b85e86 : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7dcae88925f2a8a2a2318324b12a987999f53b9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-01-10 02:14:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8debb1ad4ae7ad2201cb2b25dae78b2272d8712","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-10 11:00:07.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1dad06a7ec3681fa15b0c840205b08dd2f9db1bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 11:02:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"94e666da51c2b424c15cd3f0a7bad8477e61110b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 12:13:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9aae006d77f451bad984e2d98da703e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bbce41e8ff3e46819b28572bd29816f6 : SUCCESS in 42m 12s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9fc54d3248f34ef8bb2c2902c887fa32 : SUCCESS in 45m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b7ccbb889d674ba7a6f1b6bd7b8b9870 : SUCCESS in 51m 45s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/289574bb5e2d4d84834b4e20230dc0e6 : SUCCESS in 1h 05m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/35a5f70e183549269daf02ba82b733b5 : SUCCESS in 55m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3c4f45d54aff45eba426d50b4b126c67 : SUCCESS in 33m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73b05122ff164536802683c78701cfed : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/161bf92fc6584755a0431bde15882226 : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e192fecdb24d4075962be152125ca35e : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23a40a5949cb4c37bdf97fe0c9d7c6e1 : SUCCESS in 7m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/184e175e4053465d8357da6cdc5b5dac : SUCCESS in 35m 09s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/07ae755584b34500b92fcbd1bcd50b9f : SUCCESS in 6m 56s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e0b4330b5b424e53980058de394305cb : SUCCESS in 53m 57s","accounts_in_message":[],"_revision_number":1},{"id":"51bd9dfafb5d14991df76ab134937b1091fb113d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 12:13:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a2239f9bc1a965e5ac28a0958c69c50a87e74400","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 12:14:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d07f08ca4bb445cda3d56069f7ba0918\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7a49066d8c343fdac72b7280956a484 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"002005a440e87198e05719d17c3d39a980a552e3","revisions":{"002005a440e87198e05719d17c3d39a980a552e3":{"kind":"REWORK","_number":1,"created":"2024-01-10 00:37:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/65/904965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/65/904965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/904965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/904965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/904965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/65/904965/1"}}},"commit":{"parents":[{"commit":"d8cd8bf44802d0c2df7ec52eac007e60d44c23ad","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d8cd8bf44802d0c2df7ec52eac007e60d44c23ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-10 00:37:26.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-10 00:37:26.000000000","tz":-480},"subject":"Cleanup Heartbeat Emitter abstraction code","message":"Cleanup Heartbeat Emitter abstraction code\n\nChange-Id: Ia53230c6eaac1987c6e7cedc523079ee586dafaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/002005a440e87198e05719d17c3d39a980a552e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/002005a440e87198e05719d17c3d39a980a552e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
