)]}'
{"id":"openstack%2Fdesignate~904966","triplet_id":"openstack%2Fdesignate~master~I030c071eaacd5ef5d5a2084ffe1b177a9e4d7762","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-05-09 18:46:04.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I030c071eaacd5ef5d5a2084ffe1b177a9e4d7762","subject":"Simplified sink service and improved coverage","status":"MERGED","created":"2024-01-10 01:11:16.000000000","updated":"2024-05-09 19:44:33.000000000","submitted":"2024-05-09 19:43:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904966","meta_rev_id":"be891a02be36294c7c80c5cbfdbe78a064ec44aa","_number":904966,"virtual_id_number":904966,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-09 19:43:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"all":[{"value":1,"date":"2024-04-12 22:31:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":1,"date":"2024-04-12 22:31:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-14 22:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-05-09 18:46:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-10 02:45:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-18 01:21:07.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2024-02-18 02:55:10.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"},{"updated":"2024-04-14 22:01:11.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ec79f7919ec924be770f787051d529b95235f733","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-10 01:11:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3891ac05a09e34cc2622b6a641ff53217b1ef0e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 02:45:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40cbc68b6643465b9201a80a0c756bcc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4a22e4560b2344398ed3943892497e69 : SUCCESS in 50m 21s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7c56134721ec4b9086dff6f4b43a423b : SUCCESS in 59m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cfa59cb28e94ab0875c31781150c576 : SUCCESS in 53m 20s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2fa87d79e3ab4d6bad6ac3c580686b9e : SUCCESS in 48m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/67595905c5844baf9564d6af6e641bcd : SUCCESS in 50m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0ad960da22cf47dc950c72a713aada74 : SUCCESS in 1h 09m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/16c1684f8659448f81d71ee75bd5693c : SUCCESS in 1h 13m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2db35fe61dbc47ebae16e0e1981c871d : SUCCESS in 49m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/21941580b15c4106b8a2c6dfeec68a91 : SUCCESS in 50m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20abdb3c0b284b07aa23e07461250ef5 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a481f37aa67842908c550eeeeacf009f : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54a57191f71d45598e62fd8d64a80a06 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b300e046eb54305a32cbfe79276e0c4 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed9a5821593241859132c45c988ab37f : SUCCESS in 8m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7f1f9479d8734a20a6822ea12950bf44 : SUCCESS in 34m 15s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3ad45b0ed3d140b48a2d44bec1ebacd5 : SUCCESS in 7m 22s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/26d6c2d7b6934574a24fbe6cb505a354 : SUCCESS in 51m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ed1a2d359a644ce689e873c50fe1b650 : SUCCESS in 4m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8558eb28f690d52af523029f3306694efc053cbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-10 23:17:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f5b8337c5d6fbe916564066102d551ae7abcb7fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 00:36:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01b5b6b3eb4844c6b8ce61b8256f8208\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d16ead31532b4c868df666c75b97a1a8 : SUCCESS in 49m 59s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6e3b80df05094c2b8e6ae191ce24d064 : SUCCESS in 45m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/af266497b2634f94afcc1af62802a952 : SUCCESS in 39m 02s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b7ddedd66388460490bbad953b77e5c3 : SUCCESS in 46m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cacf056f1d9a4bd38a5198eddccc2f12 : SUCCESS in 31m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b142bf8d4ad04c509fb0f41086e3db2c : SUCCESS in 1h 09m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a79c1bac4d3a43b38867ca9d818a73bd : SUCCESS in 1h 11m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e504a65d52cd4e03b512d8e74d6d45b6 : SUCCESS in 53m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/332a3c0096bc4ecc97ab5600c9fcdd2a : SUCCESS in 47m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a247f37a1ab4ce0bd9fdc7500df84ac : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edbbc82263cc419ea760cf46ef2b13ba : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03d49ed9aa214063ad85d6a688ab9b1e : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9d7840f7ea7f46039c0a2c5f8b165bb9 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73c988bca19e470185bc6346b8ec1ed6 : SUCCESS in 7m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b49c18373ff240de90fc2af5a39e0859 : SUCCESS in 33m 22s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a1150e053e1c446d8ee32e4792d65528 : SUCCESS in 6m 26s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bab6c3a4eecc46d3bdcc8fbbc2edc157 : SUCCESS in 51m 23s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/beef47f81f084e01a7f7e44185d14bbc : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4cf9984aa684ce956524c21f59d49652263aa480","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-15 12:53:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4c9c40f06faa95b2061edaeda337f65630d137c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 14:28:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61c717101e274e3db2e2225df8a553ed\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1ff3d1f6d5d74770adea837e976e6fe3 : SUCCESS in 34m 44s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3246a094ae5747bebe2359825826825b : SUCCESS in 1h 09m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/639bab22a665493f96bd95bd118b2d18 : SUCCESS in 41m 14s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/eeb2b99672f74ea494df45fd93e7799f : SUCCESS in 50m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/822f1f54a45f44d499a519b6fc0c2f73 : SUCCESS in 44m 01s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a6f0b3d98d6146079fca07b8bb3871e8 : SUCCESS in 1h 11m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/17f3bb4adaa349d4a3da0ac7dd45eb31 : SUCCESS in 1h 14m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d158d75e95d54b2c9c7374180eb6a6c8 : SUCCESS in 54m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/72b2e8c4621942ee98c490f2a8812c70 : SUCCESS in 47m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b5b6ad5c54f4c2b8e47fd57c72a3097 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b2ae887eee8402bb97e859fbf02d0ed : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5054dd115481426e94dec9bb4ef8a175 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/929fe0e2cc004dd6b169315d2e11cc61 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c4c51ac41f545ed998ab7c09a980fa4 : SUCCESS in 8m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d7316fa2c76a4ce5ac64637c4f9c69e4 : SUCCESS in 33m 04s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e7b9604c63aa4b36a883351b4e339df1 : SUCCESS in 7m 06s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8f09efa9ed574592a847e55891cb54fd : SUCCESS in 47m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2cafa78766f24a30b640a909fafca380 : SUCCESS in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"772506cbcad9e787957faae0e96be746911e3067","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-15 17:51:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"981df0ed9c06003f4892d3486e4f840cc2eb6d8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 19:22:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab0e73671d8d4e8a8e0d793e2aed604e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e03d831a27244883b876e9312c9265ba : SUCCESS in 46m 08s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7803997e1a7c489490fd53104fbb66de : SUCCESS in 51m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd98a7a212a54959bcf01a2f03d97e5c : SUCCESS in 45m 24s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1e4bb27191a447b0818439a4e3484764 : SUCCESS in 45m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2742243f6c9348e0b91c60a60a6070d8 : SUCCESS in 43m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3c113626345a4cc89c14e29f6b0b9e18 : SUCCESS in 1h 09m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2b23b94a9eaf4c968f91622749830b96 : SUCCESS in 1h 08m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4201d878adbd4638baee09e5d73b1831 : SUCCESS in 41m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7c44fa6db2ea4e0c9357ce89ff083b2f : SUCCESS in 39m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c74144534194665ab5be7429293f02b : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c622bb5e00a84e54a7d18246a3b8a538 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb5ccd6a357f46ddadbfd0151e87a6ac : SUCCESS in 6m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f48c5859b4f1432ea8f2808f2a9112e6 : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27ef9cb5ba69439d900e04a870e31042 : SUCCESS in 8m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a0ff680c1d4347c1884b402f5b9f5424 : SUCCESS in 50m 50s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2c63db5b5a0843169e26de7b66c6a647 : SUCCESS in 6m 48s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/941b1111d1434098b8dc0ae15e311789 : SUCCESS in 48m 27s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ad3d911e0a734ed187a3c960101b9183 : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"97e63bdfa6e34c7b0ffef00e6546f3419f3e7461","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-26 02:03:39.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6ba036336f296015f61891f611b5994ff1a3e617","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 04:43:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85fa9b93587e47dc99befb726382bdc1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b47b6592ad1540fc9f7084730c918c94 : SUCCESS in 43m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/62c051fd6f60448f9aa23be2e4706eb7 : SUCCESS in 43m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/12125a6eeed2446aa2049ce8b6a7db53 : SUCCESS in 49m 03s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/417275119d4c4597bb23ec3424b3b236 : SUCCESS in 55m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6e385259710c4ca189d0e7f3259be654 : SUCCESS in 45m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9bc43256921b443fa149af12329f9aba : SUCCESS in 53m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/73b8f7b94d1f4bfa92489677bcc45dea : SUCCESS in 55m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4c5e1dd659d94517ab868c7629915e88 : SUCCESS in 39m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/47b431be9eb740c99d70434d261b0ec7 : SUCCESS in 56m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ca2e22ed950489baf4849d6c368c1be : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cbcb12a16c74fe084be366fdef56fae : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cc7047dd62d4591a702ab04e007b587 : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3fb37d083704ba79b3b674cabf342ee : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/448936779a4244098df625281a535461 : SUCCESS in 7m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4c597d5747094373b07f2078aacd4197 : SUCCESS in 33m 58s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/258d44086e5045bf8bc8806756111e05 : SUCCESS in 7m 23s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4de96c1f1c6d469098bb77bccf04cff7 : SUCCESS in 37m 11s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2c7c6bca3fd24bda9e9e4e48dad95a4c : SUCCESS in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"538ceca7ccc5aef840e043ced9941c241eff6468","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-02-04 15:23:53.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"579d059fc6c619155587388c8d9c1cf7f4b9e807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 16:38:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fa73160aa194d49ab9cc193f7c74653\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dcb03cd7bbab4016b0198b62be2b923c : SUCCESS in 47m 04s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/219b6f26b1174fbd9fe47d35889b3d33 : SUCCESS in 1h 00m 41s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/cccb5e1861b940469ac359913272bf12 : SUCCESS in 1h 00m 45s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9ef622fb43e74dc096f75ce53e13b53c : SUCCESS in 46m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f544aa6834164f2cb9b11982a3f7e2c8 : SUCCESS in 35m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/99e876050b5647639178204da1283355 : SUCCESS in 1h 05m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/95a9bd3bfaad4e8788c0ea28628645cd : SUCCESS in 1h 02m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/af19d3ae24504fe6b62737c5d6060aec : SUCCESS in 31m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/95b60e3073774207851953ee2e3fd765 : SUCCESS in 44m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c2517917c45442283c9b445d0c93c5b : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea4207697f984f20b2e5c535a9bf0377 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37ce89f646df41648e3c07d07e341ce4 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d795c8e85784118bc4cb683aa7c3946 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/088c09eb505b4ec983f5190f4f026aa4 : SUCCESS in 7m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d3b1706ffeb24fd1acc411215e51f69b : SUCCESS in 45m 40s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/eb22d548cc42404c878014761dba255a : SUCCESS in 6m 48s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4196631f77db44b68688bd82429dad33 : SUCCESS in 37m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0aa1e7d81b9c43f488f0765878b9e01e : SUCCESS in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ff58528bf6fbcfb93e3274873bfe13d50bd314aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-02-17 11:36:44.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"11aab89672173cd0b983bd60d2d34723599df3a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 12:45:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74274858389446cea13d11e6a405f02a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/840c1dcfaf004152baa9b1acd5b34804 : SUCCESS in 48m 01s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d5497babdcfe4f3c8fa979f477baa963 : SUCCESS in 1h 00m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd52b355917c4c3694beb5c8cbc8ba9c : SUCCESS in 57m 50s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8492c46857b64716aa1b77eb46210ab1 : SUCCESS in 50m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0caaec23a0f745dc8945b3f5dd2bd33c : SUCCESS in 45m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9e758f94df0a407aba93702243475d94 : SUCCESS in 57m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/679725cdad4c4d5c9386fd98e9de0a40 : SUCCESS in 49m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/45e04a55855e42f7b9a8fbca34342cd9 : SUCCESS in 46m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e8bd787cf50249bda13dceba0e6e83c5 : SUCCESS in 36m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b681aa3a62e4d9a9c9d4660efdbf87c : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2544a5b3af88480eb399f3ad9ff1fb65 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acab338e1eab4b4fbb65d93e557adaf8 : SUCCESS in 7m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e116a859594b4fcfbe91c5b02e6ced10 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bba0bfb060884ad7a4fb3a3837fe5c93 : SUCCESS in 8m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e7b1df8d53774379b6f162d0a387237d : SUCCESS in 32m 22s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/26d574a4fa4946b7887421869476cb40 : SUCCESS in 6m 05s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a856d0e8dfe64d1d81b94b53441d2cc8 : SUCCESS in 41m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c82ef5330ed44f3fbd121060fefe1044 : SUCCESS in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"89e5a7f9a669b323078337ef8394f7487a81f5d6","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-02-18 01:21:07.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"c0456f28dab18d269cb030230c7ac6df2fac3f87","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2024-02-18 02:55:10.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"a00172b4626527533ab6bac5678f16458cb13740","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-03-02 15:48:48.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"c152202f3d018d963362645026bb213336a8dd78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 17:00:40.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf194b610e594bd8857028e7dd5999b8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/642103500e334f8ba9885d02fc6dbda8 : SUCCESS in 37m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/693107ea5cca40b7ab15f11f0ba7ddd8 : SUCCESS in 58m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f8f1c6297194b52872ae6dc274827c7 : SUCCESS in 1h 01m 59s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/fa82d81370644278bb8ae6a8a993ae65 : SUCCESS in 44m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/59edcb7f087246218cf9facae1ecf95c : SUCCESS in 44m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9fc5ac3f9fe94ce0a20bed1e0da6306b : SUCCESS in 1h 04m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a5af7ef42aa94e42a87bc6a21a6a7d04 : SUCCESS in 1h 06m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/399244c6b3c2421d88604415941629e1 : SUCCESS in 33m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/af8118d0d1bf411ea41a45da21cef2ea : SUCCESS in 46m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75b6bdced14144db975c8338ed92cd7d : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66456767de0c49fda292a721150611f8 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e6db82d70b94368ab05e1168c750c64 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f362e18c9854464b89d83f4b6f6cf429 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9724c27de26417db725c9c0796afb01 : SUCCESS in 6m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b680ed74038b4f67acf2fa06475227f3 : SUCCESS in 30m 41s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/dddcec9de8594aeebd4c24e3374614c6 : SUCCESS in 8m 05s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0a5a946e8be64e38b38f089829eddd1f : SUCCESS in 37m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/76303a9521b0461b8b6fd38edd1d68b6 : SUCCESS in 5m 36s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/7472a29ee83e428db268c20bf35d917d : SUCCESS in 39m 44s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a23851eddb061d977d6372dff7b1e073a581ca7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-03-23 18:17:50.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"a836bb31bee111204fc95d49f78ac481c3d95b96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-23 19:31:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5732bfffa574d6c9481b75d555e88d8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4c096230fc444511b5890b56670a9fb6 : SUCCESS in 44m 33s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e6de0fb3535e4082b5d1a20cc144c6bf : SUCCESS in 52m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b170d434b2114070bbbd01e52a022cf7 : SUCCESS in 52m 17s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e4fbf9f01200419ca83f6ae435b82fc1 : SUCCESS in 44m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fb9673b32dff4ae9b59c7554626f1274 : SUCCESS in 47m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/eae7e91096164f90be618a139f5ce18c : SUCCESS in 1h 07m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1695b38060c648f49721cf644625c1b4 : SUCCESS in 1h 04m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7ee719498c4f4c159e2a060d81d7666a : SUCCESS in 36m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a3903bb8d91141ee82fda8c371a5687d : SUCCESS in 40m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78a49ad6f75a400297419e29cfa703d1 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e53f3a5d8da4a8e8916d0e44be57413 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2831409fb5e42fabeb9c2772ad6d720 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e6e83a1120f47c380611ecf76b12a87 : SUCCESS in 6m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c1eaaeb094334cc99a112b77f89ef213 : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15471551f7164eefa3c3b2dac5e5b1a1 : SUCCESS in 6m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b2064269f861478e9fd92266bd52f19e : SUCCESS in 32m 25s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/742228b795144d4e91eaeaef117c4764 : SUCCESS in 8m 10s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/09be40d0cedb423083275c83f5ef3209 : SUCCESS in 50m 27s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/30eff567b616444f837f68a590c49293 : SUCCESS in 4m 42s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/7e24d45684f44391a5d097cefd7b141b : SUCCESS in 33m 04s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1e7ef6af15a3f1c9889d17a2ec16877005fe1de6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-04-12 22:31:54.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"6dc48bf9876c449b7400d3e3628ce60bcbb9fc4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 23:57:26.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7fae6890860462693cc67670b546e3a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e8485ad4f38a4f388f708e5255982d4a : SUCCESS in 48m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/01decc4306e641e789808ddb81452432 : SUCCESS in 56m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/62bcb963deec4227abff825a9f9477ab : SUCCESS in 39m 37s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a1958bf42a5d4b9b85853ac799a80d2d : SUCCESS in 38m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/134588edd48149a18d80cfc7285ea042 : SUCCESS in 41m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/dd2f39f3ae104fb4b89dd8b811b744ad : SUCCESS in 48m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/be1db5a6f2f5442b812f50e74f048ee9 : SUCCESS in 44m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3054fdb62c6e4acbb8096afe0c36774a : SUCCESS in 31m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2d0abaeb412142878ea93a7f5831d2a3 : SUCCESS in 33m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a4e3530ee0a420b8fa1d2daa7f2fdcb : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5953ed7e9c264f25a895f4ed022d7dd6 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53a24377779840e68fac7c4fde12c11e : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a26d57c801824542a456e9e399037788 : SUCCESS in 5m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99426a69bf3d426693b48258966e1a29 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/412972b032434c6cb39bd317a88e2899 : SUCCESS in 6m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2050d63cf0694588bd116d10aaa0e312 : SUCCESS in 32m 07s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7110537ffebf4a02a8bf8b24ed55ea3d : SUCCESS in 6m 34s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7f8a5aa57a824e3d8013771096be72dc : SUCCESS in 48m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/82fdd1bb3be54a99b5ee1ac92177e90c : SUCCESS in 4m 17s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/31e0552d19334cb2b9e299164444bbe6 : SUCCESS in 30m 10s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d26661c6e64949fac3fb0789d55e441f48474849","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-04-14 22:01:11.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"982cceb90cbd602796964b758ba9266a8998798f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-09 18:46:04.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"92af026757f27f2c29f5bd8036188149d92e2ebf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 18:46:40.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"66fe6bac6b63c94b998f1a99aa8e71ec893cda08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 19:43:07.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/372eabf17ac246ad86a3b0e451f5f535\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/10268d107be64a73989547862c8896fd : SUCCESS in 36m 15s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/62248324fa2647f186a3825e837f4d37 : SUCCESS in 33m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fbfba04ddfba44d98b24f4eb2fa096e2 : SUCCESS in 31m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9ecb29f8bea64a5ab6a27bb3ffa1ccfc : SUCCESS in 55m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6208cc60de2c4f55974f6ef69139ca48 : SUCCESS in 36m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/01cc43b32b9d464bbcb8aafad214f2e9 : SUCCESS in 34m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f14817f1cb3f44b280d8d8782f575568 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/344efb444a2d48938381d73fd543660d : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc3880154db14bbc9ab5fd2b82407194 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67e6dace971a4e11a662a3ad2d277102 : SUCCESS in 10m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/90a42354994e4350a73f1ca99ae7e72b : SUCCESS in 43m 19s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9f2ad576fe6343039679ad7fb83e67ba : SUCCESS in 6m 26s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ef670af34e0c43bf992eab567d40c8cf : SUCCESS in 51m 05s","accounts_in_message":[],"_revision_number":10},{"id":"dbf09d364282a01dddf157bc5c2e4721bada096d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 19:43:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"be891a02be36294c7c80c5cbfdbe78a064ec44aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 19:44:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/492d5a9447954150b5ed29616286eb5e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0e0928d512641289ba75ef6ae8e917e : SUCCESS in 53s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"f314f471b0144e6d4101eece912905deff2dca07","revisions":{"53db7b6f2085276f2ed54a77215b3a9427d44ff9":{"kind":"REWORK","_number":1,"created":"2024-01-10 01:11:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/66/904966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/904966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/904966/1"}}},"commit":{"parents":[{"commit":"d8cd8bf44802d0c2df7ec52eac007e60d44c23ad","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d8cd8bf44802d0c2df7ec52eac007e60d44c23ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-10 01:09:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-10 01:09:15.000000000","tz":-480},"subject":"Simplified sink service and improved coverage","message":"Simplified sink service and improved coverage\n\nWe had some unnecessary optimizations that made the code difficult to read.\n\nChange-Id: I030c071eaacd5ef5d5a2084ffe1b177a9e4d7762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/53db7b6f2085276f2ed54a77215b3a9427d44ff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/53db7b6f2085276f2ed54a77215b3a9427d44ff9"}]},"branch":"refs/heads/master"},"4b5bb27577474879911b49be89608ec02946a6d9":{"kind":"REWORK","_number":2,"created":"2024-01-10 23:17:40.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/66/904966/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/904966/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/904966/2"}}},"commit":{"parents":[{"commit":"d8cd8bf44802d0c2df7ec52eac007e60d44c23ad","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d8cd8bf44802d0c2df7ec52eac007e60d44c23ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-10 01:09:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-10 23:16:57.000000000","tz":-480},"subject":"Simplified sink service and improved coverage","message":"Simplified sink service and improved coverage\n\nWe had some unnecessary optimizations that made the code difficult to read.\n\nChange-Id: I030c071eaacd5ef5d5a2084ffe1b177a9e4d7762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b5bb27577474879911b49be89608ec02946a6d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b5bb27577474879911b49be89608ec02946a6d9"}]},"branch":"refs/heads/master"},"2bf2129d4682213a9897acae8cbe1d6fd9bd98f6":{"kind":"REWORK","_number":3,"created":"2024-01-15 12:53:15.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/66/904966/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/904966/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/904966/3"}}},"commit":{"parents":[{"commit":"d8cd8bf44802d0c2df7ec52eac007e60d44c23ad","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d8cd8bf44802d0c2df7ec52eac007e60d44c23ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-10 01:09:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-15 12:52:32.000000000","tz":-480},"subject":"Simplified sink service and improved coverage","message":"Simplified sink service and improved coverage\n\n- Removed unecessary error handling.\n- Simplified event type handling code.\n\nChange-Id: I030c071eaacd5ef5d5a2084ffe1b177a9e4d7762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2bf2129d4682213a9897acae8cbe1d6fd9bd98f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2bf2129d4682213a9897acae8cbe1d6fd9bd98f6"}]},"branch":"refs/heads/master"},"9e93082e4f641877dab59c80efa148945c711033":{"kind":"REWORK","_number":4,"created":"2024-01-15 17:51:46.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/66/904966/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/904966/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/904966/4"}}},"commit":{"parents":[{"commit":"d8cd8bf44802d0c2df7ec52eac007e60d44c23ad","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d8cd8bf44802d0c2df7ec52eac007e60d44c23ad"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-10 01:09:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-15 17:51:39.000000000","tz":-480},"subject":"Simplified sink service and improved coverage","message":"Simplified sink service and improved coverage\n\n- Removed unecessary error handling.\n- Simplified event type handling code.\n\nChange-Id: I030c071eaacd5ef5d5a2084ffe1b177a9e4d7762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9e93082e4f641877dab59c80efa148945c711033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9e93082e4f641877dab59c80efa148945c711033"}]},"branch":"refs/heads/master"},"f2b240ce605e6a672e29b10dad374ed3df77aa14":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-01-26 02:03:39.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/66/904966/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/904966/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/904966/5"}}},"commit":{"parents":[{"commit":"cb47b4e00e1b4bc1a560b95b9ac7d1514942d879","subject":"Make edgegrid-python optional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cb47b4e00e1b4bc1a560b95b9ac7d1514942d879"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-10 01:09:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-26 02:03:39.000000000","tz":0},"subject":"Simplified sink service and improved coverage","message":"Simplified sink service and improved coverage\n\n- Removed unecessary error handling.\n- Simplified event type handling code.\n\nChange-Id: I030c071eaacd5ef5d5a2084ffe1b177a9e4d7762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f2b240ce605e6a672e29b10dad374ed3df77aa14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f2b240ce605e6a672e29b10dad374ed3df77aa14"}]},"branch":"refs/heads/master","description":"Rebase"},"ebecaa9d37b1d5e7b95e7bd6bca358645c689908":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-02-04 15:23:53.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/66/904966/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/904966/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/904966/6"}}},"commit":{"parents":[{"commit":"dd5993ad1b0633cee3269c4b2476d4acc9b672fc","subject":"Merge \"Removed unused function and improved Zone Master coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dd5993ad1b0633cee3269c4b2476d4acc9b672fc"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-10 01:09:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-02-04 15:23:53.000000000","tz":0},"subject":"Simplified sink service and improved coverage","message":"Simplified sink service and improved coverage\n\n- Removed unecessary error handling.\n- Simplified event type handling code.\n\nChange-Id: I030c071eaacd5ef5d5a2084ffe1b177a9e4d7762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ebecaa9d37b1d5e7b95e7bd6bca358645c689908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ebecaa9d37b1d5e7b95e7bd6bca358645c689908"}]},"branch":"refs/heads/master","description":"Rebase"},"3ec74336a4b5629c5f554e93e3612ebdcaa5e5e1":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-02-17 11:36:44.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/66/904966/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/904966/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/904966/7"}}},"commit":{"parents":[{"commit":"91f7d7d3fb3c39c6897f1508ce390c2dcbbd8759","subject":"Merge \"Add support for catalog zones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/91f7d7d3fb3c39c6897f1508ce390c2dcbbd8759"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-10 01:09:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-02-17 11:36:44.000000000","tz":0},"subject":"Simplified sink service and improved coverage","message":"Simplified sink service and improved coverage\n\n- Removed unecessary error handling.\n- Simplified event type handling code.\n\nChange-Id: I030c071eaacd5ef5d5a2084ffe1b177a9e4d7762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3ec74336a4b5629c5f554e93e3612ebdcaa5e5e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3ec74336a4b5629c5f554e93e3612ebdcaa5e5e1"}]},"branch":"refs/heads/master","description":"Rebase"},"d97088b9d290bbf247900ba2fdff3d41a4798b22":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-03-02 15:48:48.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/66/904966/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/904966/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/904966/8"}}},"commit":{"parents":[{"commit":"54e3ece623d018923602104d4ff8c666ccf15de1","subject":"Merge \"Add catalog zone support to devstack plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/54e3ece623d018923602104d4ff8c666ccf15de1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-10 01:09:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-03-02 15:48:48.000000000","tz":0},"subject":"Simplified sink service and improved coverage","message":"Simplified sink service and improved coverage\n\n- Removed unecessary error handling.\n- Simplified event type handling code.\n\nChange-Id: I030c071eaacd5ef5d5a2084ffe1b177a9e4d7762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d97088b9d290bbf247900ba2fdff3d41a4798b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d97088b9d290bbf247900ba2fdff3d41a4798b22"}]},"branch":"refs/heads/master","description":"Rebase"},"1e5a7005679c1eb14839e49c2ae7399ae5dc07db":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-03-23 18:17:50.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/66/904966/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/904966/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/904966/9"}}},"commit":{"parents":[{"commit":"097ffc6df181290eba1bcd7c492b1b505bc15434","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/097ffc6df181290eba1bcd7c492b1b505bc15434"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-10 01:09:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-03-23 18:17:50.000000000","tz":0},"subject":"Simplified sink service and improved coverage","message":"Simplified sink service and improved coverage\n\n- Removed unecessary error handling.\n- Simplified event type handling code.\n\nChange-Id: I030c071eaacd5ef5d5a2084ffe1b177a9e4d7762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e5a7005679c1eb14839e49c2ae7399ae5dc07db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e5a7005679c1eb14839e49c2ae7399ae5dc07db"}]},"branch":"refs/heads/master","description":"Rebase"},"f314f471b0144e6d4101eece912905deff2dca07":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-04-12 22:31:54.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/66/904966/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/904966/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/904966/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/904966/10"}}},"commit":{"parents":[{"commit":"808f0c9198a4aad558b22c3b3f839202a825f980","subject":"Cleaned up coordinator and locking byte handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/808f0c9198a4aad558b22c3b3f839202a825f980"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-10 01:09:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-04-12 22:31:54.000000000","tz":0},"subject":"Simplified sink service and improved coverage","message":"Simplified sink service and improved coverage\n\n- Removed unecessary error handling.\n- Simplified event type handling code.\n\nChange-Id: I030c071eaacd5ef5d5a2084ffe1b177a9e4d7762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f314f471b0144e6d4101eece912905deff2dca07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f314f471b0144e6d4101eece912905deff2dca07"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
