)]}'
{"id":"openstack%2Fdesignate~905817","triplet_id":"openstack%2Fdesignate~master~I8e91b06199bfbed1d6e9f8d62ddf52fafd60b7f9","project":"openstack/designate","branch":"master","topic":"bump-hacking","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-18 10:37:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8e91b06199bfbed1d6e9f8d62ddf52fafd60b7f9","subject":"Bump hacking","status":"MERGED","created":"2024-01-17 03:38:09.000000000","updated":"2024-03-18 10:38:34.000000000","submitted":"2024-03-18 10:37:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"905817-bump-hacking","meta_rev_id":"38a71ca1e32794e3c86cfee774e33e8e07cd6d36","_number":905817,"virtual_id_number":905817,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-18 10:37:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-03-18 09:28:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-02-08 21:06:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":1,"date":"2024-01-20 05:20:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-17 11:02:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-03-18 09:28:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-03-18 09:28:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-17 05:00:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-17 11:02:38.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2024-01-20 05:20:04.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-02-08 21:06:47.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-03-18 09:28:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"b531f9aec106130decea360f0551ef598c91ca1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-17 03:38:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca6a3ced0e8a5fe65b7b0e75fadf6da92cf60e3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 05:00:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9fd5afe3062438dbf690e9995e0da95\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ddec9ae30ece4ec29cc1d5d5113161f1 : SUCCESS in 4m 12s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a74820871a41421faca8d5de88e8246d : SUCCESS in 46m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3231c97e8c7d4339beab8fb8a2c57358 : SUCCESS in 1h 02m 01s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e1e26c96675456b95caf0a4ea6fd6d2 : SUCCESS in 1h 12m 19s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d501fe0c7da24db5a2bbc7fbd470e95c : SUCCESS in 33m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8848207b9ad54d5db12a71e71b514865 : SUCCESS in 50m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4117c701681143c39a5b6404114ec331 : SUCCESS in 59m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/13af41d768aa4981b7e5bab7051cdbd4 : SUCCESS in 1h 11m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/31cc673c40d547e59d72f3414f5507e6 : SUCCESS in 35m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e70c3db559584e40b64b24c2f9fbfb84 : SUCCESS in 50m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0669e0339554cce90e7ca50ed36dc5d : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41b4d0e3c72e4947a26485266308ea66 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2ab637ddd4549d8ac896ca2bf512bd3 : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f4f96cc83b64a0aae28e0d9ff992d7b : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/918c3f76e2724432bc070f144202b2df : SUCCESS in 9m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5aaf9c9bc30749b3816981344ff66950 : SUCCESS in 1m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45f919318949454fa5ecdcc07419fba7 : SUCCESS in 5m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9913d9aee18c498c84ec784b047a052d : SUCCESS in 43m 42s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/360426fa2c7941819a998d8779bffb98 : SUCCESS in 6m 42s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a6c5391b4b354ffe9307a172bb6337ea : SUCCESS in 48m 45s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2ceafc9bab294691a699eaeeba5a040d : SUCCESS in 7m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c0f34b1a1abb1f25c2c445ecf923216bb8782c1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-17 11:02:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"61d45a4f6e9435e404fad56e4b95396594cb5610","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-01-20 05:20:04.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28d68aa6609d019db9446bcfc4f1cecfead530ee","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-08 21:06:47.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0471c89ddaf8b648aea469d2550c7bcf1f54a1d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-18 09:28:58.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bac0e2b7224067fbe6f70ee9ab2f43b0de717663","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 09:29:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87fda75048c0442c1e3ab6600f114dc4f343f039","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 10:37:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0d4f6fe6c8c4004b97d0ebd6eceb3e3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/148857a9172445fe8dec7e0e15f16380 : SUCCESS in 3m 58s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7ceb5c734a214342babb4a54332f2957 : SUCCESS in 42m 08s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/bb8aabb666ff49dfa3cfb6ec9ae48abd : SUCCESS in 43m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/03051c5292f04a83be85e571890b715a : SUCCESS in 46m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3d6e0a4001fc492f90dec641f8ee4971 : SUCCESS in 1h 02m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1e6c384563c44ac5ac75e96cc2e1b949 : SUCCESS in 42m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c379fdcbf30042e392e7264153e4ba5c : SUCCESS in 44m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8d131e7a7df49bf8128645099e9859a : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/694eba67650e4e3084e3c5fd7efc7996 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a9b5fc11c964a11b023ea2e6ec59f6d : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d9ce1d56f104fc08c31c4b4119b62ab : SUCCESS in 7m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/802b1ef7397f4807917705b14dc4ffd2 : SUCCESS in 2m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6ed7dfa72264084a09a663322fa3bcc : SUCCESS in 4m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/db03e59b91b14ca891508064afb83745 : SUCCESS in 31m 42s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/01aee946ad274fd6b28790c7bc54efc9 : SUCCESS in 6m 04s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/21cd6f838f3046cd9eab60cdc1b6a61e : SUCCESS in 48m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d8ba37a84e9206271b27930673d9c4a6db38e129","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 10:37:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"38a71ca1e32794e3c86cfee774e33e8e07cd6d36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 10:38:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb41559561b64639a415c7a41c454367\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/9f97dee1b9fa439986d526934d72a4ec : SUCCESS in 44s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdf8bdf9a20f4b0a99a68f8b060ec943 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4b9e1e15ed448008a5c592ae97a0d57 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7472f515a6776dd52442f3d44c52092b88a206a9","revisions":{"7472f515a6776dd52442f3d44c52092b88a206a9":{"kind":"REWORK","_number":1,"created":"2024-01-17 03:38:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/905817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/905817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/905817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/905817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/905817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/905817/1"}}},"commit":{"parents":[{"commit":"21b266b070ab9d84e87aef8487dbc149e518c0cf","subject":"Merge \"Remove devstack support for suse\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21b266b070ab9d84e87aef8487dbc149e518c0cf"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-17 03:25:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-17 03:37:51.000000000","tz":540},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.0.x is too old. This also removes the note about old behavior\nof pip which has been changed in recent versions.\n\nNotes:\n- A few errors newly detected are also fixed.\n\n- This also fixes the compatibility with flake8 \u003e\u003d 3.0 because flake8\n  is also bumped when we bump hacking.\n\n- pretty_flake8 tool is removed because it\u0027s not commonly used but\n  needs maintenance effort to keep it in sync with newer releases\n\nChange-Id: I8e91b06199bfbed1d6e9f8d62ddf52fafd60b7f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7472f515a6776dd52442f3d44c52092b88a206a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7472f515a6776dd52442f3d44c52092b88a206a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
