)]}'
{"id":"openstack%2Fdesignate~906248","triplet_id":"openstack%2Fdesignate~master~If13e722e5ec3cdcd11461201ea20b2dc022cce10","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-23 00:09:35.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-01-22 12:54:37.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"If13e722e5ec3cdcd11461201ea20b2dc022cce10","subject":"Make edgegrid-python optional","status":"MERGED","created":"2024-01-22 07:58:19.000000000","updated":"2024-01-23 00:10:52.000000000","submitted":"2024-01-23 00:09:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906248","meta_rev_id":"1d429ae8b9970993c7c6c7d53f61e99758101a6f","_number":906248,"virtual_id_number":906248,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-23 00:09:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-22 22:33:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-01-22 12:54:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-22 22:33:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-22 09:19:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-22 10:21:23.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2024-01-22 11:22:38.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2024-01-22 22:33:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"c6e99a1f2558838ce9cbe07214faeafbec713b75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 07:58:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1f21342fb65e7c62283e0c2533e21f0bd2b2390","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 08:01:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5ec915cc806b6bfe4f0a592dc5266b28fb367109","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 09:19:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5862225d7d2c4a0885aad2e0bc1854ce\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/91df20c8378046ab8e5bebcac2f6a624 : SUCCESS in 3m 43s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0a40a8f0ae9b43dabdc9c647572ed7b7 : SUCCESS in 1m 51s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c7cd3baccb1148e2a7aed22a33ac892b : SUCCESS in 55m 23s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/afcb215b0cd8486f9350e1d779c0df33 : SUCCESS in 59m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a0979bcd02348f7b5a4e6654f3de936 : SUCCESS in 54m 56s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/78e30b31ae014e2a947ceb40f5b493bd : SUCCESS in 33m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/24e05ca5186f456a8212ed5f3104287f : SUCCESS in 44m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b795291929944a23a9052049d2dcacbe : SUCCESS in 1h 11m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5d66d0475bcb46c282322c1a0e1ca255 : SUCCESS in 1h 08m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/331406ddde5e4b5987951adcec25e332 : SUCCESS in 45m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2fe1e40063874529bc88c3f3438b8189 : SUCCESS in 32m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bcc727c78874246b557683292ee01dc : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b75a41a361c74df3b317b8b8ad2fbd65 : FAILURE in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f44ab4d08e4440f9ca33ab69f9e07a0 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b92ae8f63aed4c78a9963aec9ea60fd6 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cff565c47ca2499bb754020398210d9a : SUCCESS in 7m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/13c0a6966a9b408ebb004ff7a124eb33 : SUCCESS in 2m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed06f60f16d94ddaa5914e3e96e93dec : SUCCESS in 5m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/00f11175229041cd9f0fa548a82b3c64 : SUCCESS in 32m 34s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7af625275bee435f892254551e9286b1 : SUCCESS in 6m 47s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1d5975fee58c44d08efa7c80fc5d47ce : SUCCESS in 48m 23s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/52b8b8fe284d4511b8ea2123ef3007c2 : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eda169a3a13ef8ff25afcea85c6c4ad60b5d11ce","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-22 10:21:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fbd166aabf9e8cbd128afdd74680eadf07273b86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 10:29:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a59448fcbe68621aa0250a8f988f6cd62b086e0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 10:46:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1851e580dacbf8bc98b4e9dda52ef530a035ae28","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-22 11:22:38.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3edacab52655a9a6c802bca3a50030ca67618543","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 11:52:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"99d6d36789f32051791461de08c20c29a02c9fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 12:34:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5b32b83cddb4adc9d87f6ac33c10284\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2b2c36f491f54de98269cd9295de62a9 : SUCCESS in 3m 55s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3217ab17a4204e03a217d0dc50b400d9 : SUCCESS in 2m 33s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9bcc5ea7828a45b8a79698cc9d0ed5a6 : SUCCESS in 45m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/05bb6b8cbd60445e98edfd13f887c528 : SUCCESS in 47m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/67c707ce19a74b7b949f8b9363f92878 : SUCCESS in 39m 26s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cb4d44a09d494663a065bf29f4553d8d : SUCCESS in 44m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b942ba784dcc42569b5bb6b8efec03da : SUCCESS in 41m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/66f4a0006f9f4c55845a895fb4cb9e02 : SUCCESS in 1h 10m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/668400b7549c461fa6875b3bbde68073 : SUCCESS in 1h 15m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f94d5bd7f6a440ec9d72aabffdfaadb1 : SUCCESS in 48m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/772f0a436ce24c008f8740035d3ab0cb : SUCCESS in 47m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9700f21a881e4dfabb370593716538c3 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/decf6e83d91645139124cd04ce635b86 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c35892e1b384751a02adb728004cf4b : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13d634757a664e75a17743ec20848ce8 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d7a4b0bc8ba49f1bc560462d7daa1db : SUCCESS in 8m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f47f35274df14bffab698ac0f1dd8339 : SUCCESS in 2m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4307604c65e644fb8bda0a3fa88bb5c3 : SUCCESS in 5m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/804fa62c6907449886282db4b11aef10 : SUCCESS in 32m 40s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/76da959554b5447a8a56dab8f264b8e2 : SUCCESS in 7m 19s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b8e74b6d8bf042c2b883ffc9e274189f : SUCCESS in 46m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e85af95e540945eda34f006dfc386ce0 : SUCCESS in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"aa36511df9ea50c002e9956a8121af8137ac0875","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-22 12:39:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e1d8ee85bd2201c8a3d31cc56f72f978c930834d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 12:47:42.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"7aad16560e94ae4ea53ec6236d984760d4960a96","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-22 12:54:37.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"752e7d9bca02955a882a0300a7ad3debeb99a713","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 14:14:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6992f2ae9c764c93804ff03049d4c99d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bd3d773d396241cfbd5efb787d8aaea5 : SUCCESS in 4m 08s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2a65cd0efbf64711b5fa96465051281d : SUCCESS in 2m 55s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/47dd5a2677ed41debb1eacce9658803c : SUCCESS in 56m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/93e74b970fba4124870103ffe814c392 : SUCCESS in 48m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/35aa03828aaf49439d5ba5c2808aa842 : SUCCESS in 57m 22s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c349ecee48a647919a87ed897b3e0bf0 : SUCCESS in 47m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/12f96342c0b541f2816e11f436a0c126 : SUCCESS in 45m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/259c2ca9c04d4bab882192fdbddaa393 : SUCCESS in 1h 18m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/52662716ab3e4012bc7c6ac08b034dc6 : FAILURE in 45m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7cd65be32f3d4941a46c24d84e8afb09 : SUCCESS in 48m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e6224bb1e20440b6aa49c59bc54d827b : SUCCESS in 36m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ae2c6e02b8b4b018c8c38a6bc349088 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c9f95caf8b343fcbdf841b82bdf4e82 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f326fe14a34483abfd4d809ac00e5d2 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c29b50c7e8684caa93f4723f00f2c469 : SUCCESS in 11m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36556cd4bd4647abaa62d7a74babf9f4 : SUCCESS in 7m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0daa78355aac4fc79571266fcd8fc98c : SUCCESS in 2m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1adad9f9f868451cb5bf215c6a7fe99a : SUCCESS in 4m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9faa42f3da124f16a23d93b596d86398 : SUCCESS in 32m 19s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/261def9c501248feb2165eea47b8f544 : SUCCESS in 7m 00s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ce32bc262cd1432fb472395abc8a0299 : SUCCESS in 49m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1c6f0f0be6134dde8d3964dbe60024e1 : SUCCESS in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2e473b880e9118061620ace18925e2958b8ca9a5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 14:17:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ceec7fd70a737f050e3bdf447e8a38888d2eb6c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 15:57:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/876519a11b6b458abe752308b2e5931e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/75e70d941be44e9784bd896f1eaddc64 : SUCCESS in 4m 19s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/187e432e83364e309efe4a6d28e5d202 : SUCCESS in 2m 09s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5c3f768ccf034b40be37f193dcf08a29 : SUCCESS in 47m 28s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/888fe70e723e4d8194efee9afa9712c1 : SUCCESS in 1h 06m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/128a2cd156c045268d902c0298aa199b : SUCCESS in 1h 02m 22s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c947a2627b384c3d9e90bad96c962b61 : SUCCESS in 38m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e8651700922f4796a76734659c3e361f : SUCCESS in 44m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ea33af2431b844fc92a6307e832a62eb : SUCCESS in 1h 07m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0066ca3106724b07902eb21eb40a19a8 : SUCCESS in 1h 03m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4c14855aa736433b8fef855485f1823d : SUCCESS in 49m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/85d3d4cdaf2646a39341cf45602581e7 : SUCCESS in 50m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf10ede39f714a42a422d3337e6617a4 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/793764576ef54724a7c37e5c5570b131 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ee012ed73294e8db53121dc6b3837a6 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56979aa945b24ed9ac3c2e33bbe4d981 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba16351563f4b9ba6d008ac60522393 : SUCCESS in 8m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/75dacc3f26b84258b09d42b3f274f1a3 : SUCCESS in 2m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a45aacf73035490788ec3027668f7867 : SUCCESS in 5m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/702d83f8e58d4d788f30ed00e894ea76 : SUCCESS in 31m 34s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/221db2d6769845f79513df09ed7cc9a7 : SUCCESS in 7m 03s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/63e7b0c090384d36982f62cf4fd03178 : SUCCESS in 46m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/911dda0867d9455f8909e19c0946cfe8 : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"40c1c07668707f918008bb9743f53ec3cee01f37","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-01-22 22:33:38.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"071864679b9a11e717354ee9ad73b4d2abccbf52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 22:34:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"eef46779a2c0f43f8a0aa1b21884229902281e7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 00:09:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1075b89530064751be358d3cf4a9b42b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6a9ae2b0afba492e96f43c257cd465f8 : SUCCESS in 3m 58s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/25a420f3d5254796baf9424e7fc731c0 : SUCCESS in 2m 14s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cf1c2f4011f9473b854da65f71e18127 : SUCCESS in 49m 09s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/374543ea4e8347269da3eb3c71565824 : SUCCESS in 51m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/561d9f95501e4b7781f1bfde1f3c34df : SUCCESS in 52m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/92291f1a75dd4a379d4a2747688fbce0 : SUCCESS in 1h 11m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/876d990e053b4958a775bd85086113c0 : SUCCESS in 48m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/53a35f2f260e484bb2706faf345a66f3 : SUCCESS in 47m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0275698b27394b079fc5a87b5a4ad33f : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eda8dce0309f46ef981d9977758f3a27 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8dfcf6a628bc432fac231950834cf070 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da43ced35c654e1f97f6d09bd68c5f5a : SUCCESS in 7m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5b94d60b6e4c46ee9e9e70cc0e718b20 : SUCCESS in 2m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3be966ab76274014b68e0752f60ff7e5 : SUCCESS in 4m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/19dfe00ea1244a768f6bb469d91a07a1 : SUCCESS in 32m 11s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7d88985882fa4a4b8d77f3c0829dd343 : SUCCESS in 7m 18s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/34bff1b820cb48709cbef2f699d08e5b : SUCCESS in 49m 22s","accounts_in_message":[],"_revision_number":5},{"id":"3c74884e4e32c50f631a555ace82c2c06226a9d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 00:09:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"1d429ae8b9970993c7c6c7d53f61e99758101a6f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 00:10:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4360e745675f49ffb50d70ff60a446dd\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/5e4d9fe03c91492ca857685da98815c0 : SUCCESS in 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a86025da0494296bbdf2f6a57da76b5 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eaab7f8f9437469a9a2959a7c11bb99e : SUCCESS in 57s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"cb47b4e00e1b4bc1a560b95b9ac7d1514942d879","revisions":{"15f7492f105730bee2c9374b9e8eaedebfec8f59":{"kind":"REWORK","_number":1,"created":"2024-01-22 07:58:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/906248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/906248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/906248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/906248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/906248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/906248/1"}}},"commit":{"parents":[{"commit":"002005a440e87198e05719d17c3d39a980a552e3","subject":"Cleanup Heartbeat Emitter abstraction code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/002005a440e87198e05719d17c3d39a980a552e3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 07:56:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 07:58:07.000000000","tz":540},"subject":"Make edgegrid-python optional","message":"Make edgegrid-python optional\n\nThe edgegrid-python library is not necessary unless akamai backend is\nused, so this library should be an optional dependency to avoid pulling\nunnecessary packages.\n\nNote that akamai backend hasn\u0027t been updated since it was added 4 years\nback. Probably we can deprecate and remove it assuming it\u0027s not really\nused(or at least it\u0027s not maintained).\n\nChange-Id: If13e722e5ec3cdcd11461201ea20b2dc022cce10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15f7492f105730bee2c9374b9e8eaedebfec8f59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15f7492f105730bee2c9374b9e8eaedebfec8f59"}]},"branch":"refs/heads/master"},"50f4b3afb74e509f4da90749bf2065de796849f6":{"kind":"REWORK","_number":2,"created":"2024-01-22 08:01:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/906248/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/906248/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/906248/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/906248/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/906248/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/906248/2"}}},"commit":{"parents":[{"commit":"002005a440e87198e05719d17c3d39a980a552e3","subject":"Cleanup Heartbeat Emitter abstraction code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/002005a440e87198e05719d17c3d39a980a552e3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 07:56:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 08:01:23.000000000","tz":540},"subject":"Make edgegrid-python optional","message":"Make edgegrid-python optional\n\nThe edgegrid-python library is not necessary unless akamai backend is\nused, so this library should be an optional dependency to avoid pulling\nunnecessary packages.\n\nNote that akamai backend hasn\u0027t been updated since it was added 4 years\nback. Probably we can deprecate and remove it assuming it\u0027s not really\nused(or at least it\u0027s not maintained).\n\nChange-Id: If13e722e5ec3cdcd11461201ea20b2dc022cce10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50f4b3afb74e509f4da90749bf2065de796849f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50f4b3afb74e509f4da90749bf2065de796849f6"}]},"branch":"refs/heads/master"},"72d04bf5d84fd5d912e37dd26b7bc15db58875b4":{"kind":"REWORK","_number":3,"created":"2024-01-22 10:29:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/906248/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/906248/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/906248/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/906248/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/906248/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/906248/3"}}},"commit":{"parents":[{"commit":"002005a440e87198e05719d17c3d39a980a552e3","subject":"Cleanup Heartbeat Emitter abstraction code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/002005a440e87198e05719d17c3d39a980a552e3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 07:56:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 10:29:12.000000000","tz":540},"subject":"Make edgegrid-python optional","message":"Make edgegrid-python optional\n\nThe edgegrid-python library is not necessary unless akamai backend is\nused, so this library should be an optional dependency to avoid pulling\nunnecessary packages.\n\nNote that akamai backend hasn\u0027t been updated since it was added 4 years\nback. Probably we can deprecate and remove it assuming it\u0027s not really\nused(or at least it\u0027s not maintained).\n\nChange-Id: If13e722e5ec3cdcd11461201ea20b2dc022cce10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72d04bf5d84fd5d912e37dd26b7bc15db58875b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72d04bf5d84fd5d912e37dd26b7bc15db58875b4"}]},"branch":"refs/heads/master"},"b4fe4f66a5d83859cdc1da4200386b907273ce0a":{"kind":"REWORK","_number":4,"created":"2024-01-22 10:46:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/906248/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/906248/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/906248/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/906248/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/906248/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/906248/4"}}},"commit":{"parents":[{"commit":"002005a440e87198e05719d17c3d39a980a552e3","subject":"Cleanup Heartbeat Emitter abstraction code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/002005a440e87198e05719d17c3d39a980a552e3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 07:56:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 10:46:20.000000000","tz":540},"subject":"Make edgegrid-python optional","message":"Make edgegrid-python optional\n\nThe edgegrid-python library is not necessary unless akamai backend is\nused, so this library should be an optional dependency to avoid pulling\nunnecessary packages.\n\nNote that akamai backend hasn\u0027t been updated since it was added 4 years\nback. Probably we can deprecate and remove it assuming it\u0027s not really\nused(or at least it\u0027s not maintained).\n\nChange-Id: If13e722e5ec3cdcd11461201ea20b2dc022cce10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4fe4f66a5d83859cdc1da4200386b907273ce0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4fe4f66a5d83859cdc1da4200386b907273ce0a"}]},"branch":"refs/heads/master"},"cb47b4e00e1b4bc1a560b95b9ac7d1514942d879":{"kind":"REWORK","_number":5,"created":"2024-01-22 12:47:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/906248/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/906248/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/906248/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/906248/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/906248/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/906248/5"}}},"commit":{"parents":[{"commit":"002005a440e87198e05719d17c3d39a980a552e3","subject":"Cleanup Heartbeat Emitter abstraction code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/002005a440e87198e05719d17c3d39a980a552e3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 07:56:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 12:47:31.000000000","tz":540},"subject":"Make edgegrid-python optional","message":"Make edgegrid-python optional\n\nThe edgegrid-python library is not necessary unless akamai backend is\nused, so this library should be an optional dependency to avoid pulling\nunnecessary packages.\n\nNote that akamai backend hasn\u0027t been updated since it was added 4 years\nback. Probably we can deprecate and remove it assuming it\u0027s not really\nused(or at least it\u0027s not maintained).\n\nChange-Id: If13e722e5ec3cdcd11461201ea20b2dc022cce10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cb47b4e00e1b4bc1a560b95b9ac7d1514942d879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cb47b4e00e1b4bc1a560b95b9ac7d1514942d879"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
