)]}'
{"id":"openstack%2Fdesignate~906619","triplet_id":"openstack%2Fdesignate~master~I354bc334a355faf273b6903fa77ca9052ab4e9b2","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-25 05:35:07.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I354bc334a355faf273b6903fa77ca9052ab4e9b2","subject":"Bump hacking","status":"ABANDONED","created":"2024-01-25 03:37:16.000000000","updated":"2024-01-25 05:35:16.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0c7b135251e42d50cdbf0241a365bfade6ca6ccb","_number":906619,"virtual_id_number":906619,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-25 05:26:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-25 05:26:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57004a135f9cdfe91e603ae08f749a3daff78dcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-25 03:37:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4183e487388ff469079fc0a6511966ab3773e959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 05:26:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/904a48c4be954c2a8eb82945ea536f67\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3e9bb7d2baa6496f871663c03fb972bc : SUCCESS in 5m 18s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/03d67fb5de4c4ac2a2d5db111d233ef9 : SUCCESS in 42m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ae332947ec534372a53601eced469f89 : SUCCESS in 51m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f22ea48278fd4bba97dd6b08c603f407 : SUCCESS in 34m 50s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/255af94c9ba04a7cb42eccdd23e774fc : SUCCESS in 28m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/978aac6bda4846fcbacc72f644a7bba3 : SUCCESS in 46m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e9c3ad6c26874cc0a1a6159a43a7eb24 : SUCCESS in 1h 09m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bb38217b2ed549b099154802b170fece : SUCCESS in 46m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fb1a1478e4274c3e8bb045f7ca6e1ed3 : SUCCESS in 30m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ac2382c54f1e440a9b4511c7ff7f8d69 : SUCCESS in 51m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d20a9475e5f47988f2f7cb110ce5f44 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04771f02ea8544c3bc5cf527a48621ce : FAILURE in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db3531dbe2aa469299976af3e0270799 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c1d0a0fc3104fc7bcdd2fd78211e916 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8551d7d4fb804020805c9ce47ebde613 : SUCCESS in 8m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/42f80cc2dc0a4b16bc32d66f2510a28c : SUCCESS in 2m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dc6c4866253428f8a6d57d85887b75a : SUCCESS in 4m 53s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/86dc547aa1b04df4ade935b0ee2bdcfb : SUCCESS in 7m 13s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e1fc4c317c314f8bba37a778bcfc165a : SUCCESS in 47m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/db0aa0f482fc498a8e864f8e58a16858 : SUCCESS in 6m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da0f73759d54a16ce377ca61efb3b3eb693d27fe","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-25 05:35:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63f9b36290b30fe7192648c02c2988b415118ab4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-25 05:35:09.000000000","message":"Topic bump-hacking removed","accounts_in_message":[],"_revision_number":1},{"id":"0c7b135251e42d50cdbf0241a365bfade6ca6ccb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-25 05:35:16.000000000","message":"Abandoned\n\nin favor of https://review.opendev.org/c/openstack/designate/+/905817","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7eb210bf4c784ad5d9de3b305113698506000bad","revisions":{"7eb210bf4c784ad5d9de3b305113698506000bad":{"kind":"REWORK","_number":1,"created":"2024-01-25 03:37:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/906619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/19/906619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/906619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/906619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/906619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/19/906619/1"}}},"commit":{"parents":[{"commit":"002005a440e87198e05719d17c3d39a980a552e3","subject":"Cleanup Heartbeat Emitter abstraction code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/002005a440e87198e05719d17c3d39a980a552e3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-25 03:36:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-25 03:36:42.000000000","tz":540},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 4.1.x is too old.\n\nAlso remove the note about behavior of old versions of pip.\n\nChange-Id: I354bc334a355faf273b6903fa77ca9052ab4e9b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7eb210bf4c784ad5d9de3b305113698506000bad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7eb210bf4c784ad5d9de3b305113698506000bad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
