)]}'
{"id":"openstack%2Fdesignate~907741","triplet_id":"openstack%2Fdesignate~master~Id1b938ddc754ed07e1ba9c9cbd08c94192d5ad1a","project":"openstack/designate","branch":"master","topic":"slurp-job","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-10-23 09:12:45.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-27 14:30:22.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Id1b938ddc754ed07e1ba9c9cbd08c94192d5ad1a","subject":"Add skip level upgrade job","status":"ABANDONED","created":"2024-02-05 14:01:49.000000000","updated":"2024-10-23 09:12:45.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"5d3a9a5f079cfafc91a42ce6b89ddc530c705262","_number":907741,"virtual_id_number":907741,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-05 15:05:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-02-05 17:32:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2024-05-27 14:33:29.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-05 15:05:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-05 17:32:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-05-27 13:53:57.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"}],"messages":[{"id":"5c9c6141390d94458b9de1d838c5545fb34d748b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-05 14:01:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08988c8726fa03a5d20bdb2658db7dba63e0d89f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-05 14:05:21.000000000","message":"Topic set to slurp-job","accounts_in_message":[],"_revision_number":1},{"id":"61d383dfe1b81b757d50f6a9cdec4ca4698e7e4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 15:05:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c5d7b11cbc9461094b321311b3d07cc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f2e90337705e427fbf274d05a846c359 : SUCCESS in 44m 31s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/83056c9854ff4f668f70e9f72b51d529 : SUCCESS in 46m 00s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/34a87624eb3a4877a9803cc13f40770c : SUCCESS in 42m 36s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a5679e113aca4f3e874b3313aad334bf : SUCCESS in 36m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/34b77390ff1743069fa7285fa76ccc5e : SUCCESS in 39m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7e3851fbec864545bdc1fc99972b6090 : SUCCESS in 55m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ba2801df9a5d4359826e063b7407e91c : SUCCESS in 58m 42s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/fbcab425b4254fc2b75e69348edf1d3d : SUCCESS in 1h 01m 23s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a0af11f5fe7c44debe99132ff2b3897d : SUCCESS in 55m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e585014bbbb841e5b3a45e018a947797 : SUCCESS in 52m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d3c074d1a1434ba4be8489c2f7611df5 : SUCCESS in 41m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dabb97f0c7243ad947bfd39e8713897 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b3a84c3f6db476ba0b73b2eccfdfb2d : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a23ab26226fe4e14ab436272e2cbcdba : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac07774196b942c7968d130737379ab6 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac172c5820d6467695af07d2bb6b5d17 : SUCCESS in 7m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e5e55242f78647d69f935c501bf8b7f1 : SUCCESS in 31m 08s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ef9974891e7c41d59d020f8093d4c6ed : SUCCESS in 8m 10s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/30aac9bc73b74ee08ec16546932946d3 : SUCCESS in 49m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3fc0aaa9f0924b6691c08e0b6cc1678c : SUCCESS in 6m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0106d898acf99434a34f551718495436e8f5c75a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-05 17:32:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5f7312aa7c87149a58f23ef34c539db3c4bff1a","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-27 13:53:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68173e7dedd5e4f924353f92d05a82b577134ba6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-27 14:30:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7716120fd53254a845d496e19bd88b6bd11da41d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-27 14:33:29.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d3a9a5f079cfafc91a42ce6b89ddc530c705262","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-23 09:12:45.000000000","message":"Abandoned\n\nThis will be taken care by https://review.opendev.org/c/openstack/designate/+/932917","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd314dd719ffdfb4a874f73f86d6667c1af3e1cc","revisions":{"fd314dd719ffdfb4a874f73f86d6667c1af3e1cc":{"kind":"REWORK","_number":1,"created":"2024-02-05 14:01:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/907741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/41/907741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/41/907741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/41/907741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/41/907741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/41/907741/1"}}},"commit":{"parents":[{"commit":"002005a440e87198e05719d17c3d39a980a552e3","subject":"Cleanup Heartbeat Emitter abstraction code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/002005a440e87198e05719d17c3d39a980a552e3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-05 14:00:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-05 14:01:33.000000000","tz":540},"subject":"Add skip level upgrade job","message":"Add skip level upgrade job\n\nThe 2024.1 release will be the first SLURP release which supports\ndirect upgrade from the 2023.1 release.\n\nThis also makes the grenade job with bind9 voting, because the job\nhas been stable and bind9 is as popular option as pdns4.\n\nChange-Id: Id1b938ddc754ed07e1ba9c9cbd08c94192d5ad1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd314dd719ffdfb4a874f73f86d6667c1af3e1cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd314dd719ffdfb4a874f73f86d6667c1af3e1cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
