)]}'
{"id":"openstack%2Fdesignate~910322","triplet_id":"openstack%2Fdesignate~master~I7232b55300c8817a00e73e449e03ea6103f40701","project":"openstack/designate","branch":"master","topic":"fix-db-utf8","attention_set":{},"removed_from_attention_set":{"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2026-05-25 05:31:07.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-11-20 10:46:08.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-01-06 03:38:37.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2026-05-25 05:31:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7232b55300c8817a00e73e449e03ea6103f40701","subject":"Fix Mysql default character setting (utf8mb4)","status":"ABANDONED","created":"2024-02-27 07:50:58.000000000","updated":"2026-05-25 05:31:07.000000000","total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"2a53eab98d5abd1fd6ae98f8654c544c60c75ab8","_number":910322,"virtual_id_number":910322,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-02 08:09:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":-1,"date":"2025-01-06 03:38:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"CC":[{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-27 08:05:19.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2024-02-27 08:06:02.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-02-27 08:06:02.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-02-27 08:06:02.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2024-02-27 08:58:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8aa2e1034169725b22e06b078f8ace29098e8c4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-02-27 07:50:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f88ce1335b39fbd7899924c292ecda9304136a5","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-02-27 08:46:28.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e428af8171244a059d0caa597f6052924d961be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 08:58:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa6937b74ec4463da151fcbe207d9801\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/80571b9251da471ba812540f2583c901 : SUCCESS in 52m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1d4beadcbf1b4faea7ced2685297124c : SUCCESS in 1h 00m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3fac577192d246d3a7c079cf834e90f5 : SUCCESS in 55m 13s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/34f783b4adc64a9f83a9b29acded3899 : SUCCESS in 57m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/27cdd47b496b47419b6a9f7899f266ef : SUCCESS in 36m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/699cea11b2844207b10c69726b8d0f59 : SUCCESS in 59m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4e5146dc3c7643349905e6add9fc910e : SUCCESS in 1h 03m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/64136cb2f0ac4b91a286a340765acef7 : SUCCESS in 36m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4adb0307ff6b4055b423d71e95d3f0b8 : SUCCESS in 39m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9ef6d4c6fdc4980b531b95386826ed8 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3d192fe583640519ae931a7ba01bb3a : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9dd7a3bd27e4c628761a2439f06f1c1 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ad5f973c1074fc29bf14e7dd256017f : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b68f4eef53f64f8e89d8f5d7efffbd25 : SUCCESS in 7m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/919a55107ad349fb8cb384ec879da0a3 : SUCCESS in 36m 24s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/874b01842d834d9badda98c00ce38d74 : SUCCESS in 6m 06s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a683c01b499f4532b7483a3f5daaa856 : SUCCESS in 49m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8c2315c613f6440cb55feef9c3416d93 : SUCCESS in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31ae926ee23f06a624528db5644f64b46299fbc6","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-02-29 09:50:05.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"058ba7a876d669036435367e2ba6c8daf036b5b9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-09 05:50:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1444d376e35157b8c627a63f6a3d748d0532ba40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-11-20 03:12:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3621bd8fb8a357fc48ebccda6941887722f7c953","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-11-20 03:15:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ec18b57d6decfe84f266222edd35d9fa9ceb4e29","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-20 03:20:13.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b896f6692c1020ae13b3403028ca2bb4ce77192e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 04:38:35.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd7a969a85a7467cb7c702ecebbf6000\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d17b5bdbf6984bd68112daef4ee78539 : SUCCESS in 53m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1301d352d8894f459765fcd32c35213a : SUCCESS in 1h 04m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/498c2f6620a543d191422e62cb6108f9 : SUCCESS in 49m 47s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cdc17e8655f34b9d822bac851cc0c848 : SUCCESS in 49m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5694aa8a3eb741a7860d667f50609a8a : SUCCESS in 36m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fb54e35a51a8481bad9aabe58fdb4aae : SUCCESS in 1h 01m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2057a2a38bd845a292131f3f5f40d380 : SUCCESS in 1h 05m 19s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d841cbdf7e384a2eaa8361a44bcaea1a : SUCCESS in 1h 06m 22s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/bb5fa19aa58443668261df1d191dfaf6 : SUCCESS in 1h 05m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/078463e0314347648a8f21f17866c680 : SUCCESS in 39m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/778adc424ff94116ae94e28b13070505 : SUCCESS in 44m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/442fd85a308146c39e3a81e3d4d40be6 : FAILURE in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9551eeb41714646b7c67ad3656d58b3 : FAILURE in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd8ddd09ef6f4090affb32d1d81126a2 : FAILURE in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4952199402a840e98e910d710e9a6a29 : FAILURE in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/024a67e0a3554e46bc5df79be85e8467 : SUCCESS in 7m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0f5c4436051e4c639cd1d7191ebdf085 : SUCCESS in 33m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3f89a359670146209b8973165784f939 : FAILURE in 4m 54s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/a85d6c52d24d4845b918cb419445ebf7 : FAILURE in 56m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9adb05e1502778d9094416e7f54aa00b9afa999d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-01-02 06:49:57.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"703cc09932691cd9bacb38fc8ecdf60c7e93667f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-01-02 06:50:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e0521691d88c929c332c6ed4826ca7db68b61ab4","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-01-02 06:51:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3396e94a6d19f0164527990b49fa9024df68aafd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 08:09:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f79b8943c3dc41668025c187be8206e6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b933d12a77454aea9f22b29cee67e80d : SUCCESS in 38m 44s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/704fab5fca834933a9dcf4f104ac173a : SUCCESS in 1h 03m 32s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/02743dbbb1954013bd48ae6cecdc8ffe : SUCCESS in 41m 07s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d5e381b586a6467d90e9d47ad14d2f4e : SUCCESS in 37m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/63e487ef5f4f43c49c01b22fc227812a : SUCCESS in 51m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/34a5ed9bf0c9435fb71647e9e585240d : SUCCESS in 52m 29s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a1c669e8b654451b90d63765c96fb5d4 : SUCCESS in 55m 04s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/ab989d58cbdf4b558df8b15bef882720 : SUCCESS in 1h 11m 12s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/5b47018e8a5a47c9a2495b1815f3a0f5 : SUCCESS in 1h 04m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a22faf60cf9b46a9a7227f882f744b1a : SUCCESS in 40m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0abb014a151c4b2e9a74e16a96094fc9 : SUCCESS in 49m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f087b9a19b24efea39050022c8cb484 : FAILURE in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a1d0a091a974415975304b0c8496d6c : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0262dc49aa6458582348ddc1295f308 : FAILURE in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ed42b475388484ab0d2161b9b3363cc : FAILURE in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cfa55599cf049ee871108286dab4774 : SUCCESS in 10m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/baaac4e7149d4bcc98e89f92bd257b02 : SUCCESS in 24m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/228c7d1d5653455fa178b2b18b0f02d3 : FAILURE in 8m 01s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ab9343fbf36d489f9f40774d769c5da8 : SUCCESS in 47m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6221e74d7c7435a3bd816b63ee53008cf22fb970","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-06 03:38:37.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2a53eab98d5abd1fd6ae98f8654c544c60c75ab8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2026-05-25 05:31:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"21f750a501ca6908b6133d41ae60848d002f6e4d","revisions":{"b807373aacf08470527340024b414afe5819d212":{"kind":"REWORK","_number":1,"created":"2024-02-27 07:50:58.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/22/910322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/22/910322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/910322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/910322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/910322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/22/910322/1"}}},"commit":{"parents":[{"commit":"91f7d7d3fb3c39c6897f1508ce390c2dcbbd8759","subject":"Merge \"Add support for catalog zones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/91f7d7d3fb3c39c6897f1508ce390c2dcbbd8759"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-02-27 07:42:31.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-02-27 07:50:55.000000000","tz":480},"subject":"Fix Mysql default character setting (utf8mb4)","message":"Fix Mysql default character setting (utf8mb4)\n\nSince 8.0 Mysql uses utf8mb4 as default.\nTo use utf8, will change the new comming tables in DATABASE to use\n`utf8mb3`. And that makes new comming tables not compatible with rest\ntables within same database or tables from all other database that\nusing default setting.\nWe should make sure we uses utf8mb4 so no confusion or incompatible\nerror will raise.\n\nCloses-bug: #2033635\nChange-Id: I7232b55300c8817a00e73e449e03ea6103f40701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b807373aacf08470527340024b414afe5819d212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b807373aacf08470527340024b414afe5819d212"}]},"branch":"refs/heads/master"},"5d923419fc4cb2084d7d42a0ac4e222211b910f6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-20 03:12:41.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/22/910322/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/22/910322/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/910322/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/910322/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/910322/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/22/910322/2"}}},"commit":{"parents":[{"commit":"76a3b198cefd54404ce72232234355dde52b5d63","subject":"Merge \"Drop workaround for eventlet \u003c 0.36.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/76a3b198cefd54404ce72232234355dde52b5d63"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-02-27 07:42:31.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-11-20 03:12:17.000000000","tz":480},"subject":"Fix Mysql default character setting (utf8mb4)","message":"Fix Mysql default character setting (utf8mb4)\n\nSince 8.0 Mysql uses utf8mb4 as default.\nTo use utf8, will change the new comming tables in DATABASE to use\n`utf8mb3`. And that makes new comming tables not compatible with rest\ntables within same database or tables from all other database that\nusing default setting.\nWe should make sure we uses utf8mb4 so no confusion or incompatible\nerror will raise.\n\nCloses-bug: #2033635\nChange-Id: I7232b55300c8817a00e73e449e03ea6103f40701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5d923419fc4cb2084d7d42a0ac4e222211b910f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5d923419fc4cb2084d7d42a0ac4e222211b910f6"}]},"branch":"refs/heads/master"},"43bcf7e3f337267bc9677f3ddce2584429ee21e9":{"kind":"REWORK","_number":3,"created":"2024-11-20 03:15:46.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/22/910322/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/22/910322/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/910322/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/910322/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/910322/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/22/910322/3"}}},"commit":{"parents":[{"commit":"76a3b198cefd54404ce72232234355dde52b5d63","subject":"Merge \"Drop workaround for eventlet \u003c 0.36.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/76a3b198cefd54404ce72232234355dde52b5d63"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-11-20 03:09:56.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-11-20 03:15:36.000000000","tz":480},"subject":"Fix Mysql default character setting (utf8mb4)","message":"Fix Mysql default character setting (utf8mb4)\n\nSince 8.0 Mysql uses utf8mb4 as default.\nTo use utf8, will change the new comming tables in DATABASE to use\n`utf8mb3`. And that makes new comming tables not compatible with rest\ntables within same database or tables from all other database that\nusing default setting.\nWe should make sure we uses utf8mb4 so no confusion or incompatible\nerror will raise.\n\nCloses-bug: #2033635\nChange-Id: I7232b55300c8817a00e73e449e03ea6103f40701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/43bcf7e3f337267bc9677f3ddce2584429ee21e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/43bcf7e3f337267bc9677f3ddce2584429ee21e9"}]},"branch":"refs/heads/master"},"33e55088023d5eb2f8d585526c440b9ce2d167b3":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2025-01-02 06:49:57.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/22/910322/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/22/910322/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/910322/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/910322/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/910322/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/22/910322/4"}}},"commit":{"parents":[{"commit":"97588b86de5a431296712ac644f7704dc51a6643","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/97588b86de5a431296712ac644f7704dc51a6643"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-11-20 03:09:56.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-01-02 06:49:35.000000000","tz":480},"subject":"Fix Mysql default character setting (utf8mb4)","message":"Fix Mysql default character setting (utf8mb4)\n\nSince 8.0 Mysql uses utf8mb4 as default.\nTo use utf8, will change the new comming tables in DATABASE to use\n`utf8mb3`. And that makes new comming tables not compatible with rest\ntables within same database or tables from all other database that\nusing default setting.\nWe should make sure we uses utf8mb4 so no confusion or incompatible\nerror will raise.\n\nChange-Id: I7232b55300c8817a00e73e449e03ea6103f40701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33e55088023d5eb2f8d585526c440b9ce2d167b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33e55088023d5eb2f8d585526c440b9ce2d167b3"}]},"branch":"refs/heads/master"},"21f750a501ca6908b6133d41ae60848d002f6e4d":{"kind":"REWORK","_number":5,"created":"2025-01-02 06:50:49.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/22/910322/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/22/910322/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/910322/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/910322/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/910322/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/22/910322/5"}}},"commit":{"parents":[{"commit":"97588b86de5a431296712ac644f7704dc51a6643","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/97588b86de5a431296712ac644f7704dc51a6643"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-11-20 03:09:56.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-01-02 06:50:41.000000000","tz":480},"subject":"Fix Mysql default character setting (utf8mb4)","message":"Fix Mysql default character setting (utf8mb4)\n\nSince 8.0 Mysql uses utf8mb4 as default.\nTo use utf8, will change the new comming tables in DATABASE to use\n`utf8mb3`. And that makes new comming tables not compatible with rest\ntables within same database or tables from all other database that\nusing default setting.\nWe should make sure we uses utf8mb4 so no confusion or incompatible\nerror will raise.\n\nChange-Id: I7232b55300c8817a00e73e449e03ea6103f40701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21f750a501ca6908b6133d41ae60848d002f6e4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21f750a501ca6908b6133d41ae60848d002f6e4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
