)]}'
{"id":"openstack%2Fdesignate~915857","triplet_id":"openstack%2Fdesignate~master~Ib7c1332c998ff275452894c9aa27bff4a20d531a","project":"openstack/designate","branch":"master","topic":"stop","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-09-22 16:13:15.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2026-09-22 18:51:23.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2026-09-22 18:51:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib7c1332c998ff275452894c9aa27bff4a20d531a","subject":"Safe stop notinitialized coordination driver","status":"MERGED","created":"2024-04-15 12:51:13.000000000","updated":"2026-09-22 18:53:09.000000000","submitted":"2026-09-22 18:51:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"915857-stop","meta_rev_id":"ed9c9346a06d37844672ac8b2f3e3fb42eb318c6","_number":915857,"virtual_id_number":915857,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-22 18:51:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-09-22 17:31:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-09-22 17:31:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-15 14:11:16.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2024-04-15 14:13:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-07 16:19:42.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2024-05-07 16:19:42.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2024-05-07 16:19:42.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-05-07 16:19:42.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-05-08 14:54:53.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"7c7b072a9367eb0cf636634a03061a28249a6de1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-04-15 12:51:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73db83fa735e391b1e05f4b4ce917901ce7c8d88","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-04-15 14:11:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c687c6998269d0d032850de22bcbd26f9891a7ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 14:13:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e2b665dd7ab42498e0a3430e377f883\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a6a20f926f544536aae91347d3e55480 : SUCCESS in 39m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/62973fa3199749f7a447c2908ce26037 : SUCCESS in 45m 57s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d06d14c05b72405aafb8360a99de638c : SUCCESS in 50m 15s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/05a60b74d4da43b4a8d2b650d9c2638a : SUCCESS in 58m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ca521337bf8a4c8b89976d2cce0412a1 : SUCCESS in 42m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a515bf189a704787aab8a5232b20f0e4 : SUCCESS in 1h 15m 25s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6f6072a34eab433da012d1bcef66812b : SUCCESS in 1h 03m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/aad75452a57549a8b4a27e00877d38d3 : FAILURE in 52m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b3ea3458bcef40109fe495030ba75178 : SUCCESS in 48m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f307ea09d424bd3b899eb449ad621b7 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5f08f160dcd4d56b91cb188945f7cb4 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d6f7cf9b70b42f38becd2e175859e7c : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37c3f0012fce42d7a855c9ead56dff5b : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/448446909afb4857984b9787a755ef30 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3b89811f78940329a60479c81684326 : SUCCESS in 7m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1bd6f0ec514b4f79b97fa854ec0de693 : SUCCESS in 46m 34s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2a14d5530481442c8b59b60b33115b4b : SUCCESS in 7m 01s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f1d95f0b23ec4ca7a08ab3515b2b6292 : SUCCESS in 36m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d31d6971cbc34b6782f35a4eafb32196 : SUCCESS in 4m 44s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/d78df4a8b0524603a892847ccb339141 : SUCCESS in 40m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0486bd7310b6309d828400a4d4faa4a8d35dbae","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-04-15 14:15:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b11c87bb7cd082b789471652984aa41223aca75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 15:29:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c820750f79ac471f85873ea7fdcfa670\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e6cc2e476afc4eb19af89dd204e30725 : SUCCESS in 43m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/812d3e7fd43f40eb9aee01e13d453e88 : SUCCESS in 49m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4de56e28723413ea2087692bbdcbdb1 : SUCCESS in 52m 02s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2251e701484c42419af754b030767d7f : SUCCESS in 47m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dc13073c29bb4c2187ee2f057d9a61aa : SUCCESS in 51m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/41feb103afc2454d968faee756c1dcca : SUCCESS in 1h 08m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2a04620d54b54faf81c8eb93c88c4ec4 : SUCCESS in 1h 03m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/32a8042d84a44ababb5b637dfb32ae97 : FAILURE in 33m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1711229443864a42b92a7f768bc8463f : SUCCESS in 42m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6ed009c3feb49418d3d0ed0ac706e8c : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee35915f77d4456bb28160d7e1462639 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a89e02cce8a4fb59dac4b4b3b6f341d : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/341b7a7a373a4bbd952bf5f13d045e9e : SUCCESS in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1cd9a6b5b2f844e9ac49ae0c45a81baa : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea1c82cd49344e2c86868120e8f414aa : SUCCESS in 7m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b0efecbae69e497a8c5466ceb2887c9c : SUCCESS in 49m 03s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/01688cc52ee944bfac121db1c539a19c : SUCCESS in 6m 26s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7996c94e8256400ab4e9667770e3afe2 : SUCCESS in 40m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8590c48dc3054f5db91d1a236f12fbf5 : SUCCESS in 4m 32s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/11b6ff700838425cb7ce7ab06e4fc99a : SUCCESS in 42m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b73b77188859d68ae8168370a3267b994d282d90","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-04-15 18:58:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f575e5c7952541767ac6e61d2c6fad48e07a80e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 20:15:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/052e0cea63964ba1ae8c36cbd1362361\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b26bc718bf3540ccb3b7711d7fb84e08 : SUCCESS in 52m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/773175c1437a4c8fb436edce3cb792f9 : SUCCESS in 56m 59s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7acfed866b974ead9a65a590d1cad2d1 : SUCCESS in 40m 47s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7dcc5b0d1892486a96222ca827861174 : SUCCESS in 38m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c37096e0b2a1445489306279cd14e49a : SUCCESS in 35m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9482000de7ab480eba20a0078cb74479 : SUCCESS in 52m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2f8ec2c56ef64eaa81ff5314c6461232 : SUCCESS in 1h 10m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/292904c0b57e46689155d610261eac4f : SUCCESS in 33m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9a6e2ced38d54eec91bd460e82fb127b : SUCCESS in 33m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3df48c5f2c4d49deb84dee76fe2c065f : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8395b36971c741a484dbb63c1af1c842 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af82f80f45114848baab1f5c2aeaff9e : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c408cb185754aed9acd9ba46959ea5c : SUCCESS in 6m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d76350d6547140cdb8a2f3935e68beda : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2293f06055f3422c9c5da66f906e5780 : SUCCESS in 6m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2a5badca835d4671ab1af33e8a52d66d : SUCCESS in 31m 04s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/01afa603ef0f42dfad0fd0ce8de97c35 : SUCCESS in 6m 05s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/66c16b6b6dd4433e9dfc57b0d64a3d62 : SUCCESS in 45m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b67446d214554385876820929b65593a : SUCCESS in 5m 08s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/9fe974a668df40ef8a8e2249d69d2980 : SUCCESS in 42m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7931ccb68ce149fd0838d00d8519aaf9cff1ea92","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-08 14:54:53.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7aa2bb3adcdc32ca55daf4976be6f6004d3f90ae","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-05-22 14:51:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5659b4a9c4441eea420fa584ef6b6eb57ad73a05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-06-05 04:49:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2f17c1be0f83cfb3ea5e0605c311adf7b2eca24a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 06:03:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc0535d664324f0c89d1d5dc414c9334\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b76d6687672f4a84bbe83914d0135f71 : SUCCESS in 39m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f0bcf206aaae44c8ad0809a790606cc1 : SUCCESS in 50m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c4293efb426494499ee08eafd5bea77 : SUCCESS in 42m 53s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6b4050e6b0f849bb817bad14333e11bf : SUCCESS in 40m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9d8e1474e17f467986fcdd0173187b5b : SUCCESS in 38m 37s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/da05813df25149f58e3ffda88a66b41c : SUCCESS in 56m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/598baa949b0a4af6936c2d6d3abf0710 : SUCCESS in 1h 07m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/809e27b2a3384796ae311810bfc30897 : SUCCESS in 52m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8b18d7e02f264111bb59cfc9827c58ea : SUCCESS in 55m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d44b3d516b7149a3ad7cf428425c2947 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9eb11beb8874e67829cf809bd4f4306 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20b57809e7934892b8ddac3dd35c9c7c : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8846f7b25ca24c0ab3309010ba6e4d56 : SUCCESS in 7m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4cb1f15644914beeb41a7816bb3415b1 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc6f459719554e02a73d24f3e31459c6 : SUCCESS in 7m 08s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b7aafca86f2e46fcb87fcd4e711d05ef : SUCCESS in 6m 07s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1904fb2ba30640d5bdc967a50f64b81d : SUCCESS in 53m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/609ccdb2e5db40988b8742e347f3ca99 : SUCCESS in 6m 40s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/934fa38d449c4ebe832afb9fc8c994b2 : SUCCESS in 43m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4a77fec1ee2c80bce117e202ba8eef3440356e4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-10-30 11:35:02.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"36d868c26eba14c2be1b0132af36427549f906a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 12:58:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c97201df04b4af088db26d24f8408e6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8cf289213c3b443288dab3fc051e603d : SUCCESS in 45m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/51801d9b804942ebb8ae42a62b246152 : SUCCESS in 1h 04m 41s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6fa2400b987479bb8c668c23e55f034 : SUCCESS in 56m 52s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5996e503c2ad4cac967f8525cf8b9be9 : SUCCESS in 48m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c1566972697b434091e970c4d9757f83 : SUCCESS in 46m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/422551f1aac8421290dc48c3c785bfec : SUCCESS in 1h 00m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/61dad156933f4458aa0edd62f321f692 : SUCCESS in 1h 14m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4d673b262fa3462b89f814d5b14a046c : SUCCESS in 49m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1b29058eca4f4500b1259a2e2e31b968 : SUCCESS in 46m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e936dd39632d41b4b67e389b0f552d61 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c906d9117fe040478656e5a51c523c9a : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58007bf916004e47aa5ec9c935a18c5f : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/537b1936a63e4dab96a9147f7befc971 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95075e0a47bf4e69a9d5677c6f8a53f4 : SUCCESS in 6m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b5b23d86225c45638a5182cc26cdcaef : SUCCESS in 28m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3eb02e96917a4a33a617f8da221dee7a : SUCCESS in 4m 27s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/20310c4075d34a64b271ac50a3a6284e : SUCCESS in 35m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"74b3f0e218c6c4790609abffe36e4f39f52a9746","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-22 13:29:44.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7f0c33c980936cc019f6c021930f6e24a6e95879","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-22 14:14:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"08ad31cb7d7f991d94e6c28cc4e46881ef908178","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-22 14:15:39.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"18f810324f5691bb0e4ee55dde71c061f7014be6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 15:35:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fec38c480ad045f8828a24636fbe6254\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/54bdb3fca4154f5986b874d722e2bcb4 : SUCCESS in 41m 59s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/24b92013d3ef400aa40778731833b89f : SUCCESS in 39m 01s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/e797cfab667d408788ece8aa828a897f : FAILURE in 45m 44s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/26b32e53f86e41f98592772d1e3f963b : SUCCESS in 15m 52s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/61cf1bd1e861436b9e8f5816db062d10 : SUCCESS in 1h 13m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/66ebc51834894509a15cff6ce629aa74 : SUCCESS in 1h 07m 18s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/5c0d5cb451564159a35374bcf2426f1c : SUCCESS in 1h 09m 25s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a59e8a4f8f6b421ea74b091a5a31da2e : SUCCESS in 1h 08m 22s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/136c161085714fe0bbaee31898391f5b : SUCCESS in 35m 11s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/592e9e59cb054b5fbf043ef476650bbb : SUCCESS in 46m 29s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/fd87745f10984718b01ad9db4e209ae1 : SUCCESS in 44m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/560d270c96054bad8b911b5450b2d518 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a34ac9ed6ef3436480fdff2db6e3e7d3 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4045e0337b54f5d83a4d48905164e42 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b1cc820caab4489b36f847e326b1685 : SUCCESS in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bada1c13e5a84507a77507ca22c6994b : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9d96a8ca80f4286bcfa98e5a356d5ec : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8695f84e41c147469ca1a3feb6578b93 : SUCCESS in 4m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0166a30524e34ff89fae19be6b71b058 : SUCCESS in 32m 37s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/94b700ea608443cd9bb8f72ca89804f0 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":4},{"id":"d7be6fbd9c21ff8b5569bec254940c5a81f0d370","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-22 16:13:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"845cd7795986c4eb2d57a5863da6cf09567added","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 17:21:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4012115f87704b55b729529adc7983b1\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/0437489584204aa5b8c18c0191be3bb6 : SUCCESS in 45m 58s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/03678c2b875745eda3d3973f0c1990ab : SUCCESS in 49m 38s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/7ce30e50b2084a0b9503f4b9d29f2058 : SUCCESS in 37m 44s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/165f76c17a6942bc91a9c2751a0f4a66 : SUCCESS in 33m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3abfe07094c348489f09acd1714ea019 : SUCCESS in 1h 06m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/389e083271e74b7793ac34255d85ab2c : SUCCESS in 43m 57s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a8f5e25e442a4e68abf1bacce349b967 : SUCCESS in 1h 02m 59s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f369fa250716494c8f6ae7bff9793f84 : SUCCESS in 51m 18s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/a90cb9025f854a4eaf7f735a64ebf284 : SUCCESS in 44m 51s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/b3cc9d8d9507471995036ea7902fdc72 : SUCCESS in 31m 11s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/bb59a9298f314d2684178a2a500ef6df : SUCCESS in 47m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58f2677a76ec4b548f25a29be1d94260 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b54121c3c1746b2b5768136b303cb06 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18f249bc4b0a4a55a66655bf0751c758 : SUCCESS in 6m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4f5e5234f1e438a8897e7eb32918d5d : SUCCESS in 4m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e3d438c73b6b4f0e8bdb3d714cd5cd2c : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd161c3eb4cd41babc93badb8fe30336 : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0be4a1bf2cbd4af9abebe7b1f6151b35 : SUCCESS in 7m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5e1e46a0b9cf42e4bd429165ec4fe618 : SUCCESS in 28m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5fdf929e176b4d8ba4da9996b669f592 : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":4},{"id":"0da011c6bae610c2a4e8abf3ede14676d426ca2b","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-22 17:31:13.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0a940c849e65504df127c13e99f323aeb086350f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 17:32:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"814ff6c53778f3c70e74a93b629ac70e265d51a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:51:23.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c16e28bedcf24642865637260f72cc7a\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/678270a20a0c4ad2a00adbdcbb2fbbca : SUCCESS in 44m 36s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/513186d9771041d8848db10ca35d50af : SUCCESS in 47m 19s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/65b9beb0b57c4f2cbe5ba8a44b2f8f60 : SUCCESS in 34m 25s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/2b4a0942fca84d9daa026242aa781484 : SUCCESS in 16m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/aaf9b4d7f3bc4f9f87ae7884851f870e : SUCCESS in 1h 14m 30s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/eb10cd17354742dcb5e844f5aef4b9ee : SUCCESS in 55m 56s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/64a3f21fd0b3409db96024e2b895649b : SUCCESS in 1h 11m 41s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a32d46d2a66647a08f3875a5a3e8af74 : SUCCESS in 52m 26s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/b1a933666e064580ba5ffe0725db5b0e : SUCCESS in 45m 34s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/d80bc138326c40018c3f139890bf9059 : SUCCESS in 38m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/310280beaa9e42efa2bbb26268fe1f22 : SUCCESS in 2m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3af13786b683430192dd7e35e4ea2c38 : SUCCESS in 6m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de30cbf04d794ae5ac5b7648fb42f1af : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/162a8349a2a54489a1b682bd438319fb : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ea227552e96412eb345b25092da32e5 : SUCCESS in 3m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b84b43e7fafc46d790c0f05e5e57a7f4 : SUCCESS in 31m 16s","accounts_in_message":[],"_revision_number":4},{"id":"ff875161e64d966ec8ad453b24ca8f6b30672da5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:51:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ed9c9346a06d37844672ac8b2f3e3fb42eb318c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:53:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc77e43250d14d01b0e8de43f7e61bb3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffb0acff5f2c4baa8a66c306ab5f7356 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2705ee83949e2b883516398f88adb6f9efe22c13","revisions":{"f729f496c2443eb865885ef394f22d97c944dc9b":{"kind":"REWORK","_number":1,"created":"2024-04-15 12:51:13.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/57/915857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/915857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/915857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/915857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/915857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/915857/1"}}},"commit":{"parents":[{"commit":"808f0c9198a4aad558b22c3b3f839202a825f980","subject":"Cleaned up coordinator and locking byte handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/808f0c9198a4aad558b22c3b3f839202a825f980"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-04-15 12:50:18.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-04-15 12:51:11.000000000","tz":180},"subject":"Safe stop notinitialized coordination driver","message":"Safe stop notinitialized coordination driver\n\nIf tooz coordination is not initialized it will raise excetpion [0]\nCheck if driver is initialized before trying to stop it.\n\nCloses-Bug: #2061457\n\n[0] https://github.com/openstack/tooz/blob/0e2010d93055d94834cc5faf2733941e49fd6a1f/tooz/coordination.py#L423\n\nChange-Id: Ib7c1332c998ff275452894c9aa27bff4a20d531a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f729f496c2443eb865885ef394f22d97c944dc9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f729f496c2443eb865885ef394f22d97c944dc9b"}]},"branch":"refs/heads/master"},"3a090cfef4d15b861d217b9a76fa931318bd58db":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-05 04:49:18.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/57/915857/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/915857/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/915857/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/915857/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/915857/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/915857/2"}}},"commit":{"parents":[{"commit":"1053801c8d6ca0373cdacca5004e417a364b63b9","subject":"Merge \"Switch the CentOS 9 Stream job to use etcd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1053801c8d6ca0373cdacca5004e417a364b63b9"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-04-15 12:50:18.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-06-05 04:49:18.000000000","tz":0},"subject":"Safe stop notinitialized coordination driver","message":"Safe stop notinitialized coordination driver\n\nIf tooz coordination is not initialized it will raise excetpion [0]\nCheck if driver is initialized before trying to stop it.\n\nCloses-Bug: #2061457\n\n[0] https://github.com/openstack/tooz/blob/0e2010d93055d94834cc5faf2733941e49fd6a1f/tooz/coordination.py#L423\n\nChange-Id: Ib7c1332c998ff275452894c9aa27bff4a20d531a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3a090cfef4d15b861d217b9a76fa931318bd58db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3a090cfef4d15b861d217b9a76fa931318bd58db"}]},"branch":"refs/heads/master","description":"Rebase"},"b2311160ad0cf6b88f1213cf36ac5bb77be0e501":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-30 11:35:02.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/57/915857/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/915857/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/915857/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/915857/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/915857/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/915857/3"}}},"commit":{"parents":[{"commit":"643fa563647a5aa74ac1db84aab2d0bcffaebff0","subject":"Merge \"Revert \"Disable GLOBAL_VENV in grenade\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/643fa563647a5aa74ac1db84aab2d0bcffaebff0"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-04-15 12:50:18.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-10-30 11:35:02.000000000","tz":0},"subject":"Safe stop notinitialized coordination driver","message":"Safe stop notinitialized coordination driver\n\nIf tooz coordination is not initialized it will raise excetpion [0]\nCheck if driver is initialized before trying to stop it.\n\nCloses-Bug: #2061457\n\n[0] https://github.com/openstack/tooz/blob/0e2010d93055d94834cc5faf2733941e49fd6a1f/tooz/coordination.py#L423\n\nChange-Id: Ib7c1332c998ff275452894c9aa27bff4a20d531a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b2311160ad0cf6b88f1213cf36ac5bb77be0e501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b2311160ad0cf6b88f1213cf36ac5bb77be0e501"}]},"branch":"refs/heads/master","description":"Rebase"},"2705ee83949e2b883516398f88adb6f9efe22c13":{"kind":"REWORK","_number":4,"created":"2026-09-22 14:15:39.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/57/915857/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/915857/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/915857/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/915857/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/915857/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/915857/4"}}},"commit":{"parents":[{"commit":"643fa563647a5aa74ac1db84aab2d0bcffaebff0","subject":"Merge \"Revert \"Disable GLOBAL_VENV in grenade\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/643fa563647a5aa74ac1db84aab2d0bcffaebff0"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-04-15 12:50:18.000000000","tz":180},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-09-22 14:15:26.000000000","tz":120},"subject":"Safe stop notinitialized coordination driver","message":"Safe stop notinitialized coordination driver\n\nIf tooz coordination is not initialized it will raise excetpion [0]\nCheck if driver is initialized before trying to stop it.\n\nCloses-Bug: #2061457\n\n[0] https://github.com/openstack/tooz/blob/0e2010d93055d94834cc5faf2733941e49fd6a1f/tooz/coordination.py#L423\n\nChange-Id: Ib7c1332c998ff275452894c9aa27bff4a20d531a\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2705ee83949e2b883516398f88adb6f9efe22c13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2705ee83949e2b883516398f88adb6f9efe22c13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
