)]}'
{"id":"openstack%2Fdesignate~916134","triplet_id":"openstack%2Fdesignate~master~I2695fb7460d2e288d65bc6f6c08ecbfa312e5177","project":"openstack/designate","branch":"master","topic":"feature/fixed-ips","attention_set":{},"removed_from_attention_set":{"36834":{"account":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"last_update":"2024-04-18 08:18:36.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2695fb7460d2e288d65bc6f6c08ecbfa312e5177","subject":"Implement Reverse DNS support for Fixed IPs","status":"ABANDONED","created":"2024-04-17 18:56:46.000000000","updated":"2024-04-18 08:18:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ef7a8048dca34f632cc63ae3a79f5c9301ea7863","_number":916134,"virtual_id_number":916134,"owner":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-04-18 08:00:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-17 20:14:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c79396e83322e9ac6f7901e32685ea9d54194fc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-04-17 18:56:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51c7e8a2b67d93c219d5b76f4d6184c5fd10c16a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-04-17 19:00:35.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"827b2600657ea01964395215462c11527e8957f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-04-17 19:03:21.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7c440cb1b45f3c1f7b0a45fd90a8ecca167ea5e9","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-04-17 19:09:03.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"939b9b9bd63b1c3459d13c091ada690110436a55","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-04-17 19:09:16.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_36834\u003e\n","accounts_in_message":[{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"}],"_revision_number":3},{"id":"67e5278647ebf38999a4814ea90f75015e3df238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 20:14:35.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de62c7c3f1a249b2bcaddd4208de776d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/49b50c06eddc4ed79ba81955479f34f1 : FAILURE in 4m 53s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/18e7453cbec143a2b115df869d9e361c : SUCCESS in 42m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2a9dca49b2c9426db89a38a712844854 : SUCCESS in 51m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f99742ecc1484ef0a0ae0d63e678e2ce : SUCCESS in 51m 58s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/24bc0c7e120d477f847c9773d704cf34 : SUCCESS in 40m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/695065734fce4a9ba1ac45c5aee0e769 : SUCCESS in 52m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b08b1aae6ecd471a9ad3658c1fbd2dc1 : SUCCESS in 1h 10m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2dd0b505428d4aa486e11a2b7be05051 : SUCCESS in 1h 08m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ff7e774668034c8587c1c3be8a4a4779 : SUCCESS in 52m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ac9e530a9404454a85fda5461dfe0a79 : SUCCESS in 50m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b146c96deca04ceebe2ba1a3dbf162b8 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb184bb7233e47cd8a33ef98ff59a37b : FAILURE in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be10e8200df446f6a9383d70ea2b163b : SUCCESS in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b03740b6ca80490c83dc181ed57de2bf : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee60a64b549a4bcaad887d57d0276111 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ed5e99ee2342b390ce63d6f856c6a0 : SUCCESS in 7m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3a371fcc254f4da6b2ef85b8b68ccf86 : SUCCESS in 45m 47s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1506a8707b91441eb0b7a1d1ff4a387c : SUCCESS in 6m 27s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6bec058c2c8b4cca9aa6e5ee45c1a645 : SUCCESS in 33m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2a38661300114874806ae42a27db5a43 : SUCCESS in 5m 12s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/91f8ca5c697445aeb95aa28ce8ba8bc0 : SUCCESS in 43m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0fe4394a826edbc0d2f880490bb7c143e807c853","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-04-18 06:46:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"134af4f051c16499ae59a1f2f8db746b34508e1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-04-18 06:47:58.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"a7696599961835be2478207ca5d271fc92fb6083","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 08:00:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a8e23dd23494eeda5a7bf986b7027b3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/26ca973a493348b28c5126ce2c15cd6e : FAILURE in 4m 26s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7c8d2f783ff740e38d7925142e30bca1 : SUCCESS in 41m 57s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e249e7f3c84e461eb9ae0c81dfc80040 : SUCCESS in 44m 51s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2087f67f6634e0997e00e590c3b1fd1 : SUCCESS in 45m 15s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a74649a279f44ae885ec8c28c40f86d5 : SUCCESS in 40m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/21443b45aede4e7b8b286d8a13988704 : SUCCESS in 48m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c591827249a847e7aaabff5ea00b0601 : SUCCESS in 1h 07m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3e847b327e28459d801a1a206a7af501 : SUCCESS in 50m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/38fff58c633d4f509cdcc0dcc97fd20c : SUCCESS in 39m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f63bf5521724453497d8827e43601c48 : SUCCESS in 49m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48dec6916566446cb663de45f4c40c64 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b26f580415ad4a8baab0fc040bfe9c09 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/110ffaf79713401296d285ed88c67ac9 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d9e2a0ff9f8e48f393d7ccf07eeac19f : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8cc4404b0cdf41469154204019c1cd0b : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3751bd97d704f439b83dafcc6a98bea : SUCCESS in 8m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f1e3d49840a74e00a0bc28d9e9d4b659 : SUCCESS in 33m 44s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/04277547d2f14118a9773ea351e2092e : SUCCESS in 6m 30s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/41fc8b12a4314c72ae7ebb819a88ed23 : SUCCESS in 34m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d18414c90e614075be395c4f53c16673 : SUCCESS in 4m 22s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/3705a247718b4a74bc80d5d98d764053 : SUCCESS in 32m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ef7a8048dca34f632cc63ae3a79f5c9301ea7863","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-04-18 08:18:36.000000000","message":"Abandoned\n\ngit-review opened a new one at https://review.opendev.org/c/openstack/designate/+/916169?tab\u003dchange-view-tab-header-zuul-results-summary","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2b41f47126e646c75ab45f575779c3b9a3905cea","revisions":{"fb61a7fd6cf7803b3fc888ec19b0b862f57b73e4":{"kind":"REWORK","_number":1,"created":"2024-04-17 18:56:46.000000000","uploader":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"ref":"refs/changes/34/916134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/916134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/916134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/916134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/916134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/916134/1"}}},"commit":{"parents":[{"commit":"bbd1c106f1d5f68aee989fb8ab9c442a92f832b0","subject":"Merge \"Cleanup deprecated and ineffective api options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbd1c106f1d5f68aee989fb8ab9c442a92f832b0"}]}],"author":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-02-07 09:41:56.000000000","tz":60},"committer":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-04-17 18:53:44.000000000","tz":120},"subject":"Squash Fixed IP PTR","message":"Squash Fixed IP PTR\n\nUpdate parameters.yaml\n\nUpdate dns-api-v2-reverse-fixedips.inc\n\nUpdate record.py\n\nFix formatting\n\nChange-Id: I140f161c960b5d8637cf2d395b8635e8c8d961b1\n\nFix formatting\n\nChange-Id: I843442fdfbbe92236443cad9c1fd39e9f5c48854\n\nChange Fixed IP key\n\nChange-Id: I2f2655d12424ed61fc8fa445d83119d7b50bc247\n\nMove UUID namespace to constants and function to utils\n\nChange-Id: Iac60d9a22a17bf533cd0f5ee847f499bdc20bd12\n\nAdd regex for Fixed IP key\n\nChange-Id: Ib744d00102a782aa0092e85d664e9a832006a246\n\nRemove IP object\n\nChange-Id: I7037723d3c6412557e4f4c4e4a4d4baffeed3ab7\n\nRemove unused functions\n\nChange-Id: I6e36963934acb1f4cdd6326c315b0c432dce1a3c\n\nFix formatting\n\nDynamically retrieve external networks\n\nAdd fixed_ip unit test\n\nUpdate test_service.py\n\nRemove IPv6 for fixed IP, requires more substantial work which is out of scope for now\n\nSet IP field to v4/v6 to hopefully support IPv6 too\n\nChange some stuff for UUID\n\nRework _get_fixed_ips to network api, as designate does not use nova api\n\nRevert \"Change _get_fixed_ips strategy\"\n\nThis reverts commit 4e0a18b3adb8401ff05d4e7e5270af58471aee6d.\n\nChange _get_fixed_ips strategy\n\nCreate UUID with dedicated namespace\n\nAdd record for listing fixed ips\n\nAdd proper UUID for fixed ip\n\nRemove comma causing string to be tuple\n\nFix unit tests and formatting\n\nFirst leg of copying FloatingIP tests for FixedIP\n\nRewrite some conf related options to add FixedIP\n\nAdd FixedIP adapters\n\nReplicate various Floating logic for Fixed\n\nAdd IP constants for Fixed IP\n\nAdd IP objects to init\n\nReflect central service in rpc API\n\nRewrite central service for Fixed IP PTR\n\nAdd FixedIP object\n\nAdd Fixed IP logic to network_api\nChange-Id: Iea6f1515e9484a02aa5e9d1bee5005b6c86352b5\n\nChange-Id: I2695fb7460d2e288d65bc6f6c08ecbfa312e5177\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb61a7fd6cf7803b3fc888ec19b0b862f57b73e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb61a7fd6cf7803b3fc888ec19b0b862f57b73e4"}]},"branch":"refs/heads/master"},"f9c54db825bcfa7c02ed8840977cc0db3533da56":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-04-17 19:00:35.000000000","uploader":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"ref":"refs/changes/34/916134/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/916134/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/916134/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/916134/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/916134/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/916134/2"}}},"commit":{"parents":[{"commit":"bbd1c106f1d5f68aee989fb8ab9c442a92f832b0","subject":"Merge \"Cleanup deprecated and ineffective api options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbd1c106f1d5f68aee989fb8ab9c442a92f832b0"}]}],"author":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-02-07 09:41:56.000000000","tz":60},"committer":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-04-17 19:00:35.000000000","tz":0},"subject":"This MR implements Reverse DNS for Fixed IPs as detailed in this blueprint: https://wiki.openstack.org/wiki/Designate/Blueprints/ReverseFixedIP","message":"This MR implements Reverse DNS for Fixed IPs as detailed in this blueprint: https://wiki.openstack.org/wiki/Designate/Blueprints/ReverseFixedIP\n\nThe logic is writting to reflect the Floating IP logic as much as possible, and where possible duplicate logic has been combined together. If there are any questions or remarks, let me know! :-)\n\nWe have developed and tested this extensively on our OpenStack environment at Fuga Cloud, where it behaves as expected and we will soon be rolling this out to our customers too.\n\n--------------------------------\n\nSquash Fixed IP PTR\n\nUpdate parameters.yaml\n\nUpdate dns-api-v2-reverse-fixedips.inc\n\nUpdate record.py\n\nFix formatting\n\nChange-Id: I140f161c960b5d8637cf2d395b8635e8c8d961b1\n\nFix formatting\n\nChange-Id: I843442fdfbbe92236443cad9c1fd39e9f5c48854\n\nChange Fixed IP key\n\nChange-Id: I2f2655d12424ed61fc8fa445d83119d7b50bc247\n\nMove UUID namespace to constants and function to utils\n\nChange-Id: Iac60d9a22a17bf533cd0f5ee847f499bdc20bd12\n\nAdd regex for Fixed IP key\n\nChange-Id: Ib744d00102a782aa0092e85d664e9a832006a246\n\nRemove IP object\n\nChange-Id: I7037723d3c6412557e4f4c4e4a4d4baffeed3ab7\n\nRemove unused functions\n\nChange-Id: I6e36963934acb1f4cdd6326c315b0c432dce1a3c\n\nFix formatting\n\nDynamically retrieve external networks\n\nAdd fixed_ip unit test\n\nUpdate test_service.py\n\nRemove IPv6 for fixed IP, requires more substantial work which is out of scope for now\n\nSet IP field to v4/v6 to hopefully support IPv6 too\n\nChange some stuff for UUID\n\nRework _get_fixed_ips to network api, as designate does not use nova api\n\nRevert \"Change _get_fixed_ips strategy\"\n\nThis reverts commit 4e0a18b3adb8401ff05d4e7e5270af58471aee6d.\n\nChange _get_fixed_ips strategy\n\nCreate UUID with dedicated namespace\n\nAdd record for listing fixed ips\n\nAdd proper UUID for fixed ip\n\nRemove comma causing string to be tuple\n\nFix unit tests and formatting\n\nFirst leg of copying FloatingIP tests for FixedIP\n\nRewrite some conf related options to add FixedIP\n\nAdd FixedIP adapters\n\nReplicate various Floating logic for Fixed\n\nAdd IP constants for Fixed IP\n\nAdd IP objects to init\n\nReflect central service in rpc API\n\nRewrite central service for Fixed IP PTR\n\nAdd FixedIP object\n\nAdd Fixed IP logic to network_api\nChange-Id: Iea6f1515e9484a02aa5e9d1bee5005b6c86352b5\n\nChange-Id: I2695fb7460d2e288d65bc6f6c08ecbfa312e5177\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f9c54db825bcfa7c02ed8840977cc0db3533da56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f9c54db825bcfa7c02ed8840977cc0db3533da56"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9143439382e26601565ad7859d4cf52c97807f49":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-04-17 19:03:21.000000000","uploader":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"ref":"refs/changes/34/916134/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/916134/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/916134/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/916134/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/916134/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/916134/3"}}},"commit":{"parents":[{"commit":"bbd1c106f1d5f68aee989fb8ab9c442a92f832b0","subject":"Merge \"Cleanup deprecated and ineffective api options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbd1c106f1d5f68aee989fb8ab9c442a92f832b0"}]}],"author":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-02-07 09:41:56.000000000","tz":60},"committer":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-04-17 19:03:21.000000000","tz":0},"subject":"Implement Reverse DNS support for Fixed IPs","message":"Implement Reverse DNS support for Fixed IPs\n\nThis MR implements Reverse DNS for Fixed IPs as detailed in this blueprint: https://wiki.openstack.org/wiki/Designate/Blueprints/ReverseFixedIP\n\nThe logic is writting to reflect the Floating IP logic as much as possible, and where possible duplicate logic has been combined together. If there are any questions or remarks, let me know! :-)\n\nWe have developed and tested this extensively on our OpenStack environment at Fuga Cloud, where it behaves as expected and we will soon be rolling this out to our customers too.\n\n--------------------------------\n\nSquash Fixed IP PTR\n\nUpdate parameters.yaml\n\nUpdate dns-api-v2-reverse-fixedips.inc\n\nUpdate record.py\n\nFix formatting\n\nChange-Id: I140f161c960b5d8637cf2d395b8635e8c8d961b1\n\nFix formatting\n\nChange-Id: I843442fdfbbe92236443cad9c1fd39e9f5c48854\n\nChange Fixed IP key\n\nChange-Id: I2f2655d12424ed61fc8fa445d83119d7b50bc247\n\nMove UUID namespace to constants and function to utils\n\nChange-Id: Iac60d9a22a17bf533cd0f5ee847f499bdc20bd12\n\nAdd regex for Fixed IP key\n\nChange-Id: Ib744d00102a782aa0092e85d664e9a832006a246\n\nRemove IP object\n\nChange-Id: I7037723d3c6412557e4f4c4e4a4d4baffeed3ab7\n\nRemove unused functions\n\nChange-Id: I6e36963934acb1f4cdd6326c315b0c432dce1a3c\n\nFix formatting\n\nDynamically retrieve external networks\n\nAdd fixed_ip unit test\n\nUpdate test_service.py\n\nRemove IPv6 for fixed IP, requires more substantial work which is out of scope for now\n\nSet IP field to v4/v6 to hopefully support IPv6 too\n\nChange some stuff for UUID\n\nRework _get_fixed_ips to network api, as designate does not use nova api\n\nRevert \"Change _get_fixed_ips strategy\"\n\nThis reverts commit 4e0a18b3adb8401ff05d4e7e5270af58471aee6d.\n\nChange _get_fixed_ips strategy\n\nCreate UUID with dedicated namespace\n\nAdd record for listing fixed ips\n\nAdd proper UUID for fixed ip\n\nRemove comma causing string to be tuple\n\nFix unit tests and formatting\n\nFirst leg of copying FloatingIP tests for FixedIP\n\nRewrite some conf related options to add FixedIP\n\nAdd FixedIP adapters\n\nReplicate various Floating logic for Fixed\n\nAdd IP constants for Fixed IP\n\nAdd IP objects to init\n\nReflect central service in rpc API\n\nRewrite central service for Fixed IP PTR\n\nAdd FixedIP object\n\nAdd Fixed IP logic to network_api\nChange-Id: Iea6f1515e9484a02aa5e9d1bee5005b6c86352b5\n\nChange-Id: I2695fb7460d2e288d65bc6f6c08ecbfa312e5177\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9143439382e26601565ad7859d4cf52c97807f49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9143439382e26601565ad7859d4cf52c97807f49"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b684b0de3c3caa06771b462bd2228f6ceb7cd53d":{"kind":"REWORK","_number":4,"created":"2024-04-18 06:46:21.000000000","uploader":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"ref":"refs/changes/34/916134/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/916134/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/916134/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/916134/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/916134/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/916134/4"}}},"commit":{"parents":[{"commit":"bbd1c106f1d5f68aee989fb8ab9c442a92f832b0","subject":"Merge \"Cleanup deprecated and ineffective api options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbd1c106f1d5f68aee989fb8ab9c442a92f832b0"}]}],"author":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-02-07 09:41:56.000000000","tz":60},"committer":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-04-18 06:46:08.000000000","tz":120},"subject":"Squash Fixed IP PTR","message":"Squash Fixed IP PTR\n\nUpdate parameters.yaml\n\nUpdate dns-api-v2-reverse-fixedips.inc\n\nUpdate record.py\n\nFix formatting\n\nChange-Id: I140f161c960b5d8637cf2d395b8635e8c8d961b1\n\nFix formatting\n\nChange-Id: I843442fdfbbe92236443cad9c1fd39e9f5c48854\n\nChange Fixed IP key\n\nChange-Id: I2f2655d12424ed61fc8fa445d83119d7b50bc247\n\nMove UUID namespace to constants and function to utils\n\nChange-Id: Iac60d9a22a17bf533cd0f5ee847f499bdc20bd12\n\nAdd regex for Fixed IP key\n\nChange-Id: Ib744d00102a782aa0092e85d664e9a832006a246\n\nRemove IP object\n\nChange-Id: I7037723d3c6412557e4f4c4e4a4d4baffeed3ab7\n\nRemove unused functions\n\nChange-Id: I6e36963934acb1f4cdd6326c315b0c432dce1a3c\n\nFix formatting\n\nDynamically retrieve external networks\n\nAdd fixed_ip unit test\n\nUpdate test_service.py\n\nRemove IPv6 for fixed IP, requires more substantial work which is out of scope for now\n\nSet IP field to v4/v6 to hopefully support IPv6 too\n\nChange some stuff for UUID\n\nRework _get_fixed_ips to network api, as designate does not use nova api\n\nRevert \"Change _get_fixed_ips strategy\"\n\nThis reverts commit 4e0a18b3adb8401ff05d4e7e5270af58471aee6d.\n\nChange _get_fixed_ips strategy\n\nCreate UUID with dedicated namespace\n\nAdd record for listing fixed ips\n\nAdd proper UUID for fixed ip\n\nRemove comma causing string to be tuple\n\nFix unit tests and formatting\n\nFirst leg of copying FloatingIP tests for FixedIP\n\nRewrite some conf related options to add FixedIP\n\nAdd FixedIP adapters\n\nReplicate various Floating logic for Fixed\n\nAdd IP constants for Fixed IP\n\nAdd IP objects to init\n\nReflect central service in rpc API\n\nRewrite central service for Fixed IP PTR\n\nAdd FixedIP object\n\nAdd Fixed IP logic to network_api\nChange-Id: Iea6f1515e9484a02aa5e9d1bee5005b6c86352b5\n\nChange-Id: I2695fb7460d2e288d65bc6f6c08ecbfa312e5177\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b684b0de3c3caa06771b462bd2228f6ceb7cd53d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b684b0de3c3caa06771b462bd2228f6ceb7cd53d"}]},"branch":"refs/heads/master"},"2b41f47126e646c75ab45f575779c3b9a3905cea":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-04-18 06:47:58.000000000","uploader":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"ref":"refs/changes/34/916134/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/916134/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/916134/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/916134/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/916134/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/916134/5"}}},"commit":{"parents":[{"commit":"bbd1c106f1d5f68aee989fb8ab9c442a92f832b0","subject":"Merge \"Cleanup deprecated and ineffective api options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbd1c106f1d5f68aee989fb8ab9c442a92f832b0"}]}],"author":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-02-07 09:41:56.000000000","tz":60},"committer":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-04-18 06:47:58.000000000","tz":0},"subject":"Implement Reverse DNS support for Fixed IPs","message":"Implement Reverse DNS support for Fixed IPs\n\nThis MR implements Reverse DNS for Fixed IPs as detailed in this blueprint: https://wiki.openstack.org/wiki/Designate/Blueprints/ReverseFixedIP\n\nThe logic is writting to reflect the Floating IP logic as much as possible, and where possible duplicate logic has been combined together. If there are any questions or remarks, let me know! :-)\n\nWe have developed and tested this extensively on our OpenStack environment at Fuga Cloud, where it behaves as expected and we will soon be rolling this out to our customers too.\n\n--------------------------------\n\nSquash Fixed IP PTR\n\nUpdate parameters.yaml\n\nUpdate dns-api-v2-reverse-fixedips.inc\n\nUpdate record.py\n\nFix formatting\n\nChange-Id: I140f161c960b5d8637cf2d395b8635e8c8d961b1\n\nFix formatting\n\nChange-Id: I843442fdfbbe92236443cad9c1fd39e9f5c48854\n\nChange Fixed IP key\n\nChange-Id: I2f2655d12424ed61fc8fa445d83119d7b50bc247\n\nMove UUID namespace to constants and function to utils\n\nChange-Id: Iac60d9a22a17bf533cd0f5ee847f499bdc20bd12\n\nAdd regex for Fixed IP key\n\nChange-Id: Ib744d00102a782aa0092e85d664e9a832006a246\n\nRemove IP object\n\nChange-Id: I7037723d3c6412557e4f4c4e4a4d4baffeed3ab7\n\nRemove unused functions\n\nChange-Id: I6e36963934acb1f4cdd6326c315b0c432dce1a3c\n\nFix formatting\n\nDynamically retrieve external networks\n\nAdd fixed_ip unit test\n\nUpdate test_service.py\n\nRemove IPv6 for fixed IP, requires more substantial work which is out of scope for now\n\nSet IP field to v4/v6 to hopefully support IPv6 too\n\nChange some stuff for UUID\n\nRework _get_fixed_ips to network api, as designate does not use nova api\n\nRevert \"Change _get_fixed_ips strategy\"\n\nThis reverts commit 4e0a18b3adb8401ff05d4e7e5270af58471aee6d.\n\nChange _get_fixed_ips strategy\n\nCreate UUID with dedicated namespace\n\nAdd record for listing fixed ips\n\nAdd proper UUID for fixed ip\n\nRemove comma causing string to be tuple\n\nFix unit tests and formatting\n\nFirst leg of copying FloatingIP tests for FixedIP\n\nRewrite some conf related options to add FixedIP\n\nAdd FixedIP adapters\n\nReplicate various Floating logic for Fixed\n\nAdd IP constants for Fixed IP\n\nAdd IP objects to init\n\nReflect central service in rpc API\n\nRewrite central service for Fixed IP PTR\n\nAdd FixedIP object\n\nAdd Fixed IP logic to network_api\nChange-Id: Iea6f1515e9484a02aa5e9d1bee5005b6c86352b5\n\nChange-Id: I2695fb7460d2e288d65bc6f6c08ecbfa312e5177\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b41f47126e646c75ab45f575779c3b9a3905cea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b41f47126e646c75ab45f575779c3b9a3905cea"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
