)]}'
{"id":"openstack%2Fdesignate~916169","triplet_id":"openstack%2Fdesignate~master~Ie8dc131828a8dba7a051af612f836023d9274508","project":"openstack/designate","branch":"master","topic":"feature/fixed-ips","attention_set":{"36834":{"account":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"last_update":"2024-07-16 11:06:56.000000000","reason":"A robot voted negatively on a label"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-07-16 08:29:31.000000000","reason":"Change was marked ready for review"}},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-09-25 15:03:53.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-07-16 13:14:27.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-07-16 08:30:22.000000000","reason":"\u003cGERRIT_ACCOUNT_36834\u003e replied on the change","reason_account":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"}}},"hashtags":[],"change_id":"Ie8dc131828a8dba7a051af612f836023d9274508","subject":"Implement Reverse DNS support for Fixed IPs","status":"NEW","created":"2024-04-18 08:17:02.000000000","updated":"2026-09-25 15:03:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":48,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"557f0383a1d1927c02b0f0a0a79c75dde038dcb0","_number":916169,"virtual_id_number":916169,"owner":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-16 13:18:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":-1,"date":"2026-09-25 15:03:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-18 09:30:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-13 12:40:39.000000000","updated_by":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2024-05-13 12:40:39.000000000","updated_by":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-05-21 12:27:01.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-07-16 08:30:22.000000000","updated_by":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REMOVED"},{"updated":"2024-07-16 09:27:27.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"9905ed27ba836f7ae15e9898400f15e4e5575119","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-04-18 08:17:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b0729b29cbb3074b67ccaad2d30b7c0175c52c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 09:30:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f13fb98119884b8a94c97681acef2132\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cd90c88488f64b7fb4aa4ddd4d7579ad : SUCCESS in 4m 21s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/040440f89a9e47d7a236beb79a05a665 : SUCCESS in 37m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1442b1668566436b8d7cfd1bc78f21fa : SUCCESS in 48m 30s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b3752b4c4044502b9cf36143ff18121 : SUCCESS in 45m 03s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/aa7933b5f8754847931c9e3490bc8d97 : SUCCESS in 40m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/149858425ab04fb7814d2f5e5fd3e5ab : SUCCESS in 35m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/53197d9bc0fe4bf79f01ef15bedda8fa : SUCCESS in 1h 01m 45s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/aafc364dbe084a1593c81d4fb9198354 : SUCCESS in 50m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/498a18e5bf1b478e8c36361cf3994ea0 : SUCCESS in 38m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/798f9987b6d84bcf9534fb7d244409cf : SUCCESS in 33m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4be79de422874fdfa0d55c07e20a8d88 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a82509fb3188490baabadd5ccc772a96 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e0f5d2f0aa3411399d3da56992fdb20 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9c06210ddd0484ea9a03aa66d703ea6 : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af30b538422f463fa7a7934d100575c3 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a672283297d40c09963141627201e97 : SUCCESS in 8m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dab5f1a5098841ceaf60e2673fa28725 : SUCCESS in 37m 14s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/702c6210f83844c385dc92cf63d5a714 : SUCCESS in 5m 50s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b44a2acabad743a2852d1f5e8cfc29fe : SUCCESS in 47m 00s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2b3fb8ade7bc4505bda1a448248550d9 : SUCCESS in 4m 03s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/c8d906bfd3074b198fa4ec21ba3a78f0 : FAILURE in 44m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0f291643fa0da02ee054e66eddfa4369c391582","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-05-13 12:40:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0dc570d2539ea2ba6d552112d6af3b5e8f5fc8e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-13 13:40:57.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"55f6a79bf747bbb1b33a078227b1afe549f472af","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-05-14 07:23:09.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8062b01773bbe6a69f03dc533727f7abb59513e3","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-21 12:27:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5023ad391e0280a681ad8a0b2ad8dbf7dfa18833","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-06-06 07:31:47.000000000","message":"Patch Set 1:\n\n(9 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a7bce1c5a4db53d9f3fe9ad63ba6f0e57e1534ee","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-07-16 08:12:06.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"78663ecc40d42a32ce4c2bdc1d00cd2bd6863d5d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-07-16 08:18:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f3e1e530a52e0a687664491bd210a13bdf23419e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-07-16 08:29:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b5e7dc847de9d1490ab418a2b1a44c834b5040eb","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-07-16 08:29:31.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"98c38e64b108a29650b806f5527fd76de44d364d","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-07-16 08:30:22.000000000","message":"Patch Set 3:\n\n(14 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e5b67335302e3e16d0c275941cbda52714a56cfa","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-16 09:27:27.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8ff72a928a5c0c05f8d00edd1ad3a3bc216ef9ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 09:38:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/14ad6f9590284334914d73492082df6a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d66d58a4b04744519c2692b0eb379aa3 : FAILURE in 5m 00s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/103d86179f5b464092874c87b32f87eb : SUCCESS in 39m 56s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/45617379b6d445d8b5804c6716ec3238 : SUCCESS in 57m 50s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/40718c35137946e292f3b63aec09c9c4 : SUCCESS in 59m 06s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b032f86071ec4cfba6c16a21d5d06462 : SUCCESS in 47m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/79e7cc29c90a41dd8290770158333cf7 : SUCCESS in 51m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9e15bae7622d41b7a58cb7b55535d7c2 : SUCCESS in 55m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dc9a1a75fc5641ef820fe54d9a694db5 : SUCCESS in 1h 07m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/22d40097d7e14120b8685fced3337b87 : SUCCESS in 37m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a44b6e4f0197479dbf5165480dc7622a : SUCCESS in 44m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e7115bbc25f4e03a5b1abb4691791a8 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/633ae1db8934422692c052d4d620c953 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ae7307869844727bbbac33b899f7766 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5ea2d361b134af6bbd5c14ce768a470 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/37f0f3abc98b4b059ecfeb2eb7af91ca : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27263d1dcb9a43cb9e933a04fe5d3c97 : SUCCESS in 6m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ecc3be39d39457f8c6e2d9b8bdd2fff : SUCCESS in 45m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9752b39fd5b94af0a382fbc8d1368a00 : SUCCESS in 4m 14s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/5ee39a5d0fd74eac9662ced2f6791fa5 : FAILURE in 46m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7121869d056d36000dea931d919acc8968d80802","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-07-16 09:43:18.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"56fa20581ff0eb9436bba5f0fb0b2021a5f4ec10","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-07-16 09:43:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9edef89b2af66d5381d05c9684fba18b264c8fc2","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-07-16 09:49:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a71144855c994f8603ade247e354540b955cb53d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-07-16 09:57:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ec1216d23f34e1bc646ffd79e4c3b279d303eb06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-07-16 09:57:50.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"5dbf1d785f0f847bc5d74be93bec800d68c85002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 11:06:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f45a485982fd4dabb7afccb7a0ed68f4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cbcb33250ce14a0e97be38cf2a99f944 : SUCCESS in 4m 49s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0e4db02c9ad645a983e7b979d5457f3b : SUCCESS in 48m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/63bf727b01f843ceb37bb4a7fc8ef7f0 : SUCCESS in 49m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a95b59cd04f148e18dfe23f1ffa1a7e5 : SUCCESS in 1h 01m 43s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/186ec13511a44d64be7f0c2e4e889aaa : SUCCESS in 42m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fa2ece9ea449441488c9c5f19c1ae4ae : SUCCESS in 55m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3b3bedc2f82e4e93a7aa8f122ffe54fb : SUCCESS in 1h 05m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b284adc612d14a3ab73fcd58f470c86e : SUCCESS in 1h 02m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f29d97962f894f4680ad5607082b5f20 : SUCCESS in 37m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1fbcd6d0d8514d74a2d621e2c0b82c52 : FAILURE in 33m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3055737c888345e783e691459b93631f : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47612888ae4d4d9fa280048d86aa9c92 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c214434b4c814ad1920d466a71f6a27e : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e7bfd828297e4bc1af6b88a8ed216923 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71d9ccbc46b842ae8dbe0d9d068df738 : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34104b655a6c4e50bbea30751e011a96 : SUCCESS in 6m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/995a9de7bddd4e2f9c1c80148b7a3e9d : SUCCESS in 31m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5252b48363294bdf96680f85f9bb355c : SUCCESS in 4m 49s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/5e46c7fbaa1a4e77b2be1122a8778bf7 : SUCCESS in 43m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"78fefa5183957f858c52396f0d1a782b76c455cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-07-16 11:40:57.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"da5adba4c19fa4ad834ee143811be175e10aba2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"date":"2024-07-16 11:51:56.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"55dda8bc82d21cbe1e6a7fe642aea3f36c780c49","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-16 13:14:27.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"df599ffe389b910ced26549e7812ef41fb612172","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 13:18:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c57aa433c51041ac97cd76e7b3330e47\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c786d02e7293482fa7ade857aa4df674 : SUCCESS in 3m 53s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c7483cebabaf4786a0b56d74be2a9f03 : SUCCESS in 37m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8908db49419549c0bda800bbfd93c43b : SUCCESS in 47m 10s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/17a84bc86f2347389568f5df21f5dea2 : SUCCESS in 45m 22s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3e2490389b024b3d838318648aabdfee : SUCCESS in 39m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dfffdf524b7048aab7fb214e5a9ed1a5 : SUCCESS in 38m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9e3c1827d9ec48eb8eb3c627d394ad30 : SUCCESS in 1h 20m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a566da4f814c454c8f2d631dff951e8c : SUCCESS in 1h 13m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c21d2044eca6476da5f74580e8c2a7d0 : SUCCESS in 46m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0399d5ddad1b419fae0b667e6d105a28 : SUCCESS in 47m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9125051486c94885a2237eb54df427ab : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60396da2e48a40598e63af408cec1d3f : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/062c5a6fecf84162b727fc2a0c002ff9 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b16e64fbe174302b48b8f0196daf78d : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec77ec286b1a4371b65f539665290bd0 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46f446c53ce941aab5b5a85e80890552 : SUCCESS in 7m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/286b8013856b41e4a91c380a28704807 : SUCCESS in 49m 31s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/95d845ca0a8b4b2c914efc068086df1a : SUCCESS in 4m 34s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/699550069e9a4c7da377d411d539d0f3 : SUCCESS in 45m 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"557f0383a1d1927c02b0f0a0a79c75dde038dcb0","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-25 15:03:53.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"1ed24599e2afae3cb816522464b2f281f634b9df","revisions":{"006dc44e624c99aec037ca3a16264a6635f809f8":{"kind":"REWORK","_number":1,"created":"2024-04-18 08:17:02.000000000","uploader":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"ref":"refs/changes/69/916169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/916169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/916169/1"}}},"commit":{"parents":[{"commit":"bbd1c106f1d5f68aee989fb8ab9c442a92f832b0","subject":"Merge \"Cleanup deprecated and ineffective api options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbd1c106f1d5f68aee989fb8ab9c442a92f832b0"}]}],"author":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-02-07 09:41:56.000000000","tz":60},"committer":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-04-18 08:16:54.000000000","tz":120},"subject":"Implement Reverse DNS support for Fixed IPs","message":"Implement Reverse DNS support for Fixed IPs\n\nThis MR implements Reverse DNS for Fixed IPs as detailed in this blueprint: https://wiki.openstack.org/wiki/Designate/Blueprints/ReverseFixedIP\n\nThe logic is writting to reflect the Floating IP logic as much as possible, and where possible duplicate logic has been combined together. If there are any questions or remarks, let me know! :-)\n\nWe have developed and tested this extensively on our OpenStack environment at Fuga Cloud, where it behaves as expected and we will soon be rolling this out to our customers too.\n\nChange-Id: Ie8dc131828a8dba7a051af612f836023d9274508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/006dc44e624c99aec037ca3a16264a6635f809f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/006dc44e624c99aec037ca3a16264a6635f809f8"}]},"branch":"refs/heads/master"},"87bf6e8a88f34c81258a3f7695256501d964aa18":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-16 08:18:05.000000000","uploader":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"ref":"refs/changes/69/916169/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/916169/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/916169/2"}}},"commit":{"parents":[{"commit":"1ac55388dc46f39370591f01ad981792a7dae554","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ac55388dc46f39370591f01ad981792a7dae554"}]}],"author":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-02-07 09:41:56.000000000","tz":60},"committer":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-07-16 08:18:05.000000000","tz":0},"subject":"Implement Reverse DNS support for Fixed IPs","message":"Implement Reverse DNS support for Fixed IPs\n\nThis MR implements Reverse DNS for Fixed IPs as detailed in this blueprint: https://wiki.openstack.org/wiki/Designate/Blueprints/ReverseFixedIP\n\nThe logic is writting to reflect the Floating IP logic as much as possible, and where possible duplicate logic has been combined together. If there are any questions or remarks, let me know! :-)\n\nWe have developed and tested this extensively on our OpenStack environment at Fuga Cloud, where it behaves as expected and we will soon be rolling this out to our customers too.\n\nChange-Id: Ie8dc131828a8dba7a051af612f836023d9274508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/87bf6e8a88f34c81258a3f7695256501d964aa18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/87bf6e8a88f34c81258a3f7695256501d964aa18"}]},"branch":"refs/heads/master","description":"Rebase"},"3ed657ba398401c5e7554bd013e48b2a61a12f70":{"kind":"REWORK","_number":3,"created":"2024-07-16 08:29:21.000000000","uploader":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"ref":"refs/changes/69/916169/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/916169/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/916169/3"}}},"commit":{"parents":[{"commit":"1ac55388dc46f39370591f01ad981792a7dae554","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ac55388dc46f39370591f01ad981792a7dae554"}]}],"author":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-02-07 09:41:56.000000000","tz":60},"committer":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-07-16 08:29:13.000000000","tz":120},"subject":"Implement Reverse DNS support for Fixed IPs","message":"Implement Reverse DNS support for Fixed IPs\n\nChange-Id: I6542f4e8306f72b81c09ee43cf7517ae90dcd318\n\nImplement Reverse DNS support for Fixed IPs\n\nThis MR implements Reverse DNS for Fixed IPs as detailed in this blueprint: https://wiki.openstack.org/wiki/Designate/Blueprints/ReverseFixedIP\n\nThe logic is writting to reflect the Floating IP logic as much as possible, and where possible duplicate logic has been combined together. If there are any questions or remarks, let me know! :-)\n\nWe have developed and tested this extensively on our OpenStack environment at Fuga Cloud, where it behaves as expected and we will soon be rolling this out to our customers too.\n\nChange-Id: Ie8dc131828a8dba7a051af612f836023d9274508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3ed657ba398401c5e7554bd013e48b2a61a12f70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3ed657ba398401c5e7554bd013e48b2a61a12f70"}]},"branch":"refs/heads/master"},"a68469346413272d1404846eff6feaa45bbd9223":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-07-16 09:43:18.000000000","uploader":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"ref":"refs/changes/69/916169/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/916169/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/916169/4"}}},"commit":{"parents":[{"commit":"1ac55388dc46f39370591f01ad981792a7dae554","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ac55388dc46f39370591f01ad981792a7dae554"}]}],"author":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-02-07 09:41:56.000000000","tz":60},"committer":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-07-16 09:43:18.000000000","tz":0},"subject":"Implement Reverse DNS support for Fixed IPs","message":"Implement Reverse DNS support for Fixed IPs\n\nThis MR implements Reverse DNS for Fixed IPs as detailed in this blueprint: https://wiki.openstack.org/wiki/Designate/Blueprints/ReverseFixedIP\n\nThe logic is writting to reflect the Floating IP logic as much as possible, and where possible duplicate logic has been combined together. If there are any questions or remarks, let me know! :-)\n\nWe have developed and tested this extensively on our OpenStack environment at Fuga Cloud, where it behaves as expected and we will soon be rolling this out to our customers too.\n\nChange-Id: Ie8dc131828a8dba7a051af612f836023d9274508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a68469346413272d1404846eff6feaa45bbd9223"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a68469346413272d1404846eff6feaa45bbd9223"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e22824a038ebb4b4411278c2bfa5980fc7b4a1ef":{"kind":"REWORK","_number":5,"created":"2024-07-16 09:57:30.000000000","uploader":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"ref":"refs/changes/69/916169/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/916169/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/916169/5"}}},"commit":{"parents":[{"commit":"1ac55388dc46f39370591f01ad981792a7dae554","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ac55388dc46f39370591f01ad981792a7dae554"}]}],"author":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-02-07 09:41:56.000000000","tz":60},"committer":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-07-16 09:57:24.000000000","tz":120},"subject":"Implement Reverse DNS support for Fixed IPs","message":"Implement Reverse DNS support for Fixed IPs\n\nChange-Id: I6542f4e8306f72b81c09ee43cf7517ae90dcd318\n\nImplement Reverse DNS support for Fixed IPs\n\nThis MR implements Reverse DNS for Fixed IPs as detailed in this blueprint: https://wiki.openstack.org/wiki/Designate/Blueprints/ReverseFixedIP\n\nThe logic is writting to reflect the Floating IP logic as much as possible, and where possible duplicate logic has been combined together. If there are any questions or remarks, let me know! :-)\n\nWe have developed and tested this extensively on our OpenStack environment at Fuga Cloud, where it behaves as expected and we will soon be rolling this out to our customers too.\n\nChange-Id: Ie8dc131828a8dba7a051af612f836023d9274508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e22824a038ebb4b4411278c2bfa5980fc7b4a1ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e22824a038ebb4b4411278c2bfa5980fc7b4a1ef"}]},"branch":"refs/heads/master"},"3ab582f13c6faa6653f6647994f24e966384a910":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-07-16 09:57:50.000000000","uploader":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"ref":"refs/changes/69/916169/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/916169/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/916169/6"}}},"commit":{"parents":[{"commit":"1ac55388dc46f39370591f01ad981792a7dae554","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ac55388dc46f39370591f01ad981792a7dae554"}]}],"author":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-02-07 09:41:56.000000000","tz":60},"committer":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-07-16 09:57:50.000000000","tz":0},"subject":"Implement Reverse DNS support for Fixed IPs","message":"Implement Reverse DNS support for Fixed IPs\n\nThis MR implements Reverse DNS for Fixed IPs as detailed in this blueprint: https://wiki.openstack.org/wiki/Designate/Blueprints/ReverseFixedIP\n\nThe logic is writting to reflect the Floating IP logic as much as possible, and where possible duplicate logic has been combined together. If there are any questions or remarks, let me know! :-)\n\nWe have developed and tested this extensively on our OpenStack environment at Fuga Cloud, where it behaves as expected and we will soon be rolling this out to our customers too.\n\nChange-Id: Ie8dc131828a8dba7a051af612f836023d9274508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3ab582f13c6faa6653f6647994f24e966384a910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3ab582f13c6faa6653f6647994f24e966384a910"}]},"branch":"refs/heads/master","description":"Edit commit message"},"082a2ef04a14eaa092140e5fe142bb0278e3984a":{"kind":"REWORK","_number":7,"created":"2024-07-16 11:40:57.000000000","uploader":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"ref":"refs/changes/69/916169/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/916169/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/916169/7"}}},"commit":{"parents":[{"commit":"1ac55388dc46f39370591f01ad981792a7dae554","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ac55388dc46f39370591f01ad981792a7dae554"}]}],"author":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-02-07 09:41:56.000000000","tz":60},"committer":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-07-16 11:40:31.000000000","tz":120},"subject":"Implement Reverse DNS support for Fixed IPs","message":"Implement Reverse DNS support for Fixed IPs\n\nChange-Id: I6542f4e8306f72b81c09ee43cf7517ae90dcd318\n\nImplement Reverse DNS support for Fixed IPs\n\nThis MR implements Reverse DNS for Fixed IPs as detailed in this blueprint: https://wiki.openstack.org/wiki/Designate/Blueprints/ReverseFixedIP\n\nThe logic is writting to reflect the Floating IP logic as much as possible, and where possible duplicate logic has been combined together. If there are any questions or remarks, let me know! :-)\n\nWe have developed and tested this extensively on our OpenStack environment at Fuga Cloud, where it behaves as expected and we will soon be rolling this out to our customers too.\n\nChange-Id: Ie8dc131828a8dba7a051af612f836023d9274508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/082a2ef04a14eaa092140e5fe142bb0278e3984a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/082a2ef04a14eaa092140e5fe142bb0278e3984a"}]},"branch":"refs/heads/master"},"1ed24599e2afae3cb816522464b2f281f634b9df":{"kind":"REWORK","_number":8,"created":"2024-07-16 11:51:56.000000000","uploader":{"_account_id":36834,"name":"Sam Toxopeus","display_name":"sam@fuga.cloud","email":"sam@fuga.cloud","username":"sam@fuga.cloud"},"ref":"refs/changes/69/916169/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/916169/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/916169/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/916169/8"}}},"commit":{"parents":[{"commit":"1ac55388dc46f39370591f01ad981792a7dae554","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ac55388dc46f39370591f01ad981792a7dae554"}]}],"author":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-02-07 09:41:56.000000000","tz":60},"committer":{"name":"Sam Toxopeus","email":"sam@fuga.cloud","date":"2024-07-16 11:51:46.000000000","tz":120},"subject":"Implement Reverse DNS support for Fixed IPs","message":"Implement Reverse DNS support for Fixed IPs\n\nThis MR implements Reverse DNS for Fixed IPs as detailed in this blueprint: https://wiki.openstack.org/wiki/Designate/Blueprints/ReverseFixedIP\n\nThe logic is writting to reflect the Floating IP logic as much as possible, and where possible duplicate logic has been combined together. If there are any questions or remarks, let me know! :-)\n\nWe have developed and tested this extensively on our OpenStack environment at Fuga Cloud, where it behaves as expected and we will soon be rolling this out to our customers too.\n\nChange-Id: Ie8dc131828a8dba7a051af612f836023d9274508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ed24599e2afae3cb816522464b2f281f634b9df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ed24599e2afae3cb816522464b2f281f634b9df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
