)]}'
{"id":"openstack%2Fdesignate~918031","triplet_id":"openstack%2Fdesignate~master~I214cfc05a834d97a61d0236ab223470631bfc405","project":"openstack/designate","branch":"master","topic":"multipools","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-05-28 20:19:49.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-05-28 19:04:33.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-05-23 15:08:58.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I214cfc05a834d97a61d0236ab223470631bfc405","subject":"Fix YAMLAdapter when optional attributes are missing","status":"MERGED","created":"2024-05-03 09:49:56.000000000","updated":"2024-10-29 09:45:34.000000000","submitted":"2024-05-28 20:19:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"918031-multipools","meta_rev_id":"72f6b2258336dcedcf46993bf0c8462e58b95ea6","_number":918031,"virtual_id_number":918031,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-28 20:19:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-28 19:04:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-05-23 15:08:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-28 19:04:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-03 10:50:23.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2024-05-03 11:09:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-08 09:52:02.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-05-08 14:09:47.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2024-10-29 09:44:32.000000000","updated_by":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"reviewer":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"state":"REVIEWER"},{"updated":"2024-10-29 09:45:34.000000000","updated_by":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"reviewer":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"state":"CC"}],"messages":[{"id":"d1949ec36a4e8bb1e0a7a92ed1f16c9c3f2b3a93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-03 09:49:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed1dbb0766a2325cae1b3c3b7dae5c1d205f3462","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-03 10:50:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7de7d4fc2b394b2349648e63c78e456dc2c6f357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 11:09:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bb960619e0548d3ace5be51c9d57f8b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/93eafc16300f465dbf112bf2cdd8fac9 : SUCCESS in 46m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a1febc94e5ba4b6c8e73cb5f25f7a69f : SUCCESS in 53m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/de62aa5f199b4ca7b9fc6085a37b8350 : SUCCESS in 38m 48s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/99802be485e14993bd912d52979ac78f : SUCCESS in 38m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e4368029a6d24115a5372be01b08ba7e : SUCCESS in 39m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/14b6f59f48354af3bc94694b36c18448 : SUCCESS in 50m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5338f64aef394cbaa823038cd8ed3d67 : SUCCESS in 1h 08m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/136bb47c63094e2f9cf2ac48885c9669 : SUCCESS in 50m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/046aea48db184862b31bb4ba4c36bde2 : SUCCESS in 42m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ea7766a61e2451caf780c247822fec9 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a8bb1cfd222448b9683f70ec9f62aab : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e1eac052a40440f8a81f28d4f8d0d64 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2aa64267750b4e5096463f399190bae9 : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84b50156137141f69814666523653063 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18528def490b4de5bd3a4d5f61788140 : SUCCESS in 7m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5cf2aab505684151a31526045d5750a8 : SUCCESS in 35m 42s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cb7089e289a9452d8e1c3c87b86df2fb : SUCCESS in 5m 28s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6d94aa000bc4410c856245a086ba3e40 : SUCCESS in 37m 37s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f715005738a14f5c94e2ce1850f88316 : SUCCESS in 4m 37s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/663d6890635c42008ceabd99f7df46fa : SUCCESS in 38m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"91a4b0f816b2f1cf117395275b7db2d5f1a33f99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-06 15:03:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"652edfee26f7983056a49e1a925d0b051a256c5f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-06 15:13:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6869e0a8ac0407024fa4dac67472039627f72e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 16:14:24.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e3cc4d89a644549959273016871c042\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3c483ee345664ae9b4fb6d69cce29012 : SUCCESS in 47m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/18ea249c4a88490eb71d78c162d42d7d : SUCCESS in 54m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/92f29389261b4de19c3ce38c9e763e8c : SUCCESS in 39m 13s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6fb948f9e214439eb81578bfe19828bf : FAILURE in 36m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3becf33c624e40a7b1ded34c3b0ca87f : SUCCESS in 40m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4fc0bf56402a4302a15aae1003bb0bf8 : SUCCESS in 58m 40s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6acdc29fe9904012860db07cf1eab7ff : FAILURE in 1h 02m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2df9e41d8fa94da8bdb1bf49450fe6de : SUCCESS in 34m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a032fabdd8e14b4ba25c73fc7b0e57e4 : SUCCESS in 40m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7710210418f6443b846f3dd61bc406e8 : FAILURE in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d53619583a3425d905c7102ff050161 : FAILURE in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d493e702d73545afb295b8e8eeb1446e : FAILURE in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04b8f1e4898d43daa32cf185a471703f : FAILURE in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73b8693e73814ff8819278a6879d71b7 : FAILURE in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/427c7574cd7c4a08a845a191324c8028 : SUCCESS in 9m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2ded3757d85e40219547e85cd85da43f : SUCCESS in 45m 32s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2ffe9bd460844f8ebf887cbbd81d526f : FAILURE in 5m 25s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7d7975d8effa41d18734b11a696081fa : SUCCESS in 45m 11s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/919a06e00af04ebab077a2a93085a861 : FAILURE in 4m 32s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/6b92a88532264955aa3ec1173f551a07 : SUCCESS in 51m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"370705f5a26633fde25b3259e58027381cf60930","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-07 10:05:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf2a3ab47ed9b2ee594d1bbff86f4458d0ddb93c","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-07 14:41:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22379aec4cb93113355199f9365257e045149337","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-07 14:53:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0fa00b750edcb41dad5552619d06dbdb2da0b30c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-07 15:34:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"88e64175b9977ed8fd78b55235b087316a90f294","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-08 09:16:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"49e88965076ea0037ebc06e8059004dd83cc5c4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-08 09:16:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5ad3881897ef7bf257631db140f06736780a0284","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-08 09:54:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7f3885d314678b482b71b79294b639c4e7e98001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 10:38:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c1cfe92faa848ff8b549a83acd886f2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d814849218074854a79401529086163d : SUCCESS in 37m 50s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/46b2555484a44f31bea41885d77d1920 : SUCCESS in 1h 00m 09s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ee580229fa64a7faf1108d7538acf7a : SUCCESS in 53m 28s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c043119576d94b72b5fbabff92b47cbc : SUCCESS in 48m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d60e48c581c74e1c9352be5f8f3acf94 : SUCCESS in 48m 45s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bc02b7fd62ae44b68589fcd418076d0c : SUCCESS in 1h 01m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bcd451bbfb934949b95a47bafb45bd0f : SUCCESS in 1h 02m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f948eacc98ed49089250800a27852e4d : SUCCESS in 38m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/424eadeec022489d8d071bc256d6e5ba : SUCCESS in 32m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfa7bc0316a84e31908603c7d02b5af8 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88ab18ae5c8541b2adf4b75a3e1b76b8 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d015cce485d84c6ba50125dba0d3f837 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/090bc550deef46e49433f22b10da1612 : SUCCESS in 5m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cbba93bee1bc431c804fd93ed7d031e5 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8af1cd6b149a4bb8a1245ef52ea2dc6b : SUCCESS in 8m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c8a0c36ebccc421796efbea2686afb37 : SUCCESS in 31m 52s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/805edce3646549958ff7c0e6dda25c54 : SUCCESS in 6m 44s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d2e7f5912202477493956b43ba9e269a : SUCCESS in 36m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1dd7aedb4ab74792a1fd6584ed91f861 : SUCCESS in 4m 18s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/3d43058bac874885b4952360a7240482 : SUCCESS in 47m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8ccde2274d34816597e87e508760aabc0a6854b8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-08 14:09:47.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c8488c093fac1b484295e8c7e34729319c088cf3","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-09 10:24:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"00c7be42205fa90747dc89055a177a236a2fc8b0","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-09 10:25:35.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"48f64eda25d1f70fafecb7002a6bd45be463e2ce","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-09 11:34:48.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e07fa320140e71248cb95df56a4b80e2732a5646","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-09 12:21:11.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"aef9a08d6744c9a535ba551a8fc69c0566afc113","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-09 12:32:05.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"cc00d033c1e097b8df8d42b88b29d9a77b34472b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 13:40:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa04ec8e099e4ce39fbf6bb4dbaa8de5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/54911269bfc04a5286c3113f345d3efe : SUCCESS in 41m 58s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/636dab74a3374d38a3716bd451c50af7 : SUCCESS in 41m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/31ec737a6464416b839e647c6ebbef5e : SUCCESS in 45m 10s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1ed24e42deca4e9eb1460fcf668cfa10 : SUCCESS in 36m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8e1dd9fada63498baea2dc70d7de16c7 : SUCCESS in 34m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f4f50bf106c144cc9ea91f619eef627e : SUCCESS in 54m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8eba90e70c0d4e4aab8d7a0bc48b8d1d : SUCCESS in 57m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/42c0a316b3da48638bc192861c31daa9 : SUCCESS in 46m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f37f62181fcc45a3a082cbe5b62aafff : SUCCESS in 36m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/452fdd848e044be6be410a0fa3eba467 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca1f940b45f648518178307237a86067 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0ffe0044cdb4802ad7228dfa903b294 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8af74ad1b9ac4627ab37be781af3767b : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb6a2430840248d286a113be03a5240f : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/386dc3dbb5a840bfb04591feb4233adc : SUCCESS in 7m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0a158de9d11d4d0482effa37854790cd : SUCCESS in 46m 28s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1d034e2d49c240aeb7179ca6bd744635 : SUCCESS in 6m 39s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/098cea09c4d84e47a60eb97243dcc522 : SUCCESS in 43m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/52aaf5afe8424d0395c095d02437e181 : SUCCESS in 4m 37s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/9e968275733d4452b79a8dbb402f305e : FAILURE in 49m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4ab8f50352c0fd327c098e806add063cf148d40f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-20 17:35:48.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"67c995c711285ac6004cbb997bdcb65c509d9403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 18:48:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2880940b21664ebea7cd4c5757f8fe2a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6e3cc408cd43454eaf1e9788b5e35e31 : SUCCESS in 39m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5fb75f74ed804db1a40efa4065aa09dc : SUCCESS in 39m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f78bb02e968c40cfa74aee1139fc3d5b : SUCCESS in 43m 45s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ebc7b2f7faf44675b6c7db9125274b1f : SUCCESS in 37m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f223acf30f8140e29eba7645d2e3f2a2 : SUCCESS in 35m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c2560590a23c45878cfb55f883c16716 : SUCCESS in 55m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/762032d498174239a4381c3d7ef49698 : SUCCESS in 59m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5524a623d8434a61bd0b9763e659d616 : SUCCESS in 31m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e1c183d0f0644d3999000e3b71349ca1 : SUCCESS in 36m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11bd36609b3d4d39baabc01553c295a7 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/174a1328d78a4ebf9f7c15960111a7cf : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/606db669ab884aa99785d9f144383f4c : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab97be064cd34db8b29057a8ea90616b : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b8ff2165cd445628e1a33fb58c80577 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d468f5b80a904d808df80f32881e0fac : SUCCESS in 8m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/55b6f300d23a4e3c9bff7e2013a5b1bc : SUCCESS in 40m 06s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/dc1ee68b7b4d4b749c2f52056bfd79fc : SUCCESS in 6m 11s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fc7a3444a8b945f19fc12061ab6a2d57 : SUCCESS in 42m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0831bb8ccee44819a2fecf8446e7f662 : SUCCESS in 5m 42s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/f31034850d0f4bf18266b97f2c29f382 : SUCCESS in 36m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f4835824359392c608104ecd2a81aaf834bbdd50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-21 08:34:02.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"26d363c77bb1785c5ccf80a2e10155942d774cd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 09:43:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a457a40732f54132b37705d66f8da357\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1ce63bbaa5914e8bb36b71b3ba4ceb6d : SUCCESS in 35m 28s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/75dc363c289144b9a21b3c68b973c2c5 : FAILURE in 50m 14s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/998da391b1984eeebdffc46d8ddb64d2 : SUCCESS in 43m 28s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/efebcc357c5649ceaf7e732d06f490d7 : SUCCESS in 34m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/18dfd44229ec4e5498261c8856c064d4 : SUCCESS in 33m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/43f89f6f1cbd408e9003f6531abb457d : SUCCESS in 1h 04m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/944ce5f02d89423cb66defce92aa386c : SUCCESS in 1h 02m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1fc49df61d8044d6b2696ea6bf7f45da : SUCCESS in 51m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/aa08483e33c248b090065866edf4950b : SUCCESS in 44m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa2b801fb5a746a183797c7cf71f2628 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5252370e36c74e2286cd30168cee9eac : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/804ff92b98ed4636b187e51798be8660 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/446e38c520354f998a9ace402d58134b : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99585c0e807e417fa9278fb953f69f78 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b0d46e422b449d89f0d1ad8cfed3de6 : SUCCESS in 7m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6401563ca8124fe78e53d093d44253b1 : SUCCESS in 31m 53s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0a3261314c5749abbef076dc67ff3e2a : SUCCESS in 5m 41s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fa5e6c4d31f74294afaf8ce8cd573ff5 : SUCCESS in 45m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/87d48ac1914946e896d1e229fe976412 : SUCCESS in 4m 54s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/e3d23103531a4874b17f923df2a47bd7 : SUCCESS in 50m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3709625e44490a07067e21cba2b972a278cd2f39","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-23 00:29:01.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dc20bcce2acb5e4532ab90368249025c0dbf22b8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-23 00:29:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2295e8ea1ab55239256cab26c94bf9c1219c97bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-23 14:06:24.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"5f78d98164c3c17900c6a59ba037efaabb9b2100","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-23 14:07:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7f0d2e75321667007da25f7e98f59ff155f5e094","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-23 15:08:58.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"edeebb7b94f9f25ce7048177154d5ff9ea7001be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 15:19:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa7edab9e44a493eb16f551a9a750161\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f2eabbecca164cbe903e9971f472c8ef : SUCCESS in 47m 16s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8daff65dfd524c7cb14b28d8e9c71205 : SUCCESS in 1h 07m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c98badb182d47d59bdb1a07325a554e : SUCCESS in 43m 59s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d773c8e5d506410c850bd083f1f984c4 : SUCCESS in 38m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/036aa12231a44535bdea20b7e9f5ac7a : SUCCESS in 46m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9b3b0cdfbfff4fdda5ddf65020c47d6d : SUCCESS in 1h 06m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8ee87343bfc64988875ef8817153434d : SUCCESS in 1h 01m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/433e234093bf41e693bc260682c3856b : SUCCESS in 43m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6bb63d3748744922bd1d28d805af057f : SUCCESS in 41m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0b64361e62e48fbad100a51d98dab5e : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/623a117cc2c04b61b4f872734382ba2c : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee1b0b9dead04129bb86c7a777a2d767 : SUCCESS in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef2f2f5c1825420ea33c3ffa899b506a : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46b0343a7d3944ae96b5ab069b56f8e2 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58f38bebd3694bb694fd9a9fc959aa05 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2bc85e5435f4cf5be6ac3eaa53f5a87 : SUCCESS in 5m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1923a3e057a247f29468fc5f25e8fc52 : SUCCESS in 46m 18s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/148f1efd20bf43da9b2f496636c19f99 : SUCCESS in 6m 36s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4be2f9abfc384d7d80854dee87bf167e : SUCCESS in 44m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d1012cfcb887459d9f5812dd284a5763 : SUCCESS in 5m 57s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/b1d3edf0992e4c559d153071eabe58eb : SUCCESS in 40m 20s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6da0f77b152539f59f310fcfecff4df0bef42e96","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-28 19:04:33.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bc95779c5f3f0fa34b9eacfed4d54c996bb11e37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 19:05:08.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2d098429ec4d0ee80d817776abbfd27f3921db9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 20:19:48.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6806063692874560bef92b71138c7c98\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/47feeed365bf4af6bf88b2551a5b2398 : SUCCESS in 48m 01s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/39b80cfa14164f889414a0f5f5acc151 : SUCCESS in 45m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ed61053722fe427997c42cc6f6137dbb : SUCCESS in 49m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6813cfcd39cb4e64b640241bcfec4091 : SUCCESS in 1h 09m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c7e30852f64c4a8e9df4721decfa99ae : SUCCESS in 47m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/95966472766a417b84673fb213418f23 : SUCCESS in 36m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46269a61f2864e8a8e0e88bf69a9e34e : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/128f2b78b3c149b09d5a07daaca3d839 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46584a99e8d3417fb973e903f8917772 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69c0b3529092454192bdafdbfab7d265 : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdecade0c8024048a631c7570510cfb9 : SUCCESS in 5m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e62e11a2c19e4a8ab5489e98c55d0fc0 : SUCCESS in 41m 22s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f6b0c0643e2049ab8655e5631ddc7688 : SUCCESS in 6m 21s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/13bc844299a44df880b236ab561defea : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":7},{"id":"de05c4a5baf15a28110dabdbfc9c808133166d7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 20:19:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"8316460fd3d408d5adb2fcfbfe59c9ee7bdd5ed3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 20:20:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/036e972c139443bea4385b4a97cb3e1e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c4073014a54ef0b5cab526ebec6e16 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06eb929cd26c4d27a0ad580a1e12c439 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6cc0a7d8876291f874edaac2ff86b7d7a84a738a","revisions":{"4229f25a19456cc9107bc3740946373503717837":{"kind":"REWORK","_number":1,"created":"2024-05-03 09:49:56.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/31/918031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/918031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/918031/1"}}},"commit":{"parents":[{"commit":"b26b3aff5053ab43bc8c1eede565b0c76a00e53e","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b26b3aff5053ab43bc8c1eede565b0c76a00e53e"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 09:04:46.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 09:24:14.000000000","tz":120},"subject":"Fix YAMLAdapter when optional attributes are missing","message":"Fix YAMLAdapter when optional attributes are missing\n\nSo far, using YAMLAdapter in pools.yaml context, without providing all\nof the pools.yaml attributes altogether, failed because of YAMLAdapter.\n\nAn example could be an end user who wouldn\u0027t like to deploy catalog\nzones in its pools and therefore would drop that attribute from the\npools.yaml file, like in\ndesignate/tests/resources/pools_yaml/multiple-pools.yaml.\n\nRunning any designate-manage pool command on a file without all of the\npools.yaml attributes will fail with the following error message:\n\ndesignate.exceptions.AdapterNotFound:\nAdapter for None to format YAML not found\n\nThis patch fixes it.\n\nCloses-Bug: #2064720\nChange-Id: I214cfc05a834d97a61d0236ab223470631bfc405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4229f25a19456cc9107bc3740946373503717837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4229f25a19456cc9107bc3740946373503717837"}]},"branch":"refs/heads/master"},"9ae6441627eac177d92971f29c7b741fb0824ad2":{"kind":"REWORK","_number":2,"created":"2024-05-06 15:03:59.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/31/918031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/918031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/918031/2"}}},"commit":{"parents":[{"commit":"b26b3aff5053ab43bc8c1eede565b0c76a00e53e","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b26b3aff5053ab43bc8c1eede565b0c76a00e53e"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 09:04:46.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-06 15:03:51.000000000","tz":120},"subject":"Fix YAMLAdapter when optional attributes are missing","message":"Fix YAMLAdapter when optional attributes are missing\n\nSo far, using YAMLAdapter in pools.yaml context, without providing all\nof the pools.yaml attributes altogether, failed because of YAMLAdapter.\n\nAn example could be an end user who wouldn\u0027t like to deploy catalog\nzones in its pools and therefore would drop that attribute from the\npools.yaml file, like in\ndesignate/tests/resources/pools_yaml/multiple-pools.yaml.\n\nRunning any designate-manage pool command on a file without all of the\npools.yaml attributes will fail with the following error message:\n\ndesignate.exceptions.AdapterNotFound:\nAdapter for None to format YAML not found\n\nThis patch fixes it.\n\nCloses-Bug: #2064720\nChange-Id: I214cfc05a834d97a61d0236ab223470631bfc405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ae6441627eac177d92971f29c7b741fb0824ad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ae6441627eac177d92971f29c7b741fb0824ad2"}]},"branch":"refs/heads/master"},"e7d890ed20f7396b1a934ec1f99441f2fe8b5d7e":{"kind":"REWORK","_number":3,"created":"2024-05-08 09:16:55.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/31/918031/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/918031/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/918031/3"}}},"commit":{"parents":[{"commit":"b26b3aff5053ab43bc8c1eede565b0c76a00e53e","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b26b3aff5053ab43bc8c1eede565b0c76a00e53e"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 09:04:46.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-08 09:16:48.000000000","tz":120},"subject":"Fix YAMLAdapter when optional attributes are missing","message":"Fix YAMLAdapter when optional attributes are missing\n\nSo far, using YAMLAdapter in pools.yaml context, without providing all\nof the pools.yaml attributes altogether, failed because of YAMLAdapter.\n\nAn example could be an end user who wouldn\u0027t like to deploy catalog\nzones in its pools and therefore would drop that attribute from the\npools.yaml file, like in\ndesignate/tests/resources/pools_yaml/multiple-pools.yaml.\n\nRunning any designate-manage pool command on a file without all of the\npools.yaml attributes will fail with the following error message:\n\ndesignate.exceptions.AdapterNotFound:\nAdapter for None to format YAML not found\n\nThis patch fixes it.\n\nCloses-Bug: #2064720\nChange-Id: I214cfc05a834d97a61d0236ab223470631bfc405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7d890ed20f7396b1a934ec1f99441f2fe8b5d7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7d890ed20f7396b1a934ec1f99441f2fe8b5d7e"}]},"branch":"refs/heads/master"},"baa88827091b8c77ad766ccadb13fc2dd073e666":{"kind":"REWORK","_number":4,"created":"2024-05-09 12:21:11.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/31/918031/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/918031/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/918031/4"}}},"commit":{"parents":[{"commit":"b26b3aff5053ab43bc8c1eede565b0c76a00e53e","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b26b3aff5053ab43bc8c1eede565b0c76a00e53e"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 09:04:46.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-09 12:20:55.000000000","tz":120},"subject":"Fix YAMLAdapter when optional attributes are missing","message":"Fix YAMLAdapter when optional attributes are missing\n\nSo far, using YAMLAdapter in pools.yaml context, without providing all\nof the pools.yaml attributes altogether, failed because of YAMLAdapter.\n\nAn example could be an end user who wouldn\u0027t like to deploy catalog\nzones in its pools and therefore would drop that attribute from the\npools.yaml file, like in\ndesignate/tests/resources/pools_yaml/multiple-pools.yaml.\n\nRunning any designate-manage pool command on a file without all of the\npools.yaml attributes will fail with the following error message:\n\ndesignate.exceptions.AdapterNotFound:\nAdapter for None to format YAML not found\n\nThis patch fixes it.\n\nCloses-Bug: #2064720\nChange-Id: I214cfc05a834d97a61d0236ab223470631bfc405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/baa88827091b8c77ad766ccadb13fc2dd073e666"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/baa88827091b8c77ad766ccadb13fc2dd073e666"}]},"branch":"refs/heads/master"},"ad561565fdae85e386a70a60b033693444f39bbd":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-20 17:35:48.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/31/918031/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/918031/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/918031/5"}}},"commit":{"parents":[{"commit":"05ebd151228e56ec2754f352334612b585d2d042","subject":"Merge \"Simplified sink service and improved coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05ebd151228e56ec2754f352334612b585d2d042"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 09:04:46.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-20 17:35:17.000000000","tz":120},"subject":"Fix YAMLAdapter when optional attributes are missing","message":"Fix YAMLAdapter when optional attributes are missing\n\nSo far, using YAMLAdapter in pools.yaml context, without providing all\nof the pools.yaml attributes altogether, failed because of YAMLAdapter.\n\nAn example could be an end user who wouldn\u0027t like to deploy catalog\nzones in its pools and therefore would drop that attribute from the\npools.yaml file, like in\ndesignate/tests/resources/pools_yaml/multiple-pools.yaml.\n\nRunning any designate-manage pool command on a file without all of the\npools.yaml attributes will fail with the following error message:\n\ndesignate.exceptions.AdapterNotFound:\nAdapter for None to format YAML not found\n\nThis patch fixes it.\n\nCloses-Bug: #2064720\nChange-Id: I214cfc05a834d97a61d0236ab223470631bfc405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad561565fdae85e386a70a60b033693444f39bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad561565fdae85e386a70a60b033693444f39bbd"}]},"branch":"refs/heads/master"},"6dd9aaf13b693c831aabbfac5406757b8ea96fc6":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-05-21 08:34:02.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/31/918031/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/918031/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/918031/6"}}},"commit":{"parents":[{"commit":"15ecc90f0501f587ae08e484dbd5a11b435b1919","subject":"Increase minimum test coverage to 98%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15ecc90f0501f587ae08e484dbd5a11b435b1919"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 09:04:46.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-21 08:14:47.000000000","tz":120},"subject":"Fix YAMLAdapter when optional attributes are missing","message":"Fix YAMLAdapter when optional attributes are missing\n\nSo far, using YAMLAdapter in pools.yaml context, without providing all\nof the pools.yaml attributes altogether, failed because of YAMLAdapter.\n\nAn example could be an end user who wouldn\u0027t like to deploy catalog\nzones in its pools and therefore would drop that attribute from the\npools.yaml file, like in\ndesignate/tests/resources/pools_yaml/multiple-pools.yaml.\n\nRunning any designate-manage pool command on a file without all of the\npools.yaml attributes will fail with the following error message:\n\ndesignate.exceptions.AdapterNotFound:\nAdapter for None to format YAML not found\n\nThis patch fixes it.\n\nCloses-Bug: #2064720\nChange-Id: I214cfc05a834d97a61d0236ab223470631bfc405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6dd9aaf13b693c831aabbfac5406757b8ea96fc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6dd9aaf13b693c831aabbfac5406757b8ea96fc6"}]},"branch":"refs/heads/master"},"6cc0a7d8876291f874edaac2ff86b7d7a84a738a":{"kind":"REWORK","_number":7,"created":"2024-05-23 14:06:24.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/31/918031/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/918031/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/918031/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/918031/7"}}},"commit":{"parents":[{"commit":"f9de1e836a3bfcd508a8d309f2f11790f56ce9aa","subject":"Merge \"Fix unit tests depending on /etc/resolv.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f9de1e836a3bfcd508a8d309f2f11790f56ce9aa"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 09:04:46.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-23 14:06:17.000000000","tz":120},"subject":"Fix YAMLAdapter when optional attributes are missing","message":"Fix YAMLAdapter when optional attributes are missing\n\nSo far, using YAMLAdapter in pools.yaml context, without providing all\nof the pools.yaml attributes altogether, failed because of YAMLAdapter.\n\nAn example could be an end user who wouldn\u0027t like to deploy catalog\nzones in its pools and therefore would drop that attribute from the\npools.yaml file, like in\ndesignate/tests/resources/pools_yaml/multiple-pools.yaml.\n\nRunning any designate-manage pool command on a file without all of the\npools.yaml attributes will fail with the following error message:\n\ndesignate.exceptions.AdapterNotFound:\nAdapter for None to format YAML not found\n\nThis patch fixes it.\n\nCloses-Bug: #2064720\nChange-Id: I214cfc05a834d97a61d0236ab223470631bfc405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6cc0a7d8876291f874edaac2ff86b7d7a84a738a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6cc0a7d8876291f874edaac2ff86b7d7a84a738a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
