)]}'
{"id":"openstack%2Fdesignate~918044","triplet_id":"openstack%2Fdesignate~master~I4a38c24147eb396a91577e662d3016698fd8289d","project":"openstack/designate","branch":"master","topic":"admin_guide_multi_pools","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-01-28 17:22:09.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-01-17 03:57:30.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"31933":{"account":{"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},"last_update":"2025-01-28 17:22:09.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2025-01-28 17:22:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4a38c24147eb396a91577e662d3016698fd8289d","subject":"Add pools.yaml attributes docs section","status":"MERGED","created":"2024-05-03 12:22:26.000000000","updated":"2025-01-28 17:23:12.000000000","submitted":"2025-01-28 17:22:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":41,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918044-admin_guide_multi_pools","meta_rev_id":"c5a8933524ad270e3545de30f23c8f31beb5940f","_number":918044,"virtual_id_number":918044,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-28 17:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-01-28 16:59:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},{"value":2,"date":"2025-01-28 16:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2025-01-28 16:59:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-03 12:30:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-08 10:01:57.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-05-29 14:05:32.000000000","updated_by":{"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},"reviewer":{"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},"state":"REVIEWER"},{"updated":"2025-01-27 11:29:00.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"e386dd6bebce0875637b422e53cd87a9be03417a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-03 12:22:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c87094f2f3d74e54427e33ff0500919b0c2dd75d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 12:30:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f048eb938353495a95b690ab1befc846\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40b277d151f742aeadc2f099f06c0be1 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/598bfa8246f54f9685eea50e1d3814da : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2d624c93b0f3490e483ccf2aabf8a99000b641e8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-28 19:46:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"95bd26e4bedffd4ca38207a9e6e125cc10a5900a","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-29 13:30:13.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5d24e0fb29a404c17985e4515aa2a5763d040d2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-29 13:30:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8fa564a2a81d469c1f334efe8d47a224a7b5cb8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 13:41:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62a8d9bec201413dbcf2d84a25e22598\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/482025a55cdf4f98b97424bae411833f : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/254e5893e00e481abfd62149342fe6e0 : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":2},{"id":"df22e1cd70a47df696e77b1d6efec4eff155fee6","author":{"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},"date":"2024-05-29 14:05:32.000000000","message":"Patch Set 2: Code-Review+1\n\n(11 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d92fb11c2d05735db7363038f26eaa0659478013","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-03 09:00:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"30a8bfdf7d99ceef9056b2d6a2f81698afcfd832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-04 14:33:52.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"484f899580d500fa3eac6d4943cfb5aea74d7dfb","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-04 14:35:32.000000000","message":"Patch Set 3:\n\n(11 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b74d812d4f4f5ac60fdc228f136ba9e4a3937f57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 14:42:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c061032689a42b39c7db345dfaffdfc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a817e57e15c4e2da13ab7f2ff32da39 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87bce2e21b83445dac37ce8dc8f1920d : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":3},{"id":"e706c7b576894b81f4d821ab49b73fc20bc53e38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-30 11:32:04.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a1376697f1c7617f87a639029bba423dcc3e563a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-30 11:41:36.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"b2a2aa3dad7920b5652f3f2aa1fb8d9271d956d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-30 11:47:45.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"edc1394ffd7a5e61a828d599ea22e3cb80e1b42d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 11:57:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/543548d6496a4010b799fc4ccb61e1b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45a65f8618a2471ab2142b878e406712 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd8e9e4df4d4fa18a62bc0af6410505 : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":6},{"id":"5b1c44f9db1d6e1248f8157d82c35766ee5a00e4","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-12-16 13:40:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a6704090d0f17f09ec82642cfc6e5f271f625f50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 13:48:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4407dfa43ea941ba86a7794158424250\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76070fcd49f04b73b892c42eb7c5812e : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8992d65c4004e9795736fd3c13a1624 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":6},{"id":"14c8c2690bcf940138730cbb43bee43867a3f253","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-12-23 18:52:08.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"7d55fe153c46d85704248f8197c8dbb5b17b5386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 18:59:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/516c88847d354142bffd7b873696ab53\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a0bc31d37094194a4a716684acda55a : FAILURE in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b855ec37dd2497584d3151622bd52cf : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":7},{"id":"816f4cb490fa73ff1ca62c48a2e4b780e3622204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-12-24 14:21:34.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"d8b931bd757a3e2c6974b76234dd3520fad404c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 14:28:44.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/873a3d9a104f4ce2a48c99fb649429af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a38866ca4264fd58258a93540610526 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ece94b14a0745f0ab5bd6c047d2b9d3 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":8},{"id":"ca86b0b25fd292806125d5a7ab3c6003ea20a705","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-12-24 14:49:30.000000000","message":"Uploaded patch set 9: New patch set was added with same tree, parent tree, and commit message as Patch Set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"37b1c43f5d7f8808f2759372766178cdc715abe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 15:02:44.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00f6c6a324754beb85fa80252d174667\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d3ac587e5f149caa3d8a479c8e498a7 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cca39c1845504cf1a1d58a6204eb8d52 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":9},{"id":"6addc65efdc7510cb424b0546b840cf3c8a041a8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-14 02:26:52.000000000","message":"Patch Set 9: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"0875747271a2028daff49312277d82bb4aa92656","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-01-14 14:18:13.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"0da968a1ef2f80a837c483d04fd873db5c4110b4","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-01-14 14:18:46.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"a0d6c430fe98c37ad9fd8d9e3b1d62a751f0b4a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 14:32:54.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99d6ae8c106a46d9b163ce2f96d652cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0829a6b219364005ac8c6d7ec64a3fb8 : FAILURE in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e87e55da6a234aff81deea2446a1b139 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":10},{"id":"4bed6eec9b38e15f5e16149c879ae086008016cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-01-14 16:25:22.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"93a2d1395f87a75b82ecb8b93913d052cf6902f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 16:48:48.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e9de7d524e44132bff24d7cd9c40ad6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51a0bd7ef43d49d78f7fb6c1548be019 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b48345c92de422986d21a930a3ee707 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":11},{"id":"5680bc40582bf859cd1c2d87c1f2d5c5bfbc33ca","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-17 03:57:30.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5fc76aa784388cc28de854e99f986ab40037c184","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-01-28 11:16:46.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"d1da1f5148b0d4eda46fe0733f758f46cd5bd2ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-01-28 16:58:34.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":12},{"id":"71ea1b3f59ba124ebcc38ac281e3510b2924a007","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-28 16:59:48.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"0ee0a2628ec91e65363fc457db697ceb1d8d8d51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 17:05:07.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e7b5fc2281a43c3acfd15253ec906b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f118ef0843654e9fbbb59a4f672c0b1a : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e702ebbdd2874ab89b656ac46cf27d03 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":12},{"id":"d5a8631c065b64df0920131f28ffbf2aa471fe7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 17:05:26.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"8b87b1ef1d0f79415092dd5a777818f4deec6f69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 17:22:09.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18849f5f12df4900b0b12cbd4d4125ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ac4904042fd4aa49bea8e21d5ec0df9 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3095f4ba712748f79cefef31c098e0d2 : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":12},{"id":"26be32b006e8ed6455891a185ec1e91cc27a3034","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 17:22:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"c5a8933524ad270e3545de30f23c8f31beb5940f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 17:23:12.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c4fdeaaf7da4a579770193d9170f0b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fe2c3a982fc457ab4065a70efa770a5 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"7ac4889a04f3d068932f16c5bb637df0466eaf76","revisions":{"7eb472b5fb23624dd763f6e7e79da88fad055a49":{"kind":"REWORK","_number":1,"created":"2024-05-03 12:22:26.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/44/918044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/918044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/918044/1"}}},"commit":{"parents":[{"commit":"83b229ee08693daaa5e912198706f94970595e0d","subject":"Admin and User Guide updates for multi-pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/83b229ee08693daaa5e912198706f94970595e0d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 12:17:10.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 12:17:10.000000000","tz":120},"subject":"Add pools.yaml attributes docs section","message":"Add pools.yaml attributes docs section\n\nThis patch adds a pools.yaml file attributes docs section.\n\nThe attributes section should extend the comments that appear on the\npools.yaml draft.\n\nCloses-Bug: #2063842\nChange-Id: I4a38c24147eb396a91577e662d3016698fd8289d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7eb472b5fb23624dd763f6e7e79da88fad055a49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7eb472b5fb23624dd763f6e7e79da88fad055a49"}]},"branch":"refs/heads/master"},"59d2ebcca21fb2d9eb086a1df24a6459a9b86df9":{"kind":"REWORK","_number":2,"created":"2024-05-29 13:30:22.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/44/918044/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/918044/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/918044/2"}}},"commit":{"parents":[{"commit":"83b229ee08693daaa5e912198706f94970595e0d","subject":"Admin and User Guide updates for multi-pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/83b229ee08693daaa5e912198706f94970595e0d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 12:17:10.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-29 13:29:02.000000000","tz":120},"subject":"Add pools.yaml attributes docs section","message":"Add pools.yaml attributes docs section\n\nThis patch adds a pools.yaml file attributes docs section.\n\nThe attributes section should extend the comments that appear on the\npools.yaml draft.\n\nCloses-Bug: #2063842\nChange-Id: I4a38c24147eb396a91577e662d3016698fd8289d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d2ebcca21fb2d9eb086a1df24a6459a9b86df9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59d2ebcca21fb2d9eb086a1df24a6459a9b86df9"}]},"branch":"refs/heads/master"},"5b1554a0c2e94ef770b9aa5eaad05a206b18cc3d":{"kind":"REWORK","_number":3,"created":"2024-06-04 14:33:52.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/44/918044/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/918044/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/918044/3"}}},"commit":{"parents":[{"commit":"83b229ee08693daaa5e912198706f94970595e0d","subject":"Admin and User Guide updates for multi-pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/83b229ee08693daaa5e912198706f94970595e0d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 12:17:10.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-06-04 14:33:34.000000000","tz":120},"subject":"Add pools.yaml attributes docs section","message":"Add pools.yaml attributes docs section\n\nThis patch adds a pools.yaml file attributes docs section.\n\nThe attributes section should extend the comments that appear on the\npools.yaml draft.\n\nCloses-Bug: #2063842\nChange-Id: I4a38c24147eb396a91577e662d3016698fd8289d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5b1554a0c2e94ef770b9aa5eaad05a206b18cc3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5b1554a0c2e94ef770b9aa5eaad05a206b18cc3d"}]},"branch":"refs/heads/master"},"aab26a0ad435befad4b28bbea65604b66331da9c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-09-30 11:32:04.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/44/918044/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/918044/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/918044/4"}}},"commit":{"parents":[{"commit":"0bb890d51b3d5eb91fa2a1510c30eb3e829fc066","subject":"Admin and User Guide updates for multi-pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0bb890d51b3d5eb91fa2a1510c30eb3e829fc066"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 12:17:10.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-09-30 11:31:47.000000000","tz":0},"subject":"Add pools.yaml attributes docs section","message":"Add pools.yaml attributes docs section\n\nThis patch adds a pools.yaml file attributes docs section.\n\nThe attributes section should extend the comments that appear on the\npools.yaml draft.\n\nCloses-Bug: #2063842\nChange-Id: I4a38c24147eb396a91577e662d3016698fd8289d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aab26a0ad435befad4b28bbea65604b66331da9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aab26a0ad435befad4b28bbea65604b66331da9c"}]},"branch":"refs/heads/master"},"7b2386767f15562b67b26e344ed309d2982e6c3f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-09-30 11:41:36.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/44/918044/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/918044/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/918044/5"}}},"commit":{"parents":[{"commit":"287d1754b8cf9e9dedc43405ba9caa522cce9a5f","subject":"Admin and User Guide updates for multi-pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/287d1754b8cf9e9dedc43405ba9caa522cce9a5f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 12:17:10.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-09-30 11:41:31.000000000","tz":0},"subject":"Add pools.yaml attributes docs section","message":"Add pools.yaml attributes docs section\n\nThis patch adds a pools.yaml file attributes docs section.\n\nThe attributes section should extend the comments that appear on the\npools.yaml draft.\n\nCloses-Bug: #2063842\nChange-Id: I4a38c24147eb396a91577e662d3016698fd8289d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7b2386767f15562b67b26e344ed309d2982e6c3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7b2386767f15562b67b26e344ed309d2982e6c3f"}]},"branch":"refs/heads/master"},"b3b016f84a70486b8c507f6aaf2a3be6e756f25c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-09-30 11:47:45.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/44/918044/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/918044/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/918044/6"}}},"commit":{"parents":[{"commit":"fcc6f8641d89e601f0e00272a8457366d13518a7","subject":"Admin and User Guide updates for multi-pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fcc6f8641d89e601f0e00272a8457366d13518a7"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 12:17:10.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-09-30 11:47:41.000000000","tz":0},"subject":"Add pools.yaml attributes docs section","message":"Add pools.yaml attributes docs section\n\nThis patch adds a pools.yaml file attributes docs section.\n\nThe attributes section should extend the comments that appear on the\npools.yaml draft.\n\nCloses-Bug: #2063842\nChange-Id: I4a38c24147eb396a91577e662d3016698fd8289d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b3b016f84a70486b8c507f6aaf2a3be6e756f25c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b3b016f84a70486b8c507f6aaf2a3be6e756f25c"}]},"branch":"refs/heads/master"},"a093f15727e8a63529bba128db840823fdb76a65":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-12-23 18:52:08.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/44/918044/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/918044/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/918044/7"}}},"commit":{"parents":[{"commit":"7d6a3baf6ba1c6fdf55c357ee06d931bb837011f","subject":"Add docs guide updates for multi-pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7d6a3baf6ba1c6fdf55c357ee06d931bb837011f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 12:17:10.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-12-23 18:52:08.000000000","tz":0},"subject":"Add pools.yaml attributes docs section","message":"Add pools.yaml attributes docs section\n\nThis patch adds a pools.yaml file attributes docs section.\n\nThe attributes section should extend the comments that appear on the\npools.yaml draft.\n\nCloses-Bug: #2063842\nChange-Id: I4a38c24147eb396a91577e662d3016698fd8289d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a093f15727e8a63529bba128db840823fdb76a65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a093f15727e8a63529bba128db840823fdb76a65"}]},"branch":"refs/heads/master","description":"Rebase"},"0401decf4fb1d18c3da2f338465887f27ef3c9cb":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-12-24 14:21:34.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/44/918044/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/918044/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/918044/8"}}},"commit":{"parents":[{"commit":"0c986f6b74f5d9da812068fc29d7edcdc97c321c","subject":"Add docs guide updates for multi-pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0c986f6b74f5d9da812068fc29d7edcdc97c321c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 12:17:10.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-12-24 14:20:52.000000000","tz":0},"subject":"Add pools.yaml attributes docs section","message":"Add pools.yaml attributes docs section\n\nThis patch adds a pools.yaml file attributes docs section.\n\nThe attributes section should extend the comments that appear on the\npools.yaml draft.\n\nCloses-Bug: #2063842\nChange-Id: I4a38c24147eb396a91577e662d3016698fd8289d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0401decf4fb1d18c3da2f338465887f27ef3c9cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0401decf4fb1d18c3da2f338465887f27ef3c9cb"}]},"branch":"refs/heads/master"},"7b538859e35a07b204d2ca370491b62ff22af6f6":{"kind":"NO_CHANGE","_number":9,"created":"2024-12-24 14:49:30.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/44/918044/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/918044/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/918044/9"}}},"commit":{"parents":[{"commit":"13eee59a6589a198a3baba871cf596117444149f","subject":"Add docs guide updates for multi-pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/13eee59a6589a198a3baba871cf596117444149f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 12:17:10.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-12-24 14:49:22.000000000","tz":0},"subject":"Add pools.yaml attributes docs section","message":"Add pools.yaml attributes docs section\n\nThis patch adds a pools.yaml file attributes docs section.\n\nThe attributes section should extend the comments that appear on the\npools.yaml draft.\n\nCloses-Bug: #2063842\nChange-Id: I4a38c24147eb396a91577e662d3016698fd8289d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7b538859e35a07b204d2ca370491b62ff22af6f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7b538859e35a07b204d2ca370491b62ff22af6f6"}]},"branch":"refs/heads/master"},"dab1bbbc96136194402fe6cbfb34b8c27e5bb908":{"kind":"REWORK","_number":10,"created":"2025-01-14 14:18:13.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/44/918044/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/918044/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/918044/10"}}},"commit":{"parents":[{"commit":"c82a7fed2e6fd860f01f96b776d734ac0094aac2","subject":"Add docs guide updates for multi-pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c82a7fed2e6fd860f01f96b776d734ac0094aac2"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 12:17:10.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-01-14 14:18:02.000000000","tz":0},"subject":"Add pools.yaml attributes docs section","message":"Add pools.yaml attributes docs section\n\nThis patch adds a pools.yaml file attributes docs section.\n\nThe attributes section should extend the comments that appear on the\npools.yaml draft.\n\nCloses-Bug: #2063842\nChange-Id: I4a38c24147eb396a91577e662d3016698fd8289d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dab1bbbc96136194402fe6cbfb34b8c27e5bb908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dab1bbbc96136194402fe6cbfb34b8c27e5bb908"}]},"branch":"refs/heads/master"},"304b6e1bf009f1680c1ef37f67aee6bb10762fe4":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-01-14 16:25:22.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/44/918044/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/918044/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/918044/11"}}},"commit":{"parents":[{"commit":"97bbbf6510fae5bddb5d1be1306a91ca12eeb98a","subject":"Add docs guide updates for multi-pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/97bbbf6510fae5bddb5d1be1306a91ca12eeb98a"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 12:17:10.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-01-14 16:24:52.000000000","tz":0},"subject":"Add pools.yaml attributes docs section","message":"Add pools.yaml attributes docs section\n\nThis patch adds a pools.yaml file attributes docs section.\n\nThe attributes section should extend the comments that appear on the\npools.yaml draft.\n\nCloses-Bug: #2063842\nChange-Id: I4a38c24147eb396a91577e662d3016698fd8289d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/304b6e1bf009f1680c1ef37f67aee6bb10762fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/304b6e1bf009f1680c1ef37f67aee6bb10762fe4"}]},"branch":"refs/heads/master"},"7ac4889a04f3d068932f16c5bb637df0466eaf76":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-01-28 16:58:34.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/44/918044/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/918044/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/918044/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/918044/12"}}},"commit":{"parents":[{"commit":"4d29f0b90b1d0d2f06f1c4e7df10389140671d3f","subject":"Merge \"Add docs guide updates for multi-pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d29f0b90b1d0d2f06f1c4e7df10389140671d3f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 12:17:10.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-01-28 16:58:34.000000000","tz":0},"subject":"Add pools.yaml attributes docs section","message":"Add pools.yaml attributes docs section\n\nThis patch adds a pools.yaml file attributes docs section.\n\nThe attributes section should extend the comments that appear on the\npools.yaml draft.\n\nCloses-Bug: #2063842\nChange-Id: I4a38c24147eb396a91577e662d3016698fd8289d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ac4889a04f3d068932f16c5bb637df0466eaf76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ac4889a04f3d068932f16c5bb637df0466eaf76"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
