)]}'
{"id":"openstack%2Fdesignate~919429","triplet_id":"openstack%2Fdesignate~master~I5e2e18176a1d3cad06dfedd1972f659b6d73ad95","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-06-03 09:17:07.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-06-29 00:02:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5e2e18176a1d3cad06dfedd1972f659b6d73ad95","subject":"Improve invalidate uuid tests","status":"MERGED","created":"2024-05-18 20:06:04.000000000","updated":"2024-06-29 00:03:13.000000000","submitted":"2024-06-29 00:02:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919429","meta_rev_id":"94cc82bc49cd0586c0b9c24237c26035f4ef6979","_number":919429,"virtual_id_number":919429,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-29 00:02:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2024-06-03 09:17:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-28 22:07:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-28 22:07:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-18 21:24:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-27 11:51:46.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2024-06-03 09:17:07.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-06-28 22:07:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"67a836f6b371a6c2866860febb340013950fb66c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-18 20:06:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fa55aadc40305496798a28e04ec8ced461c86ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-18 20:07:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ae960187552302e82a34d3f699047ca8e1b16c1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 21:24:56.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb80e527f1c340ea8f779552305df44b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/90130d44af4646b7b9374408fc22e3eb : SUCCESS in 45m 59s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f03a60255b084e9db9d361c4d7f5bfce : SUCCESS in 54m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5dba1d4f2e846c189d20f8875b1c45c : SUCCESS in 32m 05s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/46a071d8e95d4915baeb5cdbdaecd6b9 : SUCCESS in 30m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/49a4e93977d144bd9387d195cc2a42bb : SUCCESS in 45m 14s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f6786e1eefa743d4b44e979aec464ff0 : SUCCESS in 47m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c4bd6571bff5475f8a6f406d96de9c6b : SUCCESS in 1h 06m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dde3d9812bd141a2854375ba0c9912e9 : SUCCESS in 46m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0af02163bd124ad5b3066ba8c65d863e : SUCCESS in 47m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c86c3a549a984ec9ad30d2b0439125e1 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f18b7645c6749e8a7871c267b407809 : FAILURE in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59f56a8bcdc944b1957add8b73a4b5ad : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b32f364131541f293f8a02f4206dd29 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4dac898a08e549db957c7e5462c5bbdb : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0af273b385124b28a79410039a2c7938 : SUCCESS in 10m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a324e2e0e09449de9fdd2d05756960f0 : SUCCESS in 31m 41s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fab8be1192394902baca42a180204213 : SUCCESS in 5m 14s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/71c44958761d4e80ab27bd3eb535ff14 : SUCCESS in 38m 32s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/26eb14a6693041a3873337e4206ead10 : SUCCESS in 4m 13s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/d3cf12f743204042b9c750b209d4298e : SUCCESS in 38m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8016ad8affe7e3bd69828f0d43d8955ce79b5817","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-19 10:00:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5092d4aa7be74db4d06d1432a1f9ac1979fa6a00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-19 10:02:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"045e94839eaaa52b8bf3ebca80a6ee3bc9070226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-19 10:44:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"83774be49d6dce6af21683d517ae43c705d7f77c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-19 11:48:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/619ed7ee6dd34067ab75034313114477\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/08d2f047a4a949be975f9b043bf6eea7 : SUCCESS in 32m 58s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/914e96f74617417aabcd10e30f7c9124 : SUCCESS in 41m 50s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8796e5de9a0541e4bf06f273ae18f9a0 : SUCCESS in 35m 35s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/eb4b895d676547a5a8d6012253ecc848 : SUCCESS in 48m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0b62c4c8a6e34914be8469db80cd28c0 : SUCCESS in 45m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cee5177c6692430186462ee82ff4f2c7 : SUCCESS in 57m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/becc58f2d0a04f149f2c892f3767b735 : SUCCESS in 58m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fdb3020dddfe46de8f6b294fe2b8161a : SUCCESS in 35m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2c80bf604e4745a1b808fab8ff55e70a : SUCCESS in 41m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/902bd7275eb14567b47f69c46b52d209 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea0b9b6d82624804a8f1c7ddb848dd6f : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9993877e41b47138ad4909888063898 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f4c502877c0643bf9a08c48beefa01ac : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a45de050ce24ebc9c815bb40da81ae2 : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c7557adf55e42ceb840512352f2c06d : SUCCESS in 7m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/374b719cc69642cea503e9f7c75ef0c6 : SUCCESS in 34m 07s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ab67bf4fa831434899acd63b0058da86 : SUCCESS in 5m 46s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6d515201f44d4ae6899bf440b7e3f536 : SUCCESS in 34m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4cc3011df723400b8c2accf3fd4e7461 : SUCCESS in 5m 20s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ecbd5136a8fd46af8952924fcca41237 : SUCCESS in 44m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6fb5dd04ea2540961a52ed87b251f083eaa85124","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-27 11:51:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"047f2a75449b477ac6073ed53d5946c73bd586e3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-27 11:53:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4f40b6ecd7ff2d838729d15867aa01344846265d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-28 18:04:32.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"3ad3d40f0ea0bd84f517ab4a8e06a69ead0bf8de","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-28 18:05:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2e40c6c2e2f84d6f20a1e520d51c41c4f950cb59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 19:28:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47f062c4f39a44a08d5a579f92849ce2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/abef323fef8643aea29326599bb46772 : SUCCESS in 34m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f7a52c86f1534297bc57c29a6ebb1a3a : SUCCESS in 48m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dff375dded34181a96e0eff0a9f2fc9 : SUCCESS in 48m 02s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0aeb93ed7cee4faea47a858a479e3bf9 : SUCCESS in 37m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/922e9cc78af144dd925d0c85226d5320 : SUCCESS in 38m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8dd0abda612842fc99f0cce3e5e3cb23 : SUCCESS in 1h 09m 40s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/872a37a8d58d4e1a8b7add93f9a3c7f0 : SUCCESS in 1h 12m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b6db30902ff04bd68ba9d56a2878adca : SUCCESS in 40m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/55bfac7e0dc64922bedb9aa3235109fd : SUCCESS in 48m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f643434aa4f445869bc9ddf068821b97 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a731696581ee42df867c408561e620a7 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97fbf637233042c3a938e447db217616 : SUCCESS in 6m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6f0488be5934817a0fa54102aa340f9 : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3cb1dc7f79b74dffbda77b8d89a33ca1 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b5fc221a3f643c1b19d69d42c97a963 : SUCCESS in 8m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/10202bc03e07470883e7d56696b342c3 : SUCCESS in 45m 47s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9ee56f3f499748708451813bd04791aa : SUCCESS in 6m 21s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/70ab2b9685654a3e9794ca783e2aab7d : SUCCESS in 44m 45s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ffd7d727fcc24e85826a2149f0225958 : SUCCESS in 5m 46s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/4789bfff7548451e8ecbddd1bde82bd5 : FAILURE in 42m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"02134f8a85bb3e0aa6176efe02588dda65e00994","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-03 09:17:07.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"15eebffee68967a7557d0203c99134601ab94d81","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-28 22:07:54.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8615267604d2dce3cf72590fadaa15347d36680f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 22:08:39.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"97f690b37b3330ed33cf5f89e22126927538b4bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-29 00:02:06.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a733831662634457933c64066a7b7f6b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/746314fbc83d4e4fb63b4150cd1abd5b : SUCCESS in 38m 50s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4fa4b4b7be4b45cf85b16dcc7aab1c0b : SUCCESS in 47m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6acc59c23c394e9e969b232dc5f710f0 : SUCCESS in 46m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e235dc81077f49e388c91cf38f94a52c : SUCCESS in 1h 18m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0b8087ea821444249c53e4572214237b : SUCCESS in 43m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e9495016fcd04a95ac4b1b9a904ab80a : SUCCESS in 42m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49359ca103a640aab227847c13cf90d8 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/317971fd70c347dda71df8703ea0cf88 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95505669626543a7a16ab6e1eeff76f9 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2f8b36ada7c489eb875baeed931f47e : SUCCESS in 6m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/26beb476bd954fc0b605f866389e61ce : SUCCESS in 42m 27s","accounts_in_message":[],"_revision_number":6},{"id":"5febfbccf424ad30cd18ccfd016361c7125fdf0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-29 00:02:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"94cc82bc49cd0586c0b9c24237c26035f4ef6979","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-29 00:03:13.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f5f1aadde414d94b2d5f0d8811ec18c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c776a195e1cd4aec89acc5a2f4b7910f : SUCCESS in 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"485920286331a424574554a71d6ba9bfd0371e71","revisions":{"ef3e38d35fc7930b48f01350ec5bb80b5019cb2d":{"kind":"REWORK","_number":1,"created":"2024-05-18 20:06:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/29/919429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/919429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/919429/1"}}},"commit":{"parents":[{"commit":"b65d7c134185e14e6791363249dfe363afe4fc55","subject":"[producer] Do not hard stop periodic task during exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b65d7c134185e14e6791363249dfe363afe4fc55"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-05-18 20:05:04.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-05-18 20:05:04.000000000","tz":120},"subject":"Improve invalidate uuid tests","message":"Improve invalidate uuid tests\n\nChange-Id: I5e2e18176a1d3cad06dfedd1972f659b6d73ad95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef3e38d35fc7930b48f01350ec5bb80b5019cb2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef3e38d35fc7930b48f01350ec5bb80b5019cb2d"}]},"branch":"refs/heads/master"},"df8417f6b66541360e1e22be3de1cb0d453ec9d2":{"kind":"REWORK","_number":2,"created":"2024-05-18 20:07:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/29/919429/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/919429/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/919429/2"}}},"commit":{"parents":[{"commit":"b65d7c134185e14e6791363249dfe363afe4fc55","subject":"[producer] Do not hard stop periodic task during exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b65d7c134185e14e6791363249dfe363afe4fc55"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-05-18 20:05:04.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-05-18 20:06:57.000000000","tz":120},"subject":"Improve invalidate uuid tests","message":"Improve invalidate uuid tests\n\nChange-Id: I5e2e18176a1d3cad06dfedd1972f659b6d73ad95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/df8417f6b66541360e1e22be3de1cb0d453ec9d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/df8417f6b66541360e1e22be3de1cb0d453ec9d2"}]},"branch":"refs/heads/master"},"a5e47ffcf1eed5b37b8a046397a8c2d354a07bbb":{"kind":"REWORK","_number":3,"created":"2024-05-19 10:00:08.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/29/919429/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/919429/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/919429/3"}}},"commit":{"parents":[{"commit":"b65d7c134185e14e6791363249dfe363afe4fc55","subject":"[producer] Do not hard stop periodic task during exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b65d7c134185e14e6791363249dfe363afe4fc55"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-05-18 20:05:04.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-05-19 09:59:49.000000000","tz":120},"subject":"Improve invalidate uuid tests","message":"Improve invalidate uuid tests\n\nChange-Id: I5e2e18176a1d3cad06dfedd1972f659b6d73ad95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a5e47ffcf1eed5b37b8a046397a8c2d354a07bbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a5e47ffcf1eed5b37b8a046397a8c2d354a07bbb"}]},"branch":"refs/heads/master"},"badb8014e02f2bf9eb86304a6cd748b88813cab8":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2024-05-19 10:02:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/29/919429/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/919429/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/919429/4"}}},"commit":{"parents":[{"commit":"1dbdcac58fdb602c9a3617d4b22ba16690e869cf","subject":"Updated Infoblox backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1dbdcac58fdb602c9a3617d4b22ba16690e869cf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-05-18 20:05:04.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-05-19 10:01:03.000000000","tz":120},"subject":"Improve invalidate uuid tests","message":"Improve invalidate uuid tests\n\n- Removed the internal is_uuid_like function and\n  replaced with oslo implementation directly instead.\n\nChange-Id: I5e2e18176a1d3cad06dfedd1972f659b6d73ad95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/badb8014e02f2bf9eb86304a6cd748b88813cab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/badb8014e02f2bf9eb86304a6cd748b88813cab8"}]},"branch":"refs/heads/master"},"cfcb8b9ef09fc0aea5b2b8fcd02f6e8c115edd5b":{"kind":"REWORK","_number":5,"created":"2024-05-19 10:44:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/29/919429/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/919429/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/919429/5"}}},"commit":{"parents":[{"commit":"1dbdcac58fdb602c9a3617d4b22ba16690e869cf","subject":"Updated Infoblox backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1dbdcac58fdb602c9a3617d4b22ba16690e869cf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-05-18 20:05:04.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-05-19 10:44:07.000000000","tz":120},"subject":"Improve invalidate uuid tests","message":"Improve invalidate uuid tests\n\n- Removed the internal is_uuid_like function and\n  replaced with oslo implementation directly instead.\n\nChange-Id: I5e2e18176a1d3cad06dfedd1972f659b6d73ad95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfcb8b9ef09fc0aea5b2b8fcd02f6e8c115edd5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfcb8b9ef09fc0aea5b2b8fcd02f6e8c115edd5b"}]},"branch":"refs/heads/master"},"485920286331a424574554a71d6ba9bfd0371e71":{"kind":"REWORK","_number":6,"created":"2024-05-28 18:04:32.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/29/919429/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/919429/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/919429/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/919429/6"}}},"commit":{"parents":[{"commit":"1dbdcac58fdb602c9a3617d4b22ba16690e869cf","subject":"Updated Infoblox backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1dbdcac58fdb602c9a3617d4b22ba16690e869cf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-05-18 20:05:04.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-05-28 18:04:04.000000000","tz":120},"subject":"Improve invalidate uuid tests","message":"Improve invalidate uuid tests\n\n- Removed the internal is_uuid_like function and\n  replaced with oslo implementation directly instead.\n\nChange-Id: I5e2e18176a1d3cad06dfedd1972f659b6d73ad95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/485920286331a424574554a71d6ba9bfd0371e71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/485920286331a424574554a71d6ba9bfd0371e71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
