)]}'
{"id":"openstack%2Fdesignate~920545","triplet_id":"openstack%2Fdesignate~stable%2F2024.1~I8fe16bdb2d8776c8c6bc5923d18ffd6a2adeb3ee","project":"openstack/designate","branch":"stable/2024.1","topic":"slurp-job","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-12-18 13:31:15.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I8fe16bdb2d8776c8c6bc5923d18ffd6a2adeb3ee","subject":"stable/2024.1 only: Add skip level upgrade job","status":"ABANDONED","created":"2024-05-27 14:32:39.000000000","updated":"2024-12-18 13:31:19.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"129bb6c2e6e070d429eb5ef28bbb6af0d089da80","_number":920545,"virtual_id_number":920545,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-27 16:04:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2024-06-03 09:23:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-27 16:04:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-03 09:23:14.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"77b053785cb538a44047c3cda8df997953dc280a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-27 14:32:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3779d6672377a4a1252d9465768814ec8135ab5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-27 14:33:01.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e4a40aaa8c370b79cbef33cb9c99b4bfa79858d6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-27 14:33:08.000000000","message":"Topic set to slurp-job","accounts_in_message":[],"_revision_number":2},{"id":"baa55f665eab7c34259fbc5038ecdc0d6422b733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 16:04:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/418428177c284925be295a50fd0d405f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a55982033ff04fe5a0f1198600885ac8 : SUCCESS in 40m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9a25f4c4a58944e5a42cff7e7597050f : SUCCESS in 1h 03m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7e0e701c2844c8fad2d576fde961302 : SUCCESS in 37m 03s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ebff0f765cee4d09b3021e10af5af9c5 : SUCCESS in 49m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8ce4ae898e994d85b78f3f756496ad10 : SUCCESS in 37m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/13277aad34f4459a8a8e6a463f2f86e9 : SUCCESS in 1h 08m 01s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d5fa0e4088da4913a416a486c622c055 : SUCCESS in 1h 07m 42s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f81dd846750e4359b5bcb6900a079863 : SUCCESS in 1h 08m 15s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3a4d445b265f45b3acda4fe9acea29dd : SUCCESS in 1h 20m 00s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cf0a1e2827364571a8913c7f9d6e302d : SUCCESS in 40m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f40857053fda4b889a9c9a0f342169f0 : SUCCESS in 47m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b63a218e2bd49e0bbdfd221728fba2e : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/777b2e5817ed443cb1d89481fde0065e : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2af4056d3ad748869e6277639b53ec3d : SUCCESS in 7m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bf63861ecd74ddeb25d4907060ccad5 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ade8fc0222d4932b145088993a243a1 : SUCCESS in 8m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/57b1916a684149f8aaf1fc0dae7e8166 : SUCCESS in 44m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/eaaac83c96bd4ff7a3abb4eedc4c1c86 : SUCCESS in 4m 54s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/b0942798b8f54827bd94fea235121c01 : SUCCESS in 34m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ea247b3928cecb731aa9b46b57d98e17c33ced4c","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-03 09:23:14.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"54c045b7c77f715b47dccc50f8278308392efaca","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-18 13:31:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"129bb6c2e6e070d429eb5ef28bbb6af0d089da80","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-18 13:31:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aea8bebed6f8926d7b12a91c7695828b6ce36235","revisions":{"19712bec6a1d9b5b12b16990b2e70b4a81b3aae3":{"kind":"REWORK","_number":1,"created":"2024-05-27 14:32:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/920545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/920545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/920545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/920545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/920545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/920545/1"}}},"commit":{"parents":[{"commit":"72777e203ec6f09757ddcd90011013eeafab36c1","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72777e203ec6f09757ddcd90011013eeafab36c1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-05 14:00:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-27 14:32:15.000000000","tz":540},"subject":"istable/2024.1 only: Add skip level upgrade job","message":"istable/2024.1 only: Add skip level upgrade job\n\nThe 2024.1 release will be the first SLURP release which supports\ndirect upgrade from the 2023.1 release.\n\nChange-Id: I8fe16bdb2d8776c8c6bc5923d18ffd6a2adeb3ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19712bec6a1d9b5b12b16990b2e70b4a81b3aae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19712bec6a1d9b5b12b16990b2e70b4a81b3aae3"}]},"branch":"refs/heads/stable/2024.1"},"aea8bebed6f8926d7b12a91c7695828b6ce36235":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-05-27 14:33:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/920545/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/920545/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/920545/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/920545/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/920545/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/920545/2"}}},"commit":{"parents":[{"commit":"72777e203ec6f09757ddcd90011013eeafab36c1","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72777e203ec6f09757ddcd90011013eeafab36c1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-05 14:00:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-27 14:32:45.000000000","tz":540},"subject":"stable/2024.1 only: Add skip level upgrade job","message":"stable/2024.1 only: Add skip level upgrade job\n\nThe 2024.1 release will be the first SLURP release which supports\ndirect upgrade from the 2023.1 release.\n\nChange-Id: I8fe16bdb2d8776c8c6bc5923d18ffd6a2adeb3ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aea8bebed6f8926d7b12a91c7695828b6ce36235"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aea8bebed6f8926d7b12a91c7695828b6ce36235"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
