)]}'
{"id":"openstack%2Fdesignate~920740","triplet_id":"openstack%2Fdesignate~stable%2F2024.1~I214cfc05a834d97a61d0236ab223470631bfc405","project":"openstack/designate","branch":"stable/2024.1","topic":"fix_yaml_adpt_missing_attrs","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-10-30 09:10:37.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-10-29 19:26:40.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"32366":{"account":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"last_update":"2024-10-29 09:45:24.000000000","reason":"\u003cGERRIT_ACCOUNT_32366\u003e replied on the change","reason_account":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-05-29 14:37:16.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I214cfc05a834d97a61d0236ab223470631bfc405","subject":"Fix YAMLAdapter when optional attributes are missing","status":"MERGED","created":"2024-05-29 13:08:23.000000000","updated":"2024-10-30 11:37:29.000000000","submitted":"2024-10-30 11:36:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920740-fix_yaml_adpt_missing_attrs","meta_rev_id":"02be9c4e793fc5581298b033741c9f360bf8316e","_number":920740,"virtual_id_number":920740,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-30 11:36:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-10-29 19:26:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-29 17:10:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2024-10-29 19:26:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},{"_account_id":36350,"name":"David","display_name":"David Hitze","email":"ubuntuone@hitze.cologne","username":"dhitze","status":"inovex GmbH"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-29 13:14:50.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2024-05-29 13:14:50.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-05-29 14:24:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-29 09:45:09.000000000","updated_by":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"reviewer":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"state":"REVIEWER"},{"updated":"2024-10-29 09:45:24.000000000","updated_by":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"reviewer":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"state":"CC"},{"updated":"2024-10-29 09:47:36.000000000","updated_by":{"_account_id":36350,"name":"David","display_name":"David Hitze","email":"ubuntuone@hitze.cologne","username":"dhitze","status":"inovex GmbH"},"reviewer":{"_account_id":36350,"name":"David","display_name":"David Hitze","email":"ubuntuone@hitze.cologne","username":"dhitze","status":"inovex GmbH"},"state":"CC"},{"updated":"2024-10-29 09:56:39.000000000","updated_by":{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},"reviewer":{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},"state":"CC"}],"messages":[{"id":"6e5fa4928458d63d5180f6a631d32cf3e4d83ae4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-29 13:08:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd5ccca96d07672e9876a31c362ec900611858fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 14:24:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cdf816292704e20a01846629cad60f8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ecf7607fd38b4cd2b67c45eed7a794b1 : SUCCESS in 47m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9935a40c2b9d444d8d2a45645b2847e3 : SUCCESS in 1h 09m 48s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3853c700b144079bd9ae1278cf44641 : SUCCESS in 52m 42s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/596e045c36514caeb79e158eff3d05fe : SUCCESS in 49m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/67e683ea79f3407d9747c655727167fa : SUCCESS in 48m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5b9f9cc817724dfbbad611c904fe83a3 : SUCCESS in 1h 06m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c80072a4094b4f5abc8596c19dc71f52 : SUCCESS in 55m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b381382a290c44b1a71cb130b03ac0ec : SUCCESS in 36m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/55e68ecda9424aca981522dda5dd2083 : SUCCESS in 37m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e6cc46ddc534d4e804b96aa02f640bd : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2856e58cc7d84b7e8b231b8e9cb74dc2 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/616f3e81a16e460db61474701c89740b : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1ddf1c597754790be18dbbec35075ec : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8dbee3d671c499db7b48d37db657535 : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26c88e1a49364d808f004158f5cfeb5e : SUCCESS in 4m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b8a2c7bc26304f8d9a2e7056c9312bdb : SUCCESS in 46m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c463497993d94b8db21093f8540b1ccb : SUCCESS in 4m 38s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/681862c6769e4edc9fd61878dbdbc749 : SUCCESS in 39m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf99f13819f66369eff7ac23d10abea81d46bbf6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-29 14:37:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f1fc62e24504888a031ec1c31467a63f4dbc11dd","author":{"_account_id":36350,"name":"David","display_name":"David Hitze","email":"ubuntuone@hitze.cologne","username":"dhitze","status":"inovex GmbH"},"date":"2024-10-29 09:47:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1213b0d5a747f24e662120464b7708b7cea078e3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-10-29 15:34:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a390129cea3fb50eff0c2a5f6e9f565b1e750270","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 15:34:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcc5aea142faac013f613a571493f139a9378681","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 15:53:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f806f297d71841b497ed19181512951a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/eebafb697dde498ba28511e3987e5862 : CANCELED\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/63b06a241c95471aaec684448e214247 : CANCELED\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/65e64f92bdb34526935d48f3b5be3048 : CANCELED\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6938e333a2e1463289a355055205a0dc : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/869c5b97f1174e51a23b77cab9bf5de2 : CANCELED\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2193538367ec46799f944a7953bcd598 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea64a6a93dda42309887062fa99a009f : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5a7095c9f46489d8a25936b68bbb695 : SUCCESS in 8m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c28938336e5d4584aaa8d5683f8b6373 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c6acad29ead448a98b34123b18c026a : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ff3870811aa465daa4f5f75453cd80e : SUCCESS in 4m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c3eb00f3512b4f76b872b1fbf2b114f0 : FAILURE in 17m 59s","accounts_in_message":[],"_revision_number":1},{"id":"39afd8c4aaa17b0f700030f66e0408716b3b457d","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-29 16:21:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da4da5a0e5f2daa04307f05d2ec114f14d4c2ac1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-29 17:10:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e0238b7af860440117af623407b434614da8d5c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 18:34:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0467d5d5b3c49539a4ca76da67b82dd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/08c208bd5ce342a083e7b81cf6101255 : SUCCESS in 39m 41s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b200edef5ee249768cf8dc88d00ac8bf : SUCCESS in 47m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/826a27acf50a4a5f8de51c56bdc67dce : SUCCESS in 43m 08s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/aeaa05cf3e954b85adb4906a0fcad5df : SUCCESS in 40m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ccf90cb43ecb42abaf101c86cce118d4 : SUCCESS in 52m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f0d4c3d29b724688b2a67b5a50838a0f : SUCCESS in 1h 17m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/881d1f7f54c94563babb5b04d7092042 : SUCCESS in 1h 03m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/427cd84224f042a8b811cf4d708454e3 : SUCCESS in 33m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b7f5af2d0e354e06bb9adc22b5f33256 : SUCCESS in 39m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/108325b57f6f419da9d2d8fd0cae6ada : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6be9f52e2e9448c7bf0341055fac9618 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df64626f6aff44109282d635f314a50e : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4dcea740617e4d81b8c3c70c2a6a3b07 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2978edff1b54a5f9ae47c43a9b2475e : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07e23d37233a409dacd0bdbc44bad772 : SUCCESS in 4m 26s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/d8b85ca3122a4d5a9b289fcf97d4a8fe : SUCCESS in 31m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/35b6e37996a94ccb821656333986bb33 : SUCCESS in 4m 25s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/6cbe7c3c25a4469baa63a9e01b87c8a8 : SUCCESS in 29m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d79ccdb30eb9a9e1cf58eb6bbd34acd49b874076","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-10-29 19:26:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"10306a43e2697542b4eda052726a11f1a5acccb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 19:27:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ed314b279df5117d2ad0d70aa3f78ef901b4bbc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 19:35:19.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/933028 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"7b4f9d26679f57622962a76886c09420e5c86d6f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-30 09:10:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e5651677a99fdc17c01f30403227bc7a60aae9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 10:17:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbaf5d82d14e43619f680677b9935ae7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1b5ecaafebdb42bf8b9205d2a52bb850 : SUCCESS in 40m 12s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9f4a126ee8df44379a198f394af53100 : SUCCESS in 43m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/80132b24ec144b5abc386cfbef319e04 : SUCCESS in 56m 43s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6afe913c59e045eeb3c8c47cd8c2aa08 : SUCCESS in 30m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5ae4dfe06a304918ac016f9f207828f1 : SUCCESS in 38m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/915ccef5e65c4088b4789f9948b259b4 : SUCCESS in 57m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b7af2f95014c4224982182d20917cb0e : SUCCESS in 59m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8e60e58c146a4a5eaf8d3b0bf3bd67c5 : SUCCESS in 51m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cad088acaf4e40a1b055d9195bef3a07 : SUCCESS in 44m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/315ecba6c0c74fcda8eff42d8fb106c0 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc20a737b2cd44aaaee690caf209246e : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a46c54482124e9992c5958a9400f051 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11dc4c6d22d6488c96f103a3cb85c1f0 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b3bd95bd5ba4472adf8b8ac6043df58 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0310ea7dbea4761ae43f3fe066e2845 : SUCCESS in 3m 48s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/ce5c31723d9e4174b7d8d496f07c557e : SUCCESS in 28m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7dd2e91345204e57b3c593955e1b6aa8 : SUCCESS in 4m 25s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/bf83ec77660641daaaa4fac283810440 : SUCCESS in 37m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"deb0ba9e76d0e17a2b530087ec5489567ab469a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 10:18:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"435b09b5103d6f151c17f47048e2d4e342a56fee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 11:36:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7c00aeae7a648528d62cf1368584402\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9094b6ee3eec444e95f67fbe7c75d6b8 : SUCCESS in 41m 42s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/33c5cd4801144c8db3d316608fe447d2 : SUCCESS in 1h 06m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1f446a73862f4a66ba4ded6bee7ba9a0 : SUCCESS in 37m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1c97b1d928a6411bb4359e2e29ce403e : SUCCESS in 58m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4e200827a277447bb8beee6f05432131 : SUCCESS in 34m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/49ee65dcc63c48a3a9fa8e46a689af9c : SUCCESS in 35m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b4e85f2323247cdbbd6a9e1634d4c7e : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1be51c02f1f2442db0fdd8678441add8 : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0caeed1ed39d47e89a7fa540e3ec1257 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48cd98dadfb24e828d80055a52d609c2 : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebcc3af1b81e4f448490bdaa4b9df689 : SUCCESS in 3m 20s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/c31a75a2fc0a485983b0805ce654c270 : SUCCESS in 28m 47s","accounts_in_message":[],"_revision_number":2},{"id":"348f1ea452dba1242a122e956de5881fc31fd4c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 11:36:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"02be9c4e793fc5581298b033741c9f360bf8316e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 11:37:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c2a33e2d4494369afa2bf1381fbe96d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e25f58c6eb7b4289840394fc60bb1551 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/142b59f94adb4e63b92b580fc7630903 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7e58bdd5254439e6e3a5ac0198de17afcf873c8b","revisions":{"316f9e4e8c13e175e72e5a26f47e1672005ba9e7":{"kind":"REWORK","_number":1,"created":"2024-05-29 13:08:23.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/40/920740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/40/920740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/920740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/920740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/920740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/40/920740/1"}}},"commit":{"parents":[{"commit":"72777e203ec6f09757ddcd90011013eeafab36c1","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72777e203ec6f09757ddcd90011013eeafab36c1"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 09:04:46.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-29 13:07:47.000000000","tz":120},"subject":"Fix YAMLAdapter when optional attributes are missing","message":"Fix YAMLAdapter when optional attributes are missing\n\nSo far, using YAMLAdapter in pools.yaml context, without providing all\nof the pools.yaml attributes altogether, failed because of YAMLAdapter.\n\nAn example could be an end user who wouldn\u0027t like to deploy catalog\nzones in its pools and therefore would drop that attribute from the\npools.yaml file, like in\ndesignate/tests/resources/pools_yaml/multiple-pools.yaml.\n\nRunning any designate-manage pool command on a file without all of the\npools.yaml attributes will fail with the following error message:\n\ndesignate.exceptions.AdapterNotFound:\nAdapter for None to format YAML not found\n\nThis patch fixes it.\n\nCloses-Bug: #2064720\nChange-Id: I214cfc05a834d97a61d0236ab223470631bfc405\n(cherry picked from commit 6cc0a7d8876291f874edaac2ff86b7d7a84a738a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/316f9e4e8c13e175e72e5a26f47e1672005ba9e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/316f9e4e8c13e175e72e5a26f47e1672005ba9e7"}]},"branch":"refs/heads/stable/2024.1"},"7e58bdd5254439e6e3a5ac0198de17afcf873c8b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-29 17:10:33.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/40/920740/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/40/920740/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/920740/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/920740/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/920740/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/40/920740/2"}}},"commit":{"parents":[{"commit":"50b04e630c0220547f63c257873f788f571974da","subject":"Stable-only: Switch neutron job to stable version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50b04e630c0220547f63c257873f788f571974da"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 09:04:46.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-10-29 17:08:32.000000000","tz":60},"subject":"Fix YAMLAdapter when optional attributes are missing","message":"Fix YAMLAdapter when optional attributes are missing\n\nSo far, using YAMLAdapter in pools.yaml context, without providing all\nof the pools.yaml attributes altogether, failed because of YAMLAdapter.\n\nAn example could be an end user who wouldn\u0027t like to deploy catalog\nzones in its pools and therefore would drop that attribute from the\npools.yaml file, like in\ndesignate/tests/resources/pools_yaml/multiple-pools.yaml.\n\nRunning any designate-manage pool command on a file without all of the\npools.yaml attributes will fail with the following error message:\n\ndesignate.exceptions.AdapterNotFound:\nAdapter for None to format YAML not found\n\nThis patch fixes it.\n\nCloses-Bug: #2064720\nChange-Id: I214cfc05a834d97a61d0236ab223470631bfc405\n(cherry picked from commit 6cc0a7d8876291f874edaac2ff86b7d7a84a738a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7e58bdd5254439e6e3a5ac0198de17afcf873c8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7e58bdd5254439e6e3a5ac0198de17afcf873c8b"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
