)]}'
{"id":"openstack%2Fdesignate~920743","triplet_id":"openstack%2Fdesignate~stable%2F2023.1~I214cfc05a834d97a61d0236ab223470631bfc405","project":"openstack/designate","branch":"stable/2023.1","topic":"fix_yaml_adpt_missing_attrs","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-11-12 17:11:52.000000000","reason":"Change was abandoned"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-11-12 17:11:52.000000000","reason":"Change was abandoned"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-05-29 14:37:08.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I214cfc05a834d97a61d0236ab223470631bfc405","subject":"Fix YAMLAdapter when optional attributes are missing","status":"ABANDONED","created":"2024-05-29 13:13:00.000000000","updated":"2024-11-12 17:11:52.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"994b4c17f0e96c7c99712d07514966cb0c9ab997","_number":920743,"virtual_id_number":920743,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-29 14:11:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2024-05-29 14:39:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-29 13:15:16.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2024-05-29 13:15:16.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-05-29 14:11:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24319f3bdaf9b7fb0cd9d478347e45c973197956","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-29 13:13:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3174a936ca775b0e875fd1022ba5c5b5d63714a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-29 13:13:50.000000000","message":"Topic set to fix_yaml_adpt_missing_attrs","accounts_in_message":[],"_revision_number":1},{"id":"94ea4c7705db76776cdd76748b24b2916bba5141","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 14:11:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ed45a5f9e6d40d09f27ac398121131c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8af1721914bd4f888439e629008be832 : SUCCESS in 45m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c5835abc067b4cf8aa1685d2cae0d07c : SUCCESS in 37m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/31b139095e804a039010219d8c006aa1 : SUCCESS in 51m 21s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2a610bc4fbe746d5b834484a9c786dc3 : FAILURE in 46m 37s (non-voting)\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/45f9a006284b4f80b2fefecc7e57c787 : FAILURE in 32m 29s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3458f9f2ecab41efa539815e1e0687b3 : SUCCESS in 43m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c973372b1acd47ac8e6dd0657ba551c2 : FAILURE in 52m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/544ad6bb275c412883cf5641677adb22 : SUCCESS in 40m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/175216d8a12b475fb09b6fbd27525915 : SUCCESS in 43m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26daafc0d9174914ba2007efd414e291 : FAILURE in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/921ab458e1ba432e8d3ff3225ddb1d83 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e75eca25002450286628a74e535fb85 : FAILURE in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e961e56ea1c644d78ea77924a59c90cf : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc79262fc600418abb1bc9a0827685d1 : SUCCESS in 5m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dc6ad053d222477a84f8ec2f485eab64 : SUCCESS in 41m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0cd674439f730a04f53a6b9d8aafc020e27a45f1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-29 14:37:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e8dc31823716823f5d4206cb378bda5fbfc9bbbb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-29 14:38:44.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"2a69a5e94626acbe7bf5f384cebfac62471f7c54","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-05-29 14:39:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"994b4c17f0e96c7c99712d07514966cb0c9ab997","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-12 17:11:52.000000000","message":"Abandoned\n\nCatalog zones was not in this release. Abandoning this patch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8439e80ecdc36262912b464a39ec13c26646b97d","revisions":{"8439e80ecdc36262912b464a39ec13c26646b97d":{"kind":"REWORK","_number":1,"created":"2024-05-29 13:13:00.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/43/920743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/920743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/920743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/920743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/920743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/920743/1"}}},"commit":{"parents":[{"commit":"ff1986cbb755f0d74a7db98623d6d17816d933a6","subject":"Fix Producer shard range ignoring the last shard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff1986cbb755f0d74a7db98623d6d17816d933a6"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-03 09:04:46.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-29 13:12:32.000000000","tz":120},"subject":"Fix YAMLAdapter when optional attributes are missing","message":"Fix YAMLAdapter when optional attributes are missing\n\nSo far, using YAMLAdapter in pools.yaml context, without providing all\nof the pools.yaml attributes altogether, failed because of YAMLAdapter.\n\nAn example could be an end user who wouldn\u0027t like to deploy catalog\nzones in its pools and therefore would drop that attribute from the\npools.yaml file, like in\ndesignate/tests/resources/pools_yaml/multiple-pools.yaml.\n\nRunning any designate-manage pool command on a file without all of the\npools.yaml attributes will fail with the following error message:\n\ndesignate.exceptions.AdapterNotFound:\nAdapter for None to format YAML not found\n\nThis patch fixes it.\n\nConflicts:\n  designate/tests/resources/pools_yaml/pools.yaml\n\nCloses-Bug: #2064720\nChange-Id: I214cfc05a834d97a61d0236ab223470631bfc405\n(cherry picked from commit 6cc0a7d8876291f874edaac2ff86b7d7a84a738a)\n(cherry picked from commit 316f9e4e8c13e175e72e5a26f47e1672005ba9e7)\n(cherry picked from commit ea7159571fe296a2bb8a620ed236bd9acab60556)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8439e80ecdc36262912b464a39ec13c26646b97d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8439e80ecdc36262912b464a39ec13c26646b97d"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
