)]}'
{"id":"openstack%2Fdesignate~920838","triplet_id":"openstack%2Fdesignate~master~Id9ed9d5b71fbd2bb8225ab376d06c29e62454853","project":"openstack/designate","branch":"master","topic":"add_tempest_plugin_devstack","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-08-28 17:39:05.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-19 19:17:41.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-08-14 15:34:23.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"Id9ed9d5b71fbd2bb8225ab376d06c29e62454853","subject":"Add designate-tempest-plugin under devstack_plugins","status":"MERGED","created":"2024-05-30 15:39:00.000000000","updated":"2024-08-28 17:41:34.000000000","submitted":"2024-08-28 17:39:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920838-add_tempest_plugin_devstack","meta_rev_id":"9d4672c00e64a6beb0a1c2ffd615755e550c0e4b","_number":920838,"virtual_id_number":920838,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-28 17:39:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-19 15:06:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-28 16:25:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-30 17:00:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-01 12:15:31.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-08-19 16:09:22.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"89f87f816b01495549fae40bef19fd1489c35e63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-30 15:39:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5c8da899413e3a016154593a5e58c8fe8ac06e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 17:00:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06dc5b2861644911bca3d89c9a27b923\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fcf03c9053af4b94a1fa8ae45932bcfb : SUCCESS in 50m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/70ccdcffb5a24c94bd2cc44729098446 : SUCCESS in 1h 04m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec222d3c318445a4a425bd85af6fba98 : SUCCESS in 48m 48s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1bc3f9affaf947b197607d4f299d7603 : SUCCESS in 48m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ace7d4d09fc945dc97a66e05c14fad8e : SUCCESS in 53m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1eaa8ef31d46496895b60b1e636da6b3 : SUCCESS in 1h 01m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9e6d48e3fc8a4594ac390a9472065625 : SUCCESS in 1h 04m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6c3e92d706c549219cc2dfedb2c6af41 : SUCCESS in 45m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/29da467d5a8747a4a6f74f1934c66e15 : SUCCESS in 50m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94e017c0341944308f578f4e9fa24bb6 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54b42ece22e44fac8b21ebbd63b15759 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4ab7cb4e121410b880033c581eb7710 : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf964244868e4e30a006880ff39e6ff5 : SUCCESS in 5m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b5197926d80469098eacb8adabb6f29 : SUCCESS in 9m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cf25f7537bc4237b73289c608d0723c : SUCCESS in 7m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3e6515bf77664881acbd8b36f7ffa056 : SUCCESS in 33m 29s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/297ca3d783c049cbb49cb3f65e47dea6 : SUCCESS in 7m 31s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b0cf143bccb9497c80e0d07ff86df9db : SUCCESS in 42m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a25b4043e9c74aca974783cb941f6e26 : SUCCESS in 4m 37s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/634fc9e2860a49fdb23163f55f730b35 : SUCCESS in 41m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b522966339d10e9c3e0635a53e5e3f11fcd61378","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-07-01 12:14:46.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"80d4f6b37fa35bca0363e6c092efcb86bb196fad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 13:22:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ac7fd5626a24112a729c79b40d77f08\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/88476374d2e44e76b52393e9ce02285b : SUCCESS in 45m 44s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e952d07b624a4115bdadba12557d313f : SUCCESS in 47m 06s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a28b975bfaa54a53bd4243309684756a : SUCCESS in 59m 16s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/85faaa18ff3f43f88ab84b9ff9e83fc5 : SUCCESS in 51m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/33055a30ed4c43ad9b64e7ac7fc719df : SUCCESS in 47m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/14c109745273491c9709a6f898811daf : SUCCESS in 1h 04m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/eb7c9c4cd6fc472085f2ef173937dcc7 : FAILURE in 7m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4938d782f7ca4b7fa537e5f303593837 : SUCCESS in 48m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/423a41a49f3e4c8e8e965d320004af0e : SUCCESS in 50m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/243389b3ba3d4b4fadc703c31ed88c76 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a6048de5286432b8fb8604ba796b85e : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edc328309b164061976ea21676272f0a : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63a3f2f185ea4ad4bbb31192d7ca3bb9 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d30c06eb702349c7bfe76c2f79ce989f : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde7cb635e674e308f36c951fa103fc9 : SUCCESS in 6m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/881feb7dd69c43a185f901cc4bcf235d : SUCCESS in 29m 38s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c6e3f7b7ee084c728efeec03bd006555 : SUCCESS in 4m 53s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/35f4f85fd63d45bd930ba006af507834 : SUCCESS in 43m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9bb04979e249898f6b921946586e50e260910362","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-07-01 14:37:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"62296dc1fd54911e14b04ed1e361017c3bc78e71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 15:57:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a40c07f3117460f90888dde8a6326a0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4cf781ec01fc40f2b7ba3d91a44feae9 : SUCCESS in 46m 54s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b2358ea6ce024624aa1ef7696b6c773e : SUCCESS in 1h 06m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/36612231a3924a499a132c1e8c43efab : SUCCESS in 58m 06s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/08eddb6dca274858810c68da4e3d5c95 : SUCCESS in 51m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0af3d05132d7438da0be683c7152eab2 : SUCCESS in 47m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/09cfb3384a074a668ad15512d1836f41 : SUCCESS in 1h 12m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9da005d215fd4f0191e2b1fe087d6096 : SUCCESS in 1h 12m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f61ee33709564669b8d6c8ccfdf48886 : SUCCESS in 32m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3c0ec2316b224f96bf268a0844d9706b : SUCCESS in 32m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01e4fac900a84e0fb3ab350f27f06b4f : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f29d172df85c4fdcbbd5659450da00b7 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce8dff2a03fc4895bf0bcbc8a59a4f89 : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76beb372f46f4954ac3525fbcb8a1160 : SUCCESS in 5m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d05f61efb184ce9897adc85d38e93f0 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0286ee8bafb6451f997891b8c2ca2f76 : SUCCESS in 6m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7f61d0a14d414823bfcc439a72a5b1a2 : SUCCESS in 30m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1078a6a65d54446eab1bcc5a36739462 : SUCCESS in 4m 38s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/78528197158b43549baa7ae8e548c14f : SUCCESS in 44m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34c1ddef4906e1a276ce2788fdae61d58fb8eb99","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-14 15:34:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df0facc0481db47ba72b16f5a5a7b596973aa63e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-19 15:06:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b483ef8ed26ebc982f2c7cd5a3629c85d0da87a9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-28 16:25:41.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7979cdfb36f880566387679cd62d3be05ea7bc73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 16:26:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"89a63fcdc645bcf314c60bef06c8c22064de74f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 17:39:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8a47f7f571d4abeb3b0563a4c60583a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c027c182f6e04dc79d60230abf079d1a : SUCCESS in 47m 28s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/65aa92b73735473ab728cf19419f50f2 : SUCCESS in 46m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4637e0aa41cb47b59de4675d62d8bbf2 : SUCCESS in 50m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8c3aa9af3525422eb16b1fc53a0a6c65 : SUCCESS in 1h 08m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e6610c2cc749493b95a9134a5c9b445b : SUCCESS in 44m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b4f1a77376c6470eafa4ab159f11b794 : SUCCESS in 46m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76a498562a004f6b93581c1e3a13a7c9 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/323815e9fe144bddba4d440b8de22556 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd0e2f3ccabe4b20859fc0d9c7b9781a : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3f9da998634519bff3efb2a50cea13 : SUCCESS in 7m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5cad830712634ae8a94a80d0ca0931af : SUCCESS in 31m 45s","accounts_in_message":[],"_revision_number":2},{"id":"dbd5e9a8e3fd8d16b59ae6f560694f37e204dc91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 17:39:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9d4672c00e64a6beb0a1c2ffd615755e550c0e4b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 17:41:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f68d4f53e0574b9380962ca300b1b04a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26c9334bfb8444d7a77b3dbf34447a20 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"706db9be728ceae0f4f3f670da029e3bcea892fb","revisions":{"dee0fac35de1f799a3525b602feefe257b808f65":{"kind":"REWORK","_number":1,"created":"2024-05-30 15:39:00.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/38/920838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/920838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/920838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/920838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/920838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/920838/1"}}},"commit":{"parents":[{"commit":"73d83a9369d92f94feff3d864af958830f68ed49","subject":"Merge \"Fix multipool sample file errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/73d83a9369d92f94feff3d864af958830f68ed49"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-30 15:37:14.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-30 15:37:14.000000000","tz":120},"subject":"Add designate-tempest-plugin under devstack_plugins","message":"Add designate-tempest-plugin under devstack_plugins\n\nSo far we did not have access to specific tempest setting entries that\nare created by devstack.\n\nThis patch will allow it.\n\nChange-Id: Id9ed9d5b71fbd2bb8225ab376d06c29e62454853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dee0fac35de1f799a3525b602feefe257b808f65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dee0fac35de1f799a3525b602feefe257b808f65"}]},"branch":"refs/heads/master"},"706db9be728ceae0f4f3f670da029e3bcea892fb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-01 12:14:46.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/38/920838/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/920838/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/920838/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/920838/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/920838/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/920838/2"}}},"commit":{"parents":[{"commit":"9048699344e461e4bf89439880f52435d4434d57","subject":"Merge \"Fix bashate errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9048699344e461e4bf89439880f52435d4434d57"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-05-30 15:37:14.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-07-01 12:07:40.000000000","tz":120},"subject":"Add designate-tempest-plugin under devstack_plugins","message":"Add designate-tempest-plugin under devstack_plugins\n\nSo far we did not have access to specific tempest setting entries that\nare created by devstack.\n\nThis patch will allow it.\n\nChange-Id: Id9ed9d5b71fbd2bb8225ab376d06c29e62454853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/706db9be728ceae0f4f3f670da029e3bcea892fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/706db9be728ceae0f4f3f670da029e3bcea892fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
