)]}'
{"id":"openstack%2Fdesignate~922598","triplet_id":"openstack%2Fdesignate~master~I5fb5497f9a6b83b8347b3bfc0bdf7a192e410082","project":"openstack/designate","branch":"master","topic":"remove-TestSkipped","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-07-04 01:23:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5fb5497f9a6b83b8347b3bfc0bdf7a192e410082","subject":"Replace use of testtools.testcase.TestSkipped","status":"MERGED","created":"2024-06-24 09:56:20.000000000","updated":"2024-07-04 01:24:32.000000000","submitted":"2024-07-04 01:23:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922598-remove-TestSkipped","meta_rev_id":"ffa7ff2a66cbb84ade3c7ab1f4a265a87d10ca3c","_number":922598,"virtual_id_number":922598,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 01:23:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-06-24 10:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-06-24 15:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-06-24 15:39:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-06-24 10:41:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-24 10:40:37.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-06-24 10:41:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-06-24 11:21:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-24 15:39:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"93716543673636b43b2bc8fa44a623ca30538d77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-06-24 09:56:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"154d0d4f5126b1137bef1cfb704e255ccc5c7269","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-24 10:41:26.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3dd7afb8d17798b57bcde95d569e26a7ac785936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 11:21:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b8edc8bca924ab88b3d6e45b3b06019\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dae6252cd73c46c19853810098b8ec65 : SUCCESS in 48m 28s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7a45e5cc9a5249ccbf029b3d1c09a7d1 : SUCCESS in 1h 03m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac5d328201a241e49aa746f82064c7c1 : SUCCESS in 57m 46s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9ea392f0177b4054a0d1474b5562a4fa : SUCCESS in 43m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4b066e8756fb4402a8ce4bf25e1b8e67 : SUCCESS in 46m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/63e5c4f1ac81417cbad96a008be71d01 : SUCCESS in 56m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/31b1d59d454b41298fd53ff9cc35d82f : SUCCESS in 1h 16m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6c48de305a604e8ba477aabd71f05b9d : SUCCESS in 54m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8b57a94076df49c0a31f359def63620a : SUCCESS in 52m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c74167c913f84bfc82b5bbbd1a9ae54b : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c032412f519149ff9299f22094226985 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/789410ef2b4844f7b97aa0caf0162b51 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b3803a2b2b643e39e79694e5bcd84e4 : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e77aa1ed215e448f85ac5bb882e1c251 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/014c19ae2a88421cab74cdefb3fb27d6 : SUCCESS in 8m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dd70e249d3e44b06bd9ab6d2a74b9026 : SUCCESS in 30m 08s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6686a97926824107b1ce73f5036caf4b : SUCCESS in 7m 06s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1f0c2a9f605e4d8ca1eec03fef836d52 : SUCCESS in 48m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/20960a617d0d45f2ad9beb19b01f0275 : SUCCESS in 5m 01s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/eec1d115f5a44d2fba748b9d001b91cd : FAILURE in 47m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"347d6903f7070fead6b7c718b3aa0441943e1067","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-24 15:39:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ccecc7e7e811a279650305a472e4ad9a892e7d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 15:39:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f209e44afb833d56fc9d9573a3b97b1451401b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 16:52:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5578ad8a96074fdea41f10e960f3ac14\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6a4ea33bda014810b4235f6257601b5a : SUCCESS in 53m 15s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/16e4b3fa80ec4f2193b1f94898336be5 : SUCCESS in 39m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/85a33d979aea45b18907a5c3b6ea23a8 : SUCCESS in 52m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/56421ddc35c4473fa888fe6336f93ee9 : FAILURE in 1h 08m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e90f184d4d6a46a6b5478488dc81bc3c : SUCCESS in 47m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/baa025e7d65c4a5a920ea2f10a3e1fdd : SUCCESS in 52m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eca235c2be84def990a62151caf512c : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d994c0cfbd94f2399842ff8d50c2529 : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8014c32c9046478d9c4c5cd50306b6f7 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36abc868282e416a922430f478b8127a : SUCCESS in 9m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a702dcf987224be08719a1b488e815d6 : SUCCESS in 33m 33s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/411492a15ce74748ad89e292f56ea8f3 : SUCCESS in 8m 41s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f44c6083f8764b2eafdb8904921b01fb : SUCCESS in 42m 52s","accounts_in_message":[],"_revision_number":1},{"id":"c288920faadbebde0e12024e647974fde4af6242","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-24 20:50:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"365238a48ca36d7c7d7551579024c3ae047663d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 22:00:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a463d9487bb471e8ecaa1633f406e29\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/53967430b0b74fc985d98330a9402450 : SUCCESS in 48m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/fcd65899641a4c8eb9f16c4f69ea183e : SUCCESS in 1h 07m 15s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/cdf74f833a3d4095af1ae4bd0871e793 : SUCCESS in 40m 48s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8697caf7aef94e44af6bc96ee4a4ae71 : SUCCESS in 43m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/70cc82506af74adfbdf7acd14b5111df : SUCCESS in 52m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f2edf9ebee6347d4aee66126c403257f : SUCCESS in 55m 46s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2db26a77350b43ffa474ed0a321b5b7d : SUCCESS in 1h 02m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d350e5f8c7ff49a189bbe8cbee6a0eae : FAILURE in 52m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1b4178e5566b4b68b5d91c8fea38eebc : SUCCESS in 36m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/615912ef9a1d449c9baf10d0ec64b787 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce80996c10bc4f9bbc9d15cc9b76d99c : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d8879f70be940198273a5c985d69c53 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30480e08ab3240a1a1c7d283a2dce0dc : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64d5ae9fe4c74f34a53f5bd914bf9959 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/504c0358c52b42db8da93d6689a67d94 : SUCCESS in 6m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/18f3c6afc3dd4193a4ec4790b8756cf7 : SUCCESS in 33m 40s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8801d9f4784248e68e062b105ae2873c : SUCCESS in 6m 08s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/36babcf4f2124a918d1e319c1feb9852 : SUCCESS in 1h 02m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/491f0688971d408e8019b39ad2fc8362 : SUCCESS in 4m 37s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/5fa994008277459395b7fb0ea517e36f : FAILURE in 48m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1600cad2de58d9c883299229b26603905fdf3fde","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-07-03 22:47:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93a71003b1ae2007fc66d7fc715ef93949d5bd25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 00:05:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d948e7fdf3140d8992cb7ef92870e79\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b33730eb6927408faa0bfd4e9b924689 : SUCCESS in 51m 46s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/13d9d021fabd4f94888f736ab7e6246d : SUCCESS in 56m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a4f5846f56c49dc80146ff0d339c04f : SUCCESS in 46m 17s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4276e9aaf77c4cc995917f5bf0af1282 : SUCCESS in 51m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8375f8edb09240a49f076a1238b56cc0 : SUCCESS in 50m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/121d1f7bf7794cf1b8a57fb234f70674 : SUCCESS in 1h 06m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/956c66d040b04118997b5c97093a5f83 : SUCCESS in 1h 12m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fbe6c84f3c2848a8a8293776b214413d : SUCCESS in 45m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/60491ab03df24f3e8fe346fa0bb46d2e : SUCCESS in 36m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/744d3888114a4cc4af3e05d8b988e3ce : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d4c60550ceb43239cf76bb07d6bb8ff : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/835e300ed9fa4c2bbe62262cebf6baff : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22a1b9ae49ea4dbe91d031337fa7cb73 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/86ff4753e9d7412180543e6e169f74c7 : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e52e3e69c0f3473aa4fdf8cf73b998ae : SUCCESS in 7m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d84fdaa47ef43329e1d8a6dca4e6be9 : SUCCESS in 34m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/efff79afadca4f569f045b0cb263a2d8 : SUCCESS in 4m 21s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/1de0bfe553974c57b4980dd7401d8d38 : SUCCESS in 35m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c173685e3e8656597b686acc5b688d34a26e9c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 00:06:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a5c531448fa02b8b7ad0285ffeb58890f3fb66e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 01:23:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f045f9d03a68431b9b2fb53dd626f761\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/034259515f69495884032e025d0e7054 : SUCCESS in 49m 55s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2b400beada5246d78d985f84f9473d9c : SUCCESS in 47m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f119af4797614a61b4a21868c546dbc3 : SUCCESS in 51m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d887d424e9704a00ada40ee8f605d58a : SUCCESS in 1h 10m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1121da720eef43809c30ecbf3f63a6b6 : SUCCESS in 1h 00m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1946b14befb74fc99dff81dfc820fee9 : SUCCESS in 51m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06d99fd4b5084cea8905a3726ce40ac8 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/202e87bb8e6e4432b0ad174438935118 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d475ec10fb442948900ff136ed0df10 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40a80e01dfe648db8b107223c2512f22 : SUCCESS in 7m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0f8cc3522adb417bada671824a5d9beb : SUCCESS in 36m 50s","accounts_in_message":[],"_revision_number":1},{"id":"792ebf1f16e52cebff1a780411eafa77db9b1dde","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 01:23:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ffa7ff2a66cbb84ade3c7ab1f4a265a87d10ca3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 01:24:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/084dacf62e6545ccb421e1e1aba42a25\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4b843308b1143f2b3d4b27975ca755e : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b468cc2b9b981034b59e343d0b2de7aa293bdbab","revisions":{"b468cc2b9b981034b59e343d0b2de7aa293bdbab":{"kind":"REWORK","_number":1,"created":"2024-06-24 09:56:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/98/922598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/922598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/922598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/922598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/922598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/922598/1"}}},"commit":{"parents":[{"commit":"097ffc6df181290eba1bcd7c492b1b505bc15434","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/097ffc6df181290eba1bcd7c492b1b505bc15434"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-24 09:56:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-24 09:56:00.000000000","tz":60},"subject":"Replace use of testtools.testcase.TestSkipped","message":"Replace use of testtools.testcase.TestSkipped\n\nThis has been deprecated for removal [1]. Use the stdlib variant\ninstead.\n\n[1] https://github.com/testing-cabal/testtools/commit/59b890db3c\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I5fb5497f9a6b83b8347b3bfc0bdf7a192e410082\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b468cc2b9b981034b59e343d0b2de7aa293bdbab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b468cc2b9b981034b59e343d0b2de7aa293bdbab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
