)]}'
{"id":"openstack%2Fdesignate~923668","triplet_id":"openstack%2Fdesignate~stable%2F2024.1~I35e03b9818851487686153bc68ad90c081e61966","project":"openstack/designate","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"36670":{"account":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"last_update":"2024-11-21 16:37:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I35e03b9818851487686153bc68ad90c081e61966","subject":"Updated Infoblox backend","status":"ABANDONED","created":"2024-07-08 16:50:51.000000000","updated":"2024-11-21 16:37:24.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":904978,"cherry_pick_of_patch_set":28,"meta_rev_id":"f42d4b7b383494c147df5ee9e2626702cd168274","_number":923668,"virtual_id_number":923668,"owner":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-08 18:14:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-07-08 16:56:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-08 16:50:51.000000000","updated_by":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2024-07-08 16:56:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-07-08 18:14:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c32c32ffdfc814dc60bd6f46bc559093eb913cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"date":"2024-07-08 16:50:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6dec996dbafc258d0d38d683380426205043b81c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-08 16:56:07.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb167986b8aa2e86ec58b95e3396dad1fa3e82d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 18:14:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa8db4e2e5064ad29f0a6234ee7b9be6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eb3c626107964dd6ab95bcc389bfd550 : SUCCESS in 2m 01s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d00cf37879834000ab46f3d5f24a80a9 : SUCCESS in 38m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/79f16cd3237a4fd4aa2f8099f822f8d1 : SUCCESS in 56m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/556ce69ad967408fa3c8c1a5b2867f6e : SUCCESS in 56m 05s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4e6768447cac46519ed699becb566c11 : SUCCESS in 48m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a3c2b869b6a843e39abc71e11c51d059 : SUCCESS in 38m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b27115557b33412e9f1a3544cf35712e : SUCCESS in 1h 05m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e35b2b9902ac4efaa8b78168f8b6b849 : SUCCESS in 1h 17m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e3c5a379dc7a4325af9f3514feeaa2f5 : SUCCESS in 39m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/406203f33c7f431fa6d862f00767bcb6 : SUCCESS in 39m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0b99a180a764e38b60c1f81da15c242 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68d9a503fa0449088c8be5bc73cd088a : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a43bcdd41514ed288b98fdcec3a9d0e : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50d0d7e0e72d43129149a1085833abaa : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd4e728cf1104263ae6fc3f099ac63d7 : SUCCESS in 8m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6ca830de321347e48b3498603aa586d7 : SUCCESS in 2m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1aabd87b88ae4a26b1d7bdfc22a6b9a3 : SUCCESS in 5m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/26edbc7dba8a4e73a1d9686e00e782eb : SUCCESS in 33m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7295d172be524f3e99fc28a9b722d111 : SUCCESS in 6m 24s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/f0c355cf992744b3bd4338d2c59a1568 : SUCCESS in 44m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f42d4b7b383494c147df5ee9e2626702cd168274","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"date":"2024-11-21 16:37:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"db5474538a24ead092c293eed559857f2f16e355","revisions":{"db5474538a24ead092c293eed559857f2f16e355":{"kind":"REWORK","_number":1,"created":"2024-07-08 16:50:51.000000000","uploader":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"ref":"refs/changes/68/923668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/923668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/923668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/923668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/923668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/923668/1"}}},"commit":{"parents":[{"commit":"72777e203ec6f09757ddcd90011013eeafab36c1","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72777e203ec6f09757ddcd90011013eeafab36c1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-16 00:35:57.000000000","tz":-480},"committer":{"name":"Jay Jahns","email":"jayjahns@gmail.com","date":"2024-07-08 16:50:51.000000000","tz":0},"subject":"Updated Infoblox backend","message":"Updated Infoblox backend\n\nThis implementation replaces the existing Infoblox driver,\nand is based on the official Infoblox Python client.\n\nThe following new configuring options are now available.\n- wapi_host, wapi_version, cert and key\n\nThe original wapi_url option still works, but can now be\nalternatively replaced by wapi_host and wapi_version.\n\nFinally, the deprecated configuration options for designate.conf\nwas removed, and all Infoblox configuration now happens in the pools\nconfiguration.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/905764\nChange-Id: I35e03b9818851487686153bc68ad90c081e61966\n(cherry picked from commit 1dbdcac58fdb602c9a3617d4b22ba16690e869cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/db5474538a24ead092c293eed559857f2f16e355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/db5474538a24ead092c293eed559857f2f16e355"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
