)]}'
{"id":"openstack%2Fdesignate~923669","triplet_id":"openstack%2Fdesignate~stable%2F2023.2~I35e03b9818851487686153bc68ad90c081e61966","project":"openstack/designate","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"36670":{"account":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"last_update":"2024-11-21 16:37:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I35e03b9818851487686153bc68ad90c081e61966","subject":"Updated Infoblox backend","status":"ABANDONED","created":"2024-07-08 16:51:44.000000000","updated":"2024-11-21 16:37:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":904978,"cherry_pick_of_patch_set":28,"meta_rev_id":"b654471105387bea0821fb01e26833b5afbbd2ad","_number":923669,"virtual_id_number":923669,"owner":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-07-08 18:17:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-08 16:51:44.000000000","updated_by":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2024-07-08 18:17:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2bd0dc48dec8bbfaa3e20c4ad109f3954525524","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"date":"2024-07-08 16:51:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* designate/backend/impl_infoblox/__init__.py\n* designate/backend/impl_infoblox/connector.py\n* designate/backend/impl_infoblox/ibexceptions.py\n* designate/backend/impl_infoblox/object_manipulator.py\n* designate/tests/unit/backend/test_infoblox.py\n* setup.cfg\n* test-requirements.txt","accounts_in_message":[],"_revision_number":1},{"id":"2a137a54687e4528f72a625a5d722279ba965097","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"date":"2024-07-08 16:57:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7677a3fb3c15bb63924e8614cd2402e71852c6a0","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"date":"2024-07-08 16:58:22.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"939426b349b864d7360ab4a4581f21136cbca06b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 18:17:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59f9150582cb44baaf35a111df23c1e9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5ba0caf7addb44cd8ac0ed853914ef8e : SUCCESS in 1m 50s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f3eab637e8a44241854bbf4dd2c689e2 : SUCCESS in 44m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c9aa96b23bf84a48871fbd769fa8194b : SUCCESS in 1h 00m 06s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2594512ba4b1457ba4d074706570304a : SUCCESS in 46m 28s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/17c27f37849c4499855b4345948873ad : SUCCESS in 44m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ff0a6a4bce1b4cb9b4854c3245d52794 : SUCCESS in 41m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a22ed5638a5744d3ae45cd92155b54ee : SUCCESS in 1h 03m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ba83b59cd759420ab722df290ba1da89 : SUCCESS in 1h 14m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a3e4df9dfdf04a34bd3c1824d191ca5d : SUCCESS in 40m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/32ccd64a80a5484ca3d7435dab4f9593 : SUCCESS in 40m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0df35ff7498d43429182a188620d5589 : FAILURE in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28bfb1d9f0414709a49ef46882a252f0 : FAILURE in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05332c21e4cf4c63948418af01adc74e : FAILURE in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4804356e852d4fcf83576dc624cc2d52 : FAILURE in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f41b094a3ba46869577ae53df419e0a : FAILURE in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2c111cc9a6e4f0698d2b2b3b89b0dfb : FAILURE in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23db7202fb6a419d84c3760a5afd53a3 : SUCCESS in 9m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0fca3a9b0bb142db8e3a2f60fe38be8b : FAILURE in 1m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc1ef97489094e849593637861bb884a : SUCCESS in 5m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/778a43883da14ca79d0e7364e7ed47e3 : SUCCESS in 43m 12s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4571df524f464749abbf0edaa6b6983d : FAILURE in 3m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b654471105387bea0821fb01e26833b5afbbd2ad","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"date":"2024-11-21 16:37:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"14f0c990f0f60093191f770e3c8fb8c2f9ad00d5","revisions":{"d476adddb55cf83593e2f8018eb022137062e195":{"kind":"REWORK","_number":1,"created":"2024-07-08 16:51:44.000000000","uploader":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"ref":"refs/changes/69/923669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/923669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/923669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/923669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/923669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/923669/1"}}},"commit":{"parents":[{"commit":"2c817b3d7f01de44023f195c6e8de8853683a54a","subject":"Merge \"Add basic storage test to test config overrides\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c817b3d7f01de44023f195c6e8de8853683a54a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-16 00:35:57.000000000","tz":-480},"committer":{"name":"Jay Jahns","email":"jayjahns@gmail.com","date":"2024-07-08 16:51:44.000000000","tz":0},"subject":"Updated Infoblox backend","message":"Updated Infoblox backend\n\nThis implementation replaces the existing Infoblox driver,\nand is based on the official Infoblox Python client.\n\nThe following new configuring options are now available.\n- wapi_host, wapi_version, cert and key\n\nThe original wapi_url option still works, but can now be\nalternatively replaced by wapi_host and wapi_version.\n\nFinally, the deprecated configuration options for designate.conf\nwas removed, and all Infoblox configuration now happens in the pools\nconfiguration.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/905764\nChange-Id: I35e03b9818851487686153bc68ad90c081e61966\n(cherry picked from commit 1dbdcac58fdb602c9a3617d4b22ba16690e869cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d476adddb55cf83593e2f8018eb022137062e195"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d476adddb55cf83593e2f8018eb022137062e195"}]},"branch":"refs/heads/stable/2023.2"},"14f0c990f0f60093191f770e3c8fb8c2f9ad00d5":{"kind":"REWORK","_number":2,"created":"2024-07-08 16:57:53.000000000","uploader":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"ref":"refs/changes/69/923669/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/923669/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/923669/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/923669/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/923669/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/923669/2"}}},"commit":{"parents":[{"commit":"2c817b3d7f01de44023f195c6e8de8853683a54a","subject":"Merge \"Add basic storage test to test config overrides\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c817b3d7f01de44023f195c6e8de8853683a54a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-01-16 00:35:57.000000000","tz":-480},"committer":{"name":"jayjahns","email":"jayjahns@gmail.com","date":"2024-07-08 16:57:47.000000000","tz":-300},"subject":"Updated Infoblox backend","message":"Updated Infoblox backend\n\nThis implementation replaces the existing Infoblox driver,\nand is based on the official Infoblox Python client.\n\nThe following new configuring options are now available.\n- wapi_host, wapi_version, cert and key\n\nThe original wapi_url option still works, but can now be\nalternatively replaced by wapi_host and wapi_version.\n\nFinally, the deprecated configuration options for designate.conf\nwas removed, and all Infoblox configuration now happens in the pools\nconfiguration.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/905764\nChange-Id: I35e03b9818851487686153bc68ad90c081e61966\n(cherry picked from commit 1dbdcac58fdb602c9a3617d4b22ba16690e869cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14f0c990f0f60093191f770e3c8fb8c2f9ad00d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14f0c990f0f60093191f770e3c8fb8c2f9ad00d5"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
