)]}'
{"id":"openstack%2Fdesignate~925531","triplet_id":"openstack%2Fdesignate~master~I9b1c487522575d4816d0b9739ebba9c10c01f271","project":"openstack/designate","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-08-02 18:20:20.000000000","reason":"Change was abandoned"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-08-01 21:38:49.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I9b1c487522575d4816d0b9739ebba9c10c01f271","subject":"Keep new RBAC disable by default","status":"ABANDONED","created":"2024-08-01 19:46:27.000000000","updated":"2024-08-02 18:20:20.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5cbd386a42212fe0c75d6a5d1bfcaec5631d192c","_number":925531,"virtual_id_number":925531,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-01 20:57:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-08-01 20:00:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-01 20:00:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-08-01 20:57:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13f13c509621471542f365625ed816fa4892bbca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 19:46:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac420f2a1e34ac589321933b1cdf3487ee6164bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 19:48:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ec64f3e4e36c68eb32895bd651aef9badcec41f5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-01 20:00:04.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7dbf85e3108cb786f23a16736e4785d181e360e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 20:57:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c25077ba6cea4cc598314d60d4a7fa9c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3ee7dc7554a346ed94f88f77c52ba7b0 : SUCCESS in 34m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8415dc0151224efb9afde058f01aa761 : FAILURE in 23m 29s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/245e1c3bcf7c4cd6a84822ab594b76f9 : FAILURE in 11m 55s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/dba05be50d984f7faf904949de24e70b : SUCCESS in 37m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2948ad2fae104f33b37d21f12473bf83 : SUCCESS in 45m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c370531cdda248f1a0f508b1406d71e1 : SUCCESS in 58m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/14881780d18d49cbb3dc4dfd588e43a7 : SUCCESS in 1h 08m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c24000ef7fdf49c2893fb16a333f7f3e : SUCCESS in 47m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1b2abd410dba40f393bb23e62faedb1e : SUCCESS in 40m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03109958c0ed4508b703164e7d65fb44 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d7695113aba4fba915e796ff634356b : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fc27c31f9594145ac54bf390fd1370f : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f4166dca0a64e0c8a0ae84232e8cd11 : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/688dfaf89c294cbfb91c373dbf1f6089 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/383c51f1acb246b3a7b7cfad0512ba61 : SUCCESS in 7m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/650483942bde4e009f5c9e14baa3eb50 : SUCCESS in 1m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bc44b58f11af47bc835298fdd6f25c53 : SUCCESS in 29m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/108f963c6b6e4e5fa78a179794c8aa61 : SUCCESS in 4m 01s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/bd71177c254e49888b655a46fc48cabf : SUCCESS in 51m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6d59dc091d1a0ecd778428fc6e71f66635ca2702","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 21:38:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"15ca3c81c9ca7b7591f375e52b14203f672ce595","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 21:39:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5cbd386a42212fe0c75d6a5d1bfcaec5631d192c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-02 18:20:20.000000000","message":"Abandoned\n\nwe are going with the new default enable https://review.opendev.org/c/openstack/designate/+/925623","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ca1a5f6a5cbb0c1d3fbe56b82c5bb44ef230e72","revisions":{"a88061f0390e7b6562a3dc4d22bb38adddb0b760":{"kind":"REWORK","_number":1,"created":"2024-08-01 19:46:27.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/31/925531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/925531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/925531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/925531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/925531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/925531/1"}}},"commit":{"parents":[{"commit":"b9466df8c2e98e2c0a93326cae6f2e77e26324aa","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9466df8c2e98e2c0a93326cae6f2e77e26324aa"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-01 19:45:23.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-01 19:46:26.000000000","tz":-420},"subject":"Keep new RBAC disable by default","message":"Keep new RBAC disable by default\n\noslo.policy has enabled the new RBAC config options\nenforce_scope and enforce_new_defaults by default[1][2].\n\nUntil Designate is ready with the new RBAC, we need to continue\ndisable them by overriding their default values.\n\nNOTE: there is no change in behaviour, Designate continue with the\nold RBAC as default.\n\nref: https://review.opendev.org/c/openstack/requirements/+/925464\n\n[1] https://review.opendev.org/c/openstack/oslo.policy/+/924283\n[2] https://review.opendev.org/c/openstack/releases/+/925032\n\nChange-Id: I9b1c487522575d4816d0b9739ebba9c10c01f271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a88061f0390e7b6562a3dc4d22bb38adddb0b760"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a88061f0390e7b6562a3dc4d22bb38adddb0b760"}]},"branch":"refs/heads/master"},"0ca1a5f6a5cbb0c1d3fbe56b82c5bb44ef230e72":{"kind":"REWORK","_number":2,"created":"2024-08-01 19:48:47.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/31/925531/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/925531/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/925531/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/925531/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/925531/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/925531/2"}}},"commit":{"parents":[{"commit":"b9466df8c2e98e2c0a93326cae6f2e77e26324aa","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9466df8c2e98e2c0a93326cae6f2e77e26324aa"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-01 19:45:23.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-08-01 19:48:47.000000000","tz":0},"subject":"Keep new RBAC disable by default","message":"Keep new RBAC disable by default\n\noslo.policy has enabled the new RBAC config options\nenforce_scope and enforce_new_defaults by default[1][2].\n\nUntil Designate is ready with the new RBAC, we need to continue\ndisable them by overriding their default values.\n\nNOTE: there is no change in behaviour, Designate continue with the\nold RBAC as default.\n\nref: https://review.opendev.org/c/openstack/requirements/+/925464\n\n[1] https://review.opendev.org/c/openstack/oslo.policy/+/924283\n[2] https://review.opendev.org/c/openstack/releases/+/925032\n\nChange-Id: I9b1c487522575d4816d0b9739ebba9c10c01f271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ca1a5f6a5cbb0c1d3fbe56b82c5bb44ef230e72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ca1a5f6a5cbb0c1d3fbe56b82c5bb44ef230e72"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
