)]}'
{"id":"openstack%2Fdesignate~925623","triplet_id":"openstack%2Fdesignate~master~I40fe2363d11558206260929fa8f8bde4c2e1ab45","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-08-02 19:47:33.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I40fe2363d11558206260929fa8f8bde4c2e1ab45","subject":"Update functional tests for new RBAC system scope","status":"MERGED","created":"2024-08-02 18:08:55.000000000","updated":"2024-08-02 22:29:06.000000000","submitted":"2024-08-02 22:28:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925623","meta_rev_id":"819a1cc9a79061ffc3324cffe8e48715c3697b75","_number":925623,"virtual_id_number":925623,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-02 22:28:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2024-08-02 18:27:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2024-08-02 20:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2024-08-02 20:31:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-02 18:27:46.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-08-02 19:35:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"354cc3fff46c2891fd2f0db22e4e69b371b5b04c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-02 18:08:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccd9976d5b9e430f257fc690cf66add493d6b966","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-02 18:27:46.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3664bb0ae60049692743362812e0273ce18c5be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 19:35:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88659cff1e1941119ff287ae756a899b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a752ecb8a1e747b7ab78d16cfded8a73 : FAILURE in 46m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f45953b749af4f53a846d556fb90fb43 : FAILURE in 14m 48s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8620cabcb23b4b92b2193b90953e0257 : FAILURE in 11m 40s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/86d9dc22f9354415a517de306548ee4c : SUCCESS in 50m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b970fa7e3a7b44b5b7d84e58294ce140 : SUCCESS in 46m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/aa8b1bc7175b4c0c96652c2ab20a22e9 : SUCCESS in 1h 26m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/faaf24684432486aab7f547dffe19881 : SUCCESS in 1h 01m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f68669c4b10c45c2ba7bcb0656feb55f : SUCCESS in 50m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5d51f26fa6e143049f1f69c93fdf2139 : SUCCESS in 50m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/401c7f5e419b4af290ac5bbed8683540 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b24583442c4045b1b3ee5b6544956847 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15c428830aec40688a2c6249b963c714 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/039c08598f1e4c74bf14a0c689c22dea : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0176f323268442d94ee02ed8fc25e7b : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab39650a59a2420b9aa11a670d16251b : SUCCESS in 7m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1af3b60f944d4947bb8ded484775b44d : SUCCESS in 32m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1cab97b91c844af0bf0013ef8ccf29b8 : SUCCESS in 4m 39s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/60902d8f0e084976b5e52e7d1a3aba39 : SUCCESS in 48m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f8581d16766bc95da3ddc4dcdd137633173ff34","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-02 19:47:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"496162ad76bef8b2a42ae2ee9ce3d952ccd8776a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-02 20:31:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b62d9f414ee2a38c435f5f31abab42ac9bd45e62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 21:11:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f02cb117e8a489cb3abca8cbc54f9ac\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cc2e4b9c93ce493cab7228a8ddb8ad29 : SUCCESS in 38m 53s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e7245feb21a74de9bf6b5da7e41a2ea2 : FAILURE in 23m 40s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dd0b64571884186972fb8bac328d9c3 : FAILURE in 8m 34s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/04214f8e32b7469da6afb4b4f3ee90bb : SUCCESS in 54m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/19b2464736cc479ab6223b97b29d0787 : SUCCESS in 34m 45s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9c454d5864354f9782686a6a08d40ce7 : SUCCESS in 1h 17m 29s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4071b61a326f4f9f8e136186f4e33643 : SUCCESS in 54m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d72399235c134282b1d0231eb6aaba91 : SUCCESS in 29m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/22fc862c9d724f75b70fceb943210b1a : SUCCESS in 37m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64651e770516413298a360b33ec1574d : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f61f5b344ca4f58b9361bf5e1163516 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f37b412b816f422d9f2e2d700db0e580 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7727aae2896b42e19ef9cb670f098a85 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/191bfba26f1445baacc38f317946196a : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63a0eb1031164a1ca836b553e8a8feb1 : SUCCESS in 8m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/32afda5a1687455e884ed36f7fd68ffd : SUCCESS in 37m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/94d24bd7c79347d2a232af634a7fac14 : SUCCESS in 4m 06s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/58794ba3bfef457182823d2d8317c6a5 : FAILURE in 46m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4193acf9c8461e0a9e632ea4d462a786ce814fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 21:12:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a2c5ee67e8a0110b52dc5f4dca28e5dfeee685d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 22:28:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90d008e48d49465fbf381e9108b1c0d5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b0143e3a55094e139751ce12ddcc1a83 : SUCCESS in 35m 47s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/82f808ebc34a45c590d6f7cca4b1bfa2 : SUCCESS in 49m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/18ca841765ab468fb06b08c5d8ed34bb : SUCCESS in 34m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8c0e654fcb2c454c8fb2e9cd65d41d9e : SUCCESS in 1h 15m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9e321ac701d544f6bc6b88e79fdd0bb0 : SUCCESS in 45m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/793e7a793b484efca612843760206efc : SUCCESS in 44m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d009e22df9e1419fa5c1a2a894660a56 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a2da9c21b3247aa80ed725b9b7b3e83 : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45387bf166e0418f8278f24bc9bdb022 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a20e4b14130540cabaf0b1a84c6f48df : SUCCESS in 6m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5b508b4cf0b441f39ac27b10351111c3 : SUCCESS in 38m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f80dca95162afb9122ea9cf4ad6d126c59361ac6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 22:28:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"819a1cc9a79061ffc3324cffe8e48715c3697b75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 22:29:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb41a0c8d21f44cc987ab0edeabeec05\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ea684b85a9d4cdfae57754c9386266c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50f686fcffd007506e0cd88788a668d4f57febc3","revisions":{"50f686fcffd007506e0cd88788a668d4f57febc3":{"kind":"REWORK","_number":1,"created":"2024-08-02 18:08:55.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/925623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/925623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/925623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/925623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/925623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/925623/1"}}},"commit":{"parents":[{"commit":"a979265c70b86c87d26109a282fec43f0d17e083","subject":"Merge \"Remove unnecessary generate_uuid wrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a979265c70b86c87d26109a282fec43f0d17e083"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-02 18:05:45.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-02 18:08:53.000000000","tz":0},"subject":"Update functional tests for new RBAC system scope","message":"Update functional tests for new RBAC system scope\n\nAnother change to the \"Secure Default RBAC\" goal is to default to\nsystem scope True, but any use of a system scope token will fail with\nan InvalidTokenScope error.\nThis patch updates the Designate functional tests to work with this new\nsystem scope setting in oslo.policy.\n\nChange-Id: I40fe2363d11558206260929fa8f8bde4c2e1ab45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50f686fcffd007506e0cd88788a668d4f57febc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50f686fcffd007506e0cd88788a668d4f57febc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
