)]}'
{"id":"openstack%2Fdesignate~925627","triplet_id":"openstack%2Fdesignate~master~I140e87e1441fdc229c0b8eca429024a98b4c96d4","project":"openstack/designate","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-11-20 10:38:51.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"33123":{"account":{"_account_id":33123,"name":"Michael Jonsson","email":"micke@secit.se"},"last_update":"2025-02-11 03:07:54.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-09-13 11:40:18.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-02-11 03:07:54.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-01-06 02:59:06.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-11-18 16:49:56.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I140e87e1441fdc229c0b8eca429024a98b4c96d4","subject":"More cleanup for the RBAC new default","status":"MERGED","created":"2024-08-02 18:26:43.000000000","updated":"2025-02-11 03:10:17.000000000","submitted":"2025-02-11 03:07:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925627-secure-rbac","meta_rev_id":"ff78d8de06fc311f4b46c1414e2d8fff2ba52269","_number":925627,"virtual_id_number":925627,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-11 03:07:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2024-11-20 10:38:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2025-01-06 02:59:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2025-02-11 01:49:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33123,"name":"Michael Jonsson","email":"micke@secit.se"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":33123,"name":"Michael Jonsson","email":"micke@secit.se"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-02 19:49:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-16 04:54:07.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":33123,"name":"Michael Jonsson","email":"micke@secit.se"},"state":"REVIEWER"},{"updated":"2024-08-16 20:21:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-09-03 19:03:37.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2024-09-11 09:13:24.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-09-13 10:51:15.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"6aa0d9fa09bd8751f6498419c4814923f3e6b8fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-02 18:26:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4d35539974a93cd45c4248abe0581def7d50bf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 19:49:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f2e02bdbba64719aad4d3d330d3e7a4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c5579b50d0d84f01bbfaf4caf71cc3a2 : FAILURE in 50m 54s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/cb1264ebf85b4eafbaf7b5d94070c73d : FAILURE in 23m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4522c352ff4540ffb73d91f9f80366e0 : FAILURE in 20m 38s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ec859d81e74a416a83d0248ae4651b9f : FAILURE in 42m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/441237a0db124581b4fd6e57f7d6a51e : FAILURE in 36m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/05eb4dd88a224bad8f11432cf14d6d3c : FAILURE in 55m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3c58212ca22c4320abff5568f1dfe373 : FAILURE in 56m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/93df77cad63d42d3bb6480ee40bbf3d1 : FAILURE in 36m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2a3853d7be9a41e69287e09acbc728b0 : FAILURE in 36m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/642ac38a67a34d8e98fd4eb70c729c03 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fccf3c7f72a472380625a5f7bc08af1 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bd7440edfe147138e377788ac9f4e2d : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95ef84ed361242b9a20aa4dbcca4b60c : SUCCESS in 6m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cee11ff5ee6d46118b8bab5c565058f5 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/074ce4fc849142e5b28f8bf419c4d555 : SUCCESS in 8m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a2817c6204694440a9e4332e2687eb26 : SUCCESS in 1m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ec0f31afb8e54476b75e254803c4e4da : SUCCESS in 5m 30s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/8e1a628cda2044d7acfe757aabeb47c8 : FAILURE in 42m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a232b9118ce8a9c3c552e3645c726018b7cfebe0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-16 02:01:50.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b35a320b2e22e1f57380b09d76271b2e080d6358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 03:49:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a77c5f60ba4483c867093dac6fa8133\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d9b26007dcdd46dbbfb194e0b7d940a5 : SUCCESS in 47m 10s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/64e38005ee204e7a8056cabfebf16138 : SUCCESS in 1h 02m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5211ad03b0b64d3ea1fe8d1254b5df6e : SUCCESS in 54m 13s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a16a7efa6aed4e7089d3d8fad0bd261f : SUCCESS in 43m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/15aac1ba7f624fa59966e26cca82961c : SUCCESS in 41m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/74690debb2ca49c6ba2fbf491e769715 : SUCCESS in 1h 12m 13s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/46402baa9b874fbf970e269d1c0fb9da : SUCCESS in 1h 11m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bd094401de4a48068c2d44f110e33998 : SUCCESS in 45m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/66487955cb5d4aa894cfa680378aca5f : SUCCESS in 50m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fdde9bd4b5040b4b3aa349bba67432f : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d8b4b38e9a845daa9aa30f951838a12 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44fd74677bf34b2caab3a150359eba99 : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36bcc9b1c7f841f0adec93198917bf8f : SUCCESS in 5m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bdefa0cf57f449fa23edbe444a12334 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa0fd79fa3db4e7d9f6eccaba6ab62df : SUCCESS in 8m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5a077d4180194787ab1cb68abecc340a : SUCCESS in 2m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/09fd5d03f12d4b63be4e73c14e38cd0e : SUCCESS in 4m 40s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/1f5537b4893d4cb59da7048d21d05b3d : FAILURE in 1h 04m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"36f4811f5b2f2abac637b144dab88ab9d8c6a8c4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-16 20:21:36.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91422ada8b9e3d7d1cdf3b2ec2c76900ad83a2b6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-30 18:26:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c0870a95e19419f025864a46867ddc96b5d7472","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-30 20:22:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03249abc43b5ebff6e7219952136896386ace61d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-03 19:03:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fd0bcdf9f23854a5b68e149742bdb4fc1040dfbc","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-09-11 09:13:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"652446a94c49b82f413b8cf1f555a11bfa095277","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-12 22:33:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"93ca9652b6a308ac6641b9956ff864382f830fd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-12 22:33:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0b9107f2d0fd019e806101b8b4a87e7d1225bec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 23:49:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f9b5b38ae744492aef0edfbdea1f9b1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b70bc11972bb4fbaa94931e50bd75765 : SUCCESS in 36m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9fbec35b8e1c4a6295c03d060b13cfda : SUCCESS in 51m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/24db371c26224c86bc8444fdabd6cbee : SUCCESS in 45m 41s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/036094a8c5854914a070bd0313bf336e : SUCCESS in 40m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/808e5f15d5d246f2a8d6aa2d27d53311 : SUCCESS in 42m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/109f1cf9383f47c2b69d9b6b200bde50 : SUCCESS in 1h 10m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bb3ed237767f44f4b99fb5ffb11999be : SUCCESS in 59m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/02da500e045b4992b9883c86047fd286 : SUCCESS in 45m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a73ca6a4862f4e898e26e53c3e8d478c : SUCCESS in 45m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1b4e165f00542589ff2ba343c9307b7 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cf0433e09e544f3bd673029a2434548 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e639209c096f4e26b01b165c93622c41 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a34d1db278624a70ab5c954ee9fa982a : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34d786ff1ea4476297e1c815e6e05576 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88ecbfcb55e7451d9b358cc0f08802a0 : SUCCESS in 6m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dc46e024d48f42699532e022c63b07cf : SUCCESS in 2m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f111acc3996d42a5870efb51b9c30772 : SUCCESS in 37m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a0e181cec99e4f0ebc07ccbbfefdda66 : SUCCESS in 4m 19s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/0a32176ee1274c76b85be1895992d145 : SUCCESS in 44m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"440ae788dae61fd8e797b92c1425d6bf08ab1bc2","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-13 10:51:15.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"177c70e71cca81c6a192580a175f737165116f87","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-09-13 11:40:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a9bdb99b765ab66f7fba8b22417ed45e6a3175bd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-09-21 16:55:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"595c2f920322c55a38dca1c28667585f22c79482","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-16 02:42:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a4318898ebe6fec40ab1f4d98527079dac2831ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 05:01:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddb9af8c44174b98a11b24d8355841d3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3d3ff593fad2439ba2519375b3f1ae4a : SUCCESS in 45m 33s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c0fd4b36d558470f97d1887d6fc2773f : SUCCESS in 1h 07m 51s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/84c983b283384e84b8b8aa7b1fdc3d65 : SUCCESS in 44m 16s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/d7b0c11611f1497bbb615f641939b298 : SUCCESS in 41m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d278dfa5bfb04ee1829d2f5b16df9cac : SUCCESS in 40m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/eb077a57210f437a95c48dea386914c7 : SUCCESS in 1h 12m 43s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2b72edaa9b1b4a8683418c8918793b2e : SUCCESS in 1h 16m 45s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a3d7640251c1475687e411e869bb915b : SUCCESS in 56m 06s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/032236ec97224c90b1f7ea4efbb658dd : SUCCESS in 1h 08m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d2068fc563e942efa6e2bd58d782e8f4 : SUCCESS in 1h 02m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f2549a2f354e4694baf19132557aa2d3 : SUCCESS in 49m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e2cbab15db54d2187e4134c834aea28 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8df4e60914644e14ae5acd7f514a1752 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f304a7c26644f60a97d46b7a038455c : SUCCESS in 6m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b64193a83c164d54a1853c7a6516ef7f : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9ecb05d76b54192a6b03b7f672fb698 : SUCCESS in 17m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/41d46e66a1a642b6b47330298b978d05 : FAILURE in 29m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7a22c674241f4b4aae018e15522b3a19 : SUCCESS in 7m 16s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/d4cc483a69b64e63a9a2f500f159c3e4 : FAILURE in 50m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"82cfef5642e0de5d7e33273415fed886b54fd668","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-18 16:49:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f17328df1cbb90fec034c0368ae8bcd4cb9b66d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 18:09:49.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a06075d70f74fc29a00ea0cbf68aa10\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c37ee90f51804b4db7f0671cfbc7caa7 : SUCCESS in 38m 48s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/fa1d0f63797c450183db8376286d4ef9 : SUCCESS in 1h 05m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/38bd3e191f3a4db6936e179b17026c8f : SUCCESS in 1h 03m 32s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/66fcdded02a549e39af7e6f6a92e2841 : SUCCESS in 52m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/038812d14e604c8ca81da83c2a076161 : SUCCESS in 43m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fe671f4643f44d37ba0225ac577f5256 : SUCCESS in 54m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3a91b8e75139405a9cdda9645dd27b8e : SUCCESS in 55m 58s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/f929039d50cd4b9585fc2856a33b6ff6 : SUCCESS in 1h 11m 53s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/13f3bd1eaff446c79da0514f1f586ca9 : SUCCESS in 1h 11m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/58cf5da61b3e4f5988a3bb3abc5351a3 : SUCCESS in 45m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2238de4b75b54c9886f5e4904509fa79 : SUCCESS in 45m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c48f07df0264a3a8eafbe2167f6b38a : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceaf238b48c14acc956e1e85f3c7a76f : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81ba15dbfd1d40189016fc335c10e477 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c42322cb7074e14b688da1095f10d92 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e11083acdaf4140a46b9caddafe74fa : SUCCESS in 7m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/27a3c3ea035947d8b27f91b37dcab121 : FAILURE in 35m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/42ce48d008454ec98d0c709cb58e777d : SUCCESS in 4m 30s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/02a6d0e820e048fcbdd360f33a374e19 : SUCCESS in 54m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bc3079d662d3cadb3a0f5c29fa9d0a0c40f2ee6d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-18 21:35:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"47c21ab04f2d843c68643783d57ad6853a38a5f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 23:03:16.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0357407a4882437e9e25f9cd60db4444\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/998e27e1cd194c76a2137c0383954691 : SUCCESS in 44m 30s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2a5bd167bb1b44e1b8ca47a8706a6366 : SUCCESS in 48m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d695f4b01144078b29ba16d21a03b43 : SUCCESS in 47m 06s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/091d6114e6514a61a0446ba0648fc188 : SUCCESS in 44m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/904e6dbbb72b4cdf84baea657dc3551c : SUCCESS in 45m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/232eb572d78f47f682adc1a7a636687b : SUCCESS in 1h 12m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fb186e3ef52d41399003a05fad02be5e : SUCCESS in 1h 01m 30s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/ce7ea2de455c467bb1e876cb0e95bd9e : SUCCESS in 58m 01s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/d89182698fea42e293911437d2237c59 : SUCCESS in 59m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/918fd6bde9904eeb963891ab3ef43183 : SUCCESS in 37m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bc0b17afb173448b83dcc4c4d3517c7f : FAILURE in 12m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dd22759aea14800b92ebb508c839cce : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/580424fdf7b144ba873cdc52324df126 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97b72c28b9804ed39de954ef9f1ce5f7 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2c1e4038477d444486f10c423239f164 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/febf18a910244964bacb71f296137e1e : SUCCESS in 6m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/74e0f3cfd88f47f18c265fed180686ed : SUCCESS in 28m 13s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e0a79e1afc1b4eb989a49ce133f7cab1 : SUCCESS in 4m 58s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/9aea120706cf4ea4aca766f2bc388d08 : SUCCESS in 37m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1d783df50267bb946f8e04ab15d221349ec87957","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-19 11:53:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"809e4c8b5330f3723239de312e0c5c47a4782180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 13:18:54.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f30cc6e9d0c64dfdaa3e07c5a643d84c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4e9e5d38301840fead2b23ab90678bbb : SUCCESS in 30m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8da7ca7fba104f53a9c251d924ace01e : SUCCESS in 49m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a69784e73db4479681991aab605eb532 : SUCCESS in 57m 02s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/2bbd4ec03f994b9b81756189380fe0e1 : FAILURE in 36m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/495f854db32142329868e6e901f041bc : SUCCESS in 48m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4d6a25d7ba594a91bb48942c423026a7 : SUCCESS in 1h 12m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/34f033f7e3044ff6b9e569a707dd54a2 : SUCCESS in 59m 45s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3c2d35ff8fe34293891704457a042e36 : SUCCESS in 1h 12m 11s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/60d1d1bcb395433281f05b0c8cbed23f : SUCCESS in 1h 06m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/91d3b2d67b314c338bd1e032fa34867f : SUCCESS in 53m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/85510c413fda463c8588adef1bdf4649 : SUCCESS in 36m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b83d1ed431a74b3787d70a9ccbd8b65e : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccda03cf03384e6aa0e90ea1d462878e : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62178720aaab400abd398e59f4dbfbe2 : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/090ee56d38524c9b8115d76a089adf40 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af80d07548434ae6b070279489c5be24 : SUCCESS in 8m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4398188b91b944feb8e827a6e5fd6639 : SUCCESS in 29m 23s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/72e8c637d7c34accb9d06c15291933f0 : SUCCESS in 5m 05s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/0ad323137de2440ab5af0df5dbdf6924 : SUCCESS in 38m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8fafbcfc1ca11092571295464fcb4186a1835c34","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-19 19:15:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5de3192c374dc7b367c5ac7f4c87b9aab3490889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 20:51:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0049ce8eccf94132ae289d4102c436a2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b470d8f5b4df45aabb3fccca317b7433 : SUCCESS in 49m 46s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e8cde029dea94a978746d20ac56c58e5 : SUCCESS in 1h 09m 59s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/73e417ac70fe41778145c0e59a94f6d8 : SUCCESS in 58m 30s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/0cc45206e45c490d829a22d99bf98c05 : SUCCESS in 41m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b673af3c02b64bcb888d8a2bc740f956 : SUCCESS in 48m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a317b30a4e9448d1aacf7ca6cf864cfa : SUCCESS in 53m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e54ba8af9844485f8744e61af5dcc5dc : SUCCESS in 1h 10m 51s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/df0edad72d62497b97357ffbfcc19970 : SUCCESS in 1h 27m 37s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/5bd5754c7d2d4dfe92c72a5c4ac725b8 : SUCCESS in 1h 11m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ead867cc41584d818a14f127c3acd166 : SUCCESS in 39m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8c42b75a23774fc88117526d2225d4a6 : SUCCESS in 42m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/172e3aafa88140afaa2687f13f77ab97 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/748f455832424261bef1a2bd0f1e837a : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/868f3509e2694265a76de991518bd350 : SUCCESS in 5m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b5f3f30e4af4a3191e67e5f8c60fb16 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee661340908641e0bfc309068fff182d : SUCCESS in 7m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b147f97e395c42389609d819f9de749d : SUCCESS in 32m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/546fd069f91d45029c02a4b13bdcbccb : SUCCESS in 4m 48s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/65761faecb7c48519cd86315b12de108 : SUCCESS in 46m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"707d677f15a370b6bef97b0379adc7c2f905218b","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-11-20 10:38:51.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6f6a09085d63fcbea68700a864050540e3fe5266","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-06 02:59:06.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f40d295677ccffc9d52cd472ac86a52233b26af9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-11 01:49:00.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e2c0d0b1b22b725fe564ea9561e28fc4f25aae1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 01:49:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a0515a2a8c80dcb7c82a06d4603a4c1edb1dd270","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 03:07:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b2cc91ffa9f4cbfb8b8e4c9671461fe\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5906fdf0ace04d8184e87a617f35b9de : SUCCESS in 47m 02s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/8ecf3d1bbdef4eb9a0cef2df713a256b : SUCCESS in 52m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8034691258614efa8f525df529b3b749 : SUCCESS in 45m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fb3fb87a48a34f7ab21c8b8ded8d2ac6 : SUCCESS in 57m 37s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/2cfb8c24a96a4e49942bd0561f309c51 : SUCCESS in 1h 12m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7ce3cc82c370460cb31ac39792269047 : SUCCESS in 43m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1a540e4a02f44d80b68c21aa85afcc26 : SUCCESS in 55m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd842f5aeb1464883ce1a219425d6df : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/787cbf6f6e0a4c2389abf64c4f49630e : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85f2fb598e0c457d99aadd1090e758e2 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f815fa0caeb4f5d94efd8131aaaa704 : SUCCESS in 7m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/653fd8da1ad24256ae37d2e93a0aaffa : SUCCESS in 27m 52s","accounts_in_message":[],"_revision_number":4},{"id":"5e036d41b3844d0fe6bc2c0ed49fc5a79a72654b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 03:07:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ff78d8de06fc311f4b46c1414e2d8fff2ba52269","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 03:10:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f861f0cdb158409b9a99e1f2b690f94f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3712a7d2d3aa46ad90a3345a3ae5f4a4 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5aa3412826e153ad8ee00697fbb8c3bf901b89d3","revisions":{"6867be867b8d9cb5f2810134c9fad329e79df5a6":{"kind":"REWORK","_number":1,"created":"2024-08-02 18:26:43.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/27/925627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/925627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/925627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/925627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/925627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/925627/1"}}},"commit":{"parents":[{"commit":"50f686fcffd007506e0cd88788a668d4f57febc3","subject":"Update functional tests for new RBAC system scope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50f686fcffd007506e0cd88788a668d4f57febc3"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:25:59.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:25:59.000000000","tz":-420},"subject":"Update oslo.policy version for new defaults","message":"Update oslo.policy version for new defaults\n\noslo.policy 4.4.0 enable the new defaults and scope\nchecks by default which means those will be enabled\nfor all the OpenStack services.\n\nThis bumps the oslo.policy version in requirement file\nalso to ship new default enabling.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/925464\nChange-Id: I140e87e1441fdc229c0b8eca429024a98b4c96d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6867be867b8d9cb5f2810134c9fad329e79df5a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6867be867b8d9cb5f2810134c9fad329e79df5a6"}]},"branch":"refs/heads/master"},"f0ae34098b47c9a591099ee84f489b17088b1633":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-08-16 02:01:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/27/925627/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/925627/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/925627/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/925627/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/925627/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/925627/2"}}},"commit":{"parents":[{"commit":"50f686fcffd007506e0cd88788a668d4f57febc3","subject":"Update functional tests for new RBAC system scope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50f686fcffd007506e0cd88788a668d4f57febc3"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:25:59.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-08-16 02:01:48.000000000","tz":0},"subject":"Update oslo.policy version for new defaults","message":"Update oslo.policy version for new defaults\n\noslo.policy 4.4.0 enable the new defaults and scope\nchecks by default which means those will be enabled\nfor all the OpenStack services.\n\nThis bumps the oslo.policy version in requirement file\nalso to ship new default enabling.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/925464\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/926455\nChange-Id: I140e87e1441fdc229c0b8eca429024a98b4c96d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0ae34098b47c9a591099ee84f489b17088b1633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0ae34098b47c9a591099ee84f489b17088b1633"}]},"branch":"refs/heads/master","description":"Edit commit message"},"81b9b44bdae2e02329151142abeb7c6085b9fd2a":{"kind":"REWORK","_number":3,"created":"2024-09-12 22:33:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/27/925627/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/925627/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/925627/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/925627/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/925627/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/925627/3"}}},"commit":{"parents":[{"commit":"a23a18bd61dd8fbf2325235a947577662d5f91e7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a23a18bd61dd8fbf2325235a947577662d5f91e7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:25:59.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-12 22:28:17.000000000","tz":-420},"subject":"More cleanup for the RBAC new default","message":"More cleanup for the RBAC new default\n\noslo.policy 4.4.0 enable the new defaults and scope\nchecks by default which means those will be enabled\nfor all the OpenStack services.\n\nAs new defaults are enabled by default, this commit\ndoes more cleanup for enforce_new_defaults flag. Remove\nenforce_new_defaults() method and also set tenant_id\nand project_id both in oslo policy target. When we\ncompletly remove the old defaults and support of tenant_id\nthen we can do tenant_id cleanup.\n\nAlso, all the jobs run with new default so modifying\nexisting rbac job to test the old defaults.\n\nBumps the oslo.policy version in requirement file\nalso to ship new default enabling.\n\nChange-Id: I140e87e1441fdc229c0b8eca429024a98b4c96d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81b9b44bdae2e02329151142abeb7c6085b9fd2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81b9b44bdae2e02329151142abeb7c6085b9fd2a"}]},"branch":"refs/heads/master"},"5aa3412826e153ad8ee00697fbb8c3bf901b89d3":{"kind":"REWORK","_number":4,"created":"2024-11-16 02:42:35.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/27/925627/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/925627/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/925627/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/925627/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/925627/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/925627/4"}}},"commit":{"parents":[{"commit":"60d4e95a5ae9b2dee719302a3371ecdb6a861bdf","subject":"Merge \"Ignore generated sample config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/60d4e95a5ae9b2dee719302a3371ecdb6a861bdf"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:25:59.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-16 02:42:33.000000000","tz":-480},"subject":"More cleanup for the RBAC new default","message":"More cleanup for the RBAC new default\n\noslo.policy 4.4.0 enable the new defaults and scope\nchecks by default which means those will be enabled\nfor all the OpenStack services.\n\nAs new defaults are enabled by default, this commit\ndoes more cleanup for enforce_new_defaults flag. Remove\nenforce_new_defaults() method and also set tenant_id\nand project_id both in oslo policy target. When we\ncompletly remove the old defaults and support of tenant_id\nthen we can do tenant_id cleanup.\n\nAlso, all the jobs run with new default so modifying\nexisting rbac job to test the old defaults.\n\nChange-Id: I140e87e1441fdc229c0b8eca429024a98b4c96d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5aa3412826e153ad8ee00697fbb8c3bf901b89d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5aa3412826e153ad8ee00697fbb8c3bf901b89d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
