)]}'
{"id":"openstack%2Fdesignate~925991","triplet_id":"openstack%2Fdesignate~master~Iae6910a1244cdd16b01e968d050455c1fe0d2d03","project":"openstack/designate","branch":"master","topic":"policy_fix","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-09-13 16:23:44.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Iae6910a1244cdd16b01e968d050455c1fe0d2d03","subject":"Fix invalid policy generation","status":"MERGED","created":"2024-08-08 15:21:51.000000000","updated":"2024-09-21 16:04:56.000000000","submitted":"2024-09-21 16:03:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925991-policy_fix","meta_rev_id":"3a34164db3cbb72c4b24d198ec70d0eab0d8ca9f","_number":925991,"virtual_id_number":925991,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-21 16:03:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-08-08 23:04:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-21 15:08:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-08 16:34:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-08 23:04:33.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"785bc12be5b9ceff0d2a74b4af0e490d6a078470","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-08-08 15:21:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f90f7eeb778e5162093e96e7171a7ee69682e536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 16:34:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8c439d83f1e42d3a80086ef87bcd699\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ceb87b4575b545069121c178737761e4 : SUCCESS in 45m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0be6f02431344c8ba8635e278d837300 : SUCCESS in 47m 09s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4b3b7d3cf924fb58e7faf3fed9c8278 : SUCCESS in 52m 16s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2425a2095a484d24b5dc7e54b0ef2659 : SUCCESS in 48m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/85c39c141f804e91a2b33400f4508f40 : SUCCESS in 47m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/10338f562e5446e78b6ac9b9cb04b0b2 : SUCCESS in 57m 08s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/01b2f8b133df4bb18033230f1ab3c6fc : SUCCESS in 1h 11m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/16ad5495ab9c4e91823f0a6ab0a9cc4b : SUCCESS in 53m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/493f1170eb03439db6a238b631270f40 : SUCCESS in 36m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d8b2169b50648ee9aa98cdcc0f2b53f : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/956bf7a898384d959a8156ed60be6e8c : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4aac0ffdd3584ceb94d44d3d81899dd0 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/851faded699049468d6af2aed561dcfa : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7caa6342ccba428b9fb4c99424bb78be : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08ff95402f5042438af189bbd37d4b85 : SUCCESS in 6m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/10699f6c7bc248d686ceebffa1c34f3f : SUCCESS in 44m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/002fe9966f00450888f6fa698bc031ed : SUCCESS in 4m 11s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/e8a779bdfff340df94c11f6c540d98a3 : SUCCESS in 56m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b083385f9adc77b7a7aa9f56b4486860797eaebd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-08 23:04:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1587bb5b88a35c882897fa3777e78bcc34f69ef","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-09-13 16:23:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07dd7e5ccce76f199c05b4294e441d68c7ecc043","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-09-21 15:08:39.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bcb62778b7e1b050bd8ff03b1800e1edf1d4d062","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 15:09:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"252cc47081cb4d2d8de6923fe9faa26498b2ca36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 16:03:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e52229dde534da49f81a3f7af29054c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ead737cf4f54400fad3b312a679b17e7 : SUCCESS in 32m 08s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/34809ae25dd44db0bfff3d640ea0c7a2 : SUCCESS in 34m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5dd6cd3b8dff4b97a59486cdf52ad60e : SUCCESS in 31m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/df1a67d155b0422f969a868ee3eaf99c : SUCCESS in 53m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/86f71ddc07b74c7aa1b38bf3aad0889d : SUCCESS in 33m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7721f4cd1d75445493aff62f59f34980 : SUCCESS in 33m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f365c2672535416eaf79ebd3865128e3 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea8d033703144520a3bccf900beba166 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/593bb55cde524296af46bb4ebc8571d5 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec2f542026164ac081b85e4d9eb655b9 : SUCCESS in 6m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8de9cfc46f0146ad93ed2c8b9043666c : SUCCESS in 30m 36s","accounts_in_message":[],"_revision_number":1},{"id":"de8e6f32f95946bd3d86420ce5f7a359c96951cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 16:03:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3a34164db3cbb72c4b24d198ec70d0eab0d8ca9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 16:04:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5959431929e541af8f20de3c07db1bee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1535d66dbc1b4eedb463f2315118de77 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f52de13a3f287854b14b8623e6b662e98f3b620","revisions":{"0f52de13a3f287854b14b8623e6b662e98f3b620":{"kind":"REWORK","_number":1,"created":"2024-08-08 15:21:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/925991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/925991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/925991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/925991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/925991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/925991/1"}}},"commit":{"parents":[{"commit":"50f686fcffd007506e0cd88788a668d4f57febc3","subject":"Update functional tests for new RBAC system scope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50f686fcffd007506e0cd88788a668d4f57febc3"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-08-08 15:21:44.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2024-08-08 15:21:44.000000000","tz":120},"subject":"Fix invalid policy generation","message":"Fix invalid policy generation\n\nChange-Id: Iae6910a1244cdd16b01e968d050455c1fe0d2d03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f52de13a3f287854b14b8623e6b662e98f3b620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f52de13a3f287854b14b8623e6b662e98f3b620"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
