)]}'
{"id":"openstack%2Fdesignate~926274","triplet_id":"openstack%2Fdesignate~master~I7a1d58d324413c1116cf0eb8939c939767b9f54f","project":"openstack/designate","branch":"master","topic":"add_tempest_plugin_devstack","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-12-24 21:30:18.000000000","reason":"Change was abandoned"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-01-17 15:12:15.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":[],"change_id":"I7a1d58d324413c1116cf0eb8939c939767b9f54f","subject":"Add designate-bind9-multipool job","status":"ABANDONED","created":"2024-08-14 10:42:22.000000000","updated":"2025-12-24 21:30:18.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d999a6ab750b7b49aea56d902582aca9d529d8f5","_number":926274,"virtual_id_number":926274,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-17 15:12:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-14 11:58:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-19 12:05:58.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"}],"messages":[{"id":"d272c483608c8ea5d643ec5351ff58331d8a098a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-08-14 10:42:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d629ec94a800385d9e795fa7e24a1712d3fcacf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-08-14 10:48:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b1f65c2299ab80d5a0304da38b7edba35c838db6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 11:58:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af623e368fb24c13b67eb3ce3a2b2894\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a2f5cbf771fa4889afc4bda567885c10 : SUCCESS in 51m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f27b07d4719847358135e08b316ea1cf : SUCCESS in 49m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/728e2747a50a409a82099cb16a8b2571 : SUCCESS in 53m 35s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4b35a706ec8b45e4ac29cb318ddf064e : FAILURE in 43m 31s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/ee1ddc87a1754e429a3e9d8117bbff7f : SUCCESS in 48m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b337a17703e64e9f81d190e74576f0af : SUCCESS in 48m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9f6fdf4a3b5449e9929ba36daf01d43a : SUCCESS in 1h 03m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e4cf533a2f9c441eba90b7e9f6c9d80b : SUCCESS in 1h 09m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9fca0a86213d4092875bb251588fbf0e : SUCCESS in 40m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c38f6b1fa64944b0b8517b8952cd8276 : SUCCESS in 54m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07367ab5ea2046dca43c32404da71298 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adffccad0b004a4e8b9bb2727aebe267 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25718af41e4f4b25bf630a81f0f88e38 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb07e71b2ab2461f9910ca6de5d66a2b : SUCCESS in 5m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ff460917fbf4b62b8079c623f726c9d : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec795aab6644c2288acaeea4f8a30c8 : SUCCESS in 7m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/205380ae57ec4d0ebf5d4cdafe549729 : SUCCESS in 31m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ee8678943e9348b0a201f3a2f6e804d2 : SUCCESS in 4m 51s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/112e768a15474ad396ee613b4f7daa94 : FAILURE in 46m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1eb5c0b596041edb36fee1f2a6b6bce9616eb27e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-08-19 16:26:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"869e46031b2546a67b934ff5b2e81a5fed0342a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 17:44:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b89cea3149fc45efb8e97f27fc0cf942\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b26c92a76a234909829155f2be63c0e1 : SUCCESS in 38m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/572ab54a9abd4173811d2182520fb5f1 : SUCCESS in 50m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4a2a05c988d4895a12039056efe1707 : SUCCESS in 52m 01s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6c6995ecea3841ceba329394dcf08bbb : SUCCESS in 47m 34s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/9f2d2f95dd9a416d9c5c0a5c55980f6c : SUCCESS in 49m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c66908d9d61049cd803f7cb3cc125a11 : SUCCESS in 47m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/10d8e6a275194bd59bdb00f55d198596 : SUCCESS in 1h 08m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d7377e8d1d0540079272e61ae724d24b : SUCCESS in 1h 11m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2c3d14ee5b6d4f2680ff0a6973469bc3 : SUCCESS in 42m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/df4477f33a2f4e168a7d25aa495ce66e : SUCCESS in 38m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2426547c474e4f4783a6864fb3769d36 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1e8d72e35c14c7daa370ae1f313af6e : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c90ede384b964b129dc988d69e66514f : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3e935806cc24283ba92ec39c6840a4e : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb07bfa575e947289dc02a29c306c074 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab65d56b897445df8ec6c100c69d01fe : SUCCESS in 6m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4bdd1fe4cc88471994b91e258274d0c3 : SUCCESS in 32m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6613681c400946f9aedd9de4a6fc8ff0 : SUCCESS in 5m 45s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/b1424da0139a40fd850dc27f958a473f : FAILURE in 46m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e68f49f175cf7583a0f15110fe302f7be0a6e9af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-08-28 16:44:05.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6dce1f1b1bc1030dc9f44ab52c6b70fabe2619a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 18:02:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f28aae58599a4aeca8d0f89198348eed\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/29aa98616ad54e65b8bdbc8b35f550ae : SUCCESS in 42m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1b0e3b3b4afb4fcbbd7a2c534948ccc5 : SUCCESS in 1h 02m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f51d30b9cbf463484cb8487029fe7dc : FAILURE in 53m 59s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/134f9a70c09e46319933f698281c0025 : SUCCESS in 49m 29s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/afedd8743b864d358c7ee45db011925c : FAILURE in 27m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/66ce93f340dd462c92900169aeac522b : SUCCESS in 46m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3a1b333beba2487ba8e322f0b4a2c19a : SUCCESS in 1h 07m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b080f63fc24740938252d71bfb406b50 : SUCCESS in 1h 11m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c333a4b37cb94a4ca35f032ef9ad005c : SUCCESS in 42m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4f0c24f8b68640e29b258c7813df5651 : SUCCESS in 50m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39e3c5deaf844e06a5b02e9a860ff1ec : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec1575034ca24b828c69b65b86d17fb7 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bfde0f928954949a4959569652c7c62 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a417304b4824ce4953771bb11726297 : SUCCESS in 5m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b6ec93f049f45d390319cabbcb8889d : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7773728fe91415b9e3ac0f647f6da5d : SUCCESS in 7m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/950c161f35a54c748eb53a3947d5a1e8 : SUCCESS in 30m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/99ff35bbcfa344439b5d8191d93f4988 : SUCCESS in 4m 31s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/6f6f7d57841f4d0f9439866b914e4e08 : SUCCESS in 47m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1531ca0ea5b9019eb41545e66509c9f1220c7e70","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-13 11:01:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a94f6180031bea9ec6ea959343eab943957c8666","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-17 15:21:04.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"dc6d94ba483c10ed04a01756f5bb90b58d126472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 16:32:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa0a052090eb42c0aa1a229bf5d9955c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dcf0bb063ca64c12a2e3f813511913ec : SUCCESS in 43m 12s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d668b67c2d684ae2b211e58de0d2d42c : SUCCESS in 58m 10s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8829fd8ca3649708d79e747f23f8ec7 : SUCCESS in 51m 58s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f21e2e72b9664378aa4b832c4f9edd6b : SUCCESS in 44m 45s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/481521c7377d461bb90a396b7a727279 : FAILURE in 23m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0349924e143d4723babddfe5e63fb5bd : SUCCESS in 44m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ee5e775e5f804bef85e14e2b79d23396 : SUCCESS in 1h 06m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/55af7929ec5a48f6ba7d485b9504261e : SUCCESS in 1h 07m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/256ffec4b67e46718bc6d971eaf9f013 : SUCCESS in 44m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/55581c8f886445fb8a3ce2dfaf6339fd : SUCCESS in 47m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20a4aa9cba0b47b7967331154f766b4b : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61d7a9404d1447c999bad6a4ea63a849 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1168399e9c9f45d2ab749862be54cb0f : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7fca82ec1f5d409b811a524744dc96fc : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0953d3d522ae4b4888a33054c6c20d21 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93254392d373485eabb4775f0d81ea1f : SUCCESS in 7m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eb19e551555f4132b7d4a41d5bb2f18f : SUCCESS in 27m 46s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/63f7ce937d504db6a267edcc730d4c8a : SUCCESS in 5m 25s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/9caccedf503f4d89a06fd4d8ecb89921 : FAILURE in 54m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b9cd976befc4c6bee98f773e31110be7e11be573","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-25 08:30:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4c8c8f987ca8f573e92fb2db0f25d52d2f350718","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 09:51:46.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d29d8f29c23a4ba09b7ce744f300ee71\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/108fd57ed8be487ca7d47e767e0fa37c : SUCCESS in 40m 16s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5b7789699ceb4e0d813baa7f37f449d7 : SUCCESS in 51m 02s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/600e0350c1344e03b1e323dad0abe916 : SUCCESS in 47m 56s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/17c2a4f7a3a74f74868a0fc8e545f800 : FAILURE in 47m 52s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/f717d2d3656e46358c223147dd74dc53 : FAILURE in 34m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d9187a59801b40d489587b02ce61e9f8 : SUCCESS in 49m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/02c96bafdedf4815b2cc79d42692b091 : SUCCESS in 1h 11m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7c20ccd8d3474958bd240be24ca9a271 : SUCCESS in 1h 14m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/431f19ce0ac34383816820c30d28142f : SUCCESS in 46m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/30860976a8004efbbf8df7d95551821c : SUCCESS in 49m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c00f346d38640208df1a186ec217bfb : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20a9dac38cf04b8296ea7d55a73a2d30 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0549d289e3f45dbb04ba807571c6462 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76a9131305a84fc88053f532a55e18e4 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a737c170c8a14e779d03dd872452cd94 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c04a8adb4df4970bfc9765133ef4a53 : SUCCESS in 8m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/72fd9402687643ef9b57e2b2810babfb : SUCCESS in 28m 12s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/12632cfab8de424f880c6968832adf7c : SUCCESS in 4m 46s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/bd1928cfcb7546e4ac9b20bbc3b8046d : FAILURE in 47m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"28dfbc5db024b8ae24e5bbaaaeacc3ef95dce470","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-25 10:09:19.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"812664c2c8349996957231c4cb0b36122b6eb1dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 12:03:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f14f5569055f4edaa637d656ddd81ad8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d400af963f994e58aed1244ca8b5460d : SUCCESS in 43m 54s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/985417ef2a9541f694742b926f76d750 : SUCCESS in 1h 06m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/834e15e7e4ce4fe087fc0019812e95f8 : SUCCESS in 51m 18s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/905bc3c59e0f4c4d996870b97b0ebf30 : SUCCESS in 50m 33s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/1390213eb8bf4e719a0bd99894c80c69 : FAILURE in 1h 08m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8d1b6a4324454a298b6df2a4f7abbd16 : SUCCESS in 47m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1f920583045541dabe43b685d0b24db0 : SUCCESS in 1h 11m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/25cede14fe56464e8187cd05bd8ca75d : SUCCESS in 1h 47m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/eb9660e2ded84643bf31609b4fb36f83 : SUCCESS in 42m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6e1334ea779d44549b2e513e01bdb29f : SUCCESS in 45m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9023c0c280e47a8a072fc93ff4d5442 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc60a30d166e48b8a83d9115a27336e5 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67c0c78c58944479ac9ec7f6df39bac8 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c49ee86f8cdf4cfca72b3e9f7cdc0fd4 : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d817d614f8a4d98abf892797ae38909 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52254c0ba73a4a678e001ede4f1fc8c6 : SUCCESS in 7m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/83fa2fed8f254fd79e95bbc404689711 : SUCCESS in 27m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1e83903030d3461c95ab3bf9b44dd6d6 : SUCCESS in 5m 42s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/2e428fd87f1d4577b83503f95457750c : FAILURE in 46m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b3d716044bcb86bf1fca2c8bbfe2fa93d8454822","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-09-27 15:41:22.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"bd91080475ca30ff86b71dd8f59cd5bf895e443c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 15:42:10.000000000","message":"Patch Set 6: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/designate-tempest-plugin on branch master.  The\nproblem was:\n\n  Job designate-bind9-multipool in openstack/designate-tempest-plugin is not permitted to shadow job designate-bind9-multipool in openstack/designate\n\nThe problem appears in the the \"designate-bind9-multipool\" job stanza:\n\n  job:\n      name: designate-bind9-multipool\n      parent: designate-bind9\n      vars:\n        devstack_localrc:\n          DESIGNATE_BACKEND_DRIVER: multipool-bind9\n        devstack_local_conf:\n          test-config:\n            \"$TEMPEST_CONFIG\":\n  ...\n\n  in \"openstack/designate-tempest-plugin/.zuul.yaml@master\", line 27","accounts_in_message":[],"_revision_number":6},{"id":"92ce0992e2825ce28e2685b2833acb6ab116ce15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-26 00:46:29.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"9456e6947ac4ffbdecd7b4a1c3bb05e3d42352a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 02:04:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72cc083fd1a947ad877d52445e2e6f4b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ec950884f6704b0088a52a3efe7a84a9 : SUCCESS in 42m 35s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/92f2e31c264b4f15a2b3dea16042a528 : SUCCESS in 50m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/844457bb17364831ae6d0a1a7f3dba65 : SUCCESS in 38m 57s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3ed8bfb287ec461191f3506136e8dda8 : SUCCESS in 46m 16s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/ddb8a44e6fd148dfa5eeba96add1f459 : FAILURE in 18m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f304b3324f114d3699e67260551cda0d : SUCCESS in 37m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9de8ec88158642d6a3010393e19bcc1d : FAILURE in 52m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/104852c5849142439e2133ce86d6d070 : SUCCESS in 56m 33s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/1075a0cdd3a64d32bc9a32eaa8a40ea7 : SUCCESS in 1h 09m 44s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/8e01c182d33846c1958ca0c378ecf57b : SUCCESS in 52m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5a01a1fd5aa344c2b8c0aea76e87fb35 : SUCCESS in 40m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7b0c4d00277a4b57aa7d855f02de79bb : SUCCESS in 49m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/065864e261de4079975343255c526807 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfcc26a7c2fc457c9cca39ad3e737b1f : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea7aeb779f66483f962e76640100d89e : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/51cdb095c4f94b15bc284ca7fd280c59 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2ac8967e50f46d6b0cc75c495125559 : SUCCESS in 6m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b07c73511f9144ae9b9f568ccee0cdbe : SUCCESS in 31m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d0da381acc3548e4bbeea2c1168c5b3a : SUCCESS in 4m 18s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/247f253a083e46ff8bb68b8f0b84fd5d : SUCCESS in 37m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"59657befa9a77b563f977ab564433c4e70c9631d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-12-19 12:05:58.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"862ca8c7156a0aaaee8b2e8a9b88119af50a8d96","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-12-19 12:58:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a00eb4c4c005f832e16f99b8a2237152afd6f886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 13:32:25.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97839a0c674246f8844ca87cf27161da\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/47ab699e834c46799eee6f1cd493c87e : FAILURE in 52m 54s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1147114660924edf9d02096bf23d6da1 : SUCCESS in 51m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/51c16769ac834a9d998c63a019a125a5 : SUCCESS in 40m 42s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5bab17b1c05e4dcfb5f4948978085e0a : SUCCESS in 41m 02s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/f070ae9626bc458fa29cf4fb659011ff : FAILURE in 17m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/86bbb70857454dfb841db63b8e8a5ecf : SUCCESS in 41m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/99472c8650f4477b909388760cf842ca : SUCCESS in 1h 00m 46s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d912b362cebe4591b88886bbac317d13 : SUCCESS in 58m 05s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/0395cabc8ad547d6a4ffcbce3c047091 : SUCCESS in 56m 22s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/23930cf0a82346d5b2a1b201bb182532 : SUCCESS in 1h 20m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e8a308af71e54556b8c71c9513e50c95 : SUCCESS in 38m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/be118dd24142470090aa8d4517706556 : SUCCESS in 53m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/372f72412ce54e3ab70e6da78677d28b : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d229129a162647d1b09fbbef84467510 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e43373e7455a439fbeb4df32bc2ffb4d : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7245c940b76e44a68d6ebdd3724b7603 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3eb08f80f7442a28d7baff239f92865 : SUCCESS in 6m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6b0cde449e534a7dba23b586f3eaef7f : SUCCESS in 33m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8de60ee3ba9b41cab7e6f2883ec7c2ac : SUCCESS in 3m 46s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/d9ec6880e8c642cfbe815f14f430f25c : SUCCESS in 46m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b595c94d85041d637888daad0fcb3b5ac2ce8bc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-01-17 15:10:37.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"f99b58d185f43d2ca993aa4b81c19c087a8a497a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 15:12:15.000000000","message":"Patch Set 8: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/designate-tempest-plugin on branch master.  The\nproblem was:\n\n  Job designate-bind9-multipool in openstack/designate-tempest-plugin is not permitted to shadow job designate-bind9-multipool in openstack/designate\n\nThe problem appears in the the \"designate-bind9-multipool\" job stanza:\n\n  job:\n      name: designate-bind9-multipool\n      parent: designate-base\n      nodeset: openstack-single-node-jammy\n      vars:\n        devstack_localrc:\n          DESIGNATE_BACKEND_DRIVER: multipool-bind9\n        devstack_local_conf:\n          post-config:\n  ...\n\n  in \"openstack/designate-tempest-plugin/.zuul.yaml@master\", line 19","accounts_in_message":[],"_revision_number":8},{"id":"d999a6ab750b7b49aea56d902582aca9d529d8f5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-12-24 21:30:18.000000000","message":"Abandoned\n\nEventually we added multipool job in another patch, this one is obsolete now","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"c029ad3ec93215c6062301f89fda27c1a71dd881","revisions":{"7f046c84c70dab2fbd88740ff31970d75379a2fd":{"kind":"REWORK","_number":1,"created":"2024-08-14 10:42:22.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/74/926274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/926274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/926274/1"}}},"commit":{"parents":[{"commit":"0e9294aca19d1909209a8ef730fda60f9914df0f","subject":"Multi-pools implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e9294aca19d1909209a8ef730fda60f9914df0f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-08-14 10:40:24.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-08-14 10:40:24.000000000","tz":120},"subject":"Add designate-bind9-multipool job","message":"Add designate-bind9-multipool job\n\nThis patch adds a job that tests multipool-bind9 deployment and tests.\n\nChange-Id: I7a1d58d324413c1116cf0eb8939c939767b9f54f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7f046c84c70dab2fbd88740ff31970d75379a2fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7f046c84c70dab2fbd88740ff31970d75379a2fd"}]},"branch":"refs/heads/master"},"b765e54b4f212871e5ed5f079c73d567102a8f47":{"kind":"REWORK","_number":2,"created":"2024-08-14 10:48:22.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/74/926274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/926274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/926274/2"}}},"commit":{"parents":[{"commit":"0e9294aca19d1909209a8ef730fda60f9914df0f","subject":"Multi-pools implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e9294aca19d1909209a8ef730fda60f9914df0f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-08-14 10:40:24.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-08-14 10:48:22.000000000","tz":0},"subject":"Add designate-bind9-multipool job","message":"Add designate-bind9-multipool job\n\nThis patch adds a job that tests multipool-bind9 deployment and tests.\n\nChange-Id: I7a1d58d324413c1116cf0eb8939c939767b9f54f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b765e54b4f212871e5ed5f079c73d567102a8f47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b765e54b4f212871e5ed5f079c73d567102a8f47"}]},"branch":"refs/heads/master"},"897590b3e0fa72e57c9375cc95e8dbc24712bf53":{"kind":"REWORK","_number":3,"created":"2024-08-28 16:44:05.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/74/926274/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/926274/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/926274/3"}}},"commit":{"parents":[{"commit":"0e9294aca19d1909209a8ef730fda60f9914df0f","subject":"Multi-pools implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e9294aca19d1909209a8ef730fda60f9914df0f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-08-14 10:40:24.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-08-28 16:44:05.000000000","tz":0},"subject":"Add designate-bind9-multipool job","message":"Add designate-bind9-multipool job\n\nThis patch adds a job that tests multipool-bind9 deployment and tests.\n\nChange-Id: I7a1d58d324413c1116cf0eb8939c939767b9f54f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/897590b3e0fa72e57c9375cc95e8dbc24712bf53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/897590b3e0fa72e57c9375cc95e8dbc24712bf53"}]},"branch":"refs/heads/master"},"b7faf47cde87d3c8b71a8dc6778ad1491a060198":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-09-17 15:21:04.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/74/926274/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/926274/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/926274/4"}}},"commit":{"parents":[{"commit":"0e9294aca19d1909209a8ef730fda60f9914df0f","subject":"Multi-pools implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e9294aca19d1909209a8ef730fda60f9914df0f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-08-14 10:40:24.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-09-17 15:21:04.000000000","tz":0},"subject":"Add designate-bind9-multipool job","message":"Add designate-bind9-multipool job\n\nThis patch adds a job that tests multipool-bind9 deployment and tests.\n\nDepends-On Ica5005cd98cc76656ccdbe1681a5c9a5a611e29e\nChange-Id: I7a1d58d324413c1116cf0eb8939c939767b9f54f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b7faf47cde87d3c8b71a8dc6778ad1491a060198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b7faf47cde87d3c8b71a8dc6778ad1491a060198"}]},"branch":"refs/heads/master","description":"Edit commit message"},"850a4611a1a0502bf735772e199ffa1c72574908":{"kind":"REWORK","_number":5,"created":"2024-09-25 10:09:19.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/74/926274/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/926274/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/926274/5"}}},"commit":{"parents":[{"commit":"0e9294aca19d1909209a8ef730fda60f9914df0f","subject":"Multi-pools implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e9294aca19d1909209a8ef730fda60f9914df0f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-08-14 10:40:24.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-09-25 10:09:19.000000000","tz":0},"subject":"Add designate-bind9-multipool job","message":"Add designate-bind9-multipool job\n\nThis patch adds a job that tests multipool-bind9 deployment and tests.\n\nDepends-On Ica5005cd98cc76656ccdbe1681a5c9a5a611e29e\nChange-Id: I7a1d58d324413c1116cf0eb8939c939767b9f54f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/850a4611a1a0502bf735772e199ffa1c72574908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/850a4611a1a0502bf735772e199ffa1c72574908"}]},"branch":"refs/heads/master"},"573df3ea041fc74910d8528cea3e48f3491c2d14":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-09-27 15:41:22.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/74/926274/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/926274/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/926274/6"}}},"commit":{"parents":[{"commit":"0e9294aca19d1909209a8ef730fda60f9914df0f","subject":"Multi-pools implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e9294aca19d1909209a8ef730fda60f9914df0f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-08-14 10:40:24.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-09-27 15:41:22.000000000","tz":0},"subject":"Add designate-bind9-multipool job","message":"Add designate-bind9-multipool job\n\nThis patch adds a job that tests multipool-bind9 deployment and tests.\n\nDepends-On: Ica5005cd98cc76656ccdbe1681a5c9a5a611e29e\nChange-Id: I7a1d58d324413c1116cf0eb8939c939767b9f54f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/573df3ea041fc74910d8528cea3e48f3491c2d14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/573df3ea041fc74910d8528cea3e48f3491c2d14"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fc58d33dd045c3d8982f4a37af05e07a1328524a":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2024-11-26 00:46:29.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/74/926274/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/926274/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/926274/7"}}},"commit":{"parents":[{"commit":"0e9294aca19d1909209a8ef730fda60f9914df0f","subject":"Multi-pools implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e9294aca19d1909209a8ef730fda60f9914df0f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-08-14 10:40:24.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-26 00:46:29.000000000","tz":0},"subject":"Add designate-bind9-multipool job","message":"Add designate-bind9-multipool job\n\nThis patch adds a job that tests multipool-bind9 deployment and tests.\n\n\nChange-Id: I7a1d58d324413c1116cf0eb8939c939767b9f54f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc58d33dd045c3d8982f4a37af05e07a1328524a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc58d33dd045c3d8982f4a37af05e07a1328524a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c029ad3ec93215c6062301f89fda27c1a71dd881":{"kind":"REWORK","_number":8,"created":"2025-01-17 15:10:37.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/74/926274/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/926274/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/926274/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/926274/8"}}},"commit":{"parents":[{"commit":"d6e3f96cd844208b665ea720837b7069bd25a4a6","subject":"Merge \"doc: Use dnf instead of yum\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6e3f96cd844208b665ea720837b7069bd25a4a6"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-08-14 10:40:24.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-01-17 15:10:10.000000000","tz":60},"subject":"Add designate-bind9-multipool job","message":"Add designate-bind9-multipool job\n\nThis patch adds a job that tests multipool-bind9 deployment and tests.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/927419\n\nChange-Id: I7a1d58d324413c1116cf0eb8939c939767b9f54f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c029ad3ec93215c6062301f89fda27c1a71dd881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c029ad3ec93215c6062301f89fda27c1a71dd881"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
