)]}'
{"id":"openstack%2Fdesignate~926446","triplet_id":"openstack%2Fdesignate~master~I4d5a48d4d0ad9c39d731c29048bcddd514b02cff","project":"openstack/designate","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-09-03 19:03:42.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4d5a48d4d0ad9c39d731c29048bcddd514b02cff","subject":"DNM: testing oslo.policy 4.4.0","status":"ABANDONED","created":"2024-08-15 21:49:41.000000000","updated":"2024-09-03 19:03:42.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"40106d2648626378984d4d290cec915ba605761e","_number":926446,"virtual_id_number":926446,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-08-16 03:44:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-16 00:02:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3e2603246a121d8f62210c6cafee6bd48cf38ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-15 21:49:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"538b3204a4b2af1acc021e41df3d6a3f3a8b6e4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 00:02:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bf08363d7bc4a75bdd3da53324aa385\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c8139b1285f74b7a8cb5ec54ec745922 : FAILURE in 51m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1db630a095014df88cabe57ed7d3c4a0 : FAILURE in 1h 03m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/60a85d6e4b3242098197ed26aed6d0e5 : FAILURE in 57m 38s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/52b96abad1224dac81bc4c1f263546b3 : FAILURE in 46m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/56289dbb029e4fd2ad1c2c9b498138ea : FAILURE in 36m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9f025e770ed147049ea64651f9132d4a : FAILURE in 1h 13m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1e578a00d0254d99be0e414175a0aa94 : FAILURE in 1h 08m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ea265de1aac741cfb4f7956b72da522b : FAILURE in 31m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3a22358bb57e4b16b8be5e3644a0661a : FAILURE in 34m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c5e4fcff2ee4634b09e14d8f22e6581 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c6c4eab67de4f468d9ef5932e8e8487 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43926820f0dd4c71abc0c2e2ca709ea6 : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af4766c80a29424eb02e6d8a0e8cbb0b : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0efc3a9f22bd4b42ba93df8cb229fc8b : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1db3b52cfef04d5fa6d481e5f91a4bb7 : SUCCESS in 7m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ab2924b5e64742b8a06b07ab9ba4d9f7 : FAILURE in 32m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/53c835ae0e254f47ab0debf07a6ffda1 : SUCCESS in 5m 37s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/61201efcf6994c72b085d5a08b43d4bf : FAILURE in 48m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fac4f7b7fee30066341b2a186e803dc48d5948a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-16 02:01:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bdd6deeaaeb118305f3f8d24fe0e1dc3e4d61294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 03:44:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/628993566b284fbba75b2f8b597b0e91\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ce9ef6acc2cd4316959b295f8229f71f : SUCCESS in 48m 53s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3bf879011f674f708ff7a8004388b393 : SUCCESS in 1h 01m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb20c5be599e4a368d8501e4b4748bd3 : SUCCESS in 1h 04m 07s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b29c666eb90c4f5bba94d41c8369e1cf : SUCCESS in 1h 06m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7fbc7989c3ca423e93692b0540ccfa26 : SUCCESS in 46m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1f49bb98bc1542449509c0b48827318c : SUCCESS in 1h 06m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3c02903fa1604d04af9cb1faa74aa288 : SUCCESS in 1h 10m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/105dae428eca40398436ec8a4329791a : SUCCESS in 58m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/98e245beaa9a446d95a997aff67b1f6e : SUCCESS in 1h 02m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16d979c0804a45b1a20e35c0006fca0f : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f277c5f968894958b125e8e89238fff6 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e04f9843e9a44dc98871c94f2eef4a0 : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e4df0e3a7e040cf9768209d4dc62362 : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73eafa4a103f48adab8f96dea8f84567 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc875570f374ab5aa9e9d293b0ff72f : SUCCESS in 9m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1d4da5b4c3f415eab90848630bab847 : FAILURE in 34m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/58147d5302724c0494957998461c214a : SUCCESS in 5m 07s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/2d95af4fd9054ab4beb336061fbd6f52 : SUCCESS in 1h 04m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e8096d8bb6d201f1e91ccf87f0e3b46b3be49a88","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-23 19:00:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"3dfc7af0192f2ad5e39b0828ca687028daf75089","tag":"autogenerated:gerrit:restore","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-30 18:14:45.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"af29928d3399c0a29509e525f4e3ee01c6c25d73","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-30 18:26:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2edc9414993494ddb025a37177c4e06a8d86b88d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 19:30:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c96010df7274bc7ac7c2e24b32e8b48\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/28e94deb8ba64810bd1ea9f72d122c16 : SUCCESS in 48m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/622ae2d03ef04ab1aeab8ed36cecfbb5 : FAILURE in 28m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2804223e8cc4d9881a4e56f91c50a85 : FAILURE in 25m 25s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/58fe5261dc974fec8746b51890821fd9 : FAILURE in 42m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/94bd18420f2248318441dddabb42d137 : SUCCESS in 41m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/76ab5513b1094ba9af1524dfd0b0aee5 : SUCCESS in 57m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/29d745419958435fbf2d4765f7b2512c : SUCCESS in 1h 08m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/349396aae2a0429a8ef0a1d84f454fc0 : SUCCESS in 49m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/93139252991c441e9438f7cc2a71835c : SUCCESS in 43m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d94a658d26cd48caa7adf01f806dcd72 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/774260cc5aa14acfaa6edd1d8ab5a39c : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0abd6237959041c2ad68c3e4179087e0 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/744553dba6ad47b09bc14b72593c1e64 : SUCCESS in 7m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69efe30d27004448959ed979dcba6f3c : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2058f928a1084df0b3cd0a814e9014d6 : SUCCESS in 7m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fc2ac878898c4aec93cc87ca542cf557 : FAILURE in 30m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/476df0e2f94042f5b5f46a57fcda977e : SUCCESS in 6m 01s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/9a5f97bc1383461bb8bb230367942bfd : SUCCESS in 36m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"40106d2648626378984d4d290cec915ba605761e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-03 19:03:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a8257ffbab11edb4257ab61beb2e38c6e820e87a","revisions":{"65233e80ad1162c445a61a85e004585fce143111":{"kind":"REWORK","_number":1,"created":"2024-08-15 21:49:41.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/926446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/926446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/926446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/926446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/926446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/926446/1"}}},"commit":{"parents":[{"commit":"6867be867b8d9cb5f2810134c9fad329e79df5a6","subject":"Update oslo.policy version for new defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6867be867b8d9cb5f2810134c9fad329e79df5a6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-15 21:49:10.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-15 21:49:10.000000000","tz":-420},"subject":"DNM: testing oslo.policy 4.4.0","message":"DNM: testing oslo.policy 4.4.0\n\nChange-Id: I4d5a48d4d0ad9c39d731c29048bcddd514b02cff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/65233e80ad1162c445a61a85e004585fce143111"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/65233e80ad1162c445a61a85e004585fce143111"}]},"branch":"refs/heads/master"},"a8257ffbab11edb4257ab61beb2e38c6e820e87a":{"kind":"NO_CHANGE","_number":2,"created":"2024-08-16 02:01:58.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/926446/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/926446/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/926446/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/926446/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/926446/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/926446/2"}}},"commit":{"parents":[{"commit":"f0ae34098b47c9a591099ee84f489b17088b1633","subject":"Update oslo.policy version for new defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0ae34098b47c9a591099ee84f489b17088b1633"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-15 21:49:10.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-08-16 02:01:58.000000000","tz":0},"subject":"DNM: testing oslo.policy 4.4.0","message":"DNM: testing oslo.policy 4.4.0\n\nChange-Id: I4d5a48d4d0ad9c39d731c29048bcddd514b02cff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a8257ffbab11edb4257ab61beb2e38c6e820e87a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a8257ffbab11edb4257ab61beb2e38c6e820e87a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
