)]}'
{"id":"openstack%2Fdesignate~927060","triplet_id":"openstack%2Fdesignate~master~I0b4ec333a3e00011c24ad1745499501fe1572e7c","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-01-06 03:48:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0b4ec333a3e00011c24ad1745499501fe1572e7c","subject":"pdns: Accept hostnames for master hosts","status":"MERGED","created":"2024-08-24 14:00:53.000000000","updated":"2025-01-06 03:49:22.000000000","submitted":"2025-01-06 03:48:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927060","meta_rev_id":"9486db84daf6b3bc1fb50589902ad0becca89465","_number":927060,"virtual_id_number":927060,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-06 03:48:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2024-09-13 11:06:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2024-08-27 16:01:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-08-26 23:51:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-01-06 02:51:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-24 15:25:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-26 23:51:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-08-27 16:01:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-09-13 11:06:27.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"6fcb656405804818c01fe611e075eac28eaa4d50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-24 14:00:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71f4aba43da96248bd48088dcdb2063a04839f97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 15:25:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b355c77c5efb4da58af8cf21c20ea45f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/561d1bd565dd49bc8993039f3acc3b8c : SUCCESS in 43m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/57114fa446d44337bf68a6736e811312 : SUCCESS in 1h 06m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bcea5432aa63489a98f035a68de65aa7 : SUCCESS in 53m 20s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/16271426dd5949d193e00387e2f58b98 : SUCCESS in 46m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0ca888e576904bc48ebba40eec0b9b26 : SUCCESS in 50m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ad1513885fb048458bc82520cf26fe21 : SUCCESS in 51m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d106f8bc4320481babe5c8e97580c6ba : SUCCESS in 1h 18m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ed63a65fa1ad496db962e7908ab1d748 : SUCCESS in 47m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/74a6bf3c6b184582849b6c1b262bd2a1 : SUCCESS in 35m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c6b3773344b4062bb4bfc1e683e1888 : FAILURE in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc5518d300e04a6aa29fc3c9a6d2a9e8 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6bc730f0a7344dfa9365289805ccf36 : FAILURE in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2adf022487374d1fae8db0c13fd56e2c : FAILURE in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c2625de943f4fc789cde7c0462e3c00 : FAILURE in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/294ac442ca5b4f35814e44fbb8ba5117 : SUCCESS in 7m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5d813ac50bbe42f2b26e3c7a3f02cba2 : SUCCESS in 30m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b0cd41e72b0c4775b3770ab5ba4a0c7e : FAILURE in 4m 16s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/47f575554bae4b6aa4ea9b09f090946a : SUCCESS in 35m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"776ffd5581eb5bcd51d9c3a0b8ef2e587124619e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-26 00:02:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d8ab3c638299d8e5c60a23d38d8ee4009b6b0b1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 01:09:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/293992ee7e6344199462bfd80cf27912\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b6af743df393420b8601f259a5986ac8 : SUCCESS in 32m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e544b687a60c4cab85fad82bb8d7021a : SUCCESS in 40m 09s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5382203d7fa24a6f9aba58addae0be97 : SUCCESS in 39m 00s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3ca8b302fc834942ae3ca379e168645a : SUCCESS in 33m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/773b33af286e45048d2be3fa0fba6fab : SUCCESS in 32m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/080b801ae3104ae386d3a25f4f9faf71 : SUCCESS in 55m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d073aa6abc2c42dd84c605f17fa92a0a : SUCCESS in 1h 06m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ea1053b9d0dc4942be45ef95ae34f145 : SUCCESS in 33m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/424736ac0bfb4ce5b2af52772d5cd9f5 : SUCCESS in 49m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b528cf9bd54f479ca8f98583e3e5f76b : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/896ee804298c47c1a4813e6f5ac857a3 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24a77787038841bf98de69cc92f541e2 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3281989dca54dd4b7c8fc77733d85dc : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae3cc06579d34bbaa58e2878612dab0a : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2827dce36c2459598b45f229b1a3977 : SUCCESS in 6m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d3c43c695c644338800ae37feff7a04e : SUCCESS in 31m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e62abdbd05574a65ae6c1d181a25c99f : SUCCESS in 4m 10s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/979aa4dc7e4b4af291c5e824a344f392 : FAILURE in 36m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"273076082bff740bdb14337747a4339e11f8c33b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-26 23:51:46.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b081986c4f6d42ebd06fec277b1ece529b719552","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-27 16:01:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"42b5dfdeb89e123d5229c966cad4731087e1f8fa","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-13 11:06:27.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a1da80adf7b4484d818ad1d5c45cd6e92a1f485","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-06 02:51:47.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6de2262836ae214b10605778245e17a108613d3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 02:52:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"120121c595cd1cd44ebe785ad49ddbb07c1f2247","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 03:48:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92f1d078b41f4cb9849ea86a8285ec14\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/64bf3b940af743e699683c625004c13d : SUCCESS in 36m 49s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/85b0ed599d1240548deafdeed9291222 : SUCCESS in 38m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/247d8b3dcf2a4aa596bf7ab1c7c42d97 : SUCCESS in 41m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c12d6b9c4ef745ecb6954bda233502d3 : SUCCESS in 51m 58s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/178d89c04fb647cdb84dbe3299da5aa7 : SUCCESS in 54m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2337b65203ff49a8803a707553919ad2 : SUCCESS in 38m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cc3c4b0b21074c5487a25624eed8265a : SUCCESS in 36m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8236e91cd21d4fac84bb04d3ff066aa8 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6a5e629f5104aaa93ce8647917f9d11 : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fdea94e3177b45e38647657749335899 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4797e8c4c250467c951ba5f4d24652f3 : SUCCESS in 7m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f0789fc1dbb34bbf8a4c300b15e58934 : SUCCESS in 32m 05s","accounts_in_message":[],"_revision_number":2},{"id":"8016293a185047560d318c3eb0b0a1d616142efc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 03:48:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9486db84daf6b3bc1fb50589902ad0becca89465","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 03:49:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff3044c92e68474fb50baf65a4c4202b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57ac94a3da6d476a810d68fc55f34313 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0fa4eb2f57d4f24cfcc0ff828ebdeb5760538354","revisions":{"f1084df7f65a0f2185662579748a9e3a4a32e667":{"kind":"REWORK","_number":1,"created":"2024-08-24 14:00:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/927060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/927060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/927060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/927060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/927060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/927060/1"}}},"commit":{"parents":[{"commit":"50f686fcffd007506e0cd88788a668d4f57febc3","subject":"Update functional tests for new RBAC system scope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50f686fcffd007506e0cd88788a668d4f57febc3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-24 13:58:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-24 13:58:33.000000000","tz":540},"subject":"pdns: Accept hostnames for master hosts","message":"pdns: Accept hostnames for master hosts\n\nin addition to IPv4/v6 addresses.\n\nChange-Id: I0b4ec333a3e00011c24ad1745499501fe1572e7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f1084df7f65a0f2185662579748a9e3a4a32e667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f1084df7f65a0f2185662579748a9e3a4a32e667"}]},"branch":"refs/heads/master"},"0fa4eb2f57d4f24cfcc0ff828ebdeb5760538354":{"kind":"REWORK","_number":2,"created":"2024-08-26 00:02:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/927060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/927060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/927060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/927060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/927060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/927060/2"}}},"commit":{"parents":[{"commit":"50f686fcffd007506e0cd88788a668d4f57febc3","subject":"Update functional tests for new RBAC system scope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50f686fcffd007506e0cd88788a668d4f57febc3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-24 13:58:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-26 00:02:06.000000000","tz":540},"subject":"pdns: Accept hostnames for master hosts","message":"pdns: Accept hostnames for master hosts\n\nin addition to IPv4/v6 addresses.\n\nChange-Id: I0b4ec333a3e00011c24ad1745499501fe1572e7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0fa4eb2f57d4f24cfcc0ff828ebdeb5760538354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0fa4eb2f57d4f24cfcc0ff828ebdeb5760538354"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
