)]}'
{"id":"openstack%2Fdesignate~927748","triplet_id":"openstack%2Fdesignate~master~Ib1b6ef79154122ef67748bd2bfd4ad1b748b00a3","project":"openstack/designate","branch":"master","topic":"fix_missing_pool_lines","hashtags":[],"change_id":"Ib1b6ef79154122ef67748bd2bfd4ad1b748b00a3","subject":"Fix missing keys in multipool pools.yaml","status":"ABANDONED","created":"2024-09-02 16:23:43.000000000","updated":"2024-09-11 14:44:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0d7b06951c4a9629076a3f8001f60974a89d7fdb","_number":927748,"virtual_id_number":927748,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-09-02 17:41:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-02 17:41:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a0cf52dd0966c3e952a147d21aae53005352f64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-02 16:23:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6ec276727a62b8790d9c37efef3692259e2d387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 17:41:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/601a2434f0354a6e8289fc04d386c6a6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/50b62cf517aa4bb888e0a36c74f95cd1 : SUCCESS in 51m 57s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f9fdb90d26cb4481ac39eec9c914dcff : FAILURE in 30m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3036581dccf14c19952cc4fc4dddfb5f : FAILURE in 13m 48s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3c1a546fbb314f1ea4147a1a84b13026 : SUCCESS in 34m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cbc5bd483a924c60a545c5582f45e121 : SUCCESS in 35m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6f402fc7ce69422ab2275809ee8d2cd6 : SUCCESS in 55m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2ecffb80dc1347c89fcc1d86a8c734ac : SUCCESS in 1h 11m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/78d0a3a4d5db4083b4b8dfd305527c12 : SUCCESS in 37m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/08cdfec76a9f4e9f955060a6db0e3d4e : SUCCESS in 52m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2974a80a7ecc4272ab4ede081c130e10 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cb06ef8f98746268a0c26c6042b3c92 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ba6d21d37144d20b53cdf16fb2a1a75 : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/99f1421f206f41e19fab9ac380e0e81a : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a61b7bb5e8b42bdb27d6409861fa439 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f6df6286caa463e9384b2f2ad866a34 : SUCCESS in 6m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e8af2e2f9b254c229f978fe2ff8dd654 : SUCCESS in 29m 55s (non-voting)\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/bbd1ceddc45c4a50a5f4c8a540af010a : SUCCESS in 3m 54s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/744678b65c684816844ff826f2916701 : SUCCESS in 36m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d7b06951c4a9629076a3f8001f60974a89d7fdb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-11 14:44:51.000000000","message":"Abandoned\n\nThe change is not needed, the config file has a reference to the key.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"09cec94cc34b9b10fd7a81c0e04e2115e6bd8ebe","revisions":{"09cec94cc34b9b10fd7a81c0e04e2115e6bd8ebe":{"kind":"REWORK","_number":1,"created":"2024-09-02 16:23:43.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/48/927748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/927748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/927748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/927748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/927748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/927748/1"}}},"commit":{"parents":[{"commit":"cfa330c0fddd356b3053dd5050c76cfb42fcda64","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfa330c0fddd356b3053dd5050c76cfb42fcda64"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-09-02 16:20:03.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-09-02 16:23:03.000000000","tz":120},"subject":"Fix missing keys in multipool pools.yaml","message":"Fix missing keys in multipool pools.yaml\n\nBoth rndc_key_file and clean_zonefile config options are missing from\nthe first pool of the generated pools.yaml when using multipool-bind9\nbackend.\n\nThis patch adds them to that pool config options.\n\nChange-Id: Ib1b6ef79154122ef67748bd2bfd4ad1b748b00a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/09cec94cc34b9b10fd7a81c0e04e2115e6bd8ebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/09cec94cc34b9b10fd7a81c0e04e2115e6bd8ebe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
