)]}'
{"id":"openstack%2Fdesignate~927792","triplet_id":"openstack%2Fdesignate~master~I477cc96519e7396a614f92d109867222207ec388","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-09-03 16:40:31.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-09-03 16:40:31.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-09-03 16:15:18.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-09-03 15:34:14.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I477cc96519e7396a614f92d109867222207ec388","subject":"Add \"admin\" role to the designate user created by devstack plugin","status":"MERGED","created":"2024-09-03 08:54:04.000000000","updated":"2024-09-13 17:02:44.000000000","submitted":"2024-09-03 16:40:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927792","meta_rev_id":"2407d38684e6f700f289d744676d56cc9d283fad","_number":927792,"virtual_id_number":927792,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-03 16:40:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2024-09-03 15:18:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2024-09-03 15:34:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-09-03 16:15:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2024-09-03 15:34:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-03 09:35:23.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-09-03 09:35:23.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-09-03 09:35:23.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-09-03 10:37:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-03 15:18:06.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"f76935fc531c65409102a030c7f130c41f6f39e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-09-03 08:54:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b391cc0d7f819d722bdcfbd6249139d846ae26e1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-09-03 09:33:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aaf5fc4aa2f1d20c4a1cd497a7d512a09d35462d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 10:37:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ce0eecf03aa4611970e396ffb97aa18\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cbf05b31cfe74d509e50c49050a2ecae : SUCCESS in 49m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d72e0d23dbd747ee841f8562d00f9dcc : FAILURE in 29m 57s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d98bef1c9737461d905fea851d98876b : FAILURE in 26m 08s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0a3b6c05ad4e4c1193436a185f23272a : SUCCESS in 52m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c98c53a3538847b1bb9dcaa66250cf88 : SUCCESS in 42m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2233d51cada9406b993cdd6845b197da : SUCCESS in 1h 11m 03s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d42defb474da41fdb89f82e22e42fb9c : SUCCESS in 1h 06m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f72475d8f2c84685855ff3d2767eb1ad : SUCCESS in 50m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/35f49641f68d43c3a405992f930325cb : SUCCESS in 49m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3d7914cbe4c4822a916df988feca594 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1a4a412df234ac08aa7f7752f6a3e1e : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b6c18c218fe4ef6a60408ea216a39a4 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3cbc7e6726754b458bd499cb41c7d67a : SUCCESS in 5m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4be3dc621b84b51a10b161dcdc3f1e7 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc79b6bf877e4a70b1ed4082b453a818 : SUCCESS in 6m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4b4fd653366a42c5b15cf0a5efe8fbe1 : SUCCESS in 30m 48s (non-voting)\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e160eb5827214f9f97bc129e774315ea : SUCCESS in 4m 51s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/94fb8e225b50471689d7affaae07bfe2 : SUCCESS in 45m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d9d559c09dd921d6590e156cafcf95f98d6e93e5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-09-03 15:18:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d1752cd040a3e09582e02b6e67e23fea3069d08b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-09-03 15:34:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58b2a080f7dae7e87390675d2d4f804282963d57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 15:34:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb44a17f3fa79378a5195d94d3863b6a66f447ca","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-03 16:15:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb15321ddbd6fd82c32716cfd18faf042d91f98e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-03 16:18:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4fc30cbadef1dd5fd8b877e550bbcf643f9bbb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 16:40:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ad3610ba4d04d64918bbc6f0172db87\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/de908ba4545d40508a790e60a26a344d : SUCCESS in 41m 45s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6453c4de947746a9981b10e1cbca8c81 : SUCCESS in 42m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c7d4de652fb446d8afd1fdf9bf51ab86 : SUCCESS in 44m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/78a8e5e7a19540c99da615437fe1c559 : SUCCESS in 1h 03m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0e3ade26cf1d49a1961f5345aff6c205 : SUCCESS in 40m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0e242bb9f71b45b8b153c1c99f834ff9 : SUCCESS in 48m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16b50c58e8414f7582a3692cf7e6c86c : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9f382602391438f9b129a88b3516d14 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b647a0dcecb14d0eb922353dfa6600af : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc425c3388524b60a7072c8e102a2431 : SUCCESS in 7m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5cac7fd546ee4ad39d32aab2f98fe11d : SUCCESS in 32m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5f56644d65f6f293e9026986b174c1c05698ec6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 16:40:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5e7889dc9c97565704eec1f66772f8179bd848e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 16:41:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b09495ed4c248f88fd9ab50687d3974\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/577edc33b40f4a4bbc60b43fabdc3187 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1},{"id":"2407d38684e6f700f289d744676d56cc9d283fad","tag":"autogenerated:gerrit:revert","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-13 17:02:44.000000000","message":"Created a revert of this change as I2a35a1e958173b887872663637af6aef4362b1c5","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4388f00d267c4090b7de6bc94da9e2970abdf0cc","revisions":{"4388f00d267c4090b7de6bc94da9e2970abdf0cc":{"kind":"REWORK","_number":1,"created":"2024-09-03 08:54:04.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/92/927792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/92/927792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/92/927792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/92/927792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/92/927792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/92/927792/1"}}},"commit":{"parents":[{"commit":"cfa330c0fddd356b3053dd5050c76cfb42fcda64","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfa330c0fddd356b3053dd5050c76cfb42fcda64"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-09-03 08:49:04.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-09-03 08:54:00.000000000","tz":120},"subject":"Add \"admin\" role to the designate user created by devstack plugin","message":"Add \"admin\" role to the designate user created by devstack plugin\n\nService user with name \"designate\" had only \"service\" role up to now but\nit seems that with oslo.policy 4.4.0 where \"enforce_new_defaults\" is set\nto True by default, this breaks integration between Neutron and\nDesignate as e.g. Neutron\u0027s creation of the recordset fails with\nForbidden exception as this seems to be allowed only for admin user or\nshared or primary zone.\n\nThis patch adds also \"admin\" role for this \"designate\" service user to\nworkaround that issue, at least until Designate will support \"service\"\nrole usage with Secure RBAC policies.\n\nCloses-Bug: #2078518\nChange-Id: I477cc96519e7396a614f92d109867222207ec388\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4388f00d267c4090b7de6bc94da9e2970abdf0cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4388f00d267c4090b7de6bc94da9e2970abdf0cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
