)]}'
{"id":"openstack%2Fdesignate~927840","triplet_id":"openstack%2Fdesignate~master~Ie514e256f0d52d76188c93e6991daf29df5b2e33","project":"openstack/designate","branch":"master","topic":"add-blacklist-crud","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-01-06 03:07:19.000000000","reason":"Change was abandoned"},"37216":{"account":{"_account_id":37216,"name":"이현규","display_name":"hyungyulee","email":"dlgusrb3456@naver.com","username":"hyungyulee"},"last_update":"2024-09-25 14:48:09.000000000","reason":"\u003cGERRIT_ACCOUNT_37216\u003e replied on the change","reason_account":{"_account_id":37216,"name":"이현규","display_name":"hyungyulee","email":"dlgusrb3456@naver.com","username":"hyungyulee"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-01-06 03:07:19.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-01-06 03:07:19.000000000","reason":"Change was abandoned"},"35119":{"account":{"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"},"last_update":"2025-01-06 03:07:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie514e256f0d52d76188c93e6991daf29df5b2e33","subject":"fix blacklist list api-ref","status":"ABANDONED","created":"2024-09-03 14:57:03.000000000","updated":"2025-01-06 03:07:19.000000000","total_comment_count":14,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"01d57fe8ba7bfd2fe1410ee51ece6e22d9de7187","_number":927840,"virtual_id_number":927840,"owner":{"_account_id":37216,"name":"이현규","display_name":"hyungyulee","email":"dlgusrb3456@naver.com","username":"hyungyulee"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-09-09 17:11:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2024-09-12 17:50:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":-1,"date":"2024-09-10 10:23:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"}],"CC":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-03 15:54:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-05 12:20:54.000000000","updated_by":{"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"},"reviewer":{"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"},"state":"REVIEWER"},{"updated":"2024-09-09 16:30:50.000000000","updated_by":{"_account_id":37216,"name":"이현규","display_name":"hyungyulee","email":"dlgusrb3456@naver.com","username":"hyungyulee"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2024-09-10 10:23:14.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-09-13 10:00:04.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"}],"messages":[{"id":"19cd4a2a6adafddf2a31551b57abb4683d70be66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37216,"name":"이현규","display_name":"hyungyulee","email":"dlgusrb3456@naver.com","username":"hyungyulee"},"date":"2024-09-03 14:57:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aef44a929143f581b5820ce9b5e437c0ab51f394","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 15:54:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e591b5ce794464dafe09e1b3ba43e59\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7e58e0e3f22743e49c3d8c38e7ca0088 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10d555de38804f1bbbd2873d5c452f64 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e22832bf3a6f49ad89ecf8f5dbfc8e5a : SUCCESS in 8m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9a5aa2ad9b734e18b14ec8d8c336dfdd : SUCCESS in 31m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f089da1ad7c0b0f73ff853fc3405724575e506e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37216,"name":"이현규","display_name":"hyungyulee","email":"dlgusrb3456@naver.com","username":"hyungyulee"},"date":"2024-09-04 03:17:52.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7f9d02ab0319c873c6772363ee0ea7063be5e0ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 03:51:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6450ae0a4ec944f3a613402a60fdc6ea\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/09457e1384034f759f0b495ec5030b2e : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54dcf0a8fc4f45c4ba6a1ce18b72a0e8 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d843c6874e4afeb3da48401bba345f : SUCCESS in 7m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/db80ededca674fe49356d3d985932e3f : SUCCESS in 32m 05s","accounts_in_message":[],"_revision_number":2},{"id":"110907be2fe9861b6ee1c12cec2e558a9b3796ba","author":{"_account_id":35119,"name":"jihyun huh","email":"huhji.elha@gmail.com","username":"jhhuh"},"date":"2024-09-05 12:20:54.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"14c3ca252f2e2390e86e6a6c6540d006a0fee136","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37216,"name":"이현규","display_name":"hyungyulee","email":"dlgusrb3456@naver.com","username":"hyungyulee"},"date":"2024-09-09 16:27:47.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"bb9c1c0875edc6755b1255b734069f207b4af35e","author":{"_account_id":37216,"name":"이현규","display_name":"hyungyulee","email":"dlgusrb3456@naver.com","username":"hyungyulee"},"date":"2024-09-09 16:30:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b2b49722a9261fb4440bd86f0f697a13b55582b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 17:11:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e308235d32c4f96ba4e7ca6b44cfad2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/49a71feaf46b4f5284192c321a98e953 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac5289f78954f6e9f04887422f57520 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/466223e3832940f79465da06301b4d50 : SUCCESS in 8m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f21d24ca1d5a4996925bd2bba3f83682 : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":3},{"id":"bbb6ddabf95a08399f20836fc52e793d0e7aa82c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-10 10:23:14.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a2c739c66d81c03424188706e2743c3b7a918f94","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-12 17:50:04.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"785d117a2280967c67896d1dee0009c31d8b3853","author":{"_account_id":37216,"name":"이현규","display_name":"hyungyulee","email":"dlgusrb3456@naver.com","username":"hyungyulee"},"date":"2024-09-13 04:42:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"47ba1da124cc1ce16614b6f877a2b61b7b81444e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-13 10:00:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f7e8ec8da1b4e0b8c222ad6a84548edb2bcd8a98","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-13 10:05:16.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"98f6f16c04d220940ea718dfd827aa1e80252578","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-13 10:05:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"73dae03595f5a4a1222384d1fad802d9ac9ca2b4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-13 11:10:49.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d7ffb472d4ec93ac6e6b00959516014f1f868b86","author":{"_account_id":37216,"name":"이현규","display_name":"hyungyulee","email":"dlgusrb3456@naver.com","username":"hyungyulee"},"date":"2024-09-25 14:48:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"01d57fe8ba7bfd2fe1410ee51ece6e22d9de7187","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-06 03:07:19.000000000","message":"Abandoned\n\nReplaced by https://review.opendev.org/c/openstack/designate/+/930479","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cf62722c6a6e2795f23cbfa33ccfb4cfddbff810","revisions":{"da2d1dcad07cf1031ee3c27fc687379e57a09eef":{"kind":"REWORK","_number":1,"created":"2024-09-03 14:57:03.000000000","uploader":{"_account_id":37216,"name":"이현규","display_name":"hyungyulee","email":"dlgusrb3456@naver.com","username":"hyungyulee"},"ref":"refs/changes/40/927840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/40/927840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/927840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/927840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/927840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/40/927840/1"}}},"commit":{"parents":[{"commit":"cfa330c0fddd356b3053dd5050c76cfb42fcda64","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfa330c0fddd356b3053dd5050c76cfb42fcda64"}]}],"author":{"name":"dlgusrb3456","email":"dlgusrb3456@naver.com","date":"2024-09-03 14:50:19.000000000","tz":540},"committer":{"name":"dlgusrb3456","email":"dlgusrb3456@naver.com","date":"2024-09-03 14:55:31.000000000","tz":540},"subject":"title: fix blacklist list api-ref","message":"title: fix blacklist list api-ref\n\nChange-Id: Ie514e256f0d52d76188c93e6991daf29df5b2e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/da2d1dcad07cf1031ee3c27fc687379e57a09eef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/da2d1dcad07cf1031ee3c27fc687379e57a09eef"}]},"branch":"refs/heads/master"},"2d73ac52fc0c112258c000ea6468102542797c09":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-04 03:17:52.000000000","uploader":{"_account_id":37216,"name":"이현규","display_name":"hyungyulee","email":"dlgusrb3456@naver.com","username":"hyungyulee"},"ref":"refs/changes/40/927840/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/40/927840/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/927840/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/927840/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/927840/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/40/927840/2"}}},"commit":{"parents":[{"commit":"cfa330c0fddd356b3053dd5050c76cfb42fcda64","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfa330c0fddd356b3053dd5050c76cfb42fcda64"}]}],"author":{"name":"dlgusrb3456","email":"dlgusrb3456@naver.com","date":"2024-09-03 14:50:19.000000000","tz":540},"committer":{"name":"dlgusrb3456","email":"dlgusrb3456@naver.com","date":"2024-09-04 03:15:09.000000000","tz":540},"subject":"fix blacklist list api-ref","message":"fix blacklist list api-ref\n\nI fix designate list function api-ref because dns blacklist api do not appear list function\n\nChange-Id: Ie514e256f0d52d76188c93e6991daf29df5b2e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2d73ac52fc0c112258c000ea6468102542797c09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2d73ac52fc0c112258c000ea6468102542797c09"}]},"branch":"refs/heads/master"},"cf62722c6a6e2795f23cbfa33ccfb4cfddbff810":{"kind":"REWORK","_number":3,"created":"2024-09-09 16:27:47.000000000","uploader":{"_account_id":37216,"name":"이현규","display_name":"hyungyulee","email":"dlgusrb3456@naver.com","username":"hyungyulee"},"ref":"refs/changes/40/927840/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/40/927840/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/927840/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/927840/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/927840/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/40/927840/3"}}},"commit":{"parents":[{"commit":"cfa330c0fddd356b3053dd5050c76cfb42fcda64","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfa330c0fddd356b3053dd5050c76cfb42fcda64"}]}],"author":{"name":"dlgusrb3456","email":"dlgusrb3456@naver.com","date":"2024-09-03 14:50:19.000000000","tz":540},"committer":{"name":"dlgusrb3456","email":"dlgusrb3456@naver.com","date":"2024-09-09 16:27:01.000000000","tz":540},"subject":"fix blacklist list api-ref","message":"fix blacklist list api-ref\n\nI fix designate list function api-ref because dns blacklist api do not appear list function\n\nChange-Id: Ie514e256f0d52d76188c93e6991daf29df5b2e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cf62722c6a6e2795f23cbfa33ccfb4cfddbff810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cf62722c6a6e2795f23cbfa33ccfb4cfddbff810"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
