)]}'
{"id":"openstack%2Fdesignate~929660","triplet_id":"openstack%2Fdesignate~master~I35049ee395edfbcb3e86004c2b95436fb9e4a789","project":"openstack/designate","branch":"master","topic":"oseibert-fix-soa-address","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-08-06 10:52:50.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"31975":{"account":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"last_update":"2025-08-06 11:53:00.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-08-06 11:53:00.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-07-01 19:46:58.000000000","reason":"removed on reply"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2025-08-06 11:53:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I35049ee395edfbcb3e86004c2b95436fb9e4a789","subject":"Convert email address to/from SOA","status":"MERGED","created":"2024-09-17 15:19:51.000000000","updated":"2025-08-06 11:53:54.000000000","submitted":"2025-08-06 11:53:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929660-oseibert-fix-soa-address","meta_rev_id":"809b89c98dd0411c5e572efb45023f41395af3ac","_number":929660,"virtual_id_number":929660,"owner":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-06 11:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-08-06 10:52:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-01 19:45:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-07-01 19:46:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2025-08-06 10:52:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"reviewers":{"CC":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-17 16:40:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-19 10:55:44.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2024-09-19 10:58:44.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-09-21 16:38:09.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2024-09-23 09:51:13.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-01-15 21:12:14.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2025-01-15 21:12:14.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-03-15 18:11:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-05-28 13:52:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"432022f7a6d288ca2bd252cb8c01b256dea18746","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-17 15:19:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6894bf6a3dba682f6709d6d503924b4c9e5fee3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 16:40:33.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a8b139f26a24672b979ed34cd015834\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b0a8a8a56c7a4c23a7befee5a4957564 : SUCCESS in 47m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/64b91319a0484840806f2b79fccd93ca : SUCCESS in 55m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/72db58c26d1e4bccaaa624a5c29bb2d3 : SUCCESS in 1h 02m 11s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cc3afd4a6d6f48bdb73f9bd428395efd : SUCCESS in 48m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5ee08a3be379478da25fed1f238ec334 : SUCCESS in 39m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f3db30e093e54674bbd85cb774759bb6 : SUCCESS in 1h 01m 01s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6b8384028d784f51a195c0bcf6e7ee00 : SUCCESS in 1h 14m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6390f4e2151a415d990b70dc48ea0ce6 : SUCCESS in 47m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/85fed18baba1484280d0d9d5aa2fb769 : SUCCESS in 49m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7be0ded351374d429c3df1e9eb2616ab : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fa6ab5b398d46aa95d354cbb2f5c256 : FAILURE in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d2a1325b5134333a3ee72dba9ec368a : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1da1648059049169dbbd9642c3e6605 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8998efc290a145b1aa6c26277fe2763a : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f956311a6114a02be76a529b440f1c7 : SUCCESS in 7m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/719247d64a334301befe2c9d9c2e52d4 : SUCCESS in 28m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/27a664ac777b44e6835c8b649cd592a8 : SUCCESS in 4m 11s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/24620dc628014ae98c3fd67044bf1a33 : SUCCESS in 58m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9ff6fe20e03cd4a1582aa874554c3edb8944de3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-18 08:00:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"28b17b94a100b3bb3ebe44623152819c33e9492a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 09:05:33.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81c2dd8f5cae43a9bdae3512c2125c33\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/56dbaac2744e4680803682e3e86eac6a : SUCCESS in 44m 41s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6ea8271d2e244120951691db97cfa8d2 : FAILURE in 52m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f48272496a6478190c40982496e1169 : SUCCESS in 54m 42s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/80e21d1c5cca432f8a122bb621c3a55c : SUCCESS in 41m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/41983258956248b79f90231ffb90165c : SUCCESS in 34m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/592ee867e0c34c4fb17a39289e76a3b0 : SUCCESS in 1h 00m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0e321165297d41f5972665610e31785f : SUCCESS in 1h 02m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cea3366026204f308471b157f24f5afc : SUCCESS in 42m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/faa1cd14ec184e7e83852070e38eae6c : SUCCESS in 54m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04600e6495c8499d92f871859f3ef1f7 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ea7452002df4c2fbc758b52ba0b5e96 : FAILURE in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/342656168a7f47d5a1b78f7bac6f2c12 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be13967bad6244e1b74e7c04ec87cf5c : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e57d884c212e413798099bc63b0142f6 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124c64c1cfb245e296dd7c86586a01b6 : SUCCESS in 6m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c78c9b771e9e4e3ca75a0f6a7df69270 : FAILURE in 33m 30s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/09706e0b797747cca6ff1287be4d9d20 : SUCCESS in 4m 36s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/52ab9d7083764f81807594184411bea1 : SUCCESS in 44m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a2e52886db60a6880b20f4062ef7c7a170be4b9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-18 10:07:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"80b4f5323741c4fa174091698ed7355921a6f099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 11:23:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d80a977229243f196a042827888b91e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a919b0deadb9412f8f71c88d2f338707 : SUCCESS in 38m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9bb4443bc3df49809306f25270ad8f22 : SUCCESS in 50m 15s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2f0e5b5e3314e5399675802d5ff51c7 : SUCCESS in 48m 54s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/54269ee50b8a451387b43ac152879f0f : SUCCESS in 40m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c2f6bd708c644be5a33e6e81cf7d51fd : SUCCESS in 45m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9f962d302a1941bea43afced451e2fab : SUCCESS in 1h 11m 40s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2e425348482b42f781bc555626341a6a : SUCCESS in 1h 10m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/171e5ce448514f30b7ac7efdf0f46f48 : SUCCESS in 47m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/45be82a0c465400db786ed397b228f06 : SUCCESS in 44m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/beebbc4d8e834b2ea4a06b324b303af0 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f0a3858080d4505a6dcb1bd2dc389fd : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/195ecf5a187947ceada279048114cc06 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0af41ba4489c43e2b0785cc304400534 : SUCCESS in 6m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d57544f735e948fd9fe1f686c536573b : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4884f8960359403f83dcd96026a44ef0 : SUCCESS in 8m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b8d0d53a53ca4779b3cea652e8208b99 : FAILURE in 31m 45s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7cbde406cf6c4d4b827db3a36ded21df : SUCCESS in 4m 14s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/8e4c4a206ca1402ea122b335be74a02f : SUCCESS in 46m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"21cd62e30b7d6e76a71c72bdaf504be96b8754da","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-18 12:28:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c72844f5e83ef7c540a66772ffb247ab771ba833","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-19 10:55:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3344b00bbd56838436a5b6d33ff7add965a5f27e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-19 10:56:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a9ba06e6d0d67b21e383fdc2a2639903ff0599e9","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-19 10:58:44.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"025f7d986e8184a53fb8ce2ada5c41738ad69392","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-19 11:10:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"49d19fa04e992e9748060ee5004ac1727f5353a1","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-19 12:00:29.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d505bfdc2233d52ecc1d09e30a2d20028344618b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-19 12:00:56.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"543bcecc39e2ead8f4bb4cce90da87124dd67016","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-19 12:02:18.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"a1f6d5ab290fac7467c301e0c903e8a1b28011e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-19 13:03:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9b5c12b5f72e47ce18a58614c291c8d6a98d1f31","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-19 13:42:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0bc29c2a138237dd5937e98ef10146310387ec75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 14:18:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18fcb57215124657ad35cfdf73d6d181\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a30e267b8ac34e6485d8a403d2113d2c : SUCCESS in 42m 49s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/58efdd7b586b4c84ac17c84bdb8242da : SUCCESS in 57m 42s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d0149419023488f9d9c860f863a338c : SUCCESS in 49m 31s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/fe58218187d84b7eaa778f381d76353d : SUCCESS in 41m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a5e7b0fc1a6e4239ad2f17cf1b25edd5 : SUCCESS in 46m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/53c983a545d242d496eea0a9b8ab5c03 : SUCCESS in 1h 06m 09s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/16ffb46929d54c7faf8e3c0f2e430d14 : SUCCESS in 1h 13m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ab8e6cc82ab942ce88fec25f944856b6 : SUCCESS in 43m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3b44428a4db048f6a32dd2863d04904b : SUCCESS in 44m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdea7a156d084e8ab51627ec10af6d83 : FAILURE in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4918392698e41df9ebad498bf0b576b : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d297d484d994f4db8905cf7700a7350 : FAILURE in 7m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6a6201741a54b66be4d58b101f097a4 : FAILURE in 5m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2695651d423a4ac588e3f320bb52bbbb : FAILURE in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e2839599874c7e87e3de71acdf779c : SUCCESS in 8m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c49a46c934d749d4b1aebe5784e86aac : SUCCESS in 34m 30s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5e9ca52ea86c4060a52f6b8fab94829f : FAILURE in 6m 29s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/0f92e52ebb034f8b9f067c30b09d4f25 : SUCCESS in 51m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a19cd59feed8648d516e364ec7042adfa9ccffce","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-19 15:57:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2a7295228187ad810e917ac9d1f44dc46064c0fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-19 15:58:15.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"1d08b8f28c59abe38c8810d734a353bedda84bf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 17:19:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db043458f7f641e281270e2de3fa2e94\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c946497a25bb40a7a2b8cae6e5c1d332 : SUCCESS in 46m 08s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8c98dd3e53654d14bfbee745855c1440 : SUCCESS in 1h 02m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b99ca16976b44488ad744cf523a423a6 : SUCCESS in 50m 01s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/11dfcd6b7ad442b09f94b0d3232d7ba1 : SUCCESS in 46m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/22f41f2ef1b0400990093fd2e4e0879c : SUCCESS in 42m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b68e639f51114ab9bd23bc26707e8e5e : FAILURE in 1h 18m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cc6880677cc54596b0b8eb5cd4e6a91a : SUCCESS in 1h 10m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b2f17b1a5b284f87a09d81b44908fb6b : SUCCESS in 39m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e02fb5b0d88249d5b35f173dd15c92a7 : SUCCESS in 41m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e12b08ab7df74653b712848faa65b955 : FAILURE in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a7e340ff2c54cd0a766e0867d9306be : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f81e37e30a2d4a3cae656a6ba4926646 : FAILURE in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d80254777d1f4577a65748e026c486ad : FAILURE in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b68467c5da2c4ae086482ef001642169 : FAILURE in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a69a0a346c524941a45b228be741af36 : SUCCESS in 7m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e1dc5408f23f4fde8cbcdea9122563ac : FAILURE in 28m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6391b22a580e45738e231466bfcebab1 : FAILURE in 4m 18s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/f33dd4f3fc51439a87573db220aab915 : SUCCESS in 52m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4181a3f8c0e3f6aef1a0518d22f6b6bd07add6fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-20 08:50:44.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"44182e023fe50ad04102d35297432545c7db6ae4","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-20 08:52:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1a03878e8f97ce2dbb7196b2199004e8418a94c8","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-20 09:38:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8b3f4cbaf27b3f5e48fa095d0b4df80cdbdbb4b3","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-20 09:40:26.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0c84800c43995feb09a41dbcc5fffd06c181a5e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 10:16:07.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da75d3a800014844aac6c7291aa3a261\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ab31a38e48ec4bc6970e288fa44be3d0 : SUCCESS in 43m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/107368ec89e2495f95aef7e6188e587d : SUCCESS in 1h 02m 26s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d360510ea845465f89684c3974efc38f : SUCCESS in 49m 30s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e2e2001d2f3a4d239efe6c686bf2ca3a : SUCCESS in 45m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e823a0aed92643059ea508621c866cd2 : SUCCESS in 41m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4529dfa6594b476f9b9286797aaacde8 : SUCCESS in 1h 16m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0ee3e1444065449197e2900b39a6de76 : SUCCESS in 1h 00m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/79d63380c7c44ce2a74da18bd622b242 : FAILURE in 43m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2538d411364b4dc2a991b6c1f6ec446b : SUCCESS in 41m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d79df2569fb4e8a831a0be267020c4a : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/239cdf0d84274cbdad15f1a49174346e : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3b9e294394b4841ac45a9905268de9d : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea666a0e65f94afb85b3264c8e1e7d06 : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5dab9cd7b92544f0aadffa2fce6753db : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6521c46ece344a32b59b241cf477df14 : SUCCESS in 5m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/59d3c6f5fbaa41fd9e037bdf0168d744 : SUCCESS in 32m 38s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cc592ed4ee034cc0abe0a59eaff4b2d1 : SUCCESS in 4m 45s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/67ab06ba097146cc95a436a8fb3a9fcf : FAILURE in 43m 07s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f1fc8a2c2a105534c303b1f6fa6ecea10876165f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-09-21 16:38:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b440bb28ed79b1344529816f3656c675850a2446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 17:43:55.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b85c5c203675434986a4976b5c4d0441\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e54f011341a943a698fb060f52bcb772 : SUCCESS in 47m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/fc2dff953a2748e697deeb2f03a450cd : SUCCESS in 1h 00m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c47e6cc19624dc48a25b1166a151f37 : SUCCESS in 49m 24s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/12ad16f7b7e3400b8a26328926c718df : SUCCESS in 44m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1fa74737992f4651adb7117ba12db48c : SUCCESS in 37m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/eb564b54d105403ea4b464b39d71d72d : SUCCESS in 54m 22s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/949243b2e7494238aef1dfb17f01bb9d : FAILURE in 56m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5ba09559d9644242a4068e05493124e7 : SUCCESS in 48m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1320f6ea2ae3458e95d964c0aeadae1d : SUCCESS in 47m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fccc75c330a94da48e09412651cf690b : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd7323551f074fda9ff4c9a823cf25aa : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b61c6dbafc445babd85f12905aeaee0 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef26ba1353354be38b946e0193a9d5c7 : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1bbaf6b224d54475a422edd8caa65316 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5d6d593af245b5a459f0142af359e4 : SUCCESS in 6m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2c1c8a9b3a2a4c3586e8e9085eae477d : FAILURE in 48m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/84db4060e7314c4b9b788f81237e74fe : SUCCESS in 4m 30s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/1a86db90da024c04bb63cc1ee47176bd : FAILURE in 43m 52s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ed774867a251314568f7ff0feb6cdf7916aeaa7c","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-09-23 09:51:13.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"8f2d40c481288f8ef05055c5106450b567bf3e9a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-09-23 11:49:07.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0b67a79f03f8a9a0f3c03008dc8e0a826bd001f6","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-23 14:57:04.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"2f0c87f53471ca91dbea06b958177be80a0a61c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-23 14:57:51.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"589e54be6dc99d798efc179582bfca8850c3c44c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 16:21:29.000000000","message":"Patch Set 9: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3af084a306f46bbb4fdbc1a8592ae37\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/acca61975def40a19f08afe9e7a0ef1c : SUCCESS in 53m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ae9c2832ff92427681033d0b48e20bd2 : SUCCESS in 1h 00m 48s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9af21cf83b0948c3bda0b54a84480d1b : SUCCESS in 56m 56s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0c59aa125f0241a694be47ae4d5c96dc : SUCCESS in 54m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3dc1537374514eaa85835a016ee5c642 : SUCCESS in 39m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b4196043b20b480982cbd093c397f347 : SUCCESS in 1h 17m 08s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/20d128179f6b401c97c8ae3fd281775e : SUCCESS in 57m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bd324c860fc748d8a27f21b9a0c8890d : SUCCESS in 50m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bbcd609fd1db46feb2bafbda728f3f33 : SUCCESS in 42m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbdfa7d3cbb04bb2a4ef22d8bee3d44e : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf554bceb1b34688947ae4e54c64f9e9 : FAILURE in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3442622c4e446e89e11c531588ba4aa : SUCCESS in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e92bbf4ec4d4b8d994313ee8c35a112 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f264963343c041faa61a2916ccb25cc8 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f97c0037beaa4965aa446f091ece28e4 : SUCCESS in 5m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/12456f4c1b974879837768d6fa9bf945 : SUCCESS in 31m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b0bee413ff8f4cde96fa4f41b2f58657 : SUCCESS in 4m 25s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/12a28583ccd64a71b51d7fe1d5576891 : SUCCESS in 48m 44s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8d5d389c080748e74ba55d04b925988d4cbb18de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-09-24 14:06:54.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"6f378aa670ca7a59ea6ff5562847dfa6eed6ab78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 15:27:13.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b76ca4bcf1da460a87c7cb6400906332\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1f20b0316b0540a7aca28ef07ee3d15f : SUCCESS in 50m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a2139f890aee468297b0d49ece127fc1 : SUCCESS in 1h 03m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4dae7607347479da7b1b1c2c3014cdb : SUCCESS in 56m 59s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/dce88e59d9004365915d31e044da9d7e : SUCCESS in 50m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/05082bbbddaf41a3b84a495df39c112b : SUCCESS in 39m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b95f3fef29e445239b4775cd822d254a : SUCCESS in 1h 11m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3ab2fb3593524a7cb76f0608611f3c3d : SUCCESS in 1h 01m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/411708899aaf4f5a845351b705bdba3d : SUCCESS in 42m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/45bc00fa82c942a7a28b16bbc4980ce6 : SUCCESS in 37m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b904c894c1a546c5a637ffe2377a4f35 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bea41c5b7574490da073a884c0daf23f : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e08206994d548d69f5e794048cd574e : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf37655afa1446ef8446fcc08d951159 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ecbd98a45dda4d64895d40135b4247bc : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1901715cc854d3bb771ae0210f94d66 : SUCCESS in 4m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b0bcaa2ddec84ce8b563e75e73fe60b9 : SUCCESS in 28m 11s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/61598f4f6f62465fa343dfb0596d94bb : SUCCESS in 4m 52s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/1a6830f507a443918ace6a0843afbcf1 : SUCCESS in 50m 25s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9f2876862aaae69f6a3c83b4afaff5f9818529d0","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-08 11:22:27.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"22a938f6be4116a7aaf31ec69304efa78942e7bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-10-10 09:04:53.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"2900003535b7267622d771cc5a9207175a5b9ba7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 10:28:57.000000000","message":"Patch Set 11: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4c510eab21e4d62b251ba68b56710e4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/acb6782a023e4267be8df3b043cd8b81 : RETRY_LIMIT in 1m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7b9017431c2e4876933dc6615f1855ed : RETRY_LIMIT in 1m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac0ed578d66b41219a58d0873e493e84 : RETRY_LIMIT in 1m 55s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/31eb32c50308459a879b45e0dcf03d05 : SUCCESS in 55m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cfb0994d3f3b4514b5f2ae4f5510836a : SUCCESS in 46m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f78a2f01acef40998e15064dbb2937c7 : SUCCESS in 1h 09m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7aa495b88bef4329926d9f55a3d1c075 : SUCCESS in 1h 06m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a8627a7088e8442981cbce018e204368 : SUCCESS in 54m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8e4270a452f6456c960a2fa85d5ba360 : RETRY_LIMIT in 1m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39b33cdeba7245a58041b41d7fc2997c : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73b07eefce58462586cc805c138b63c1 : FAILURE in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c469acf18029483eb59669cf3ea68721 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/706cde40d8854f90b29dbb1b908ec648 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8f93376d85242b4874083460f38489d : SUCCESS in 8m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6a0831e8ad08447b88e24cf42d5ff542 : SUCCESS in 30m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c2320d73bc094101a891e78ab7e316c9 : SUCCESS in 4m 55s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/f0c497dec6924e8abeafdcde61ad4580 : SUCCESS in 46m 24s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"edda4756b5e43659bc8b6b60d4f5760b07a9b43f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-10-10 15:46:34.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"f0e779bb909864b79889910e4d8f77d577411243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 16:53:59.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/922cf4b657c14402bd03d667f6f939c3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bcd2d9c3912d457d9b8e7ad82c6ec72c : SUCCESS in 46m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/281781aeba2e419e89b0a062b64761ce : SUCCESS in 1h 02m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b17868b95de747328e5e927067801c82 : SUCCESS in 54m 33s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b8c213fcade74ca3937ca139277c0219 : SUCCESS in 50m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/427eecfaa1684ffca8997c4f231a5fcc : SUCCESS in 49m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f830ffb6203d44d6a9b96cb0aec4e5f3 : SUCCESS in 58m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/436b3ef0e94042fda8ac27b7e9903173 : SUCCESS in 57m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3c8941e065f1410d82dd6398abb60180 : SUCCESS in 38m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5b4854ab709d4b4294942fcf078c751a : SUCCESS in 53m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abab2899b22b4197b481449f21b34c5b : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7170849335484702a5fc93281e8fe5e7 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1d6b5c0be2e4170af7799726aa840c6 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79bc3ec734a84a2c856813cde92871fe : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06f354493bc64b25a95a71e0847cde01 : SUCCESS in 8m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3bff82f441214d1692183b49a3dd78b3 : SUCCESS in 35m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e1b4609001bc415d8da5da1af219163d : SUCCESS in 4m 16s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ba00e8fe93bf4225af961ef5b680c27b : SUCCESS in 51m 50s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"bc4c59f4e3df525da80a64f1b1a3aa50a14c66e6","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-11 11:28:28.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"66865f032eab40d16cfb9680c7badeba4b9f9e45","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2024-10-23 14:41:14.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"64b4f728eef00aa9d53a9fb74194ab91f3886ba3","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2025-01-15 16:22:43.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"736debd50e187104633208528736f5878ba5315c","author":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"date":"2025-05-28 13:33:24.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"125c4ab3e9a6fbe548d3c56615fcb0deafb1407c","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-28 13:46:22.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"826d4c044538a9b0edf03bed5180bae529d3f884","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-28 13:52:04.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"be451ecaef9a9a890510a1e6282038ba9fbbc83f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-06-03 13:59:27.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"bf3734f8d3f31a4548dbf5a1ca10deab71914581","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 15:45:26.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68595888dc6344878c98b08d8d6b1a53\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/383b7eb204f642ef9d0a134e12979483 : SUCCESS in 57m 48s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/5658a9e6de7043f7bb6d6e82e6946aae : SUCCESS in 44m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b1ea2586d51342958e1d4e14d5f6de8e : SUCCESS in 46m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f9c49ef1fd1a43d880b513d8f6b1495b : SUCCESS in 1h 02m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a161aa68a44a448a9c051fd7f1d07e13 : SUCCESS in 1h 07m 56s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/b1bc1828da7c406b91e6f8c457a409c6 : SUCCESS in 1h 04m 12s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/cf51900833a045cb84f74ff8372b5251 : SUCCESS in 1h 08m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2e60cc2bdfcb4222bd6c9c18376ba005 : SUCCESS in 44m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/15951f76da7f4aa58af00fbe94eaaf5c : SUCCESS in 42m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c2c91177576479081276fe471224dc9 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07cacc2e79a248da96947910e413d7e9 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74b5e76473c04acfa129e6d570c846a1 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e0fdf5f4020452a9518b13bdb72b43e : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94756b7fccda4a5085566fb504ec2e22 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b6aa3cbf4f54237bb889af47a5eb7ab : SUCCESS in 10m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/816f9e4cc9604535bbf103b809cc8d14 : SUCCESS in 33m 27s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fee95f85eaa240c7aefb901c48686c54 : SUCCESS in 5m 40s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/0cf2ae6ddb414c10a87e3efa91b9f853 : FAILURE in 41m 34s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"1e8af4c8c86c80dfc3ce24eeea315e180902b3ab","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-07-01 19:41:46.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"6220ebbee95b5ca40a9d49f3f151a3461e6ab0b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-07-01 19:45:17.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"efe7559aa800e157e491e4c568418112004f28e3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-07-01 19:45:41.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"81703672f7efa32e158ebdc8a33ac91f23ebc953","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-01 19:46:58.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"f95820dc8d92ca3c575b1f388db6d549c6d71cfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 20:52:01.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52c677128e48496fad3da498f86c8e80\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/71c93a718cc84ca897d6f268667ba427 : SUCCESS in 44m 40s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/77ec6965a35c4f4b8ecf3359c8801d0b : SUCCESS in 43m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/97fd87a0841b41bb80a01242e9e70e41 : SUCCESS in 45m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3e84f25fe5124521a0795ffc91908e84 : SUCCESS in 1h 01m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d12dd3d9ac194f36aa4cbf9caf3ac58b : SUCCESS in 1h 00m 36s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/e18e360340534e6a9b3a4218f92ea80e : SUCCESS in 1h 01m 56s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/de9eb716ea7f44f5a944ee54de7a4095 : SUCCESS in 1h 00m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ba77ce6df8b3443c961db2692380cac4 : SUCCESS in 41m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4a05c206cfad4975ba78c3d7b25aa1f6 : SUCCESS in 39m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66681565aee34c13bef01052e42e5e1a : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba9d311eeaf04b4094777b9e3f17f907 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e371271c28140a0ab08bd6b025def10 : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d775eacc6cbc40bd9d1d97792a99d538 : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60a940d837e340f98e3aa3d0a0080e6c : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61e203e50aae436aa9df49042c34970e : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/844d7c0a4e284b85a58e89833ae3035a : SUCCESS in 3m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eba2d43b7baf4f73ae988fafd4bd4bcc : SUCCESS in 33m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c09ac2ba8dca4a32855ff4182a82f1f5 : SUCCESS in 3m 24s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/192bc2cfc8c246438388e8104179f723 : SUCCESS in 39m 25s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"d0d7b8305290705dc0a5c251a5d3bc26db508019","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-08-06 10:52:50.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"2909717ed361654da158399306f25ba4a8d6a139","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 10:53:16.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"d1907f4bd9d08bccac9ded835d2cfaffc462e10c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 11:53:00.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81a23dccc57946fd820278eb26acb58c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/af3282e39483445da13addd114a0aa95 : SUCCESS in 47m 58s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/feb9107651f44a0a9bb63f752e449820 : SUCCESS in 41m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f856af3aefa646b4868b08c90123f67e : SUCCESS in 44m 30s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/45831eeecd94481eb0cb1be2651b9786 : SUCCESS in 59m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1f4345deb50a432abf6280db07499809 : SUCCESS in 40m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1aa30710c4c54b0580ad63fad599f48c : SUCCESS in 49m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef512fa576634d8d8016d44558f8efef : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c2ad13b3cf844a7a714344e6cbb6411 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c035bcffab6948658326418cec978a79 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f1f1bc7728f49d29c25616604595b28 : SUCCESS in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4036f2a40c4c4f71b7b534a81fbd99c0 : SUCCESS in 4m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1e1133d2e984a74991464ca69a9134b : SUCCESS in 30m 05s","accounts_in_message":[],"_revision_number":13},{"id":"69bcd62e751f6837d390f9899aa7386b0b20f282","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 11:53:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"809b89c98dd0411c5e572efb45023f41395af3ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 11:53:54.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed98cfa4b90b471ea56d0d48df68d32a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd4d36eb16804ffaa4bf4b28f289a9a8 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c8163524b834421a5378aad6c7a94e0 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"66a8987b494849c29b9e317f46d42aee148fc10f","revisions":{"3304cbd5b9e77dc98909c98485313632f907b931":{"kind":"REWORK","_number":1,"created":"2024-09-17 15:19:51.000000000","uploader":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"ref":"refs/changes/60/929660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/929660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/929660/1"}}},"commit":{"parents":[{"commit":"a23a18bd61dd8fbf2325235a947577662d5f91e7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a23a18bd61dd8fbf2325235a947577662d5f91e7"}]}],"author":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-17 14:58:41.000000000","tz":120},"committer":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-17 15:19:50.000000000","tz":120},"subject":"Convert email address to/from SOA","message":"Convert email address to/from SOA\n\nincluding dots in the local-part of the address (which are to be escaped\nwith a backslash).\n\nOne of the locations in the bug report cannot be found any more.\nInstead I improved the conversion from the SOA email address to a normal\nemail address form.\n\nCLOSES-BUG: 1835409\nChange-Id: I35049ee395edfbcb3e86004c2b95436fb9e4a789\nSigned-off-by: Olaf Seibert \u003co.seibert@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3304cbd5b9e77dc98909c98485313632f907b931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3304cbd5b9e77dc98909c98485313632f907b931"}]},"branch":"refs/heads/master"},"79fa3d751f7eb7d050e1a580b7df3f5707a87138":{"kind":"REWORK","_number":2,"created":"2024-09-18 08:00:05.000000000","uploader":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"ref":"refs/changes/60/929660/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/929660/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/929660/2"}}},"commit":{"parents":[{"commit":"a23a18bd61dd8fbf2325235a947577662d5f91e7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a23a18bd61dd8fbf2325235a947577662d5f91e7"}]}],"author":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-17 14:58:41.000000000","tz":120},"committer":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-18 07:59:56.000000000","tz":120},"subject":"Convert email address to/from SOA","message":"Convert email address to/from SOA\n\nincluding dots in the local-part of the address (which are to be escaped\nwith a backslash).\n\nOne of the locations in the bug report cannot be found any more.\nInstead I improved the conversion from the SOA email address to a normal\nemail address form.\n\nCLOSES-BUG: 1835409\nChange-Id: I35049ee395edfbcb3e86004c2b95436fb9e4a789\nSigned-off-by: Olaf Seibert \u003co.seibert@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79fa3d751f7eb7d050e1a580b7df3f5707a87138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79fa3d751f7eb7d050e1a580b7df3f5707a87138"}]},"branch":"refs/heads/master"},"14c93b9193f4f222786e88fbb67f319d67e2fd1f":{"kind":"REWORK","_number":3,"created":"2024-09-18 10:07:34.000000000","uploader":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"ref":"refs/changes/60/929660/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/929660/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/929660/3"}}},"commit":{"parents":[{"commit":"a23a18bd61dd8fbf2325235a947577662d5f91e7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a23a18bd61dd8fbf2325235a947577662d5f91e7"}]}],"author":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-17 14:58:41.000000000","tz":120},"committer":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-18 10:07:28.000000000","tz":120},"subject":"Convert email address to/from SOA","message":"Convert email address to/from SOA\n\nincluding dots in the local-part of the address (which are to be escaped\nwith a backslash).\n\nOne of the locations in the bug report cannot be found any more.\nInstead I improved the conversion from the SOA email address to a normal\nemail address form.\n\nCLOSES-BUG: 1835409\nChange-Id: I35049ee395edfbcb3e86004c2b95436fb9e4a789\nSigned-off-by: Olaf Seibert \u003co.seibert@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14c93b9193f4f222786e88fbb67f319d67e2fd1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14c93b9193f4f222786e88fbb67f319d67e2fd1f"}]},"branch":"refs/heads/master"},"5092cf0431122f2a79482f8186b4a626173b3afd":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-09-19 12:00:56.000000000","uploader":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"ref":"refs/changes/60/929660/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/929660/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/929660/4"}}},"commit":{"parents":[{"commit":"a23a18bd61dd8fbf2325235a947577662d5f91e7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a23a18bd61dd8fbf2325235a947577662d5f91e7"}]}],"author":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-17 14:58:41.000000000","tz":120},"committer":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-19 11:39:52.000000000","tz":120},"subject":"Convert email address to/from SOA","message":"Convert email address to/from SOA\n\nincluding dots in the local-part of the address (which are to be escaped\nwith a backslash).\n\nOne of the locations in the bug report cannot be found any more.\nInstead I improved the conversion from the SOA email address to a normal\nemail address form.\n\nCLoses-Bug: #1835409\nChange-Id: I35049ee395edfbcb3e86004c2b95436fb9e4a789\nSigned-off-by: Olaf Seibert \u003co.seibert@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5092cf0431122f2a79482f8186b4a626173b3afd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5092cf0431122f2a79482f8186b4a626173b3afd"}]},"branch":"refs/heads/master"},"72072d920ff1424116ae1bc30619a51aff1889fa":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-09-19 12:02:18.000000000","uploader":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"ref":"refs/changes/60/929660/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/929660/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/929660/5"}}},"commit":{"parents":[{"commit":"a23a18bd61dd8fbf2325235a947577662d5f91e7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a23a18bd61dd8fbf2325235a947577662d5f91e7"}]}],"author":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-17 14:58:41.000000000","tz":120},"committer":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-19 12:02:08.000000000","tz":120},"subject":"Convert email address to/from SOA","message":"Convert email address to/from SOA\n\nincluding dots in the local-part of the address (which are to be escaped\nwith a backslash).\n\nOne of the locations in the bug report cannot be found any more.\nInstead I improved the conversion from the SOA email address to a normal\nemail address form.\n\nCloses-Bug: #1835409\nChange-Id: I35049ee395edfbcb3e86004c2b95436fb9e4a789\nSigned-off-by: Olaf Seibert \u003co.seibert@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72072d920ff1424116ae1bc30619a51aff1889fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72072d920ff1424116ae1bc30619a51aff1889fa"}]},"branch":"refs/heads/master"},"54a9a41a26dc6a21d54a9f6313b0ce5e75be5fb9":{"kind":"REWORK","_number":6,"created":"2024-09-19 13:03:46.000000000","uploader":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"ref":"refs/changes/60/929660/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/929660/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/929660/6"}}},"commit":{"parents":[{"commit":"a23a18bd61dd8fbf2325235a947577662d5f91e7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a23a18bd61dd8fbf2325235a947577662d5f91e7"}]}],"author":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-17 14:58:41.000000000","tz":120},"committer":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-19 13:01:38.000000000","tz":120},"subject":"Convert email address to/from SOA","message":"Convert email address to/from SOA\n\nincluding dots in the local-part of the address (which are to be escaped\nwith a backslash).\n\nOne of the locations in the bug report cannot be found any more.\nInstead I improved the conversion from the SOA email address to a normal\nemail address form.\n\nCloses-Bug: #1835409\nChange-Id: I35049ee395edfbcb3e86004c2b95436fb9e4a789\nSigned-off-by: Olaf Seibert \u003co.seibert@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/54a9a41a26dc6a21d54a9f6313b0ce5e75be5fb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/54a9a41a26dc6a21d54a9f6313b0ce5e75be5fb9"}]},"branch":"refs/heads/master"},"26b26a2150e64a8d8924c27abe73018cdb58fef4":{"kind":"REWORK","_number":7,"created":"2024-09-19 15:58:15.000000000","uploader":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"ref":"refs/changes/60/929660/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/929660/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/929660/7"}}},"commit":{"parents":[{"commit":"a23a18bd61dd8fbf2325235a947577662d5f91e7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a23a18bd61dd8fbf2325235a947577662d5f91e7"}]}],"author":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-17 14:58:41.000000000","tz":120},"committer":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-19 13:08:57.000000000","tz":120},"subject":"Convert email address to/from SOA","message":"Convert email address to/from SOA\n\nincluding dots in the local-part of the address (which are to be escaped\nwith a backslash).\n\nOne of the locations in the bug report cannot be found any more.\nInstead I improved the conversion from the SOA email address to a normal\nemail address form.\n\nCloses-Bug: #1835409\nChange-Id: I35049ee395edfbcb3e86004c2b95436fb9e4a789\nSigned-off-by: Olaf Seibert \u003co.seibert@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/26b26a2150e64a8d8924c27abe73018cdb58fef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/26b26a2150e64a8d8924c27abe73018cdb58fef4"}]},"branch":"refs/heads/master"},"4a8cbd3f175d526e86fd1f17df709c38bb7d3026":{"kind":"REWORK","_number":8,"created":"2024-09-20 08:50:44.000000000","uploader":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"ref":"refs/changes/60/929660/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/929660/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/929660/8"}}},"commit":{"parents":[{"commit":"a23a18bd61dd8fbf2325235a947577662d5f91e7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a23a18bd61dd8fbf2325235a947577662d5f91e7"}]}],"author":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-17 14:58:41.000000000","tz":120},"committer":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-20 08:47:38.000000000","tz":120},"subject":"Convert email address to/from SOA","message":"Convert email address to/from SOA\n\nincluding dots in the local-part of the address (which are to be escaped\nwith a backslash).\n\nOne of the locations in the bug report cannot be found any more.\nInstead I improved the conversion from the SOA email address to a normal\nemail address form.\n\nCloses-Bug: #1835409\nChange-Id: I35049ee395edfbcb3e86004c2b95436fb9e4a789\nSigned-off-by: Olaf Seibert \u003co.seibert@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4a8cbd3f175d526e86fd1f17df709c38bb7d3026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4a8cbd3f175d526e86fd1f17df709c38bb7d3026"}]},"branch":"refs/heads/master"},"4f906184aa97bab7d5de0971e1ed258f815da176":{"kind":"REWORK","_number":9,"created":"2024-09-23 14:57:51.000000000","uploader":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"ref":"refs/changes/60/929660/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/929660/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/929660/9"}}},"commit":{"parents":[{"commit":"a23a18bd61dd8fbf2325235a947577662d5f91e7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a23a18bd61dd8fbf2325235a947577662d5f91e7"}]}],"author":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-17 14:58:41.000000000","tz":120},"committer":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-23 14:57:42.000000000","tz":120},"subject":"Convert email address to/from SOA","message":"Convert email address to/from SOA\n\nincluding dots in the local-part of the address (which are to be escaped\nwith a backslash).\n\nOne of the locations in the bug report cannot be found any more.\nInstead I improved the conversion from the SOA email address to a normal\nemail address form.\n\nCloses-Bug: #1835409\nChange-Id: I35049ee395edfbcb3e86004c2b95436fb9e4a789\nSigned-off-by: Olaf Seibert \u003co.seibert@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f906184aa97bab7d5de0971e1ed258f815da176"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f906184aa97bab7d5de0971e1ed258f815da176"}]},"branch":"refs/heads/master"},"278bdb493c40aea6ddeb08bbb8bcecb06764a3fb":{"kind":"REWORK","_number":10,"created":"2024-09-24 14:06:54.000000000","uploader":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"ref":"refs/changes/60/929660/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/929660/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/929660/10"}}},"commit":{"parents":[{"commit":"a23a18bd61dd8fbf2325235a947577662d5f91e7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a23a18bd61dd8fbf2325235a947577662d5f91e7"}]}],"author":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-17 14:58:41.000000000","tz":120},"committer":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-24 14:06:41.000000000","tz":120},"subject":"Convert email address to/from SOA","message":"Convert email address to/from SOA\n\nincluding dots in the local-part of the address (which are to be escaped\nwith a backslash).\n\nOne of the locations in the bug report cannot be found any more.\nInstead I improved the conversion from the SOA email address to a normal\nemail address form.\n\nCloses-Bug: #1835409\nChange-Id: I35049ee395edfbcb3e86004c2b95436fb9e4a789\nSigned-off-by: Olaf Seibert \u003co.seibert@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/278bdb493c40aea6ddeb08bbb8bcecb06764a3fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/278bdb493c40aea6ddeb08bbb8bcecb06764a3fb"}]},"branch":"refs/heads/master"},"7c5f5d06b85c6f384a7980fae9cecc89be633589":{"kind":"REWORK","_number":11,"created":"2024-10-10 09:04:53.000000000","uploader":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"ref":"refs/changes/60/929660/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/929660/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/929660/11"}}},"commit":{"parents":[{"commit":"a23a18bd61dd8fbf2325235a947577662d5f91e7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a23a18bd61dd8fbf2325235a947577662d5f91e7"}]}],"author":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-17 14:58:41.000000000","tz":120},"committer":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-10-10 09:04:33.000000000","tz":120},"subject":"Convert email address to/from SOA","message":"Convert email address to/from SOA\n\nincluding dots in the local-part of the address (which are to be escaped\nwith a backslash).\n\nOne of the locations in the bug report cannot be found any more.\nInstead I improved the conversion from the SOA email address to a normal\nemail address form.\n\nCloses-Bug: #1835409\nChange-Id: I35049ee395edfbcb3e86004c2b95436fb9e4a789\nSigned-off-by: Olaf Seibert \u003co.seibert@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c5f5d06b85c6f384a7980fae9cecc89be633589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c5f5d06b85c6f384a7980fae9cecc89be633589"}]},"branch":"refs/heads/master"},"0a8a3dea437c20e080f3338b1f2cccf1a8334f1b":{"kind":"REWORK","_number":12,"created":"2024-10-10 15:46:34.000000000","uploader":{"_account_id":31975,"name":"Olaf Seibert","email":"o.seibert@syseleven.de","username":"oseiberts11"},"ref":"refs/changes/60/929660/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/929660/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/929660/12"}}},"commit":{"parents":[{"commit":"a23a18bd61dd8fbf2325235a947577662d5f91e7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a23a18bd61dd8fbf2325235a947577662d5f91e7"}]}],"author":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-17 14:58:41.000000000","tz":120},"committer":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-10-10 15:45:56.000000000","tz":120},"subject":"Convert email address to/from SOA","message":"Convert email address to/from SOA\n\nincluding dots in the local-part of the address (which are to be escaped\nwith a backslash).\n\nOne of the locations in the bug report cannot be found any more.\nInstead I improved the conversion from the SOA email address to a normal\nemail address form.\n\nCloses-Bug: #1835409\nChange-Id: I35049ee395edfbcb3e86004c2b95436fb9e4a789\nSigned-off-by: Olaf Seibert \u003co.seibert@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a8a3dea437c20e080f3338b1f2cccf1a8334f1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a8a3dea437c20e080f3338b1f2cccf1a8334f1b"}]},"branch":"refs/heads/master"},"66a8987b494849c29b9e317f46d42aee148fc10f":{"kind":"REWORK","_number":13,"created":"2025-07-01 19:45:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/60/929660/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/929660/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/929660/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/929660/13"}}},"commit":{"parents":[{"commit":"a23a18bd61dd8fbf2325235a947577662d5f91e7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a23a18bd61dd8fbf2325235a947577662d5f91e7"}]}],"author":{"name":"Olaf Seibert","email":"o.seibert@syseleven.de","date":"2024-09-17 14:58:41.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-07-01 19:45:16.000000000","tz":0},"subject":"Convert email address to/from SOA","message":"Convert email address to/from SOA\n\nincluding dots in the local-part of the address (which are to be escaped\nwith a backslash).\n\nOne of the locations in the bug report cannot be found any more.\nInstead I improved the conversion from the SOA email address to a normal\nemail address form.\n\nCloses-Bug: #1835409\nChange-Id: I35049ee395edfbcb3e86004c2b95436fb9e4a789\nSigned-off-by: Olaf Seibert \u003co.seibert@syseleven.de\u003e\nSigned-off-by: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/66a8987b494849c29b9e317f46d42aee148fc10f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/66a8987b494849c29b9e317f46d42aee148fc10f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
