)]}'
{"id":"openstack%2Fdesignate~930729","triplet_id":"openstack%2Fdesignate~master~Ia10b37af7a4e95236cfce4c7d95d32e6eae7a69b","project":"openstack/designate","branch":"master","topic":"remove-eventlet-workaround","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-19 14:56:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia10b37af7a4e95236cfce4c7d95d32e6eae7a69b","subject":"Remove workaround for eventlet \u003c 0.27.0","status":"MERGED","created":"2024-09-27 15:08:02.000000000","updated":"2024-11-19 14:57:10.000000000","submitted":"2024-11-19 14:56:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930729-remove-eventlet-workaround","meta_rev_id":"b84eb1a83347f77bf1a3b18448cb08bf958f041d","_number":930729,"virtual_id_number":930729,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-19 14:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-11-14 18:30:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-18 16:49:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2024-11-14 18:30:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-27 16:14:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-14 13:29:42.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2024-11-14 18:30:30.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-11-15 09:33:05.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"}],"messages":[{"id":"52d703134db7f84445d880ec1f3bb7c08cc49a46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-27 15:08:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b07ac62bb97d7c3bffdce91a7bd17489b02c989","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 16:14:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4979567e0c7648f98104508a99419397\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/651acc9562804851ba64cbc0328a8b6a : FAILURE in 44m 04s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d829598f3a794fb5bd934bdb3ce68004 : SUCCESS in 59m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae0cc7dfdb2b4b50a9b08709477786b8 : SUCCESS in 48m 35s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cb06457aa2d84bf38d321ff22c375a29 : SUCCESS in 51m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a357955ebe49482b881e08c2da2f0db1 : SUCCESS in 44m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e192047ee94640768e53d927237caa32 : SUCCESS in 1h 01m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/51dc2580cbae453ba1bc5d6e7d8cf1cb : SUCCESS in 58m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/59da3d7099c54700912179f5eb506d59 : SUCCESS in 46m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bc83d0b82e6841349ca03f4691984fcb : SUCCESS in 47m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d08aba48d012497288c8dd91f9ab1bb2 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff0fa68732e448c68f425bf5abe1f419 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/489d761dd5474c4e867329eed8265e9a : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44dfe160170f498db057bf2cec60dc68 : SUCCESS in 6m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5355ca5b61f24892995f966b09c479ff : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8490e188cbe64f0cb25c9c17cde7526f : SUCCESS in 9m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/98edd5cfd0db4ffa9f504134de11c4f6 : SUCCESS in 1m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5779b9c728d3400cb0e5eb44be012da9 : SUCCESS in 41m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2d489916600b4407998e2dfee029e5e7 : SUCCESS in 4m 25s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/cf740e778ca54ffa9f15ef12485300c8 : SUCCESS in 45m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b25f82def3cf83e0af4d6002c868b167d6a879ff","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-27 16:21:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed9fb275ffc1165875da554055380f5c39cbdc6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 17:41:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ad4dc4f89bf432ba64f4ebca8460eb4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e14e564ebe574ed9842eb81d2c0a96a8 : SUCCESS in 36m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a8ee26805e594573b48268613a6e5b99 : SUCCESS in 49m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b75399252ff48f392809d4cdfc0bb64 : SUCCESS in 39m 25s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/037838f7dc5247cbb4d1fd2c6f2a1831 : SUCCESS in 38m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9d1470b510bd45b1b6cbf496b1856247 : SUCCESS in 38m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f87a0e9c9b9b4e54b20a0ebbcb705c2b : SUCCESS in 58m 08s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5ed77965dadf48d3b5a0e6c384f440ed : SUCCESS in 1h 10m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8e6c272a40e84d05b1cab12388ad719c : SUCCESS in 36m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d4df7253d34c4fac89d1c13d0f87cc9d : SUCCESS in 49m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e386a0ca7c8344918d8fc842b0d589bb : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcfdaf6fe9e948f5886024cad3394dee : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c1ff2fe070f4e1386025051ba38ca41 : SUCCESS in 6m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1010d3d21d764ccc99e98db5f8b79acc : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c883046641a4e1c9d7a17b76302d21f : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7c8de74c0e04ed59dbdb1c0d448392f : SUCCESS in 9m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/93742790b98a47a59175766f48432806 : SUCCESS in 1m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf6950a8a2b84c8a9a5651d5bfbb793d : SUCCESS in 25m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/54335a35aa4f4bcc9aaeefcfa4c1cea1 : SUCCESS in 4m 32s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/949bdd97532c435cbddff9e701540c60 : SUCCESS in 42m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11ab7ce7f45286068b3ec69daed607b52dc78636","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-14 13:29:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b273a8cc131cf502a97221a9b2ea6f5154dcbed3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-14 13:29:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":2},{"id":"073c86db4eef3cd3d01a703743d325e5bfac64bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 14:46:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a4fc4224ba244d6a8ac1acd6e9baeae\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ada7a02b49354053803e532d63080535 : SUCCESS in 44m 19s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7c72cd476af34faea32fd6a200b4ad85 : SUCCESS in 46m 12s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1492d46fb23e42b196327ff147bfbd06 : SUCCESS in 39m 19s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8bd5650fb8054f1e9ace7ae542c50be6 : SUCCESS in 48m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b10b2ab503f145a8a3720895dddc2447 : SUCCESS in 47m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fecb229299794bb7a18607f936922d60 : SUCCESS in 1h 08m 35s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7ec0064feb004de6847705bfda4a11a5 : SUCCESS in 56m 57s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/f04f7b1cbe3949a1b276ba7860d1e2e6 : SUCCESS in 1h 09m 53s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/91dce402384043e99ca6ebdb20338113 : SUCCESS in 54m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6d35c5abda604252a154cdadc3d35fdc : SUCCESS in 46m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0e70538d35154025b97d9bc38d32deab : SUCCESS in 50m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ab4eaa1e01948fabfefcf591afba7dd : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afc4e669d2c34f5c8915798d9ad5e1a6 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b09b51ae6fb243f7be3c0d02a9658072 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fcf1e5661e8c43159e84cdde4e839d14 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5799995c0814490dac7aedac489c7f67 : SUCCESS in 4m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9408e17af7224d8fa74d7b7bd31a1187 : SUCCESS in 1m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0cb4b0df9e2749c59d5f6f0cae408347 : SUCCESS in 27m 59s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/35697365fbf745bf896e5920864a49de : SUCCESS in 4m 40s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/743b865a788747988851f91257cf43f5 : SUCCESS in 50m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"07f835d1abbb1ec013f19afdb154ecc31576cead","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-14 18:30:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"623b45df37f6de878848f5c6451db4c47c02e97e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 18:31:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"06628ca0d547dcfdfd20514db883385c3620cc49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 19:10:25.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84bddf14303d4d58b06b95596fa41126\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/199b60490ff740b4b4f8dedfe19dfd71 : SUCCESS in 37m 18s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2d7d47eb7c3b48969d87b306e48e6dfd : CANCELED\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/23e949367779439f96a569d96fbbaf19 : SUCCESS in 36m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/257d4e18a36a4e01981aa3bef7a780c2 : CANCELED\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/99024fa2304e480ba0347feeb0d4f162 : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fad762f009354d128824533c5489354e : CANCELED\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d6d58993de87478a929a9e5fcc51963c : SUCCESS in 37m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50817e91c73e4091ba8b97a0f1e9c57d : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4f7a7e125974263bee4e54f448bd998 : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68562fe0b9f44f908976dab726524e94 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be5e21a836ca4125be2bd7028e7483a7 : SUCCESS in 7m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bb1caa3fa8b645ebab545a29f6143510 : SUCCESS in 1m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bc9de009795649e397b6169b56bfc65b : FAILURE in 37m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ffca7fa49aecb109d17cc0868dc7584e8ecfea57","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-15 00:07:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"17c876c9966fc4c5cf67d8392bc074c3fa9fa89e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 01:16:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea86f527ce26409398ca0b3ed8fdae68\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c429188736a740caa2489d085592fa29 : SUCCESS in 52m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/424514a465db443c81cd0850e34b8c87 : SUCCESS in 59m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8c472ec7f584e3dae223659278e8785 : SUCCESS in 57m 38s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/72769a8d006340ffaee63d05df72df81 : SUCCESS in 48m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4b7d7d884cd848c19ee6ce820fe45a65 : SUCCESS in 52m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d2cd09bdbf3142e0a86b57d9a8622a0f : SUCCESS in 1h 00m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5e0b34b876714e97aa412f9bf97fc136 : SUCCESS in 59m 15s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/93ebb1a77ecb4a2797656d1e54edc8af : SUCCESS in 54m 43s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/3d6a5346d0fe44bdb7fd59e0872f8c60 : SUCCESS in 1h 00m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b9fda7eef0e94ed9b6b662ee017670d1 : SUCCESS in 33m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e7deb32150514ca6b291fbe2988fa178 : SUCCESS in 47m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e171dcb775b4f689db95e739b8a6c21 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86d72249928b4302beb92ab68955138e : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4672ba790e6549709bbf6dcac9e05802 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/86fa228a49ba44ac9e313bc871cc2898 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/353756ec2c9b405ca1b0ff0304d19983 : SUCCESS in 5m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d5e548ca38a54d25bcf96d1ef9dde069 : SUCCESS in 1m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/45472422c59d42b28f33537739012bd8 : FAILURE in 41m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8b754653a4a44a068039b7bd6c449477 : SUCCESS in 4m 35s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ef10db8864ef437f9463642e85622752 : SUCCESS in 51m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b9d620cef733f3ffa9744089a30bcac917eae6e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-11-15 09:33:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"397d1f32aef7449be59cd7c24c5fa9d5becf886a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 11:05:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dca00a69092a4a06aa612ed132c5e83f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c0ddd7ee7e4f4e73b3ab3d494f9e60cc : SUCCESS in 37m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6e5195ce101447d49a954ed4c7d1dcd0 : SUCCESS in 53m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc37899da15c4bbeaae34bfb9b39f9ab : SUCCESS in 38m 05s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6361403bc85941c59d42ed25b3de0f23 : SUCCESS in 44m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f1d5f42016214d318e3f351f7656f729 : SUCCESS in 47m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/358f545fbf934bf2b12df12143854bd1 : SUCCESS in 1h 09m 29s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b91f4bfd75d447d6910057e9268dda96 : SUCCESS in 1h 18m 31s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/59edca34210048ed8585ee9a4c606245 : SUCCESS in 53m 43s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/9870c4247dc944da9e6fdc49748c4d68 : SUCCESS in 1h 08m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/114806d74c5749f9b827ac44a49515dc : SUCCESS in 33m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/aa1a889b9bf44e5b96a9caa686aa1e68 : SUCCESS in 37m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c929974402347d6920eef1d00f64014 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af6eb285df044b039578f440d9272a94 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e3b87b71dda42f390f57ed2f670426b : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/33105bc7d53d45b8880e78ea582196b3 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8852a17918141608222257a95a94d49 : SUCCESS in 6m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f6be50c346c14276be3d8897efacee77 : SUCCESS in 1m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5942d2aa0e5849a3b0fec527487103e4 : FAILURE in 24m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/751ba0b0ce5048d2a61e0489b3f9ccde : SUCCESS in 4m 25s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/95f01f7e4d3249d8966564a0ccd6fc6a : SUCCESS in 35m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b09efc86a69b1daca0dcc2604e945c45f7f96ee6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-18 16:49:19.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa5b0824a9783e056089186f037987b5e9c82fd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 18:14:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3de08e9fda2d4bd488719c8620578461\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/10185952511543c7a4e17f2ba6d63eec : SUCCESS in 40m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/120e19dc639b447391828d02f2e092e4 : SUCCESS in 40m 59s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6622c365104f4a52b049732eb55ddd5b : SUCCESS in 55m 22s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/da44770f037640f4bb754f22f94f3129 : SUCCESS in 39m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/990815aa498644b7908c8b5fde465b1a : SUCCESS in 44m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a89715c0e6034295bd944b3e81edefb7 : SUCCESS in 55m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/caa5a4120d314dd0b86ae01a27e94612 : SUCCESS in 1h 18m 12s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/241262b095324cde8338e706cd534b6f : SUCCESS in 53m 40s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/1e4dfc70a73a4eb9bea06d4076f4a447 : SUCCESS in 1h 10m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/93e9853015404bd4bb6095a02fc816c1 : SUCCESS in 43m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f6a5854ab88943388ce8e9aa34b9d136 : SUCCESS in 52m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79bad664a4e64d62a7c50d2497d4a860 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/348b8ede43a4476cbff3a0bd0eceb29e : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53a369c192b04fd7b5230fc246dac97b : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/892587a6e8d74d94aef2c4fefe5d104b : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9193edb791324f4daac776c8651b5326 : SUCCESS in 7m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0d3864acc3a44269987d191acc455761 : SUCCESS in 1m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e1e9a7de12849738da028e96b1add70 : FAILURE in 27m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/07c0b97a07f64059914fff296dd52dfe : SUCCESS in 4m 26s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/29e1b123773b44d096d3b3b7065f050b : SUCCESS in 43m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bc17c0d91b4f5a198935dee856d16d9a6dcae658","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-18 21:35:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7571fda9f886380167a9ed4d52084d3b5eb5c71b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 23:05:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dbae666ee624ab1bc280f0b7a9b5345\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bdd0778d0d024d40bf13b3c8913c5150 : SUCCESS in 46m 48s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/13b917d7aa214ee2aa4dce0d3b9b9436 : SUCCESS in 1h 07m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0c1cf6bcf95472aaabb6303e1a9423f : SUCCESS in 53m 25s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/215909863abb4769a1b24a6cb06d893c : SUCCESS in 44m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d3c9ce14e86c4a3eb322e13144ba0c89 : SUCCESS in 44m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bacbdfde06bb43129f7905ecc506d4e7 : SUCCESS in 1h 17m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/80b43e8e320c49988f83f34db3c58ab6 : SUCCESS in 53m 46s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d76bf638b04646a4929b0ab54f863a0c : SUCCESS in 1h 07m 08s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/ad02a43ea8d64eeca8c3bb4e536f154d : SUCCESS in 1h 07m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/706a20089eac4a6a8a521ecec6ee8f29 : SUCCESS in 34m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/930b1f383e4a434fb0e6372c748ad028 : SUCCESS in 43m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57d6171102234daf9ce2bd8b318d3039 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c37ffedbbb874fbbb8dee388d03bdc59 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0342238750ca435e8c5ccbf79aacd53b : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e1c90a607484d8e8d5f289ac0d8e6cc : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/009ede6735c6490d9259dd8e1f642cab : SUCCESS in 6m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c6bab74d0da146099b46a5f5de9a1e0f : SUCCESS in 1m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/66bc73e0316641debf5b6dbcb8139965 : SUCCESS in 27m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/412174840b7249e2a36aa7bf7e558acd : SUCCESS in 4m 30s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/544fc0254f4846568e30029e9d58cf93 : SUCCESS in 33m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4be4b130a3e2e068afa984cdc69786c130c33da6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 23:05:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"db0542672487177fe9037fa4d0a09ba9804833ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 23:17:16.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3fbfdd4d7094311b751d41bb050f777\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aa1cee17b2344c578860416f6bdd70f5 : CANCELED\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/bdf6fe5b576549f9833ce7c169046efc : CANCELED\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/98a63715b59f4eb7b0350354c2791804 : CANCELED\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f06aebaf93a3455f924a6a83c2c484db : CANCELED\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/9046f9b043464a86b5c71dc3972fcb2e : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/16fef9998f0540eaabb21159da4c3162 : CANCELED\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7ca1937f81a84614925c12faa074adbf : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45370692287e4f66975c38faf99e3364 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4f616374d2240999220ba50d98316ef : SUCCESS in 5m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7bafb70b30934fe0a04f705cb2d5f92e : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbbe450ac93345abb3ee5a2c983e0b8b : CANCELED\n- requirements-check https://zuul.opendev.org/t/openstack/build/7275c3a3eef04fdd9035ba8c40875a8a : SUCCESS in 1m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0a9877af7ca14edc83c376ab195462fb : FAILURE in 11m 07s","accounts_in_message":[],"_revision_number":2},{"id":"28b30d26f79e58e6fc9eb637d5ac273dfc3123b8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-19 01:35:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"90b99f28c05ca941f438dca37786be6fcde38951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 02:56:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5bb0d003f01e4660b9e3959d65b6735b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fcba9995bbad4117b526c1350ab3671b : SUCCESS in 48m 57s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/237d6c796cda474fa2ec573f9d9bb5c6 : SUCCESS in 1h 02m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d4f0dcd77f04204a5017898181ac520 : SUCCESS in 58m 39s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3f2b2cda0ef74aab9804140ca71ae73b : SUCCESS in 45m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d6a56da24a714d7cac45fbcf248e27d6 : SUCCESS in 51m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/79d7ec4af7dc4acd862ddc305d84edf1 : SUCCESS in 1h 11m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ebb0ac5fc1b04717880f3500ad3bbd86 : SUCCESS in 1h 11m 50s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/5999dd0fdbbd4f988558c4ae3dbb2c71 : SUCCESS in 1h 00m 29s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/58f427c916fe42c59285775bbf3fa42d : SUCCESS in 55m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8f8720c8127843388f71a27c72f2b0b0 : FAILURE in 43m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/14c1a5cd93474a279367cbb26b026fc6 : SUCCESS in 51m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aac646f254d54745b65dd043f10901fa : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cce933ea2e64861ba76eee615d5a395 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cc7bb6a3fb241f4849bd728127886af : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6becf2a221a2494496f401bbd9822aa5 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66577b98d57d4f0590bcdc685cdd0c72 : SUCCESS in 7m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/da77586469f24c6e9a8f407b88d9ea58 : SUCCESS in 1m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d3da1837c4a947adb839add39bb8a983 : SUCCESS in 30m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/611f5dc9961e40f1b3577b6a44ffb79e : SUCCESS in 4m 04s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ec1c1048cfd7416da2f4c17454eb273b : FAILURE in 43m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"55fc2767ff81fe8cec8115a83058d74563993b8f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-19 03:00:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9436e52f98a105bfb9cdd3b3e8f9d39aa60d22a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 04:16:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01f98e38ac2442f493eacee90f32b775\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f76770f271104a9386e1ffa9b43e017c : SUCCESS in 48m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e65f147f331a4f6d9682f465160f8807 : SUCCESS in 1h 05m 20s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7d2db41a7e543afbf241426ba907d12 : SUCCESS in 56m 10s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8648a85660324dcca7edc3ec5d679818 : SUCCESS in 40m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/de91b98ffe1a40c1b4cd4d040c7da660 : SUCCESS in 36m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/01ac17902d564ed4ab42a9e1aeb2ab1a : SUCCESS in 1h 11m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c56865e799894e10b3f7da683ed42273 : SUCCESS in 1h 14m 59s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d0c5d442a1ec46329896c571cb99aac3 : SUCCESS in 1h 09m 23s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/6a8e4078363548169620403ce2fac7af : SUCCESS in 1h 08m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2144d8d02b7444159280590dca7fb479 : SUCCESS in 48m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e6b15c9889b4416bb1a14fbca0347925 : SUCCESS in 37m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cee8d8b8ced44a486005ba33afaf29b : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71255b5b709d472aa18868c8f4ca8ebb : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78d19f45095f430da8478b0545af97a8 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b2a4a5521bf4d639012401a13a8301b : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa970dea11954e8893e352e74662ba8c : SUCCESS in 7m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7e69cdb99f0a4b77955c2a29f538045d : SUCCESS in 1m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d759601afddf421c9ed595d92587a3e3 : SUCCESS in 34m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a7b189142c0948bea4c39ce9354f8666 : SUCCESS in 5m 34s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/79e6aa4828a643389b4fc965c962a077 : FAILURE in 48m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0e5e7b104b80a955cf4bea2e0a53face5a208b9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 04:17:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bf3e43d811ecab9e55ed0585509eba2cf6e2ec2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 05:12:30.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d17d9b89534f438cb332909088b780af\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e36417543fb94f80a8af2da2374ee9e0 : CANCELED\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1bc472c5479c4d5881dad3a4d62a5a6d : SUCCESS in 49m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fdeaabcf78d74cb2804a2d150b508434 : CANCELED\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dd25a8a3ff5e4a8dab69ee586fa17fae : CANCELED\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/d032a1c21cc84ae683aac34f4bfc6dbb : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d701251edb9e454cbb84fdc635607817 : SUCCESS in 36m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d31289155cd24e278f3c00816f818958 : FAILURE in 50m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91bd08a63f4f4eac92a165acff516796 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06b23a323b2a476aad5f19bd4b0bd908 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ba0c93c3babd4cddb630dd2cd841f2d7 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e6ad44d432c456ebdf91848f64dda6e : SUCCESS in 6m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7a4b3df9e3ed44c78d2b9d2305ed3204 : SUCCESS in 2m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ab6d5d4a39a47258c286950660a1190 : SUCCESS in 30m 44s","accounts_in_message":[],"_revision_number":2},{"id":"a73e8743b0e1c9921eb0bb6b0a83b464a50c8346","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-19 11:52:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4e365223e3ea980ca9243ab04a428c192d4aace","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 13:09:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34c5215502fd48f0a1c9b4a2006b3422\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0fc44b06415f49c893386af74c28fc6f : SUCCESS in 41m 33s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/01e41f28ee104ca6848619fe2ad5b379 : SUCCESS in 1h 02m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7d341c7d26048c09a208c4b3f73bd91 : SUCCESS in 54m 39s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1c4c44ebb4c745eb9fad33542e4ad6f1 : SUCCESS in 41m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2280640ffcae4f9792e64d3d621a48f8 : SUCCESS in 38m 06s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/60beea7c920346a9939fcd57870933df : SUCCESS in 1h 07m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9a1dd2929a99429f8bb719729041bc70 : SUCCESS in 1h 06m 54s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/55a9e45c4709434eac43fcda57497d19 : SUCCESS in 1h 08m 29s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/b61a80696dea4bfaae8dc085322758b1 : SUCCESS in 59m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/89a042fed8ec424194737e3d0139b4c2 : SUCCESS in 47m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9ca69b81ba2344e5a2266bef0c479fbf : SUCCESS in 45m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05e358cfbb7d44f881380bb0287007e1 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82c0ba4dbd4c4af192caca219338eecc : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44d4e98b92774d6aa97f709589403bdf : SUCCESS in 5m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/880ef7c7bf4542039347a33c5d49e2ef : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24d044695b79451a86bc5d13d8a34585 : SUCCESS in 7m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9fd92e7764dd4adaa09ccebaacf3e73b : SUCCESS in 1m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a6d3143b6b2c41bd8d3881328471669d : SUCCESS in 27m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6a8465af9dee4625808719be0cd72691 : SUCCESS in 6m 11s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/55226225006647af816826ad464a9e75 : SUCCESS in 38m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"98be6bc2bbe5c6cef43399e53566aa069ffe750b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 13:09:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b8db9a80ee3c4ff33aeb1ea1935922f649955075","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 14:56:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb4f3a637fe74db49a92c6e4ef225d02\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a49d18f335b3467f8c1dc76350b12fa6 : SUCCESS in 41m 16s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/feb70d7a69f442fa8f1064209f969422 : SUCCESS in 50m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e445989b7c5d4ffab4cd552f6068f3fb : SUCCESS in 52m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/92999bbd3e904eb3aa50aee17f3755af : SUCCESS in 1h 14m 04s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/d0efd66ba6eb4428958c9757b21e0ae7 : SUCCESS in 1h 03m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/09d45ff5317d4decb10802d700036da4 : SUCCESS in 48m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0c0a2693bcda4af2b43c5d970de8c0d7 : SUCCESS in 46m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7d226fedfc64b9781dbd0cae73ae9e7 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f806b2300ef84d1e8d8a2587ef7e1b53 : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/62001dfc216b4927adc203bf810cd71d : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4062bc2804b245169e84066f0d1cb826 : SUCCESS in 7m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f6d15fee792b439d80217f7a5c08e5c4 : SUCCESS in 1m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b0f7e1cc8efb46f1ade543ec8521954e : SUCCESS in 29m 37s","accounts_in_message":[],"_revision_number":2},{"id":"cde1ba6c2900d25b2220bc5c473ec08f754ef77e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 14:56:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b84eb1a83347f77bf1a3b18448cb08bf958f041d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 14:57:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fc5c87b3ce748ee91b84d7c0371d8a8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5fb5c4010dc49d7a59958232a66f90f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3ae0695629d03b1b6bf49568da3c1cf044018a26","revisions":{"2f2b4506b748a02390e028c9e1e83fe52deed0c4":{"kind":"REWORK","_number":1,"created":"2024-09-27 15:08:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/930729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/930729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/930729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/930729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/930729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/930729/1"}}},"commit":{"parents":[{"commit":"c64083d45d5d91c063edc2f828440e79e69d07b7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c64083d45d5d91c063edc2f828440e79e69d07b7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-27 15:07:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-27 15:07:50.000000000","tz":540},"subject":"Remove workaround for eventlet \u003c 0.27.0","message":"Remove workaround for eventlet \u003c 0.27.0\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ia10b37af7a4e95236cfce4c7d95d32e6eae7a69b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2f2b4506b748a02390e028c9e1e83fe52deed0c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2f2b4506b748a02390e028c9e1e83fe52deed0c4"}]},"branch":"refs/heads/master"},"3ae0695629d03b1b6bf49568da3c1cf044018a26":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-14 13:29:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/29/930729/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/930729/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/930729/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/930729/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/930729/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/930729/2"}}},"commit":{"parents":[{"commit":"13c910ae9144123a9f0aa77454c4982ce4a1f6fb","subject":"Merge \"Add test for zone import with different min_ttl\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/13c910ae9144123a9f0aa77454c4982ce4a1f6fb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-27 15:07:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-14 13:29:49.000000000","tz":0},"subject":"Remove workaround for eventlet \u003c 0.27.0","message":"Remove workaround for eventlet \u003c 0.27.0\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ia10b37af7a4e95236cfce4c7d95d32e6eae7a69b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3ae0695629d03b1b6bf49568da3c1cf044018a26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3ae0695629d03b1b6bf49568da3c1cf044018a26"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
