)]}'
{"id":"openstack%2Fdesignate~932323","triplet_id":"openstack%2Fdesignate~master~I862e857e3bb969390275fab4cff5d243aa8e596b","project":"openstack/designate","branch":"master","topic":"add_pool_attr_header","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-09-12 14:29:51.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"hashtags":[],"change_id":"I862e857e3bb969390275fab4cff5d243aa8e596b","subject":"Allow to schedule DNS Zone to a non-default pool during Zone Import","status":"ABANDONED","created":"2024-10-14 11:29:00.000000000","updated":"2026-04-13 09:20:47.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8d5e4f13c2cf8e9702c3d840983f6f428715a022","_number":932323,"virtual_id_number":932323,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-09-12 15:48:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-14 12:36:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35f441c835cca45ea9912ad2549a0892574b154c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-14 11:29:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbf1711aa8dbc4feda9623c9a1c4dc1ec002136f","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-14 11:31:19.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"dc30261bfde3bdcddfff3333d335b927da72ca92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 12:36:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c77a4e5f92e44ee836fff8554d55408\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a0ead1b77bae43c6a7262e7e26a22e25 : SUCCESS in 49m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/aec8484203a1414e891ecc6a6f8a7df9 : SUCCESS in 55m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee891dba780d4e23966da94500e2ae3c : SUCCESS in 50m 30s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9bf89a52418f4cb9bd3ee786ef7f249b : SUCCESS in 58m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/84f28f7d29524bd290c3acc3cf0e2c69 : SUCCESS in 41m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0fd7d5bb51ee46ca89ee1ca7eb52db7a : SUCCESS in 59m 09s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a5120476dddf4f7e825d086a5bab5c22 : SUCCESS in 1h 06m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/630b892191d74145abce40a4623fe772 : SUCCESS in 44m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/831607f938c54aa2a4bace7f25de71c6 : SUCCESS in 47m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0295b9299f0a45b2b930bc567d7d0f53 : FAILURE in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce84af0a5ff246479b6dae67777d2e79 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2eb5b2e5207a4f1db49144a60e82c142 : FAILURE in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db096405bd394301b1b303f23d355072 : FAILURE in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81046da6837b4c3fb5d5bdffc7848c8c : SUCCESS in 7m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ebc1bb30d6ca415cb604c653dbd60080 : SUCCESS in 34m 16s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d2bb8f7458cc4135a25413e0bb019d23 : FAILURE in 4m 49s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/3af2ff27d40e4fb1b5e737ef3ada546a : SUCCESS in 47m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82f845b04f5f727c299ad0256deeda04f9d1fc78","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-14 15:32:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6b0d446028556bbfca97172ca0a41f2759bba5ed","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-14 15:32:10.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"d024740eb02c703a2d3c55b77ce996313f24ea71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 16:43:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a74e1a8c32d44048a2465fa4cfdb544f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6c203a184b6e41fdb9e020da9bc5ce4a : SUCCESS in 47m 04s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/837170a3b79a4eaeb56c80f35cddf47d : SUCCESS in 1h 05m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4ebc4a3c92548bd97a0de3586703b47 : SUCCESS in 59m 17s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8b87db11fa794f468373a9af33757fdd : SUCCESS in 48m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7d338bb9031f44819cbe67977968127b : SUCCESS in 43m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a6c083099f5340fd91050c6c2c3ea131 : SUCCESS in 1h 09m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/63b6a07afc854917b2e47089add8f1ea : SUCCESS in 1h 04m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/44d464c55a794986b0c992da3f42e8ef : SUCCESS in 37m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a1357d86e5014ea2885555dc420dbf2d : SUCCESS in 32m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d07a3507b4044cf7b91593b47cc4326a : FAILURE in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1382ba4e85db42efa825bba63ff2fc08 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6975fe5214d84597b370a4558d8023bd : FAILURE in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79db93e14c9c42bd9181cea51d227b51 : FAILURE in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c39645f80c4b63a808bb9ac876639b : SUCCESS in 6m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3677b8103e384e3bb41d763e6c4f0db8 : SUCCESS in 27m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6a5c5199f06f448ea014674a0335baa3 : FAILURE in 4m 37s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/b958c23ab3f14fef90ef8e3923a46240 : SUCCESS in 37m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"992a23b84f0f68c3a537b74d0cd442218dba5a0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-15 10:06:52.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5dd4dc2bb8c212d7e985bb724deee6ea0b361c42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 11:11:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df183e01fbb2437c8f49035e5e01194a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d51369393c1147058c18109f7282c96c : SUCCESS in 46m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9e2e3aa63ee34d7f8c4ff7c24397601e : SUCCESS in 1h 02m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d5a7db886f84609a1d1c181486acf40 : SUCCESS in 50m 27s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b2b5b466cce449b2b47bc8c242aa606a : SUCCESS in 36m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a90ab03bc1194708a2fb3ba627f1dbd8 : SUCCESS in 40m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/011544003dd14e33998c53ad7c6b72c5 : SUCCESS in 59m 43s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8d88d9c58807455c98649b85aaab97b6 : SUCCESS in 57m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/def4acc0bcb24091bdf7f1fef28d3f51 : SUCCESS in 38m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/afd9846a1e0d4f92bfee96b5f887124f : SUCCESS in 45m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/967126a7de724a4e9794b1767ffcc70e : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea0bfc8c96a64c418f5c92eaaa74c107 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e5708d5d61041d78390b69bd4bb01f9 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ad993822c4646e5aac34c50e836c4c7 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d2c9251eb53479ea4a36a033dd7289f : SUCCESS in 6m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/344f829ea51e4def88e9a2cac279da41 : SUCCESS in 38m 37s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ccd7d94f12d74964bd1ecffc9e740a6f : SUCCESS in 3m 41s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/8ce23b9c5f464f1d901f2ffd718e2cb5 : SUCCESS in 50m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd00263617eae39714d9cbb29166190c6e48ac8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-15 13:18:03.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"43e10bdaf05b665ffd694cf430fe8ff5bb579cf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 14:37:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f90b98873c24c009c00c8ac2d576331\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/de90240e37124c65a35d92e09531edfe : SUCCESS in 50m 34s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1d11a6d8cbae44c6a30835cadbf7e451 : SUCCESS in 50m 52s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/505b887ec6634369b71fb5fa91ca7521 : SUCCESS in 54m 42s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/84a4d70f4f0241e2bcaa2857881f03e4 : SUCCESS in 35m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/de394c9726804812b76739423ce56d1e : SUCCESS in 37m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fabf3ce0e98a439a8e5027c0636dfdde : SUCCESS in 1h 01m 29s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/de1ffac4c5b64ea289c88a3a6371b211 : SUCCESS in 1h 06m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5c9f87853bf5493e86d9034e7fc9f364 : SUCCESS in 38m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d8226d704333437b955ccd5236ecb2f0 : SUCCESS in 43m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01c4d583b270412eade67f13cd00f610 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f06aee192e6464791ae7caa3916dd75 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddbe1331b72e40c7925716c32549857f : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/47609136803c4dddbf1e8fa106a26cb7 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e92d3d42b6b4e5395411297b9c21996 : SUCCESS in 8m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/16f71dba75d645d2baf04cba37f1797c : SUCCESS in 29m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/dfa30cbc700a4664ab286f184761df0c : SUCCESS in 5m 05s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ce42c755828449e38d743e89d7128ddd : SUCCESS in 38m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c06b1e2bc9bfa076351c53aba5eca653bd455977","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-12 14:28:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"f9b41c84620a25dd22da387546bc29e04320b4da","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-12 14:29:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2f4d8bdfabc89e78a8a289bb7a31fc8c2cc04a04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 15:48:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74e16f3e957a4a978627ee0b88b89e36\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/84ca206bd41a40d3adfb267de26188f4 : SUCCESS in 1h 13m 50s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ba36e40275db4b0f91909a1c85b8bb97 : SUCCESS in 1h 07m 51s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/275eb3de6a05405f83b71723ae993905 : SUCCESS in 53m 52s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/26177281c31f4787a96ae616e8841cd9 : SUCCESS in 1h 10m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/175e52aadc974d958dcafc8ca9dc8b56 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd1b9122b714fe8a8cc9db90358f546 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4e3009bf04d4113b1c6c1f9c63d6231 : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4bbb19c7f8b4426082fe36d3386c998c : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/437ca5d224d04118af7437486b4bb94c : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/323820b162bc4193a915b2e28277706b : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/108c7bf038b94a3ea823b891e801f177 : SUCCESS in 4m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7311b17854aa451387ae0b68f60df280 : SUCCESS in 30m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c86acd0a621c488e89924ca712f71b29 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":5},{"id":"8d5e4f13c2cf8e9702c3d840983f6f428715a022","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-13 09:20:47.000000000","message":"Abandoned\n\nThis is probably a better approach https://review.opendev.org/c/openstack/designate/+/983408","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ee8021857d613b6c17aa4b03df8021038978e81d","revisions":{"80d6c080b07b5a544c7955dec66becb6b2816b71":{"kind":"REWORK","_number":1,"created":"2024-10-14 11:29:00.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/23/932323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/932323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/932323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/932323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/932323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/932323/1"}}},"commit":{"parents":[{"commit":"65169cde6f0ef4260b7b0db9a4a1f3d61ad15bc0","subject":"Merge \"Remove deprecated and ineffective neutron options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/65169cde6f0ef4260b7b0db9a4a1f3d61ad15bc0"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-10-14 09:45:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-10-14 11:28:49.000000000","tz":120},"subject":"WIP Allows to schedule DNS Zone to a non-default pool during Zone Import","message":"WIP Allows to schedule DNS Zone to a non-default pool during Zone Import\n\nThis commit adds an option to specify designate zone_attributes during\nzone import via \u0027X-Designate-Zone-Attributes\u0027 header.\nIf a header was given along with import request then the zone will\nreceive the \u0027zone_attributes\u0027 attribute exactly as it happens during\nnew zone creation with \u0027zone_attributes\u0027 given in json attributes.\n\nCloses-Bug: 1690184\nChange-Id: I862e857e3bb969390275fab4cff5d243aa8e596b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/80d6c080b07b5a544c7955dec66becb6b2816b71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/80d6c080b07b5a544c7955dec66becb6b2816b71"}]},"branch":"refs/heads/master"},"924df40b3bea771793ef8d62bbc4bfa1664a779a":{"kind":"REWORK","_number":2,"created":"2024-10-14 15:32:02.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/23/932323/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/932323/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/932323/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/932323/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/932323/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/932323/2"}}},"commit":{"parents":[{"commit":"58429834cbb3e3be28b435672bcc4504d08f8dbd","subject":"Merge \"fix blacklist list api-ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58429834cbb3e3be28b435672bcc4504d08f8dbd"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-10-14 09:45:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-10-14 15:30:53.000000000","tz":120},"subject":"Allows to schedule DNS Zone to a non-default pool during Zone Import","message":"Allows to schedule DNS Zone to a non-default pool during Zone Import\n\nThis commit adds an option to specify designate zone_attributes during\nzone import via \u0027X-Designate-Zone-Attributes\u0027 header.\nIf a header was given along with import request then the zone will\nreceive the \u0027zone_attributes\u0027 attribute exactly as it happens during\nnew zone creation with \u0027zone_attributes\u0027 given in json attributes.\n\nCloses-Bug: 1690184\nChange-Id: I862e857e3bb969390275fab4cff5d243aa8e596b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/924df40b3bea771793ef8d62bbc4bfa1664a779a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/924df40b3bea771793ef8d62bbc4bfa1664a779a"}]},"branch":"refs/heads/master"},"d6eab385961bd7670e16d97a18a80e3f39da484c":{"kind":"REWORK","_number":3,"created":"2024-10-15 10:06:52.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/23/932323/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/932323/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/932323/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/932323/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/932323/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/932323/3"}}},"commit":{"parents":[{"commit":"58429834cbb3e3be28b435672bcc4504d08f8dbd","subject":"Merge \"fix blacklist list api-ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58429834cbb3e3be28b435672bcc4504d08f8dbd"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-10-14 09:45:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-10-15 10:06:46.000000000","tz":120},"subject":"Allows to schedule DNS Zone to a non-default pool during Zone Import","message":"Allows to schedule DNS Zone to a non-default pool during Zone Import\n\nThis commit adds an option to specify designate zone_attributes during\nzone import via \u0027X-Designate-Zone-Attributes\u0027 header.\nIf a header was given along with import request then the zone will\nreceive the \u0027zone_attributes\u0027 attribute exactly as it happens during\nnew zone creation with \u0027zone_attributes\u0027 given in json attributes.\n\nCloses-Bug: 1690184\nChange-Id: I862e857e3bb969390275fab4cff5d243aa8e596b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6eab385961bd7670e16d97a18a80e3f39da484c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6eab385961bd7670e16d97a18a80e3f39da484c"}]},"branch":"refs/heads/master"},"00f1ac431b825697712eff32936911d759c3ff4d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-10-15 13:18:03.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/23/932323/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/932323/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/932323/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/932323/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/932323/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/932323/4"}}},"commit":{"parents":[{"commit":"58429834cbb3e3be28b435672bcc4504d08f8dbd","subject":"Merge \"fix blacklist list api-ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58429834cbb3e3be28b435672bcc4504d08f8dbd"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-10-14 09:45:39.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2024-10-15 13:18:03.000000000","tz":0},"subject":"Allow to schedule DNS Zone to a non-default pool during Zone Import","message":"Allow to schedule DNS Zone to a non-default pool during Zone Import\n\nThis commit adds an option to specify designate zone_attributes during\nzone import via \u0027X-Designate-Zone-Attributes\u0027 header.\nIf a header was given along with import request then the zone will\nreceive the \u0027zone_attributes\u0027 attribute exactly as it happens during\nnew zone creation with \u0027zone_attributes\u0027 given in json attributes.\n\nCloses-Bug: 1690184\nChange-Id: I862e857e3bb969390275fab4cff5d243aa8e596b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/00f1ac431b825697712eff32936911d759c3ff4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/00f1ac431b825697712eff32936911d759c3ff4d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ee8021857d613b6c17aa4b03df8021038978e81d":{"kind":"REWORK","_number":5,"created":"2025-09-12 14:28:31.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/23/932323/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/932323/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/932323/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/932323/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/932323/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/932323/5"}}},"commit":{"parents":[{"commit":"035453ffd1ba0f574a1d7f23b458222c0fd8fc68","subject":"Restructure Designate CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/035453ffd1ba0f574a1d7f23b458222c0fd8fc68"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-10-14 09:45:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-12 14:28:07.000000000","tz":0},"subject":"Allow to schedule DNS Zone to a non-default pool during Zone Import","message":"Allow to schedule DNS Zone to a non-default pool during Zone Import\n\nThis commit adds an option to specify designate zone_attributes during\nzone import via \u0027X-Designate-Zone-Attributes\u0027 header.\nIf a header was given along with import request then the zone will\nreceive the \u0027zone_attributes\u0027 attribute exactly as it happens during\nnew zone creation with \u0027zone_attributes\u0027 given in json attributes.\n\nCloses-Bug: 1690184\nChange-Id: I862e857e3bb969390275fab4cff5d243aa8e596b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee8021857d613b6c17aa4b03df8021038978e81d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee8021857d613b6c17aa4b03df8021038978e81d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
