)]}'
{"id":"openstack%2Fdesignate~932334","triplet_id":"openstack%2Fdesignate~master~I37a856f847f16367fb43723a15c54b88942e5b64","project":"openstack/designate","branch":"master","topic":"change_resp_code_mult_pools","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-09-25 15:22:35.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-10-28 10:33:15.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-09-25 15:22:35.000000000","reason":"Change was abandoned"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2026-09-25 15:22:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I37a856f847f16367fb43723a15c54b88942e5b64","subject":"Fix MultiplePoolsFound response code","status":"ABANDONED","created":"2024-10-14 13:52:24.000000000","updated":"2026-09-25 15:22:35.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a3fbfc770f97c50e602bcae406acd78fe7e08899","_number":932334,"virtual_id_number":932334,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-14 15:11:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-10-28 10:33:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-14 15:11:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-15 10:08:35.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2024-10-15 10:08:35.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-10-15 10:08:35.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"d62d33a2cdf2aca5c4f5c9e1fedf9102e2737dbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-14 13:52:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a16233a3e79637d6c29e86f26cfaef9c6a5e86cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-14 13:54:18.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5854239e6d226e2f957feeead96c81c223d33376","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 15:11:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4493975c3da9473ab485dbe52100ebe1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dce29e87ce7b44be9f57a23c77dd4e74 : SUCCESS in 51m 56s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d89b43b665824ccba48c52a6c38202c3 : SUCCESS in 1h 07m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f076ebc62b61475cb144fe88db7fb7c7 : SUCCESS in 52m 14s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1aee2ead174a432a90158f6446e3e348 : SUCCESS in 49m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/170a717ef2e644bc9dcaf1a852dca4d3 : SUCCESS in 43m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/89c63b874a5244fd9c595384ed8bcd28 : SUCCESS in 1h 13m 08s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a36bf32fa5494f44a04f668011f0f916 : SUCCESS in 58m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/abe0f6fd121640e28064444c9c58aaf7 : SUCCESS in 51m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/794555f456e940ca8254148e72ca70f5 : SUCCESS in 41m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c156df3d4d35453a9b151d708fbd1932 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b6c6c2938fa4c56a398002e1dcc3477 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37661e80d9b748f4a683ef8459a9dc08 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f0c33d458814cddb545b1301cc470fe : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d61f574eb31c41c0b95907259f21a4f7 : SUCCESS in 6m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ce3468854124d7c858bca975accfba2 : SUCCESS in 34m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4566065c5271418ca006f5fb28413a0c : SUCCESS in 4m 15s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/d781b92a4d414afdb0f92fc481b47b06 : SUCCESS in 47m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c538a495bb20af088d207a276e0ca6ba8a076047","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-28 10:33:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a3fbfc770f97c50e602bcae406acd78fe7e08899","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-25 15:22:35.000000000","message":"Abandoned\n\nThis patch stayed without any activity for almost 2 years, I am abandoning it","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"197ad790ea162dcbfd43cbd0bef3c8f9217e6423","revisions":{"46c154b55e9c210ecd8e2401b15907a03cc0a8de":{"kind":"REWORK","_number":1,"created":"2024-10-14 13:52:24.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/34/932334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/932334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/932334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/932334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/932334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/932334/1"}}},"commit":{"parents":[{"commit":"58429834cbb3e3be28b435672bcc4504d08f8dbd","subject":"Merge \"fix blacklist list api-ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58429834cbb3e3be28b435672bcc4504d08f8dbd"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-10-14 13:50:35.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-10-14 13:50:35.000000000","tz":120},"subject":"Fix MultiplePoolsFound response code","message":"Fix MultiplePoolsFound response code\n\nCurrently, when a user creates a zone and multiple pools match the zone\u0027s\ncriteria, 500 response code is being returned. This is misleading, as\nthe issue is not a server-side error but rather a conflict arising from\nmultiple matching pools. We should return 409.\n\nChange-Id: I37a856f847f16367fb43723a15c54b88942e5b64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/46c154b55e9c210ecd8e2401b15907a03cc0a8de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/46c154b55e9c210ecd8e2401b15907a03cc0a8de"}]},"branch":"refs/heads/master"},"197ad790ea162dcbfd43cbd0bef3c8f9217e6423":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-10-14 13:54:18.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/34/932334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/34/932334/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/932334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/932334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/34/932334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/34/932334/2"}}},"commit":{"parents":[{"commit":"58429834cbb3e3be28b435672bcc4504d08f8dbd","subject":"Merge \"fix blacklist list api-ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58429834cbb3e3be28b435672bcc4504d08f8dbd"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-10-14 13:50:35.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-10-14 13:53:42.000000000","tz":120},"subject":"Fix MultiplePoolsFound response code","message":"Fix MultiplePoolsFound response code\n\nCurrently, when a user creates a zone and multiple pools match the zone\u0027s\ncriteria, 500 response code is being returned. This is misleading, as\nthe issue is not a server-side error but rather a conflict arising from\nmultiple matching pools. We should return 409.\n\nCloses-Bug: 2084441\nChange-Id: I37a856f847f16367fb43723a15c54b88942e5b64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/197ad790ea162dcbfd43cbd0bef3c8f9217e6423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/197ad790ea162dcbfd43cbd0bef3c8f9217e6423"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
