)]}'
{"id":"openstack%2Fdesignate~932688","triplet_id":"openstack%2Fdesignate~master~I4d9c26e25851be0a955f4172a530eff266a89331","project":"openstack/designate","branch":"master","topic":"grenade-venv","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-21 15:46:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4d9c26e25851be0a955f4172a530eff266a89331","subject":"devstack: Fix manage.py execution with GLOBAL_VENV","status":"MERGED","created":"2024-10-18 10:31:53.000000000","updated":"2024-10-21 15:47:32.000000000","submitted":"2024-10-21 15:46:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932688-grenade-venv","meta_rev_id":"8d289e46ff0e5de8aba8a0f76c416706c65732d4","_number":932688,"virtual_id_number":932688,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-21 15:46:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2024-10-21 08:41:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2024-10-21 10:29:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-21 14:45:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-21 14:45:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2024-10-21 10:29:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-18 11:37:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-21 08:41:58.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-10-21 10:29:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-10-21 14:45:13.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"b4721eea37ef7ad536f00c3c0a04459c394f4cb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-18 10:31:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13444adb82d6faa2b61d6995e363c6113a1e9ce3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 11:37:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f68289dfe28347fea15a42a94409985e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/29665af353564bf7a9dea665e8e0f5fb : SUCCESS in 44m 04s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/17df03d1b6db41dbb8ecc3092324c7f5 : SUCCESS in 55m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a958d42b5fcd455eb117f830f8f046a4 : SUCCESS in 59m 58s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0c89d28384fd4f44b95fd596638c558f : SUCCESS in 45m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3837f884722d45ec875ed0b7dc586c3f : SUCCESS in 43m 30s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f149290757c440d195db3ceb502bd9d3 : FAILURE in 25m 35s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f12d5e60d66a411096183fd921ff0fff : FAILURE in 17m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/26120608aac647439275bd78c1ce5b62 : SUCCESS in 42m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1db6cb50e9b14b158a21c307d8ca1d85 : SUCCESS in 48m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5154859788da457db4604ceb5ab47c1d : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/692b0cb033cf46239c698c56a5cb1fe8 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99a2ade09d504cbc92d7978184b0e653 : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af32b6d1c48f4089aae9d95f135685a8 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ca033963c84e4f89cf3fac65a046f7 : SUCCESS in 7m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f0b2cecdc67a4aae84aa55e6a800c351 : SUCCESS in 38m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/10bccca9941a4cd6b80f3a5be377282e : SUCCESS in 4m 52s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/d2dce7528ff643bd85643d86a4c82c68 : SUCCESS in 46m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"818070c02bca651a0115e22c015a150d1e58d06f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-18 11:45:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7dea53ed03fa2b4798b19d471adef26f8e3bd3ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 13:11:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b4eaebc18c74dc68d2a845ce7b3223b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9f511488ee9042f3a6f7e867821b4831 : SUCCESS in 36m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/35ac71fa96684a21ae426ccf5239694f : SUCCESS in 42m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/098047eb645f4ef89fc37c6312392e0f : SUCCESS in 56m 09s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/53a7a031c4fa43b8aa61b8da6281a29d : SUCCESS in 43m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bacd369f195843cea37888a51b036b89 : SUCCESS in 47m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/349149e9e9134f43868b0aef1d7b2eb3 : SUCCESS in 1h 15m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4e737d3f30c3487ab46ff1bce55919ff : SUCCESS in 1h 18m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b57893692b3040f9a268c5aa3e216a21 : SUCCESS in 43m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/20ce095cec7f490cbe8119b70d314cff : SUCCESS in 50m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d75c62698c6247e09372691638cdcf9e : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1768b9f78434e568d645276e304cb11 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2443d356447e4ffba50abb0bbd751bb5 : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/def315e0d4754856874389c95f017f40 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/697e86873aa2424a889c91ba836a2230 : SUCCESS in 5m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e1cc4256237246ebab48aa771cdd97fc : SUCCESS in 40m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/399f624e4fb84c5989764ee2668eafea : SUCCESS in 4m 29s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/9d9fa040119e42f2bd0cbab3105aca64 : SUCCESS in 51m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d30a210c40b228f332502d2776d02f7c6578d438","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-18 14:20:20.000000000","message":"Topic set to grenade-venv","accounts_in_message":[],"_revision_number":2},{"id":"4a07dce43e10ddd95ff903cec83e8b758ea4932f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-21 08:41:58.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c0d5f6377933b16219756c111d645749208719a4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-21 10:29:49.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"47bea376e5194f4f29ff0d49b51ca76b6837c336","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-10-21 14:45:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dc95f658e82a0696a3cba900cdeff81fb818abe6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 14:45:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"18c6b58f23daf2a11e8b87519b82835316d24f5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 15:46:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a13c0098186d4094923e6d8fc7fa34b5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/be21c14ceeb54f929c488b1fb1baabe7 : SUCCESS in 37m 29s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/daf20367c17b4146a577da77950ffa15 : SUCCESS in 40m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e4ecaa728faa45c5b788ab119e033a38 : SUCCESS in 53m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0ca3676d30ab4dfbab2eec6270388661 : SUCCESS in 1h 00m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e11291d3d461477da66a8954dbd24e30 : SUCCESS in 35m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3462bfc902b64c7a857db705119248f7 : SUCCESS in 44m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7e16befccd5479b8720a2830158ab55 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3592d618801145ac95da35a41dfe9fdb : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b26326dabed4c1596dc71f460d2e9d9 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/507cec8dd8f74162b523f344bef0f430 : SUCCESS in 6m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/930b760cec1c4d6382dc78d7d9da0c48 : SUCCESS in 30m 40s","accounts_in_message":[],"_revision_number":2},{"id":"85829f38d8145537c9106477dab8c7a6aaed290c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 15:46:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8d289e46ff0e5de8aba8a0f76c416706c65732d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 15:47:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4005d6cf05b344c3befdb443922535fb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2c62694ec1408cbe340dd5510168d7 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0fa798fe58bb3bed7f133112e1fa38c11cc18da7","revisions":{"c3828b915d7ced943f9f3f80e1d95193929fd456":{"kind":"REWORK","_number":1,"created":"2024-10-18 10:31:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/932688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/932688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/932688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/932688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/932688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/932688/1"}}},"commit":{"parents":[{"commit":"65169cde6f0ef4260b7b0db9a4a1f3d61ad15bc0","subject":"Merge \"Remove deprecated and ineffective neutron options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/65169cde6f0ef4260b7b0db9a4a1f3d61ad15bc0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-18 10:30:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-18 10:30:46.000000000","tz":540},"subject":"devstack: Fix manage.py execution with GLOBAL_VENV","message":"devstack: Fix manage.py execution with GLOBAL_VENV\n\nWe need to use the specific python path to run the command with\nall software installed in global virtualenv.\n\nChange-Id: I4d9c26e25851be0a955f4172a530eff266a89331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3828b915d7ced943f9f3f80e1d95193929fd456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3828b915d7ced943f9f3f80e1d95193929fd456"}]},"branch":"refs/heads/master"},"0fa798fe58bb3bed7f133112e1fa38c11cc18da7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-18 11:45:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/932688/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/932688/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/932688/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/932688/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/932688/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/932688/2"}}},"commit":{"parents":[{"commit":"4b787aaba50ba59ecb9650d297d190ecbb951d49","subject":"Disable GLOBAL_VENV in grenade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b787aaba50ba59ecb9650d297d190ecbb951d49"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-18 10:30:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-18 11:45:24.000000000","tz":0},"subject":"devstack: Fix manage.py execution with GLOBAL_VENV","message":"devstack: Fix manage.py execution with GLOBAL_VENV\n\nWe need to use the specific python path to run the command with\nall software installed in global virtualenv.\n\nChange-Id: I4d9c26e25851be0a955f4172a530eff266a89331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0fa798fe58bb3bed7f133112e1fa38c11cc18da7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0fa798fe58bb3bed7f133112e1fa38c11cc18da7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
